)]}'
{"id":"openstack%2Foctavia~896389","triplet_id":"openstack%2Foctavia~stable%2Fwallaby~I453ff385620f488a3e43d4fda01634cdce3be5d4","project":"openstack/octavia","branch":"stable/wallaby","attention_set":{},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2023-09-26 10:10:51.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I453ff385620f488a3e43d4fda01634cdce3be5d4","subject":"Fix race condition in members batch update API call","status":"MERGED","created":"2023-09-25 12:07:20.000000000","updated":"2023-09-26 10:12:15.000000000","submitted":"2023-09-26 10:10:51.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"896389","meta_rev_id":"ac7738620a1fbba9a3b62a1e6a42b5545bdc87da","_number":896389,"virtual_id_number":896389,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-09-26 10:10:51.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":2,"date":"2023-09-25 15:52:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-09-26 08:06:05.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-09-26 08:06:05.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-25 14:06:32.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-09-25 15:52:30.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2023-09-26 08:06:05.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"}],"messages":[{"id":"46096364d533b7006c82d6dd685faa343570dfcc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-09-25 12:07:20.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"04c98e7156f2b6510c4abf06d4ce2ff30e94d7e2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-25 14:06:32.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/31e5f33358414f70bb660f702a8e769c\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4948ec462f2c4fc1b4415a647ca4aebe : SUCCESS in 9m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/37daa5c861604e2a92d97a6fe94d1134 : SUCCESS in 5m 06s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/ddee2404836247e8a5b4676e1e0ff56b : SUCCESS in 5m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d01ded6a1f8e4f1d9f5206fb6e0628bf : SUCCESS in 4m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d11367fb3d46418fa071d1ec5ca752bf : SUCCESS in 5m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/335366eca99f42b4ae249469ee216bb6 : SUCCESS in 11m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3cf0900af27f407abe90737cbe4ebf28 : SUCCESS in 4m 41s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/9717e2c73e3a4a6a89576bb7e5f0b9b0 : SUCCESS in 6m 37s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/62d2e527818344eca5c4876f427c8920 : SUCCESS in 7m 52s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/5ef6c3a2e1ee4ec4ad648cef7bd7b57a : SUCCESS in 4m 30s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/31bd04c7f87a4048943bea10001c8c2f : SUCCESS in 9m 41s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/01520534d7964588a305b87506bb9123 : SUCCESS in 1h 01m 08s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e0bf85528f3b43089bf44132087a9bef : SUCCESS in 1h 55m 20s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/4246fe05a87742bcace59ed63845b902 : SUCCESS in 45m 07s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/8790fad8e0ca445bb9f6b8157dd6bb0e : SUCCESS in 40m 43s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2216ba487a8c47d086037c711d223a90 : SUCCESS in 33m 53s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/edcb62a8f76b43939d549e0722408716 : FAILURE in 13m 40s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/2918891516ff47cfbc11f598a96e009d : SUCCESS in 1h 50m 45s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/1ba25e76882e4c20b4fe188f46b83bae : FAILURE in 9m 53s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/bdd79a952e27494e94aafc2545c70fbb : FAILURE in 5m 36s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/33b35d168ce04fedb9bc32034d0dd23c : SUCCESS in 1h 52m 18s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/dddb6ab1d43a4a16bf1edb7a66d24972 : SUCCESS in 1h 40m 39s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/daa2a4efdd4540299205bf02ae200eb9 : NODE_FAILURE Node request 300-0022313630 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7b5c4ac1d283bfd5671b3d0fc6d57d927ef3b42e","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-09-25 15:52:30.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4b7546afbfe366eab6962c282ed63ba34cc63b1d","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-09-26 08:06:05.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"b4ce32db4c98145774cbaeac93efd479211b358e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-26 08:06:41.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4c8bd3a9c7eac1bc352a635525c358b1205d5635","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-26 10:10:51.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/db2654c2bdb740bbbc3a72260038ee23\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bca99506dfdf482083346b504595b109 : SUCCESS in 4m 19s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/19697f9c0c084ca48c24e253b31411fd : SUCCESS in 5m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/86981d8bbb3c4b329c4f4966189b3bb1 : SUCCESS in 4m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eeca11b0599c4c96a77ea595e84fabe6 : SUCCESS in 12m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0fd49437190c4ef0b32d06e075ba7e52 : SUCCESS in 5m 18s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/e11d3288b91e4884be5a58e4c81088d2 : SUCCESS in 4m 42s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/18465fa69d39458cae870f489fe73ec0 : SUCCESS in 8m 14s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/df86b27e411c4b4a96c132efb8e4824d : SUCCESS in 1h 06m 45s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/94c5146f3dc64b0cb72ec0d9b5117c5d : SUCCESS in 2h 02m 58s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/5ea3a974074c4db1a0697e7f199748c5 : SUCCESS in 46m 25s\n- octavia-v2-dsvm-spare-pool https://zuul.opendev.org/t/openstack/build/a0716b88a559465e8e1f17679c7fd432 : SUCCESS in 34m 37s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/41f60def3b3c4749a3d9be808fd90fab : SUCCESS in 38m 32s","accounts_in_message":[],"_revision_number":1},{"id":"c9ef89d6fe1ae980e6cab608b265aa25cdc0d28c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-26 10:10:51.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"ac7738620a1fbba9a3b62a1e6a42b5545bdc87da","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-26 10:12:15.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e45ebe11e2604018b80917078073c36c\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/81fbd1ac0e2d4dd28063ea98f73d61ed : SUCCESS in 1m 04s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bb7b8bb92c164a268ae378b7d504a420 : SUCCESS in 50s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"77e14a8d790bfb0facaf2ee56c1f798e58269b13","revisions":{"77e14a8d790bfb0facaf2ee56c1f798e58269b13":{"kind":"REWORK","_number":1,"created":"2023-09-25 12:07:20.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/89/896389/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/89/896389/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/89/896389/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/89/896389/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/89/896389/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/89/896389/1"}}},"commit":{"parents":[{"commit":"60ffac310dc7b520ec20c11f3c1fed1f1cb5af9e","subject":"Drop lower-constraints.txt and its testing","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/60ffac310dc7b520ec20c11f3c1fed1f1cb5af9e"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-09-15 07:42:26.000000000","tz":-240},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-09-25 12:07:16.000000000","tz":120},"subject":"Fix race condition in members batch update API call","message":"Fix race condition in members batch update API call\n\nFix a potential race condition, the member batch update API call\ncomputes a delta between the current state of the members in the DB and\nthe members in the requests. But it fetches the members from the DB\nbefore locking the load balancer, it means that another Octavia service\nmay still update the list of members, then the computation of the\nchanges may be incorrect.\n\nThe list of members is now queried in the locked section.\n\nCloses-Bug: #2036156\nChange-Id: I453ff385620f488a3e43d4fda01634cdce3be5d4\n(cherry picked from commit 8a47577b7668e6dc82e048d937d1344c667e5124)\n(cherry picked from commit 8a5bed4cebd70c1c721a3916dc79f2b19343e445)\n(cherry picked from commit 64d9bdb97556652aa05a34276778260b541ad9c6)\n(cherry picked from commit 4c97b585ce6e519a5ff090c2cd97fab0e801be0c)\n(cherry picked from commit 231398ca5fee7b825646eeed84523dbfcb414438)\n(cherry picked from commit c6dd946b996fb6e5720892db25c2ad0a217b8bb2)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/77e14a8d790bfb0facaf2ee56c1f798e58269b13"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/77e14a8d790bfb0facaf2ee56c1f798e58269b13"}]},"branch":"refs/heads/stable/wallaby"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
