)]}'
{"id":"openstack%2Foctavia~896506","triplet_id":"openstack%2Foctavia~master~I1dd3227e8f06b4e2b1888614979c97fd116f6a31","project":"openstack/octavia","branch":"master","attention_set":{},"removed_from_attention_set":{"36229":{"account":{"_account_id":36229,"name":"Natz cheng","display_name":"Natzcheng","email":"chenggang01@inspur.com","username":"Natzcheng"},"last_update":"2024-01-02 13:52:12.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I1dd3227e8f06b4e2b1888614979c97fd116f6a31","subject":"Cleanup py27 support","status":"MERGED","created":"2023-09-26 11:28:40.000000000","updated":"2024-01-02 13:53:22.000000000","submitted":"2024-01-02 13:52:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"896506","meta_rev_id":"a8ee74856b88d3be93927638ee6132bc122df0f7","_number":896506,"virtual_id_number":896506,"owner":{"_account_id":36229,"name":"Natz cheng","display_name":"Natzcheng","email":"chenggang01@inspur.com","username":"Natzcheng"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-01-02 13:52:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":2,"date":"2023-10-09 07:45:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-10-11 10:20:03.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"all":[{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-10-11 10:20:03.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"disliked":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"all":[{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2023-10-11 12:34:15.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","value":-1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-09-26 14:07:33.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-10-09 07:45:28.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2023-10-11 10:20:03.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"},{"updated":"2024-01-02 08:37:09.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"CC"}],"messages":[{"id":"f37bcd161790e1eb00ae2b67df7252af9940ce61","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":36229,"name":"Natz cheng","display_name":"Natzcheng","email":"chenggang01@inspur.com","username":"Natzcheng"},"date":"2023-09-26 11:28:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"67315112b40a1001221f58a9df99aae999e291ce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-09-26 14:07:33.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/050c44e5d40f4d2396682b0337010006\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/abddffb9974a42d6880594035a4b21b7 : SUCCESS in 2m 03s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2f62a03cdf584ea7bed821f59e3c3cec : SUCCESS in 11m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/654fdea3f8a94b16bc6e21ffccbb38cd : SUCCESS in 5m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e36d7a10a688490da2db436ee7004152 : SUCCESS in 4m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ed5a67f2bf164705859f25e0ffe9e19a : SUCCESS in 5m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/2c23d86933844ea8b4464025ac101dc8 : SUCCESS in 4m 44s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/84819698deb0418a9e721f76414c9a9a : SUCCESS in 5m 40s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/37e2e9ed579142598fc838a6b9f28e5e : SUCCESS in 11m 44s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/12a7d9b388fc4477a94e186f6d993722 : SUCCESS in 4m 39s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/ade1fc8adee8401fa0625a4202e6c966 : SUCCESS in 7m 49s\n- octavia-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/19fe47af2a594d59b708cc9127aa5ebf : SUCCESS in 8m 23s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/c163211beeb14b208f99c5be3d227875 : SUCCESS in 5m 01s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/f51b0c6906344926a1b21106628b4b7c : SUCCESS in 6m 19s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/cdc14b4265bd41cb9c3dd1bd30f5bd5c : SUCCESS in 7m 25s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/61ecb6567fc64d39944f65ccdde85632 : SUCCESS in 6m 06s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/089aebc6ade442219b3738207b13713f : SUCCESS in 1h 09m 49s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/f76d169d3d9049f38ec199d33c169e3f : SUCCESS in 1h 33m 12s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/dff4f84269da41fd839db65f1be2d7fd : SUCCESS in 56m 14s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/6b3a45eb6f6a48cc9a87d1cf63037c4a : SUCCESS in 1h 42m 55s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/366b1d03f9c7416183d4c2d2f8a3c412 : SUCCESS in 1h 16m 41s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/e8632c8d4e6342ff8c0b26899d2c9efd : SUCCESS in 45m 16s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/fd068276fd814d65b338130a29e820a2 : SUCCESS in 52m 03s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/6481789ccb2742419508f6f9bafc8c8d : SUCCESS in 1h 12m 08s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/89eb6fab125f4deda92efebd215eec28 : SUCCESS in 39m 15s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/16d091e078ce425db899d0d8f698a13b : TIMED_OUT in 2h 33m 25s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/f27c1df8a91949b7b664a1da37e02da5 : SUCCESS in 2h 26m 54s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/2400b4a1cc7747aca44f1183ba9d4d2e : TIMED_OUT in 2h 32m 57s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/ca72eb60df5247feb3518603c3a46437 : TIMED_OUT in 2h 34m 03s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"fc23fd86c426ca6aada5b7d67eb3bfdeb407d748","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-10-09 07:45:28.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"2e6e83c96daea314acb0e4d3fa0ec76a05b158da","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-10-11 10:20:03.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4ae515cd72d9a384f9ce74f3cb59fa08c1015c6b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-11 10:20:45.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"816b1ff1162ed6576b900b701bd4d3e241f4a69a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-11 11:27:19.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9efb6fb4a005460fb60710e34aa18378\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/bac3e04f2d024c1393cb48c3636e15f9 : SUCCESS in 2m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3a1a05de4aa0448f8f9b5db30440edd1 : SUCCESS in 5m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d960ed54588d4e56ae23e3f053091108 : SUCCESS in 5m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5e97397c97c84e2eb3108255e2bc0816 : SUCCESS in 5m 02s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/417d9c23a9cf4cd098b6b64f4e64bec7 : SUCCESS in 5m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/26ef37490b684006b2180503302d6a32 : SUCCESS in 12m 06s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/a6a931ae532544a6b5a0126abf854c1e : SUCCESS in 4m 51s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/73d3eea359624d0c8e38d3676b6d5ab1 : SUCCESS in 6m 51s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/ebcf231a41cd43958b4dc84a4a0179da : CANCELED in 49m 30s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/13d1fe80d5b54a57857c000015db0acc : CANCELED in 50m 16s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/f8f350b7ea1049a0b7f16523bc086996 : CANCELED in 35m 26s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/17aec3375b2c4ee7b53eb370b8a5c0c6 : FAILURE in 50m 10s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/393c2f6b493c4ae195d7f33c5c21a9de : CANCELED in 49m 33s","accounts_in_message":[],"_revision_number":1},{"id":"b9a3ce5fe77c6270f27bcdfb263c9aa911cb973b","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-10-11 12:34:15.000000000","message":"Patch Set 1: Backport-Candidate-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"88169dff52a56e95d13c71ece4a7c374b49cf007","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-11 15:38:58.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/5eb6065d25184689b714800c86f5ccfb\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/2d802f0e69284da88a93a66e28f5e217 : SUCCESS in 2m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b0e33bade4c3422fab5473310876f03a : SUCCESS in 11m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b55f16aa115b4bba80aba0d74e458b71 : SUCCESS in 5m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/456540a1181248bd9f02f2c12282be7b : SUCCESS in 4m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f8d246a2c9f244f0aab4c51736cfd28e : SUCCESS in 4m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1e532bc687ee4267bf7ad2547a1c2702 : SUCCESS in 5m 56s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3a08d3b2326b4a259fdaca01560075a4 : SUCCESS in 5m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e042ea7b40e74fc2bb72aea18b86c496 : SUCCESS in 11m 39s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/30af0f4a88214c848abf4140d7bc48f4 : SUCCESS in 4m 58s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/c38ea09a039e4e479acf68b1acc2c2a3 : SUCCESS in 8m 00s\n- octavia-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/193587aac6c14076b547c5292fa0f2e8 : SUCCESS in 11m 21s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/462d639525b54edc821864827c014cab : SUCCESS in 5m 06s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/e1c1949ccea1405b8f00fc7d45fb80bc : SUCCESS in 7m 05s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/1382ea587e284914be4f486eb1fa1804 : FAILURE in 7m 05s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/4ff95cf2341c4f9ca77745037255a792 : SUCCESS in 8m 17s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/7f68159c57014b3e90a5e97a24ed88fd : SUCCESS in 1h 19m 39s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/75710dae97324d9385fe85813039cc83 : SUCCESS in 1h 25m 35s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/c1af71d51cb44abda37dd35f317244fc : SUCCESS in 1h 02m 17s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/b89ca59aabc14c9ca525fcde4bf29a0e : SUCCESS in 1h 28m 07s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/c78d78fafeee45b593e3d866b158a360 : SUCCESS in 1h 17m 27s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/8a3d8df8b5be4253ac3d7bcbe64f14df : SUCCESS in 43m 41s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/294acade24fe4a63ab4f7de6bdbe6a4c : SUCCESS in 48m 10s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/dfe017b0aec04b578e297ccf13572604 : SUCCESS in 37m 37s (non-voting)\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/9d27f89997a84595be773ffc5d1a48ba : SUCCESS in 1h 02m 01s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a52926cef3414fea8cb225af6cd69790 : SUCCESS in 46m 00s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/0cd9624753b44c57aeb43a37d37a4690 : SUCCESS in 2h 09m 33s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/5b46c873e4a241dcaa743ff9935fdb1e : SUCCESS in 3h 00m 51s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/db6c9fbfdcd346d29f1a7872f1625928 : SUCCESS in 2h 16m 50s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/388a61376c4a4388bb5dc95e8be9cf49 : SUCCESS in 2h 34m 42s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7914d9fa4396dd4f21efef5030798d05f5ae243b","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-10-24 12:51:41.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d7355405e5fad36d913127aeeb1170ad45786294","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-24 15:38:28.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7c8b893f51074c01819c7c5fe6f9f673\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/1689947d444e4cf080ad55117411ef85 : SUCCESS in 1m 51s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5fd66f0be60a417782fe73fc7a68cc28 : SUCCESS in 11m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/28575aadef414fd5aa9aaf9417558c89 : SUCCESS in 5m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b637c4ed67e247ff8f3cae87f9ce7f1e : SUCCESS in 4m 57s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7d76b3c16b214b3db75228b62e3b35d3 : SUCCESS in 5m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ff476d5c509b46cfab9212f5061cd779 : SUCCESS in 9m 39s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/8aac55c088dd434aaf10f54609cda4f8 : SUCCESS in 5m 45s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/1a4a53f992ed4294ae32e64e425ff1e4 : SUCCESS in 6m 45s\n- octavia-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/c77011f995f84d04b8f3ee52c898e8fb : SUCCESS in 8m 41s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/e461af6965ef4dedb0b92817f895e2c5 : SUCCESS in 4m 06s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/54172b02cca64477a2ffb93bf8e78fb9 : SUCCESS in 7m 33s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/e20c9c9951a74b0a9c4470788dfed6c3 : SUCCESS in 6m 36s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/84a77ded640e4b19860c74e6ac48c427 : SUCCESS in 8m 45s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/9683a804688340e19a6b5cac65753bb7 : SUCCESS in 1h 17m 04s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/3a364313704349b8a03fdeac9d28ef12 : SUCCESS in 1h 35m 54s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/8abb9afc4e9a4bd294b5f08b74bddcc2 : FAILURE in 1h 15m 16s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/e8f2a4d5d3854735a383179ee80bb684 : SUCCESS in 1h 57m 48s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/6b21324362c34847af706c6280bf5049 : SUCCESS in 1h 19m 33s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/ec8975b1319c492b8a80f0f30eaaf087 : SUCCESS in 37m 00s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/a804adac62e84887a2506d21a594c40b : SUCCESS in 41m 18s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/73effeeb18b64ec3b27891578c66343a : SUCCESS in 37m 59s (non-voting)\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/018726009e9c413e935c66ab96467287 : SUCCESS in 1h 13m 12s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0c1656beb81d4aa6be0716864695d623 : SUCCESS in 35m 11s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/05264ffd05d6417dbd612309bca546f9 : SUCCESS in 2h 15m 30s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/96c9191240cd4236accedcc3811a6bc4 : SUCCESS in 2h 44m 38s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/7b9cce55b34343af9acfb04d81c05602 : SUCCESS in 2h 11m 59s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/18a70783578a4af494d2f0f30917c3da : SUCCESS in 2h 29m 00s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"caaafd15e2660b665ec5e63d123140bc81dd123f","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-01-02 08:37:09.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4a99940dc3a914bc716220446e759452c9b044b8","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-01-02 08:37:54.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b563ff7182d9a362a0c01c5209f597cc94b8a728","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-02 11:21:31.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ec06b7c53459440b9c293d006508fd3b\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/d5f1c1796c55496a9c3e43c95a1ac6ad : SUCCESS in 1m 49s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/13775dcd4d754eeabd9c36fcf07ed86f : SUCCESS in 9m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d83e18331d664a48b7ed055d14150672 : SUCCESS in 4m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cb7da303805641b4912004933cd20552 : SUCCESS in 4m 36s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0cf5f72593bd48f583ae5c17635903c7 : SUCCESS in 5m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7f921b0f49cf4660a48e8fb2e2423864 : SUCCESS in 13m 04s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/90c338a8e4874672908ad7e11a744859 : SUCCESS in 5m 03s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/ff662287bbad4b8c91d11eae6809a3cc : SUCCESS in 8m 35s\n- octavia-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/ed6cf00447d843d1a407f89e71f8150e : SUCCESS in 9m 51s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/a2ac606ed2ce476eb2d660b7186a9f53 : SUCCESS in 4m 48s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/f3272f9a79964bc5966628bac00d21de : SUCCESS in 8m 23s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/bf7257274d1b4285aec573ebffc7acd5 : SUCCESS in 8m 44s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/f77c4a9a411146b9879ae4406a145da5 : SUCCESS in 8m 15s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/f2d17b9f133c4094b3898cdda856a66c : SUCCESS in 1h 16m 19s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/f5808af36112485eab54fdb333da1a75 : SUCCESS in 2h 05m 16s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/568bc05cd81a480287509f7a2d16e4a6 : SUCCESS in 1h 21m 28s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/477443bfccfa4141a2b2b48f55e646ac : SUCCESS in 2h 06m 37s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/3bab9b4b46ae4306ba0644781daddb35 : SUCCESS in 1h 39m 24s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/dd92b5dc51bf44b7856728aa1a086178 : SUCCESS in 41m 40s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/5c4a2b9b00d849f183eea87744435f52 : SUCCESS in 47m 00s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/5f587181058c40ec912ec519eabaeffd : SUCCESS in 48m 35s (non-voting)\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/ac097b69da544439a3ef6be2dedbfb5a : SUCCESS in 58m 59s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a851bfd2e3c344acb06c89c0de3c0c13 : SUCCESS in 43m 18s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/2a36ab4ac3f44c71880ed6c5d957d358 : TIMED_OUT in 2h 32m 40s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/ab4fb9af5b834997bdbc2983cfa6f525 : SUCCESS in 2h 39m 43s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/629fe2ec498b4b138cbe5f061b1d8c6e : SUCCESS in 2h 09m 59s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/6fd035363941440f8a3d98a12117542c : SUCCESS in 2h 34m 21s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8f5fc8cbcdbc79558244791c3ca233999c648df2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-02 11:22:06.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"c41f1b95a444eddb41b5da0e4c01e1612653fc89","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-02 13:52:12.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9b6a9dd5abca4cc2a22c305af4dc8911\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/82458a0529d74fb581751276c6f73e96 : SUCCESS in 2m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0c6d275296e448c193f322048201a4d0 : SUCCESS in 4m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/079d1a5c41fb467daa027aacca1664d0 : SUCCESS in 5m 03s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0b6a153f68df4143bb0eac070d9b0c0a : SUCCESS in 5m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b32caf47932e42b1bc0d0924dbce2921 : SUCCESS in 11m 31s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/12fb737ec2ae4e8ba26fa8fe4050d649 : SUCCESS in 4m 25s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/d327241c26504b4ba73ceca95406186c : SUCCESS in 9m 15s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/bad1e7ce8b854777ac16b36d5ae8fdf0 : SUCCESS in 58m 12s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/7e9f5445bffe48a59651e59f4f1b32e1 : SUCCESS in 2h 13m 57s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/b13fadaebefc4ead9715a34659e9d1c8 : SUCCESS in 1h 11m 32s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/2ac94419bb4b4612b58067f77708befd : SUCCESS in 36m 21s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/767e0cb0166b43ad8fed80a610ccfccb : SUCCESS in 41m 09s","accounts_in_message":[],"_revision_number":1},{"id":"5463120b07332c309044b0b7ddb6f724812fc604","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-02 13:52:12.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"a8ee74856b88d3be93927638ee6132bc122df0f7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-02 13:53:22.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e90b61d00bf0481190d9f7a380e2fd02\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d27b5f18569b481993788d235de7c020 : SUCCESS in 53s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"95774d47ffcda5a2623b27775a7631b453eacfff","revisions":{"95774d47ffcda5a2623b27775a7631b453eacfff":{"kind":"REWORK","_number":1,"created":"2023-09-26 11:28:40.000000000","uploader":{"_account_id":36229,"name":"Natz cheng","display_name":"Natzcheng","email":"chenggang01@inspur.com","username":"Natzcheng"},"ref":"refs/changes/06/896506/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/06/896506/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/06/896506/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/06/896506/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/06/896506/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/06/896506/1"}}},"commit":{"parents":[{"commit":"025bfdd758ea6521a28dfaf40d916ef6f7421752","subject":"Merge \"Fix race condition in members batch update API call\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/025bfdd758ea6521a28dfaf40d916ef6f7421752"}]}],"author":{"name":"Natzcheng","email":"chenggang01@inspur.com","date":"2023-09-26 11:28:11.000000000","tz":480},"committer":{"name":"Natzcheng","email":"chenggang01@inspur.com","date":"2023-09-26 11:28:11.000000000","tz":480},"subject":"Cleanup py27 support","message":"Cleanup py27 support\n\nChange-Id: I1dd3227e8f06b4e2b1888614979c97fd116f6a31\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/95774d47ffcda5a2623b27775a7631b453eacfff"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/95774d47ffcda5a2623b27775a7631b453eacfff"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
