)]}'
{"id":"openstack%2Foctavia~896995","triplet_id":"openstack%2Foctavia~master~I020ded6e1735ea56da46c5170292ba88ad639103","project":"openstack/octavia","branch":"master","attention_set":{},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2023-10-11 12:42:08.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I020ded6e1735ea56da46c5170292ba88ad639103","subject":"Fix error when deleting LB with broken amp","status":"MERGED","created":"2023-10-02 06:57:57.000000000","updated":"2023-10-11 12:43:51.000000000","submitted":"2023-10-11 12:42:08.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"896995","meta_rev_id":"c109e3d123ab9114e01f6699bf07f88e01ea5993","_number":896995,"virtual_id_number":896995,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-10-11 12:42:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"recommended":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"all":[{"value":1,"date":"2023-10-03 14:46:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":1,"date":"2023-10-02 14:53:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"value":2,"date":"2023-10-02 22:06:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-10-11 10:28:05.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-10-11 10:28:05.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-10-11 10:28:05.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-10-02 09:43:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-10-02 14:53:57.000000000","updated_by":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"reviewer":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"state":"REVIEWER"},{"updated":"2023-10-02 22:06:11.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2023-10-03 14:46:04.000000000","updated_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"reviewer":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"state":"REVIEWER"},{"updated":"2023-10-11 10:28:05.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"}],"messages":[{"id":"bf4c711e640955679fa19aa6208b060fa3ccfcd1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-10-02 06:57:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"47e105a3a58c04a9fc96e390a66e78e23d7a35fa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-02 09:43:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c13665c1561f47f08b1f45b99adcf380\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/cf49d266f98143d28dcee96a38b69b23 : SUCCESS in 9m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/aa7a2d7fb0744f6a811efe8bb6e4e292 : SUCCESS in 5m 30s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/26f5fb986c1f4540983c09c497b2ff8a : SUCCESS in 4m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/95ab219d128345bebb21e2e26ee3529b : SUCCESS in 4m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7fc55b6f31cb4a7e940fa12ed0bb3c00 : SUCCESS in 4m 48s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/922a7d20ac814741814814d5298ee568 : SUCCESS in 5m 30s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ed2c670d68a64873aabfe1fd0e203a6b : SUCCESS in 10m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/17003751bdec4d84aeeacb98accc91b8 : SUCCESS in 5m 08s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/e4db8f51f09246328ed30fca6a7b04b9 : SUCCESS in 4m 14s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/cdf6857eaaa649fea1268381fcbc5397 : SUCCESS in 7m 24s\n- octavia-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/a1150cad43f940ef9a0ca740ded1ad9a : SUCCESS in 7m 58s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/0b470dc95720470a94422b63a3e8472b : SUCCESS in 4m 44s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/60f44a9233da4591be86ac88cea7a2f6 : SUCCESS in 6m 28s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/c2e417ca4bdf4c978f41532a417f8e9e : SUCCESS in 7m 45s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/d9a8dbf2403b47f59e1af026ce20026e : SUCCESS in 7m 00s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/e100fc9a26d04d20b1834b805d7a3bdd : SUCCESS in 1h 19m 41s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/13a28d23a7d94be2a8b40ee41190a8eb : SUCCESS in 1h 20m 35s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/8cb91ff3271146d18aad26598874177b : SUCCESS in 52m 03s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/1af72920b9a34893b1ad080a1bfefd04 : SUCCESS in 1h 23m 30s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/58a4d5b557a84380bf54757dc403b9ad : SUCCESS in 1h 07m 02s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/b10fcc8ef9344f10a7da2d4faf250f8f : SUCCESS in 33m 37s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/981988fff4334c52a2681f8327f37e45 : SUCCESS in 43m 02s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/3816d90aca7e4be4bf7c5ffe80b02af9 : SUCCESS in 58m 27s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/12d19c6687f641139b7a140fdf558cfa : SUCCESS in 35m 54s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/ddb38929201a45ee85630ad8c4478a1f : SUCCESS in 2h 07m 21s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/8ec32302aa114b5aad578c1b9db0db14 : SUCCESS in 2h 42m 58s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/ffff8c45bebf4ad0a665c0164bc35d96 : TIMED_OUT in 2h 33m 49s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/63d4bfdaabf74a4998db07c2f4bcfb41 : SUCCESS in 1h 56m 54s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6e9fb2bf7e1d3b9bbdb4e913e572afbbf344142d","author":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"date":"2023-10-02 14:53:57.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"9879af1c65ced8c3c2bb9576b5a927deef783252","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-10-02 22:06:11.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"d71145053bbdaab905b01fa0823d24d011723ba8","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2023-10-03 14:46:04.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6d6c426c955e0ba7e0e38cae8f03a1992895c0bc","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-10-11 10:28:05.000000000","message":"Patch Set 1: Backport-Candidate+2 Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"45adc0349e6fb67e679db2c9f6b3369e15d5ef48","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-11 10:28:46.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"015ffecd1ae3bd84be02387a8b4a8e4c145f68fc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-11 12:42:08.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3d2a72530691408ca27e4fed8e3fd22d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5062d8e2292b492d87a13007f80ca5b4 : SUCCESS in 6m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7bfdfacabfc24c2c8af37a5376c34dad : SUCCESS in 5m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2569c3abb25241c098201ce096b1a578 : SUCCESS in 6m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ac361e656c4046f2b97ed14b8ff2a8ca : SUCCESS in 6m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4b0bba1e1b1a482baaf0447a9ef67c31 : SUCCESS in 11m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/423fca3795554786be8e919e72064178 : SUCCESS in 5m 35s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/5d3fa71b9c3747818b25717c068498a1 : SUCCESS in 4m 40s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/19d2141697074c3bb8aecec5a1c860fe : SUCCESS in 8m 06s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/bda29667e85d4e9fa0243de49e6dafdd : SUCCESS in 1h 27m 48s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/1d07839b5d524916b638ce8b532910a6 : SUCCESS in 1h 30m 45s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/0e832ea9f55949b396b7b58fbe849f57 : SUCCESS in 1h 02m 23s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/68536b4e86514d9085b02f8b46561ca3 : SUCCESS in 43m 52s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/3fcbd339bb9e49a7be5d49cae68b71a3 : SUCCESS in 49m 50s","accounts_in_message":[],"_revision_number":1},{"id":"1def5d9559a3bba18cb0c36f8d89178393a3ddab","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-11 12:42:08.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"c109e3d123ab9114e01f6699bf07f88e01ea5993","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-11 12:43:51.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fde3953ca19045d892db871ddffe1376\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7843bb5c6c3f414ab6ff26bfcc1155a0 : SUCCESS in 1m 23s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4db85c33bbea4e59b01b07bbcf188cb3 : SUCCESS in 47s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"96870916e0dcd9492f719b1a3ad5d7be15570310","revisions":{"96870916e0dcd9492f719b1a3ad5d7be15570310":{"kind":"REWORK","_number":1,"created":"2023-10-02 06:57:57.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/95/896995/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/95/896995/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/95/896995/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/95/896995/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/95/896995/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/95/896995/1"}}},"commit":{"parents":[{"commit":"05616ab7cc342c20ad7ed486d67b488c437bb0b9","subject":"Merge \"Update master for stable/2023.2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/05616ab7cc342c20ad7ed486d67b488c437bb0b9"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-10-02 06:50:44.000000000","tz":-240},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-10-02 06:56:41.000000000","tz":-240},"subject":"Fix error when deleting LB with broken amp","message":"Fix error when deleting LB with broken amp\n\nIf a load balancer has some broken amps (amphora was unsuccessfully\ncreated, entry exists in the DB but with missing fields), the deletion\nof the load balancer fails because octavia is calling delete_port with\nNone.\n\nCloses-Bug: #2037951\nChange-Id: I020ded6e1735ea56da46c5170292ba88ad639103\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/96870916e0dcd9492f719b1a3ad5d7be15570310"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/96870916e0dcd9492f719b1a3ad5d7be15570310"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
