)]}'
{"id":"openstack%2Foctavia~897466","triplet_id":"openstack%2Foctavia~master~I4a6e4830d0e916b1af1a11dd10097980a57d97ea","project":"openstack/octavia","branch":"master","attention_set":{},"removed_from_attention_set":{"34380":{"account":{"_account_id":34380,"name":"Maximilian Stinsky","email":"maximilian@stinsky.com","username":"mstinsky"},"last_update":"2025-07-31 17:24:26.000000000","reason":"Change was submitted"},"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2025-07-31 15:32:15.000000000","reason":"\u003cGERRIT_ACCOUNT_29244\u003e replied on the change","reason_account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}}},"hashtags":[],"change_id":"I4a6e4830d0e916b1af1a11dd10097980a57d97ea","subject":"Fix race condition in cascade delete","status":"MERGED","created":"2023-10-05 15:15:38.000000000","updated":"2025-07-31 17:25:56.000000000","submitted":"2025-07-31 17:24:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"897466","meta_rev_id":"7faba500a52d20c9702c20356fc24612b7bf4d90","_number":897466,"virtual_id_number":897466,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-07-31 17:24:25.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34380,"name":"Maximilian Stinsky","email":"maximilian@stinsky.com","username":"mstinsky"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":2,"date":"2025-07-30 16:19:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":2,"date":"2025-07-31 15:32:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34380,"name":"Maximilian Stinsky","email":"maximilian@stinsky.com","username":"mstinsky"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":1,"date":"2025-07-31 15:32:15.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34380,"name":"Maximilian Stinsky","email":"maximilian@stinsky.com","username":"mstinsky"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":2,"date":"2025-07-31 15:32:15.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34380,"name":"Maximilian Stinsky","email":"maximilian@stinsky.com","username":"mstinsky"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":34380,"name":"Maximilian Stinsky","email":"maximilian@stinsky.com","username":"mstinsky"}],"CC":[{"_account_id":5112,"name":"Seyeong Kim","email":"seyeong.kim@canonical.com","username":"xtrusia"},{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-10-05 17:42:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-10-06 13:10:24.000000000","updated_by":{"_account_id":34380,"name":"Maximilian Stinsky","email":"maximilian@stinsky.com","username":"mstinsky"},"reviewer":{"_account_id":34380,"name":"Maximilian Stinsky","email":"maximilian@stinsky.com","username":"mstinsky"},"state":"REVIEWER"},{"updated":"2024-02-29 20:10:21.000000000","updated_by":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"reviewer":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"state":"CC"},{"updated":"2025-06-24 07:13:29.000000000","updated_by":{"_account_id":5112,"name":"Seyeong Kim","email":"seyeong.kim@canonical.com","username":"xtrusia"},"reviewer":{"_account_id":5112,"name":"Seyeong Kim","email":"seyeong.kim@canonical.com","username":"xtrusia"},"state":"CC"},{"updated":"2025-07-30 16:19:57.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"1f59603c92b8e53908da7b83348a1c811ee3af46","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-10-05 15:15:38.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"28711b5f0c56d984d236a3591ba802221c42549d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-05 17:42:13.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ccf25a3be91747acaa075072915bc80c\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f464cc7b37df42bfbe839180197d507d : SUCCESS in 10m 25s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/02f4c289a534468ea4b7f837b1665404 : SUCCESS in 5m 48s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0dbc9f63f2db4cc78aaa309135a77fbf : SUCCESS in 4m 35s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f3344234de6241e68a3cd8ea80b2f158 : SUCCESS in 5m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/72cbc52ad6c54dc4abc10a04a94e08d8 : SUCCESS in 5m 59s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6b11ccc7618c432a8ab5de0449b54d15 : SUCCESS in 6m 00s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3ef67351f2454b7fbe5edc5b6ad6411c : SUCCESS in 9m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/96f88eadb59b467b9fe6af14335284a5 : SUCCESS in 5m 29s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/9811b46f909242d5875137b19b124acc : SUCCESS in 5m 35s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/44507878bfda45d8b9f82c1f384e0074 : SUCCESS in 8m 43s\n- octavia-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/a8198d65917e4d60b922bdfeefadccf7 : SUCCESS in 11m 35s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/b8df73d0123142fb8e44991a5e0e60bd : SUCCESS in 4m 57s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/d7561aa081bb419084169a3506e0803e : SUCCESS in 8m 22s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/03f6f8c8f8b24958b25c2cecf6bd99c2 : SUCCESS in 8m 00s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/d5a18b33b4694fb9b7c9a7a21426f01c : SUCCESS in 10m 25s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/988a7e890c024ffca64c5c5c759d8773 : SUCCESS in 1h 19m 57s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/691ede030b00473fb7d5b46c8bfdc70f : SUCCESS in 1h 26m 23s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/272ea7dbf09b43de9bfcfbdcdaded401 : SUCCESS in 1h 11m 04s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/f57a592b99004933989be3355a6dfc73 : SUCCESS in 1h 50m 19s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/41962006f4094bbbb0faac65374c667a : SUCCESS in 1h 13m 11s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/31e216462204442894eb9fdbd3c9dc80 : SUCCESS in 50m 38s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/48c9efc87a4a497d87aafcca0305b06d : SUCCESS in 46m 17s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/389ae5c63cb249548b5338ca398de6a2 : FAILURE in 47m 09s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/3070312b923848e28bc291912c3b6fc3 : SUCCESS in 1h 04m 22s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/08f9b6fa37df4f68922eb08eb8f044a2 : SUCCESS in 42m 42s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/6f01e2f15c5b4d6e8ea6e9e51963b869 : SUCCESS in 2h 24m 31s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/f7f049b5140f42c581a192ca713efb55 : SUCCESS in 2h 17m 58s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/c7e0acece41141d59b5207a872d39d84 : SUCCESS in 1h 41m 44s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/70378df5cd9a482996dc06e8dc135da0 : SUCCESS in 2h 18m 18s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"f21e35f5916a0eb9cb061bd767bd0b8cffdd412d","author":{"_account_id":34380,"name":"Maximilian Stinsky","email":"maximilian@stinsky.com","username":"mstinsky"},"date":"2023-10-06 13:10:24.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8d3ff2335a500ca375f2e365ae8540bf8f795c84","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-10-12 12:59:45.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"de5c0b42a887220f146ecd148dabb19ae330724d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-12 15:52:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/01d9bf57353746ae82ce82d3d854e782\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1d7db17375ce410e9f07d8ae28425be4 : SUCCESS in 12m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ce1e04047478474c850ed11d1259d03a : SUCCESS in 5m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3e4ef605470f44edbdceb240204f9787 : SUCCESS in 4m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a8d32bcdb30f4e2783432465bc736638 : SUCCESS in 5m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/264576c1da1d44ab92c04ae2d0a86ff0 : SUCCESS in 4m 36s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4a552d04d6e04894abadafb8804f3369 : SUCCESS in 5m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/389c99e942d04dd3b45e48695fa0ea27 : SUCCESS in 12m 11s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9cccd69f970c44f286ce10db600d3daa : SUCCESS in 5m 03s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/a5479c82257a415282555cfb73e813b3 : SUCCESS in 5m 00s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/5c587db360e245c8b301fe7ce2362a52 : SUCCESS in 8m 16s\n- octavia-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/4ac24f4722b944568f33abff3d2ebbe6 : SUCCESS in 10m 53s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/d6cc6caedc3646dca5e77d55d2b4f5d3 : SUCCESS in 4m 18s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/d076374d4a9f49e49291a06d57c9f19f : SUCCESS in 6m 33s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/d184a263d6ee4c948edb1bd018115297 : SUCCESS in 6m 51s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/baf01774ed8e4863ab4f69db08ab572a : SUCCESS in 6m 49s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/c9a75ffd8eb34d47aec8b849f549c929 : SUCCESS in 1h 24m 39s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/adf7fadcad2c4cd4ad3da0998d265750 : SUCCESS in 1h 41m 40s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/259cb91fe6d6430cb6816a56d02e602f : SUCCESS in 1h 01m 31s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/ca892931cc1145ab8bb11ac74ae66c17 : SUCCESS in 1h 46m 48s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/40e4a589d45b4785bc773121250f5eb7 : SUCCESS in 1h 15m 35s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/0ed816f293c64949a89ce7453aef91ef : SUCCESS in 37m 53s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/ab83d9e3bbed4c448e7fa9e31867ad78 : SUCCESS in 40m 56s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/91070682f10f4a87b4d2f356f93d5ee7 : SUCCESS in 44m 41s (non-voting)\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/54072953056f4917acd56829a53633ee : SUCCESS in 1h 06m 54s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1e2ff82195a14819a67a402c38ec18e4 : SUCCESS in 39m 44s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/8b6ff127d01946fba85c2be555add60a : SUCCESS in 2h 16m 21s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/4baee8c9f8684754bfb0549010b36f10 : SUCCESS in 2h 49m 03s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/f3993e9818274c53b68fa123f386b6d9 : SUCCESS in 2h 05m 10s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/81ef2b08f9a24a13ad137e11175fc84f : SUCCESS in 2h 31m 44s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"27e9c3368f8b6e523ad16e5acd908a7f80391dfc","author":{"_account_id":19298,"name":"Nicolas Bock","email":"nicolas.bock@canonical.com","username":"nicolasbock"},"date":"2024-02-29 20:10:21.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"803e19a6b0756714486045271c92d7f9e4fefbd7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2025-07-30 07:22:20.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"ddde375f9168134247f17449da92fbd5cd6a1cdd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-30 10:27:21.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5e9f4df34d3d4f27b2ea573e7d034c10\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/817fc6e0684f4282ad721a2f7f6ae6fd : SUCCESS in 10m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d3ea83b5e830424888739cec29b122be : SUCCESS in 4m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c9f4a48ebff149099a31177149c61031 : SUCCESS in 2m 53s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/00ac91a0ee804df59a00fb76dffc6b21 : SUCCESS in 4m 23s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/950d9114adfe474cb6dc7e735cdeb007 : SUCCESS in 4m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/db7c5119a9274de0bfe987fd41403015 : SUCCESS in 5m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d1469ecc9140488f8ef4e5ff3aac5db6 : SUCCESS in 2m 40s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/f58553a7d55c44739c931a2568e0747a : SUCCESS in 5m 03s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/e807de5c68604bb49d18db3457b54d02 : SUCCESS in 6m 59s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/f4dcafc111cb4322a6cf9bd38d02318d : SUCCESS in 2m 00s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/29d93b8d9d59435887b255af1aa22680 : SUCCESS in 4m 31s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/2dee5c64a515406d87b541b9a7d0e3b2 : SUCCESS in 4m 35s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/1e4a00364785468fb695702112476137 : SUCCESS in 59m 04s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/d9cc749b248644839934b50f0678a26e : SUCCESS in 1h 52m 07s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/af406f73102b4c6a936425df9e33845c : SUCCESS in 1h 08m 55s\n- octavia-v2-dsvm-scenario-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3065256b6bc244008b5fe31b59b5bfed : SUCCESS in 1h 24m 47s (non-voting)\n- octavia-v2-dsvm-scenario-non-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/dab6f0faec774e2fbf25b65179995c3b : SUCCESS in 1h 07m 37s (non-voting)\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/8d281a2b63484de19f826283d4c1a3de : SUCCESS in 1h 59m 08s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/02bc07daef7b487a93fefec344bb0850 : SUCCESS in 1h 03m 42s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/8505e6d7cc3c43029ccad75156791a9e : SUCCESS in 36m 53s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/4a562724968a4864a150e55b309bd913 : SUCCESS in 42m 59s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1e0e65e09bc248cb8e011443b97ae83b : SUCCESS in 35m 26s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/b51dbcf843014ff3bf2ebc831052807b : TIMED_OUT in 2h 33m 11s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/3dc5a4bfcce445abb4e4292ef833415c : TIMED_OUT in 3h 03m 54s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/3916610b956b43ccb064444d60dc2b99 : SUCCESS in 1h 43m 41s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"1e57fc184efffb5acb70bd46088f6c3d654b29ed","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2025-07-30 16:19:57.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"8b3564541ada81e126f832419c9f3ecb1c91ce42","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2025-07-31 15:32:15.000000000","message":"Patch Set 2: Backport-Candidate+2 Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"380fe86962802c34cdb0c6655f22ab080b1df3c5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-31 15:32:43.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"c7b4da2902e56eead966d22a7b4e3760fa27af11","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-31 17:24:25.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c8d486a4639a449592102569d570a2f5\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/58fb08d6b22447cb86def6cd32b4aa25 : SUCCESS in 5m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7d4d8af4e58c4fbfaaf7c054ec9f16ec : SUCCESS in 4m 17s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6d54977b6cd74c8dbe3fd3f295e01846 : SUCCESS in 5m 37s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cb56d53ce10b4c99886538795e671287 : SUCCESS in 14m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3666915c164f4d78a6099203b3c5a0e6 : SUCCESS in 4m 57s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/28006bbe590d41138cd17899e8da1490 : SUCCESS in 6m 00s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/c18e99c0de1a4c679fe64daa2d6eff1a : SUCCESS in 7m 42s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/235129ab814542a8a1eadc029a0b58aa : SUCCESS in 1h 27m 01s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/7fbed1ac89f14a5383ea0c336c593306 : SUCCESS in 1h 50m 49s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/ea4e32bd008c40f294611022ac655012 : SUCCESS in 1h 05m 36s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/73cb6f9fab064f2e9dd36dd39a30f57e : SUCCESS in 35m 04s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/185756b367504eb094ae09a0a90ff78e : SUCCESS in 42m 56s","accounts_in_message":[],"_revision_number":2},{"id":"5048b10374e7129abf8c82d22a7f750f2ff5494c","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-31 17:24:26.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"7faba500a52d20c9702c20356fc24612b7bf4d90","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-07-31 17:25:56.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/086aaa09ced14d789768dc44bde1ee25\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b382fdcfe7c04b3e88980d0fa6aea87f : SUCCESS in 51s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9a2b70f1cc0240f08c0977a66b052fbe : SUCCESS in 45s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"5c802aad949e4abd878f326c53123ddf8a299be0","revisions":{"e24725c4f96020cd345e4d8f7b9ce681b615a2de":{"kind":"REWORK","_number":1,"created":"2023-10-05 15:15:38.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/66/897466/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/66/897466/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/66/897466/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/66/897466/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/66/897466/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/66/897466/1"}}},"commit":{"parents":[{"commit":"8b275c9703fa610e6c9bb47645c4befbc848d6d4","subject":"Merge \"Fix Zuul warnings with regexps\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8b275c9703fa610e6c9bb47645c4befbc848d6d4"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-10-05 15:13:57.000000000","tz":-240},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-10-05 15:14:00.000000000","tz":-240},"subject":"Fix race condition in cascade delete","message":"Fix race condition in cascade delete\n\nupdate_vip was called multiple times concurrently when cascade deleting\na load balancer with many listeners, it may trigger a race condition\nwhen fetching, computing and updating the SGs.\nCalling update_vip for each listener is not necessary, it\u0027s now called\nonly once, that fixes the race condition and optimize the delete flow.\n\nCloses-Bug: #2038529\nChange-Id: I4a6e4830d0e916b1af1a11dd10097980a57d97ea\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e24725c4f96020cd345e4d8f7b9ce681b615a2de"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e24725c4f96020cd345e4d8f7b9ce681b615a2de"}]},"branch":"refs/heads/master"},"5c802aad949e4abd878f326c53123ddf8a299be0":{"kind":"REWORK","_number":2,"created":"2025-07-30 07:22:20.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/66/897466/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/66/897466/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/66/897466/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/66/897466/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/66/897466/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/66/897466/2"}}},"commit":{"parents":[{"commit":"9a4fb1ed882d7688c5d697d8deb4695c6def899c","subject":"Use only spaces in example response","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9a4fb1ed882d7688c5d697d8deb4695c6def899c"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-10-05 15:13:57.000000000","tz":-240},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2025-07-30 07:22:14.000000000","tz":120},"subject":"Fix race condition in cascade delete","message":"Fix race condition in cascade delete\n\nupdate_vip was called multiple times concurrently when cascade deleting\na load balancer with many listeners, it may trigger a race condition\nwhen fetching, computing and updating the SGs.\nCalling update_vip for each listener is not necessary, it\u0027s now called\nonly once, that fixes the race condition and optimize the delete flow.\n\nCloses-Bug: #2038529\nChange-Id: I4a6e4830d0e916b1af1a11dd10097980a57d97ea\nSigned-off-by: Gregory Thiemonge \u003cgthiemon@redhat.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5c802aad949e4abd878f326c53123ddf8a299be0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5c802aad949e4abd878f326c53123ddf8a299be0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
