)]}'
{"id":"openstack%2Foctavia~897519","triplet_id":"openstack%2Foctavia~master~I6b5c5d6c651dac223a205409c49085faf78cff15","project":"openstack/octavia","branch":"master","attention_set":{},"removed_from_attention_set":{"36379":{"account":{"_account_id":36379,"name":"Nguyễn Ngọc Hiếu","display_name":"ZiTechDev","email":"zitech.dev@gmail.com","username":"zitechdev201"},"last_update":"2023-10-09 02:23:38.000000000","reason":"\u003cGERRIT_ACCOUNT_36379\u003e replied on the change","reason_account":{"_account_id":36379,"name":"Nguyễn Ngọc Hiếu","display_name":"ZiTechDev","email":"zitech.dev@gmail.com","username":"zitechdev201"}},"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2023-10-09 06:50:37.000000000","reason":"\u003cGERRIT_ACCOUNT_29244\u003e replied on the change","reason_account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}}},"hashtags":[],"change_id":"I6b5c5d6c651dac223a205409c49085faf78cff15","subject":"Disable seqdiag in the spec docs","status":"MERGED","created":"2023-10-06 07:13:11.000000000","updated":"2023-10-09 10:51:22.000000000","submitted":"2023-10-09 10:48:55.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"897519","meta_rev_id":"d52b0d570250488a7d27303bed90d5812c00dba5","_number":897519,"virtual_id_number":897519,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":36379,"name":"Nguyễn Ngọc Hiếu","display_name":"ZiTechDev","email":"zitech.dev@gmail.com","username":"zitechdev201"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-10-09 10:48:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"recommended":{"_account_id":36379,"name":"Nguyễn Ngọc Hiếu","display_name":"ZiTechDev","email":"zitech.dev@gmail.com","username":"zitechdev201"},"all":[{"value":1,"date":"2023-10-09 02:23:38.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":36379,"name":"Nguyễn Ngọc Hiếu","display_name":"ZiTechDev","email":"zitech.dev@gmail.com","username":"zitechdev201"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-10-06 16:43:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":36379,"name":"Nguyễn Ngọc Hiếu","display_name":"ZiTechDev","email":"zitech.dev@gmail.com","username":"zitechdev201"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-10-06 17:43:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":36379,"name":"Nguyễn Ngọc Hiếu","display_name":"ZiTechDev","email":"zitech.dev@gmail.com","username":"zitechdev201"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":36379,"name":"Nguyễn Ngọc Hiếu","display_name":"ZiTechDev","email":"zitech.dev@gmail.com","username":"zitechdev201"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-10-06 07:56:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-10-06 10:00:49.000000000","updated_by":{"_account_id":36379,"name":"Nguyễn Ngọc Hiếu","display_name":"ZiTechDev","email":"zitech.dev@gmail.com","username":"zitechdev201"},"reviewer":{"_account_id":36379,"name":"Nguyễn Ngọc Hiếu","display_name":"ZiTechDev","email":"zitech.dev@gmail.com","username":"zitechdev201"},"state":"REVIEWER"},{"updated":"2023-10-06 16:43:46.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"1dffe5c80635430f28543eb9e3673920fd44d341","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-10-06 07:13:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6ab71647c8d1595220bfff643722abd7f1500918","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-10-06 07:15:39.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"1937f83421cd096521b27987646f404806f39685","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-06 07:56:27.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1ed729459c8d44dc9eba6ac1e1c5d934\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/ffa8cae7ab5448dc87166b6c4ac99399 : SUCCESS in 4m 57s\n- requirements-check https://zuul.opendev.org/t/openstack/build/f86315b5f43f42f19521170cb3b7fcc9 : SUCCESS in 2m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4f3810c5ccb74c0897de78753b6cea26 : SUCCESS in 5m 02s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bc041cdbca6d45ff990434bf83e3118b : SUCCESS in 10m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4438946b24e7494a84043be932185881 : SUCCESS in 5m 25s","accounts_in_message":[],"_revision_number":2},{"id":"53d6154e2fd762ed39383c6193a74d47f437bc5d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-10-06 08:10:29.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"fc92450600a9d3d548c88846e78281e5dc9fbc42","author":{"_account_id":36379,"name":"Nguyễn Ngọc Hiếu","display_name":"ZiTechDev","email":"zitech.dev@gmail.com","username":"zitechdev201"},"date":"2023-10-06 10:00:49.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"d7fc1d88473e7e8af35139ef6e25257771d7fe44","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-06 10:41:24.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3665aeba8bac4231aad15f8728c2ccba\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/81f6f6f80c14459da70bb3efdeb745cf : SUCCESS in 5m 41s\n- requirements-check https://zuul.opendev.org/t/openstack/build/e1c3ee24e67c483d9cbbe5530f78596e : SUCCESS in 2m 11s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ff60f1e49b9e48969d5213d1deb3d765 : SUCCESS in 11m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5e49320049d74d6fbaea945e7e039084 : SUCCESS in 5m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/db829ac714d1487cb45e5953629c9b15 : SUCCESS in 5m 36s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0e9ba1b5f3b54ff8b224bdc1b4290d46 : SUCCESS in 4m 52s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/42736f49f37d45298990043bc198d9dd : SUCCESS in 4m 51s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1f0afed234d34a1c8e9cabd943e65d41 : SUCCESS in 4m 49s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e4416345523a4fd0b014c9b5711ad11b : SUCCESS in 12m 07s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8a9087a6121e416c9c0087e3318ebdec : SUCCESS in 4m 58s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/b82c1cd1b2c643a693e70691375f6226 : SUCCESS in 5m 23s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/ce217bc74e2840e08073cd4a32c0b8dc : SUCCESS in 8m 15s\n- octavia-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/747262da3db9406087781322f4a24097 : SUCCESS in 9m 38s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/4b3ee122759a456fb6411207be27da80 : SUCCESS in 4m 47s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/2512a8ddaea24f10b10e3ed154d1724b : SUCCESS in 7m 37s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/b8ec55955303456686f69df76837474b : SUCCESS in 8m 13s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/61771fce438b449a8d70feaf1b61846c : SUCCESS in 6m 38s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/b64f076b905b4c20af5e70c5c81fc97b : SUCCESS in 59m 19s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/275131b852b94da88bebed096658fbb3 : SUCCESS in 1h 25m 26s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/61f420c526814484a519fa5b94d9e819 : SUCCESS in 1h 07m 12s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/332f8b9320aa438493f8156e0beb1e46 : SUCCESS in 1h 29m 18s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/c1170d22bee443eb9a559065b2cff91c : SUCCESS in 58m 00s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/0da44875194a40b198bb065a7b3c7873 : SUCCESS in 45m 20s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/07f37f1e150940488a39de3cffe5ee7c : SUCCESS in 37m 23s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/d1c03f4487f84ce5b6c6ad1422e80523 : FAILURE in 36m 47s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/b4ed99ce8a0b479b89e8374988da4114 : SUCCESS in 1h 10m 21s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/75a3761ce72a40038b1522ab30515c4a : SUCCESS in 39m 56s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/2ec3ca6b610340b99b55025591bccf0b : SUCCESS in 1h 48m 30s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/afa2e4cec43e4ab29d6ecb64086c9de5 : SUCCESS in 2h 29m 24s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/1e9d3ada897c4d5dab0bb43ca34f5eab : FAILURE in 1h 52m 04s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/ed7f16ddcaf34f29a01013cd7155419e : SUCCESS in 2h 15m 35s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"489d91a44dc898b95e26932279f33ca87d4cfe7b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-10-06 13:03:19.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"8aed0b77dc528749327088d14b509bb3d567fbab","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-06 15:26:38.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8746b799603c4f669429f592f2f0910a\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/1b6229ebe2224d2787ecdcb154c6fdd0 : SUCCESS in 5m 06s\n- requirements-check https://zuul.opendev.org/t/openstack/build/c6a17bdff6f04a77a412e0944150ebcf : SUCCESS in 2m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e211afd2c93c49d8b9e12d0943128be0 : SUCCESS in 9m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/882099d67ad94562be280029dbe221b8 : SUCCESS in 4m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/571e3e3d4315427ebe7fecce0937f09e : SUCCESS in 4m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0e7f16ea6db84708867dd5fd985b1951 : SUCCESS in 5m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f1a0c08853e94274b536e73d8e3ff421 : SUCCESS in 4m 42s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/959f0f266ec844838b1767c93b9c96ea : SUCCESS in 5m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5b191f5c569c426c9d06152101c6e7b5 : SUCCESS in 9m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a2f9f7c35f75466cb8e0c1af113c04ba : SUCCESS in 5m 13s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/4145bfce1c2c41ae82d5e87b45b09bf6 : SUCCESS in 5m 02s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/af50fc7e1e8a481eb0601241fd218f46 : SUCCESS in 7m 05s\n- octavia-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/b491297c0ebe4848ab8092e3d878bf1d : SUCCESS in 12m 05s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/114f7e00f94d4f5d8ffb38198fe42d8c : SUCCESS in 4m 40s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/57d527991fe441479fde668f791dd2de : SUCCESS in 7m 11s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/9f10b36396c9477ebe426af111686b7c : SUCCESS in 6m 24s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/95a751a605a94ede9190bdc46c8935a4 : SUCCESS in 7m 47s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/fd09318ee4bb49beaf214e04c5a510c4 : SUCCESS in 1h 30m 53s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/0f3742025d9a490e91ad4fb51f473b31 : SUCCESS in 1h 37m 11s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/64d1a65b78f04c86bb0742dcb5d85758 : SUCCESS in 1h 16m 09s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/1174a32f319247998702f50866d858fa : SUCCESS in 1h 43m 32s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/b93524fc55fa4e9cbd8e1bd5f64eb1f9 : SUCCESS in 1h 23m 42s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/ffb5683ed1764ae9a5690c23fe21b7da : SUCCESS in 41m 17s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/d76f605a52ff405fb3a407ac19e7bf04 : SUCCESS in 49m 38s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/c3e9947cca564f8493cad9a0f53d2ef7 : FAILURE in 42m 57s (non-voting)\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/b517f0f4465d449a91a242a255f1cdde : SUCCESS in 1h 06m 57s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a6760c5dc2194abba0b8eba8ffdada6f : SUCCESS in 37m 18s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/607108d594d74b4fba71773b8072b0b4 : SUCCESS in 1h 48m 11s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/9331442c960f44a5a07186afca15527b : FAILURE in 2h 21m 20s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/2500370690854a36be8a632aaa6304f3 : SUCCESS in 1h 39m 25s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/f6177128404943439186d760dc8953a8 : SUCCESS in 2h 12m 05s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"c7731ff501040e70cddcb3058e50092c1d169756","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-10-06 16:43:46.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"325da823e00bf18053111d5b2d275890d2ea6728","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-10-06 17:43:07.000000000","message":"Patch Set 4: Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"6615472de6c65e3e525dfd5b52823f3aee3278ec","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-06 17:44:31.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"43e92f38461cf4e7e2b248a3c21cb5173eed13a4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-06 18:48:08.000000000","message":"Patch Set 4: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6068d9769838495ba4b940efaa4873c5\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/d1c2722b9ace4de8a584eec5865b9fa5 : SUCCESS in 4m 50s\n- requirements-check https://zuul.opendev.org/t/openstack/build/6f30b826f0f64d508f4b4422c247c291 : SUCCESS in 2m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/016ed4313d7441e183917558cac7b197 : SUCCESS in 4m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/99c3c777a17944318db9ae756ed5145b : SUCCESS in 4m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e42a072f8e744e4394a0c7d789d97f75 : SUCCESS in 4m 23s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d35348d131da48d7b6673a082da96c1d : SUCCESS in 5m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/081ee1ff5b34421e9231344175a2d088 : SUCCESS in 11m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/11817fba524c4fb5be105f5e7e3d36d0 : SUCCESS in 4m 59s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/ae1424b7ac774ad289f41b10bc9aa521 : SUCCESS in 4m 59s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/ab54d80aac564ed0bce3aac78a07fa95 : SUCCESS in 6m 48s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/db40862ef858496caacaf7167ae825f4 : CANCELED\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/213f9498f9ec47d99045464f68337f98 : CANCELED\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/5753eeeb674f43598e9b0e5cf8714520 : FAILURE in 1h 02m 49s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/93942369f7f24e32a32acada5a8c0441 : SUCCESS in 39m 47s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/670f7fc242444d3bb293f158e8287033 : SUCCESS in 44m 00s","accounts_in_message":[],"_revision_number":4},{"id":"1b4ee298319aa61c8d087c2935004845f0cb1cba","author":{"_account_id":36379,"name":"Nguyễn Ngọc Hiếu","display_name":"ZiTechDev","email":"zitech.dev@gmail.com","username":"zitechdev201"},"date":"2023-10-09 02:23:38.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"a1baa61b1d0d5115ab3ca05265c3dd23f28c8709","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-10-09 06:50:37.000000000","message":"Patch Set 4:\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"e01a6f5448f1bc1d624b0514a89eea4e610b5cca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-09 09:27:36.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c3938e2b1cab4726862cd504f76ded1a\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/1aab72f2c96740c79e893a2843219b7b : SUCCESS in 6m 13s\n- requirements-check https://zuul.opendev.org/t/openstack/build/785eba5f9e6f4473a0953297f01eaaee : SUCCESS in 1m 59s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/91c8413bcbad4ec0a75a68f630035e61 : SUCCESS in 11m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7fc48c73aa09409f8f916be6ea4a7090 : SUCCESS in 4m 27s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5a5ac9cdf1304ded8995962468a7d04e : SUCCESS in 4m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/39c7dfe07f894e6e8c05d79a81a39dd5 : SUCCESS in 4m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/84b1f9e3d1344e43b9e647ce0986e30f : SUCCESS in 4m 38s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/dafa2f80155f499494e1ac6db452711b : SUCCESS in 5m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/253aaa0c7e86416ca774465f8c58e2f5 : SUCCESS in 9m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3b7cf605bc7e4b3da8f2b6cf10e36f48 : SUCCESS in 5m 20s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/950074e99ae1463898ccb4103b74949f : SUCCESS in 4m 56s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/3fe623207bf8493c896e0cd4081be2ed : SUCCESS in 7m 30s\n- octavia-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/9858c117273c474a82c5534acea6359d : SUCCESS in 11m 19s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/99ac809912694b4f9995eb59ba8ff491 : SUCCESS in 4m 42s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/fc4ae81c4ca544e19e099cacbc173d9d : SUCCESS in 7m 56s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/4bdd84f628db42b8bd0cfdd5a2965940 : SUCCESS in 7m 25s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/962594fa2dfd4f889e37239e401888dd : SUCCESS in 7m 39s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/47994a3ce696450890f55b72c0bff817 : SUCCESS in 1h 21m 06s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/a6ba2d1487fc4d4fa9011c02706fbb48 : SUCCESS in 1h 35m 04s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/ac1d5eb57aab4124b6dc5410869f15b6 : SUCCESS in 1h 14m 23s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/b05afcf8c68a4f97a6723a7a91ed0dac : SUCCESS in 1h 53m 42s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/cf96bd7f3f7a4d249aaa27044e520d54 : SUCCESS in 1h 22m 49s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/56286c09c9064490a9b8b954d976172f : SUCCESS in 50m 16s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/eac2733a6b7d403dab6aabf66d61d8a9 : SUCCESS in 57m 14s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/0dee15d1cbbc450babdc59f700b5077f : FAILURE in 1h 02m 44s (non-voting)\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/76043738bd6d407dbd8947760dc76c18 : SUCCESS in 1h 12m 57s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/19a5669a98b045eb89572471ef91d63c : SUCCESS in 41m 30s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/5e35fd7993994d07ae3935ea10bedcff : SUCCESS in 2h 03m 41s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/2c5e9c9b261444579bdaaa732be532a1 : FAILURE in 2h 10m 03s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/4280d726c57f4f8d95859a60997627b3 : SUCCESS in 1h 46m 28s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/c292c5fcf1d24dd7b44c278f598b283c : TIMED_OUT in 2h 34m 39s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"d3b344873511968a5c85b2d38095ab52ea93c9f9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-09 09:28:20.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"4587c8e0018c1e101b3af1f3a5aaf044bf71f0ff","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-09 10:48:55.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e498b09ad65646e59010ea5b8abbfc34\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/36d537882cdf4b289a4be2b1c25a1644 : SUCCESS in 4m 51s\n- requirements-check https://zuul.opendev.org/t/openstack/build/2f8a95fb73574aad8dc45b2cd6f55f9f : SUCCESS in 2m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b8e30f7db9be4d21a4c57a1b54c9bb71 : SUCCESS in 4m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e9a15118400e496e8ea0946745026dd6 : SUCCESS in 4m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a480eef6c68e4c709ab546462c93eb0f : SUCCESS in 4m 35s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6bc13d77598242249f4b2b4dd05815ee : SUCCESS in 4m 36s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/210810042cee464db6ad8700e49c9610 : SUCCESS in 12m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/534649e9531442ff8481342481737468 : SUCCESS in 5m 10s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/f619c4dfcd2e4df69b58f7860a038866 : SUCCESS in 4m 16s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/866e966077bd44909fc271ab4fa4d392 : SUCCESS in 7m 47s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/28a821212e1b47b490b9f59f1b78f6e7 : SUCCESS in 1h 20m 27s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/06e3f1274c4e49a1bebc36c2f4893727 : SUCCESS in 1h 17m 07s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/1947cb401c964a7c9bcaf67dcc7d8ff5 : SUCCESS in 55m 29s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/f9d15099a836427984f1f524d94a9610 : SUCCESS in 36m 18s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/801aea7b72a546db94db04678434c735 : SUCCESS in 40m 32s","accounts_in_message":[],"_revision_number":4},{"id":"39b8873764ed575f7c0b84db264f3ae721f17ac3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-09 10:48:55.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"d52b0d570250488a7d27303bed90d5812c00dba5","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-09 10:51:22.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d4f4eb2baf624889bd4586637880fec3\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/6765df6815854538a4ef569d60401c6b : SUCCESS in 39s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9aee17df19024d2cb11a75597b41cd99 : SUCCESS in 59s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d540bb2a462f4605b8f2334c1b0665ee : SUCCESS in 44s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"c4a162ab40023884d33ecb3cfe3f47801d6023bc","revisions":{"238f69530401ef063d6bf4607df49987399d4aec":{"kind":"REWORK","_number":1,"created":"2023-10-06 07:13:11.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/19/897519/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/19/897519/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/19/897519/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/19/897519/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/19/897519/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/19/897519/1"}}},"commit":{"parents":[{"commit":"8b275c9703fa610e6c9bb47645c4befbc848d6d4","subject":"Merge \"Fix Zuul warnings with regexps\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8b275c9703fa610e6c9bb47645c4befbc848d6d4"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-10-06 07:12:17.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-10-06 07:12:23.000000000","tz":120},"subject":"Disable seqdiag in the spec docs","message":"Disable seqdiag in the spec docs\n\nThe tls-data-security specs use seqdiag to describe communication between\nsome components.\nseqdiag is unmaintained and doesn\u0027t work with recent Pillow releases (10.0.1).\nDisable the generation of those diagrams as they are part of old specs.\n\nChange-Id: I6b5c5d6c651dac223a205409c49085faf78cff15\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/238f69530401ef063d6bf4607df49987399d4aec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/238f69530401ef063d6bf4607df49987399d4aec"}]},"branch":"refs/heads/master"},"995d51582bd6a0d2fea3105f5e7aa58b5389c675":{"kind":"REWORK","_number":2,"created":"2023-10-06 07:15:39.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/19/897519/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/19/897519/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/19/897519/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/19/897519/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/19/897519/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/19/897519/2"}}},"commit":{"parents":[{"commit":"8b275c9703fa610e6c9bb47645c4befbc848d6d4","subject":"Merge \"Fix Zuul warnings with regexps\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8b275c9703fa610e6c9bb47645c4befbc848d6d4"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-10-06 07:12:17.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-10-06 07:15:20.000000000","tz":120},"subject":"Disable seqdiag in the spec docs","message":"Disable seqdiag in the spec docs\n\nThe tls-data-security specs use seqdiag to describe communication between\nsome components.\nseqdiag is unmaintained and doesn\u0027t work with recent Pillow releases (10.0.1).\nDisable the generation of those diagrams as they are part of old specs.\n\nAlso remove the dependency on seqdiag\n\nChange-Id: I6b5c5d6c651dac223a205409c49085faf78cff15\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/995d51582bd6a0d2fea3105f5e7aa58b5389c675"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/995d51582bd6a0d2fea3105f5e7aa58b5389c675"}]},"branch":"refs/heads/master"},"0f25422341f1dbce13eabf86b912e637227ef567":{"kind":"REWORK","_number":3,"created":"2023-10-06 08:10:29.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/19/897519/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/19/897519/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/19/897519/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/19/897519/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/19/897519/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/19/897519/3"}}},"commit":{"parents":[{"commit":"8b275c9703fa610e6c9bb47645c4befbc848d6d4","subject":"Merge \"Fix Zuul warnings with regexps\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8b275c9703fa610e6c9bb47645c4befbc848d6d4"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-10-06 07:12:17.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-10-06 08:08:07.000000000","tz":120},"subject":"Disable seqdiag in the spec docs","message":"Disable seqdiag in the spec docs\n\nThe tls-data-security specs use seqdiag to describe communication between\nsome components.\nseqdiag is unmaintained and doesn\u0027t work with recent Pillow releases (10.0.1).\nDisable the generation of those diagrams as they are part of old specs.\n\nAlso remove the dependency on seqdiag\n\nPartial-Bug: #2026345\n\nChange-Id: I6b5c5d6c651dac223a205409c49085faf78cff15\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0f25422341f1dbce13eabf86b912e637227ef567"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0f25422341f1dbce13eabf86b912e637227ef567"}]},"branch":"refs/heads/master"},"c4a162ab40023884d33ecb3cfe3f47801d6023bc":{"kind":"REWORK","_number":4,"created":"2023-10-06 13:03:19.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/19/897519/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/19/897519/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/19/897519/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/19/897519/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/19/897519/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/19/897519/4"}}},"commit":{"parents":[{"commit":"8b275c9703fa610e6c9bb47645c4befbc848d6d4","subject":"Merge \"Fix Zuul warnings with regexps\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8b275c9703fa610e6c9bb47645c4befbc848d6d4"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-10-06 07:12:17.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-10-06 13:02:46.000000000","tz":120},"subject":"Disable seqdiag in the spec docs","message":"Disable seqdiag in the spec docs\n\nThe tls-data-security specs use seqdiag to describe communication between\nsome components.\nseqdiag is unmaintained and doesn\u0027t work with recent Pillow releases (10.0.1).\nDisable the generation of those diagrams as they are part of old specs.\n\nAlso remove the dependency on seqdiag\n\nDisable octavia-grenade-skip-level which will be fixed in another commit\n\nPartial-Bug: #2026345\n\nChange-Id: I6b5c5d6c651dac223a205409c49085faf78cff15\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c4a162ab40023884d33ecb3cfe3f47801d6023bc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c4a162ab40023884d33ecb3cfe3f47801d6023bc"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":36379,"name":"Nguyễn Ngọc Hiếu","display_name":"ZiTechDev","email":"zitech.dev@gmail.com","username":"zitechdev201"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
