)]}'
{"id":"openstack%2Foctavia~897768","triplet_id":"openstack%2Foctavia~stable%2F2023.2~Ice5aff062a771de65c68a52754dacce6f0ba7907","project":"openstack/octavia","branch":"stable/2023.2","attention_set":{},"removed_from_attention_set":{"31664":{"account":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"last_update":"2023-10-10 14:22:42.000000000","reason":"\u003cGERRIT_ACCOUNT_31664\u003e replied on the change","reason_account":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}},"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2023-10-10 15:52:53.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ice5aff062a771de65c68a52754dacce6f0ba7907","subject":"[stable/2023.2] Remove octavia-grenade-skip-level job","status":"MERGED","created":"2023-10-10 08:03:13.000000000","updated":"2023-10-10 15:54:59.000000000","submitted":"2023-10-10 15:52:53.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"897768","meta_rev_id":"e1621e084fdf09dcfc02d1e4d4dc289c5ad20953","_number":897768,"virtual_id_number":897768,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-10-10 15:52:53.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"recommended":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"all":[{"value":1,"date":"2023-10-10 14:22:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":2,"date":"2023-10-10 14:22:57.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-10-10 08:07:56.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":1,"date":"2023-10-10 14:22:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-10-10 08:07:56.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"},{"updated":"2023-10-10 10:10:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-10-10 10:27:24.000000000","updated_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"reviewer":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"state":"CC"},{"updated":"2023-10-10 14:22:42.000000000","updated_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"reviewer":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"state":"REVIEWER"}],"messages":[{"id":"6a0daa6888a50e3ff9cfc371dd798c1842e1e28c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-10-10 08:03:13.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"be921c66a723a7b30fb530c74dee8d67d7df2c1f","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-10-10 08:07:56.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"ee8290863889262fa73e914d2eddfb13860eaace","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-10 10:10:24.000000000","message":"Patch Set 1: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1239207e16cd4f1a9ab3ffd20a4ad0a2\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f70378d59f1c442ba18cde81d865a896 : SUCCESS in 11m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b53de4d72bbe483b8875888b557cec46 : SUCCESS in 5m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4475325dcbc74b3aa2d997f50f287eee : SUCCESS in 5m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fd1989586dba4ed686d6a0832416afeb : SUCCESS in 5m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4c1bc1b947f24c8ba474e304f43d76b9 : SUCCESS in 6m 16s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/45a6ac60719a4da2bd31b637acab9bf4 : SUCCESS in 5m 42s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/04dee6b4e7e74325991a51628106d4f9 : SUCCESS in 11m 09s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/12d14f9b51644e18a83d6ca72d27b2ee : SUCCESS in 5m 13s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/34bde82baa2a42fe891aaf1c998dc36b : SUCCESS in 7m 47s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/738a0469fed1404895eb64b3bc956dce : SUCCESS in 5m 24s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/3145c1363335454abaa1dd26c65a2621 : SUCCESS in 6m 51s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/64327a4f824041f99a03d65227d5471e : SUCCESS in 7m 53s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/ebc45e0223fb40d78327e1846e9c7dfd : SUCCESS in 8m 15s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/eb8d2542c6144d09baf84d6ffc662c74 : SUCCESS in 1h 28m 46s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/088a7fecd08e4d04b61076d6efc2a616 : SUCCESS in 1h 31m 31s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/1fe585720e49430b8bc10e3313e931e9 : FAILURE in 52m 58s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/0ab033e8c3b74d6486524839aa9aafc0 : SUCCESS in 1h 33m 49s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/b6b1b401b8b04fdf85f8260589068446 : SUCCESS in 1h 08m 41s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/251d243d072d4fcbb54662504120982e : SUCCESS in 39m 39s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/1f292fae670c454089f6e953ac027dda : SUCCESS in 45m 26s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/95f58a1d87dc41bb81818a1e9b63bcae : FAILURE in 10m 43s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b242af5f54c64407a4a25669e875167f : SUCCESS in 31m 11s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/7b8ba67a5a334e6d8c0bf57bd2fc88c6 : SUCCESS in 2h 02m 35s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/65c157543a7f401da0cab92661bc9261 : SUCCESS in 1h 52m 31s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/a9399154504041ef835a90c4cf9a379f : SUCCESS in 1h 45m 37s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/39f5ff3e51c74d55a35507e48da984bf : FAILURE in 9m 45s (non-voting)\n\nWarning:\n  Zuul encountered a deprecated syntax while parsing its configuration\nin the repo openstack/octavia on branch stable/2023.2.  The problem\nwas:\n\n  All regular expressions must conform to RE2 syntax, but an\n  expression using the deprecated Perl-style syntax has been detected.\n  Adjust the configuration to conform to RE2 syntax.\n  \n  The RE2 syntax error is: invalid perl operator: (?!\n\nThe problem appears in the following job stanza:\n\n  job:\n      name: octavia-tox-functional-py39-sqlalchemy-tips\n      parent: openstack-tox-functional-py39\n      description: |\n        Run tox python 3.9 functional against master of sqlalchemy and related projects\n      # The job only tests the latest and shouldn\u0027t be run on the stable branches\n      branches: ^(?!stable)\n      required-projects:\n        - name: openstack/oslo.db\n        - name: github.com/sqlalchemy/sqlalchemy\n          override-checkout: main\n        - name: github.com/sqlalchemy/alembic\n          override-checkout: main\n\n  in \"openstack/octavia/zuul.d/jobs.yaml@stable/2023.2\", line 161, column 3","accounts_in_message":[],"_revision_number":1},{"id":"5936ea2ae3b525b4e4c7f4f474559d815ad7886a","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2023-10-10 10:27:24.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"479a4fce8e7d63d2dc9429a80c94875e07516a47","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2023-10-10 11:31:56.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"87c9efcdeab7aa637d90a654f3c9a1bcf184ea48","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-10-10 11:40:00.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"2cec8209a5bd79dda1334c144ffdd95de95a4ec6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-10 14:13:47.000000000","message":"Patch Set 1: Verified+1\n\n(2 comments)\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3d6174094eac4da49ea80e4e14541e2c\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f32ba5d3c15e44c0aac2e4229c3c24d6 : SUCCESS in 12m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/19cd5b6649f245d78e9b2951a2588185 : SUCCESS in 5m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/66456f7a0b1647ee9f0a0c7e301a8668 : SUCCESS in 5m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3669664a2f804dc697f944e993387e99 : SUCCESS in 4m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b4d72a85eda14562bd6329f3fec9d8f5 : SUCCESS in 4m 49s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d4059cbadd1c4fe28038562911c3f3a0 : SUCCESS in 4m 56s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ef2896efde794d2991cdb9e6b687ed55 : SUCCESS in 12m 36s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/e78cf85844704e32bfb8ee9a3e4060fc : SUCCESS in 5m 02s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/acd1b6e4120147a29047151884930cba : SUCCESS in 7m 45s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/f76b0a143ec24cd38726a2d4584d6223 : SUCCESS in 4m 48s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/9adbc9b63f04472e9d89b58b5702089f : SUCCESS in 7m 27s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/eb44a7fa7e8c4fc5b68a24ff278c304f : SUCCESS in 8m 01s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/52aaa103a3db4f888a9e5bd738ea8290 : SUCCESS in 6m 36s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/ac284c447ddc4de09c97bf91fb99628e : SUCCESS in 57m 46s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/aa0c0d6159eb42b78581aab380d47379 : SUCCESS in 1h 19m 59s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/34bcff4cfc514a6ba0e9c0ac10a1b46e : SUCCESS in 59m 09s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/74bb70b40e23494aa27db46c487b4164 : SUCCESS in 1h 19m 33s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/8f5181e778e14892be64493e77d7d71d : SUCCESS in 1h 01m 38s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/f331399edb1f4558a2f80aac0a7f3232 : SUCCESS in 40m 34s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/fb57c7f6c88c49c3aa9ebf803df3ac56 : SUCCESS in 40m 17s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/29e1e9f130d543058e870556e9f96e72 : FAILURE in 12m 49s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3b98f0b57c2c4f838c90da7b9707996d : SUCCESS in 36m 34s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/c7c9290135464f9097b534148c652a87 : SUCCESS in 1h 57m 31s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/19ad9a14544d401b8c9e6faa4ded5494 : SUCCESS in 2h 39m 31s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/04d8028cccb1479ca151316ae2df7bff : TIMED_OUT in 2h 34m 54s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/92659ad0c4cf4b91b72bcc13228e483b : FAILURE in 11m 19s (non-voting)\n\nWarning:\n  Zuul encountered a deprecated syntax while parsing its configuration\nin the repo openstack/octavia on branch stable/2023.2.  The problem\nwas:\n\n  All regular expressions must conform to RE2 syntax, but an\n  expression using the deprecated Perl-style syntax has been detected.\n  Adjust the configuration to conform to RE2 syntax.\n  \n  The RE2 syntax error is: invalid perl operator: (?!\n\nThe problem appears in the following job stanza:\n\n  job:\n      name: octavia-tox-functional-py39-sqlalchemy-tips\n      parent: openstack-tox-functional-py39\n      description: |\n        Run tox python 3.9 functional against master of sqlalchemy and related projects\n      # The job only tests the latest and shouldn\u0027t be run on the stable branches\n      branches: ^(?!stable)\n      required-projects:\n        - name: openstack/oslo.db\n        - name: github.com/sqlalchemy/sqlalchemy\n          override-checkout: main\n        - name: github.com/sqlalchemy/alembic\n          override-checkout: main\n\n  in \"openstack/octavia/zuul.d/jobs.yaml@stable/2023.2\", line 161, column 3","accounts_in_message":[],"_revision_number":1},{"id":"e230539f3aee85748e1ee1c785be5c7e2c0f0f48","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2023-10-10 14:22:42.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"251d29bcf9ccf40a0401a6bb14e6460e4ae9d16a","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-10-10 14:22:57.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"817b3d2498f6f3f9f7bef7fce584edf62d330273","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-10 14:23:42.000000000","message":"Patch Set 1: -Verified\n\n(2 comments)\n\nStarting gate jobs.\nWarning:\n  Zuul encountered a deprecated syntax while parsing its configuration\nin the repo openstack/octavia on branch stable/2023.2.  The problem\nwas:\n\n  All regular expressions must conform to RE2 syntax, but an\n  expression using the deprecated Perl-style syntax has been detected.\n  Adjust the configuration to conform to RE2 syntax.\n  \n  The RE2 syntax error is: invalid perl operator: (?!\n\nThe problem appears in the following job stanza:\n\n  job:\n      name: octavia-tox-functional-py39-sqlalchemy-tips\n      parent: openstack-tox-functional-py39\n      description: |\n        Run tox python 3.9 functional against master of sqlalchemy and related projects\n      # The job only tests the latest and shouldn\u0027t be run on the stable branches\n      branches: ^(?!stable)\n      required-projects:\n        - name: openstack/oslo.db\n        - name: github.com/sqlalchemy/sqlalchemy\n          override-checkout: main\n        - name: github.com/sqlalchemy/alembic\n          override-checkout: main\n\n  in \"openstack/octavia/zuul.d/jobs.yaml@stable/2023.2\", line 161, column 3","accounts_in_message":[],"_revision_number":1},{"id":"b46b81f7fddc12938d88b9917f3b7639ca1e6a0e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-10 15:52:53.000000000","message":"Patch Set 1: Verified+2\n\n(2 comments)\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c48ead1dde1548e09d3f707dad294279\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2f4df84fe2fb4f2e9cff911fcc8f757f : SUCCESS in 6m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1533a697198b4a02bc419c31b555f67d : SUCCESS in 6m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/72251b4d202a409c89d08a679dd9b6e7 : SUCCESS in 6m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c166ae7df98845cea72a536e671e8961 : SUCCESS in 5m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7fda27828b334270b399b689c2997f9a : SUCCESS in 11m 11s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/a8251d8a84b947338919037cb6e31a24 : SUCCESS in 6m 03s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/71d6b87a044b42589f436e40b5141754 : SUCCESS in 7m 48s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/901d8b3438d2477d939873b61d8954ad : SUCCESS in 1h 19m 42s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/2b265816616e4e0e8d4b430ab64d1f79 : SUCCESS in 1h 27m 29s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/429f70b7215c4480903df3576ca52586 : SUCCESS in 1h 12m 34s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/448b53de3ae34c02b2b0405a09d76e16 : SUCCESS in 43m 05s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/e5f606cf793b4c68bd64ac082e59eee9 : SUCCESS in 45m 53s\n\nWarning:\n  Zuul encountered a deprecated syntax while parsing its configuration\nin the repo openstack/octavia on branch stable/2023.2.  The problem\nwas:\n\n  All regular expressions must conform to RE2 syntax, but an\n  expression using the deprecated Perl-style syntax has been detected.\n  Adjust the configuration to conform to RE2 syntax.\n  \n  The RE2 syntax error is: invalid perl operator: (?!\n\nThe problem appears in the following job stanza:\n\n  job:\n      name: octavia-tox-functional-py39-sqlalchemy-tips\n      parent: openstack-tox-functional-py39\n      description: |\n        Run tox python 3.9 functional against master of sqlalchemy and related projects\n      # The job only tests the latest and shouldn\u0027t be run on the stable branches\n      branches: ^(?!stable)\n      required-projects:\n        - name: openstack/oslo.db\n        - name: github.com/sqlalchemy/sqlalchemy\n          override-checkout: main\n        - name: github.com/sqlalchemy/alembic\n          override-checkout: main\n\n  in \"openstack/octavia/zuul.d/jobs.yaml@stable/2023.2\", line 161, column 3","accounts_in_message":[],"_revision_number":1},{"id":"c82dc7b53a4676cb52d3eca766d305cf1d827a70","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-10 15:52:53.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"e1621e084fdf09dcfc02d1e4d4dc289c5ad20953","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-10 15:54:59.000000000","message":"Patch Set 1:\n\n(2 comments)\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/567e050160a04e0ebca2e0ee7843bb7f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c07f10921e844a3699eb58b8df64e2c3 : SUCCESS in 40s\n\nWarning:\n  Zuul encountered a deprecated syntax while parsing its configuration\nin the repo openstack/octavia on branch stable/2023.2.  The problem\nwas:\n\n  All regular expressions must conform to RE2 syntax, but an\n  expression using the deprecated Perl-style syntax has been detected.\n  Adjust the configuration to conform to RE2 syntax.\n  \n  The RE2 syntax error is: invalid perl operator: (?!\n\nThe problem appears in the following job stanza:\n\n  job:\n      name: octavia-tox-functional-py39-sqlalchemy-tips\n      parent: openstack-tox-functional-py39\n      description: |\n        Run tox python 3.9 functional against master of sqlalchemy and related projects\n      # The job only tests the latest and shouldn\u0027t be run on the stable branches\n      branches: ^(?!stable)\n      required-projects:\n        - name: openstack/oslo.db\n        - name: github.com/sqlalchemy/sqlalchemy\n          override-checkout: main\n        - name: github.com/sqlalchemy/alembic\n          override-checkout: main\n\n  in \"openstack/octavia/zuul.d/jobs.yaml@stable/2023.2\", line 161, column 3","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"a04e0f5205511192833fcad008544122cc2b38ee","revisions":{"a04e0f5205511192833fcad008544122cc2b38ee":{"kind":"REWORK","_number":1,"created":"2023-10-10 08:03:13.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/68/897768/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/68/897768/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/68/897768/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/68/897768/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/68/897768/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/68/897768/1"}}},"commit":{"parents":[{"commit":"853d0f69e43a16e675103b515a82ec90d83e222f","subject":"Update .gitreview for stable/2023.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/853d0f69e43a16e675103b515a82ec90d83e222f"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-10-10 08:02:57.000000000","tz":120},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-10-10 08:02:57.000000000","tz":120},"subject":"[stable/2023.2] Remove octavia-grenade-skip-level job","message":"[stable/2023.2] Remove octavia-grenade-skip-level job\n\ngrenade-skip-level is not supported on .2 branches\nRemove it from the job list\n\nChange-Id: Ice5aff062a771de65c68a52754dacce6f0ba7907\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a04e0f5205511192833fcad008544122cc2b38ee"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a04e0f5205511192833fcad008544122cc2b38ee"}]},"branch":"refs/heads/stable/2023.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
