)]}'
{"id":"openstack%2Foctavia~897772","triplet_id":"openstack%2Foctavia~stable%2F2023.2~I458dd6d6f5383edc24116ea0fa27e3a593044146","project":"openstack/octavia","branch":"stable/2023.2","attention_set":{},"removed_from_attention_set":{"31664":{"account":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"last_update":"2023-10-11 11:27:18.000000000","reason":"Change was submitted"},"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2023-10-11 11:27:18.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I458dd6d6f5383edc24116ea0fa27e3a593044146","subject":"Retry to set loadbalancer prov status on failures","status":"MERGED","created":"2023-10-10 09:15:01.000000000","updated":"2023-10-11 11:28:29.000000000","submitted":"2023-10-11 11:27:18.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"897772","meta_rev_id":"9919f9eaba4cb7d7675903502b96511d0631bd30","_number":897772,"virtual_id_number":897772,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-10-11 11:27:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"recommended":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"all":[{"value":1,"date":"2023-10-10 13:52:02.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-10-11 10:07:21.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-10-11 10:07:21.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-10-10 11:20:34.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-10-10 13:52:02.000000000","updated_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"reviewer":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"state":"REVIEWER"},{"updated":"2023-10-11 10:07:21.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"}],"messages":[{"id":"94e116ece3cbdd7ad9203f691ffd1a7fae8c7331","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-10-10 09:15:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"233f2cf9113dc70a3257da0f75e2d7dbce0e3c87","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-10 11:20:34.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/100d9aa3ab7642e58e81611f957225aa\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bee11d1ebd614bfe8ae68760b31d5d92 : SUCCESS in 11m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/533a97cde45f4c97a7ec0e988caa391a : SUCCESS in 5m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c820c236ec1147dbb0f60f1953410ef9 : SUCCESS in 4m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c4c64ea90c904143bc49fbb897849d5f : SUCCESS in 6m 27s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0b08f2049406447cad0e0439a0c0e852 : SUCCESS in 4m 35s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/2d0438afff8e420b8a441f4d4349c570 : SUCCESS in 6m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ce1e39b4f138413c8e77fb7194dfb516 : SUCCESS in 11m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/31ef6afb8bcf45d1a376a6be265aed9d : SUCCESS in 6m 38s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/2b6d22c784cd4c61b6b401f6caa367b8 : SUCCESS in 5m 44s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/9881e2e3a458480696479e04f4aaaeb3 : SUCCESS in 7m 11s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/64311b76d1e24fdbb1ac2ba8501ff183 : SUCCESS in 5m 29s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/1287fcd40bd54f6ebe156a270027984d : SUCCESS in 7m 12s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/8f6a6e16776a4388baeca80d99b6c512 : SUCCESS in 7m 35s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/23f4e944de8a4bfabb6359545ea66fb2 : SUCCESS in 9m 41s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/2301b5995ee04050a9788b088384cbbb : SUCCESS in 1h 23m 27s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/f4b710ac76a04c48bc8879fc81bf1261 : SUCCESS in 1h 52m 10s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/79edc556d6384c71b6fe89a93071642f : SUCCESS in 1h 11m 22s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/2506d5d6e0934c6d92c3508875322240 : SUCCESS in 1h 51m 21s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/16e1113ed35a4211bde040aa0717422a : SUCCESS in 1h 07m 03s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/43c8c7189cd648c3a54072a29f342f6e : SUCCESS in 34m 17s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/bf65a6f332b645cc93d64a2ced9d9e8e : SUCCESS in 40m 23s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/226d29ff6f5a4b7ca30a9d76db8fe3b4 : FAILURE in 25m 24s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/ac320db512344cdfb19a3aad30e63e9e : FAILURE in 11m 01s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/457907082c7a461599b404734c75c0aa : SUCCESS in 32m 14s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/34088c13321f4044bb0ac0535de1d74d : SUCCESS in 1h 57m 46s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/59ebb6eb0406464dacf3cd1ea5388c77 : SUCCESS in 2h 03m 20s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/30b10bcbc4b740cb8befe49fdc8966d4 : SUCCESS in 2h 00m 09s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/464765122dca4bcbb4cced9cd243db3d : FAILURE in 11m 15s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7bd7b9ec0e791b5f11e2c028104c7c1c551b2648","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2023-10-10 11:33:00.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"699ba43368443649dc570d41854db10176a0c5f1","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2023-10-10 13:52:02.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f7665d2532c77deb5a4cc870fde6a20802607dd2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-10 14:53:00.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b6202a14a762456394249be4989de66e\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/31672c08dafe4a128da72311ca80532c : SUCCESS in 9m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/87e97138348c454ab61d7e233b7d790a : SUCCESS in 4m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/47d7a708968f4583bfeaf074d4dc7476 : SUCCESS in 4m 42s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/917c6f1642894946a59e278ad2651897 : SUCCESS in 4m 45s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e9ca167d09d44603a93cc944a37c9cc5 : SUCCESS in 5m 04s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/735799dcfb6440e9858203089c7919d2 : SUCCESS in 6m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cb07179959cd4c2dade8989c79d210df : SUCCESS in 12m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b6c9633fd20449cca658d39e1152dce6 : SUCCESS in 6m 18s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/6edf67d1e70640c38669a4bd91dd8ab2 : SUCCESS in 5m 36s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/f7d99b96945a438ab0a0118aeccefabc : SUCCESS in 9m 40s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/145cc9af0c504a6d8aeddea45cbc03cf : SUCCESS in 6m 02s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/b9aeb809d8c54ffc8c65df56e1ef2905 : SUCCESS in 7m 25s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/22613fb909484a5192c20cb4c665d69c : SUCCESS in 9m 09s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/c7fb030a18384f2fafdba08b2c26ea6c : SUCCESS in 7m 32s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/70828cebb8394a9f935d7af19e00d90b : SUCCESS in 1h 29m 32s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/b78d848844a1459a8f2ff61f7d1a9dc8 : SUCCESS in 1h 14m 48s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/b51a2988327d4320bc967827228e9374 : SUCCESS in 56m 08s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/7de5bfafafec4b41a19a1375bd5c5371 : SUCCESS in 1h 20m 18s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/2b6d0aa93b42457eb3162b01792b1f98 : SUCCESS in 57m 20s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/707509a1508946c3bf3dbb90c19f8090 : SUCCESS in 40m 49s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/7bbe0908f8f64e3f87aef7e652e47ba5 : SUCCESS in 46m 38s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/6a9822cf0b4946769abf28c97877a33e : FAILURE in 12m 49s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/dc3605fbf7c74d2f87ea3836e6649ac0 : SUCCESS in 42m 14s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/bdc49fa181c74d1d9529a6aaf05c4664 : TIMED_OUT in 2h 33m 56s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/fbe8982808e04f0e92e10e3cccd4eb00 : TIMED_OUT in 3h 04m 15s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/cd03470255b04623afa9b4ab144f4ab2 : SUCCESS in 2h 06m 13s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/4c3a724d05ef40b4b32ffd6f5b084f41 : FAILURE in 8m 06s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"f0a7e4bdfe12fdcb1336702775e493d3ac9a3cf2","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-10-11 10:07:21.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"63b57ac49347cd74cf9ffee84f229d9173cff02c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-11 10:08:00.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"26695e87b1e22aaedce0130f6e208d7a994f0551","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-11 11:27:18.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a58747e727be4644a3b380358ade7c72\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cdb7b23686a54b24b2c5bd92c22906a7 : SUCCESS in 6m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f161615cdb5244aea4cacedc452f909e : SUCCESS in 4m 46s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8ff583d46c4d4a669954fa3bafe8b53e : SUCCESS in 5m 49s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/e35ee4cf3a764cf9b1f25e788d3613ac : SUCCESS in 4m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/58fdd66813714b96b77fa6bb8b115ade : SUCCESS in 12m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6cfac8b606634f1495300621f9fe8235 : SUCCESS in 5m 56s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/060e961439c94d8c830d72719870fa0f : SUCCESS in 4m 52s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/e7d24186280442728753d76838b04e88 : SUCCESS in 8m 49s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/300636bc3a744a61b6e455b3a1ad27ae : SUCCESS in 1h 18m 36s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/db9e70a70cfa409ab1497b2973bbc095 : SUCCESS in 1h 16m 19s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/b3699a0a16804af682a8bb3eeb648629 : SUCCESS in 1h 03m 25s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/4c82d3858a774d6ca4f60dd8c6eca094 : SUCCESS in 41m 54s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/1ef72238bcf0494fadf84dc186f4cd36 : SUCCESS in 49m 16s","accounts_in_message":[],"_revision_number":2},{"id":"84928cfd7317470befb4e14120c13023ee42ce50","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-11 11:27:18.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"9919f9eaba4cb7d7675903502b96511d0631bd30","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-11 11:28:29.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/62a66fc6298e43d4a079ced96a7a570f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/92566c154f7e47bb8a05a39449d0dc9e : SUCCESS in 52s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d426a528dc3a4f2f8f9d6796bce3122f : SUCCESS in 40s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"ef4b4d5007c51e13b26f243b94c3075dc092794b","revisions":{"96782e2c543bfb488c7629e9ba2cf009b2a6b033":{"kind":"REWORK","_number":1,"created":"2023-10-10 09:15:01.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/72/897772/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/72/897772/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/72/897772/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/72/897772/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/72/897772/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/72/897772/1"}}},"commit":{"parents":[{"commit":"853d0f69e43a16e675103b515a82ec90d83e222f","subject":"Update .gitreview for stable/2023.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/853d0f69e43a16e675103b515a82ec90d83e222f"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-09-25 11:48:01.000000000","tz":-240},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-10-10 09:14:50.000000000","tz":120},"subject":"Retry to set loadbalancer prov status on failures","message":"Retry to set loadbalancer prov status on failures\n\nIn case of DB outages when a flow is running, an exception is caught and\nthe flow is reverted. In most of the flows, the revert function of the\nfirst task\u0027s (the last to be reverted) unlocks the load balancer by\nsetting its provisioning status (to ERROR or ACTIVE, depending on the\nflow), but it fails if the DB is not reachable, leaving the LB in\na PENDING_* state.\nThis commit adds tenacity.retry to those functions, Octavia retries to\nset the status during ~2h45 (2000 attempts, 1 sec initial delay, 5 sec\nmax delay).\n\nCloses-Bug: #2036952\nChange-Id: I458dd6d6f5383edc24116ea0fa27e3a593044146\n(cherry picked from commit be91493332786365b8e997fcf88779a12d1ae130)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/96782e2c543bfb488c7629e9ba2cf009b2a6b033"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/96782e2c543bfb488c7629e9ba2cf009b2a6b033"}]},"branch":"refs/heads/stable/2023.2"},"ef4b4d5007c51e13b26f243b94c3075dc092794b":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-10-10 11:33:00.000000000","uploader":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"ref":"refs/changes/72/897772/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/72/897772/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/72/897772/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/72/897772/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/72/897772/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/72/897772/2"}}},"commit":{"parents":[{"commit":"a04e0f5205511192833fcad008544122cc2b38ee","subject":"[stable/2023.2] Remove octavia-grenade-skip-level job","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a04e0f5205511192833fcad008544122cc2b38ee"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-09-25 11:48:01.000000000","tz":-240},"committer":{"name":"Omer Schwartz","email":"oschwart@redhat.com","date":"2023-10-10 11:33:00.000000000","tz":0},"subject":"Retry to set loadbalancer prov status on failures","message":"Retry to set loadbalancer prov status on failures\n\nIn case of DB outages when a flow is running, an exception is caught and\nthe flow is reverted. In most of the flows, the revert function of the\nfirst task\u0027s (the last to be reverted) unlocks the load balancer by\nsetting its provisioning status (to ERROR or ACTIVE, depending on the\nflow), but it fails if the DB is not reachable, leaving the LB in\na PENDING_* state.\nThis commit adds tenacity.retry to those functions, Octavia retries to\nset the status during ~2h45 (2000 attempts, 1 sec initial delay, 5 sec\nmax delay).\n\nCloses-Bug: #2036952\nChange-Id: I458dd6d6f5383edc24116ea0fa27e3a593044146\n(cherry picked from commit be91493332786365b8e997fcf88779a12d1ae130)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ef4b4d5007c51e13b26f243b94c3075dc092794b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ef4b4d5007c51e13b26f243b94c3075dc092794b"}]},"branch":"refs/heads/stable/2023.2","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
