)]}'
{"id":"openstack%2Foctavia~897777","triplet_id":"openstack%2Foctavia~stable%2F2023.1~I458dd6d6f5383edc24116ea0fa27e3a593044146","project":"openstack/octavia","branch":"stable/2023.1","attention_set":{},"removed_from_attention_set":{"31664":{"account":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"last_update":"2023-10-12 18:52:04.000000000","reason":"Change was submitted"},"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2023-10-12 18:52:04.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I458dd6d6f5383edc24116ea0fa27e3a593044146","subject":"Retry to set loadbalancer prov status on failures","status":"MERGED","created":"2023-10-10 09:35:48.000000000","updated":"2023-10-12 18:53:17.000000000","submitted":"2023-10-12 18:52:04.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"897777","meta_rev_id":"5e841949c045274797c88d4ce336c6a8cf6325ab","_number":897777,"virtual_id_number":897777,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-10-12 18:52:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":2,"date":"2023-10-12 17:11:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":2,"date":"2023-10-11 10:13:05.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":1,"date":"2023-10-12 17:11:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-10-10 11:38:05.000000000","updated_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"reviewer":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"state":"REVIEWER"},{"updated":"2023-10-10 12:30:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-10-11 10:13:05.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"},{"updated":"2023-10-12 17:11:18.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"cbc3c61af18fb18a1d3dd93c4096a7e9e89bf47b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-10-10 09:35:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"dd3d12356da79309adf3418d0c4a7202552018e8","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2023-10-10 11:38:05.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4b8e85fd4e5de15976ae06831bbb07bccf1fad33","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-10 12:30:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/679bcef4baa04edd9217bc73b230fcb9\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f713ad328fea4fd889af060d3505d642 : SUCCESS in 10m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/212634d2610b45cb84e841a88325c101 : SUCCESS in 5m 42s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0102007c59194b398ef0b3c1e5f6cb04 : SUCCESS in 5m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/45108b1308764430bc10dae38241e10b : SUCCESS in 4m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/df9be08816a14dd0adc7782a0d612fa6 : SUCCESS in 13m 24s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/548c4b173c5042a4ab5bfd55d0c219bf : SUCCESS in 6m 54s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/9ceb0ae0d57d47f2a9e9e825c8ab2125 : SUCCESS in 5m 10s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/a7a96d72cc524577822345d6ddb18151 : SUCCESS in 9m 15s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/871ffc5f25044eca97c3d84b6e8213c9 : SUCCESS in 5m 41s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/0f9b47b0cdf84997adb0cc07eae344ef : SUCCESS in 7m 08s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/db07c858f1f042ba8edb4a7a4182989b : SUCCESS in 6m 51s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/36d4299193eb4b2ab3e557163379f428 : SUCCESS in 8m 25s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/82a78424b01a448f85cac62a812c6c2f : SUCCESS in 1h 37m 47s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/c420fc5e9f4041c495f6015c0561fb1a : SUCCESS in 1h 26m 50s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/126ac82c05c7456d98ece2e0df59daf2 : SUCCESS in 1h 14m 37s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/fac86be8dc404b3cba17af1027c350b4 : SUCCESS in 41m 56s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/4eaa495c586d4c558cb87447c95994ff : SUCCESS in 46m 57s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/cf0469f1f0344d939cecbc606300a422 : SUCCESS in 55m 40s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/24b8de6966e34112aa237dce74eef683 : FAILURE in 13m 44s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f8a36d9d4d104782b83f81c34a657260 : SUCCESS in 44m 31s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/b9781d7a8f16421caa03d7078289ca1f : SUCCESS in 1h 52m 36s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/3b5c1878222d4f4dbf9574b38faa5763 : SUCCESS in 2h 52m 36s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/b67d7b07204b4ff887759fc83545c27c : SUCCESS in 1h 59m 04s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/d8d60157ff74458a8690d9d705ac76bc : TIMED_OUT in 2h 33m 07s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/2475075f2e65454aad66c432da0432d1 : SUCCESS in 1h 46m 38s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/afb455980dce4a29860084e3ba400707 : FAILURE in 7m 38s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9cfe0319420d410d1c7bf7d4de833a3e5e0cc3c3","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-10-11 07:47:49.000000000","message":"Patch Set 1: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ff33e75facf50efd1769e88b0832343ac1cd60fb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-10-11 08:06:01.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":2},{"id":"1c9edceed7a9886443b612cb2b25d7e4fb59c283","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-10-11 10:13:05.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"d67981296cf8b3c582c2e4cc3eebf4d5a9e3cc6c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-11 11:04:53.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2fbb0e11e76f4be697c0bdc7aac152ca\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/578b51f2c1c8452296a59ce16cc4837f : SUCCESS in 10m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a7818349381844358505d65cdbdb57ae : SUCCESS in 5m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e057cd57c9e9448cb5d86dbb70387917 : SUCCESS in 4m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f1c838e6eb2c45f39c3434bdb750363e : SUCCESS in 5m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d30fcabb9bb54ca9ac239d6b83357c1d : SUCCESS in 13m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1ba3e0aed1d749da8f8fb6cba76055f8 : SUCCESS in 5m 46s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/135d896c81ec4557a22d9a9179755708 : SUCCESS in 6m 27s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/7c18ea7001bf48bfb4dbb77b666c6b02 : SUCCESS in 7m 57s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/9128b5012aab49d095765f20030ffcac : SUCCESS in 4m 44s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/5f0453f5d58a41cdb152f0ec713253e8 : SUCCESS in 7m 03s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/b5c0399cb9e54c57bbc2febc52d1cc25 : SUCCESS in 9m 33s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/fe9430eae4454d54b494327cc8c72200 : SUCCESS in 8m 08s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/4f1b1b0e78b94230a7ad888f517749ab : SUCCESS in 1h 34m 14s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/5b54c6ae0bf64507b7bb0cc933eae9f3 : SUCCESS in 1h 22m 16s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/c938246310cb498f96072e7f2ae238ff : SUCCESS in 1h 09m 08s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/7c8882467b7f4a94a0115983b7dcb27c : SUCCESS in 39m 50s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/5993e57a0348437ba175ce52a191864d : SUCCESS in 38m 40s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/abf69e8d01f745498dd6db52695933f1 : FAILURE in 1h 50m 42s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/03c9d377f1ec46d3863021a4299ab057 : SUCCESS in 1h 03m 25s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4b5eb14bf1714172a384bfd7a737656f : SUCCESS in 31m 46s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/bb12486ab1a04495a662555ab47f8be0 : SUCCESS in 1h 51m 14s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/ce8883ee4ae145f69cb4d9822bb2659f : SUCCESS in 2h 17m 10s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/f65a46310f28468bbfbdbb689ae69f75 : SUCCESS in 2h 26m 36s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/214cbc59cb8b4a1dac9f980420278b5a : SUCCESS in 2h 00m 58s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/44457e416a8a4c6f95c741cc4d788fe2 : TIMED_OUT in 2h 34m 29s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/4df8e339a91c4f0b9fe8b866fba45ed3 : SUCCESS in 2h 24m 22s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"8376f48824e411b41f67c60452fa6f46cbb6da3e","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-10-12 13:41:32.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"e65e8c8af339fbc9eda5bd14feabef298f3030ce","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-12 16:33:52.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3ac162e49201409f84fbfb65e721bcf6\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7dd9b37a035b4b3586281d62c948c266 : SUCCESS in 12m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f7bcb786b7da4b4fa02c9c86a25607dc : SUCCESS in 4m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/68fd9b752a224f61ad10c10af78c639d : SUCCESS in 5m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/97e30a86670c4b508ea6c3c207b8f590 : SUCCESS in 4m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b5478ee99f9b476bbd97aed44eb78c5f : SUCCESS in 11m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/31688a690d694df5b7326bcd0b158ad6 : SUCCESS in 5m 31s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/02a3f243ed7444ee812165f5dabc96fa : SUCCESS in 5m 21s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/0871149c5fd2446f8aa1306d9f9aefed : SUCCESS in 9m 04s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/c63dcad13bdd45e5a50f248d6d66e70f : SUCCESS in 5m 01s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/b38b08644e5c4c799962553aa3b17f9e : SUCCESS in 6m 41s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/63ad72c1cec14241947905b1647429e4 : SUCCESS in 7m 38s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/8f43625a89a44d5ea7d66be3b0f34710 : SUCCESS in 7m 30s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/da85691e6a264ea3913cd4430f70573f : SUCCESS in 1h 39m 05s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/717bd6fa0dc74435ace06ff66cf06819 : SUCCESS in 1h 47m 52s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/32afdbe1be894fa998cf222f62235653 : SUCCESS in 1h 15m 30s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/4a1c97a9aa5a42589b003cbd9a5406d0 : SUCCESS in 44m 00s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/327e0e44ae35491abf4b6980937bbd3f : SUCCESS in 54m 43s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/a40474d8e5194cbbb40bd313d9f66d08 : SUCCESS in 54m 59s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/46b7c9396a9542f6a7d3ec488ad1654b : SUCCESS in 1h 08m 22s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/98e7101c78ff48aeab9951275c4bc8b9 : SUCCESS in 42m 30s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/9d62af3738ac43628e41e6a0c4547e6a : SUCCESS in 2h 29m 41s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/d1aa38e82b754afca3a0a4262e15ba69 : SUCCESS in 2h 48m 34s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/7cb7d93c84f4424785156af60b08b162 : TIMED_OUT in 2h 34m 30s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/628ccab68d964022ab9328229a8c2348 : SUCCESS in 1h 48m 25s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/0ae8f1e1a05049fa84a13c6ff706994f : SUCCESS in 1h 53m 03s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/b878af170f024413911ee232e1f118d7 : SUCCESS in 2h 06m 25s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"223561cb2b13b67d755b327ec2bd82c1ba697f29","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-10-12 17:11:18.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f360afd6adb192fe3e5a048d15548ae343b6dc13","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-12 17:11:56.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"047b2def8baeb37ea93b5016915d4befc4fb17c0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-12 18:52:04.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4140d38289c44c969e1dcf2b419fb39d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/010a150cc683462f9904972162bf887f : SUCCESS in 5m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0c5a76ee6d164ef8ae06317b5a7c5031 : SUCCESS in 5m 10s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8406ec09b39444228c30c46d2798a645 : SUCCESS in 4m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9a299a61df494dc8b5a7323fba12cad1 : SUCCESS in 9m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d9eff1dc314f43a39abd8b7319a87c8d : SUCCESS in 5m 03s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/850aee4c3886481eb08998048e0dd99a : SUCCESS in 5m 20s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/ee0fdd64121a495cb92be6b8510094da : SUCCESS in 7m 12s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/63d2267e6cee473b98102081b7e8878c : SUCCESS in 1h 06m 49s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/9e388f20f52d4b1181fc961f51c238be : SUCCESS in 1h 38m 20s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/64d95662207b429ab527c9a7be6f5964 : SUCCESS in 1h 16m 17s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/c9e61db675ad46e59c9ce34156c2f818 : SUCCESS in 36m 24s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/c8313e355d0f41a78c7ce9069a9a2c7d : SUCCESS in 54m 53s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/5f168addfb7f4a8b9b12612d18dc57c0 : SUCCESS in 48m 03s","accounts_in_message":[],"_revision_number":2},{"id":"dd4cd7b061ca697e8dcfbc06358fb427ef70b2e1","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-12 18:52:04.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"5e841949c045274797c88d4ce336c6a8cf6325ab","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-12 18:53:17.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ec20927dd6c2423bb97e925ee4c923c9\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a2d37250023049b4999b9873a35a6859 : SUCCESS in 55s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8d76541d166e4b52a0f7df2f4bf4216f : SUCCESS in 38s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"57833dbdad964f8a7861a0ab1d2847159f483577","revisions":{"b78cc9c3dd832e0988e9bae4a4974686bcc4af7e":{"kind":"REWORK","_number":1,"created":"2023-10-10 09:35:48.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/77/897777/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/77/897777/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/77/897777/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/77/897777/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/77/897777/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/77/897777/1"}}},"commit":{"parents":[{"commit":"64d9bdb97556652aa05a34276778260b541ad9c6","subject":"Fix race condition in members batch update API call","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/64d9bdb97556652aa05a34276778260b541ad9c6"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-09-25 11:48:01.000000000","tz":-240},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-10-10 09:31:54.000000000","tz":120},"subject":"Retry to set loadbalancer prov status on failures","message":"Retry to set loadbalancer prov status on failures\n\nIn case of DB outages when a flow is running, an exception is caught and\nthe flow is reverted. In most of the flows, the revert function of the\nfirst task\u0027s (the last to be reverted) unlocks the load balancer by\nsetting its provisioning status (to ERROR or ACTIVE, depending on the\nflow), but it fails if the DB is not reachable, leaving the LB in\na PENDING_* state.\nThis commit adds tenacity.retry to those functions, Octavia retries to\nset the status during ~2h45 (2000 attempts, 1 sec initial delay, 5 sec\nmax delay).\n\nConflicts:\n\toctavia/tests/unit/controller/worker/test_task_utils.py\n\nCloses-Bug: #2036952\nChange-Id: I458dd6d6f5383edc24116ea0fa27e3a593044146\n(cherry picked from commit be91493332786365b8e997fcf88779a12d1ae130)\n(cherry picked from commit 96782e2c543bfb488c7629e9ba2cf009b2a6b033)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b78cc9c3dd832e0988e9bae4a4974686bcc4af7e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b78cc9c3dd832e0988e9bae4a4974686bcc4af7e"}]},"branch":"refs/heads/stable/2023.1"},"57833dbdad964f8a7861a0ab1d2847159f483577":{"kind":"REWORK","_number":2,"created":"2023-10-11 08:06:01.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/77/897777/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/77/897777/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/77/897777/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/77/897777/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/77/897777/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/77/897777/2"}}},"commit":{"parents":[{"commit":"64d9bdb97556652aa05a34276778260b541ad9c6","subject":"Fix race condition in members batch update API call","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/64d9bdb97556652aa05a34276778260b541ad9c6"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-09-25 11:48:01.000000000","tz":-240},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-10-11 08:03:05.000000000","tz":120},"subject":"Retry to set loadbalancer prov status on failures","message":"Retry to set loadbalancer prov status on failures\n\nIn case of DB outages when a flow is running, an exception is caught and\nthe flow is reverted. In most of the flows, the revert function of the\nfirst task\u0027s (the last to be reverted) unlocks the load balancer by\nsetting its provisioning status (to ERROR or ACTIVE, depending on the\nflow), but it fails if the DB is not reachable, leaving the LB in\na PENDING_* state.\nThis commit adds tenacity.retry to those functions, Octavia retries to\nset the status during ~2h45 (2000 attempts, 1 sec initial delay, 5 sec\nmax delay).\n\nNote: stable/2023.1 and older, the patch also includes modifications for\nv1/tasks/lifecycle_tasks.py\n\nConflicts:\n\toctavia/tests/unit/controller/worker/test_task_utils.py\n\toctavia/controller/worker/v1/tasks/lifecycle_tasks.py\n\nCloses-Bug: #2036952\nChange-Id: I458dd6d6f5383edc24116ea0fa27e3a593044146\n(cherry picked from commit be91493332786365b8e997fcf88779a12d1ae130)\n(cherry picked from commit 96782e2c543bfb488c7629e9ba2cf009b2a6b033)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/57833dbdad964f8a7861a0ab1d2847159f483577"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/57833dbdad964f8a7861a0ab1d2847159f483577"}]},"branch":"refs/heads/stable/2023.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
