)]}'
{"id":"openstack%2Foctavia~897778","triplet_id":"openstack%2Foctavia~stable%2Fzed~I458dd6d6f5383edc24116ea0fa27e3a593044146","project":"openstack/octavia","branch":"stable/zed","attention_set":{},"removed_from_attention_set":{"31664":{"account":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"last_update":"2023-10-12 19:25:29.000000000","reason":"Change was submitted"},"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2023-10-12 19:25:29.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I458dd6d6f5383edc24116ea0fa27e3a593044146","subject":"Retry to set loadbalancer prov status on failures","status":"MERGED","created":"2023-10-10 09:36:34.000000000","updated":"2023-10-12 19:26:53.000000000","submitted":"2023-10-12 19:25:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"897778","meta_rev_id":"f2ec6c3822fb48318da7f5a171f5f84e826dec2d","_number":897778,"virtual_id_number":897778,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-10-12 19:25:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-10-12 17:11:42.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":2,"date":"2023-10-11 10:15:13.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-10-12 17:11:42.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-10-10 11:42:29.000000000","updated_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"reviewer":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"state":"REVIEWER"},{"updated":"2023-10-10 11:48:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-10-11 10:15:13.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"},{"updated":"2023-10-12 17:11:42.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"f6658e48cdd960390800b9670011563c96d30ad3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-10-10 09:36:34.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"97b25478759de93ad2dcec325e0cae3cb22e3e1d","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2023-10-10 11:42:29.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"82e52ebf82f31942f22f5abf37c44b4cc0abb9bf","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-10 11:48:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9e7e40626c7149169ca5d83ca9f9b4d3\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c9a73721f85040c2bdb61e09b733f1ca : SUCCESS in 11m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/604efc5f58524ed8bf53303b76b4ec7b : SUCCESS in 6m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7b953205e3434a9c89dc97e1089f36f0 : SUCCESS in 5m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3cba5776f27543c680314788ddfc3acb : SUCCESS in 6m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/07ef3ec6d0f3413cbbb17db02ecc9387 : SUCCESS in 5m 39s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b202493092a14376b07d2bc32a5592bf : SUCCESS in 12m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4ded1685204d4f68be6737e64573011d : SUCCESS in 5m 59s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/a0447b4e2e69454fbffbdd5f6637f909 : SUCCESS in 5m 22s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/39d2853b3ec54b62be09f935acce5814 : SUCCESS in 11m 41s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/d36a021ce6a94f1ebbcf25a82c0a2007 : SUCCESS in 5m 15s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/b0eeab16b5154393a6e98a482e362a2f : SUCCESS in 7m 34s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/40a468941d164c3888b680db366b828c : SUCCESS in 8m 11s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/e1cdec9f7908422b86fc61e6b4c07e4d : SUCCESS in 7m 58s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/98443133006c42cf802bef8622bb52e1 : SUCCESS in 1h 31m 31s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2a390d7e9ca04cfc8782bc8b2d3cc8b1 : SUCCESS in 1h 43m 05s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/eb9f053392ed42538a8abe353921de8d : SUCCESS in 38m 42s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/8493aa408fd74ccfbf46015865248e99 : SUCCESS in 37m 06s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/d994c60419074bbd849b8e58f96b27c9 : FAILURE in 12m 46s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d9f9d224357a420f98935bff47004b1b : SUCCESS in 33m 25s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/f4df7c61631042d7b97c399a54c28d69 : SUCCESS in 1h 46m 47s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/46ebd077975b424ebffb96149fbaab3d : SUCCESS in 2h 04m 47s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/100701baa3e04314b75d43967fb58fe0 : SUCCESS in 1h 46m 16s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/94e088b68a4943d0a177b2b8b2f9b788 : SUCCESS in 1h 42m 16s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/1a5029953228463ba3aedc4c6be325e5 : SUCCESS in 1h 59m 30s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/a23c81bc560a4e08b43b13381197de6b : FAILURE in 10m 41s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"eb95289a2cd5a284c71b0b34607b03bcc6be26b3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-10-11 08:06:57.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"32e649a6c270b08fcaffe390e83b666419cd0e5e","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-10-11 10:15:13.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"0469137db55f3947ff5475cedaf57836a81e55b1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-11 10:41:16.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/205f4cc72aad4e36bd3b4056c6c4a23d\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0f47fea3ec004cee9fc848532128d70c : SUCCESS in 10m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d04d13aa4e524b96827b75ebad416e54 : SUCCESS in 5m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7bcc4df19022493b95ef437333fc53b9 : SUCCESS in 5m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2a8855a002f54f16826f24989bc0fba3 : SUCCESS in 4m 37s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/58fac272b4214146bb5c7fa443b0c3c4 : SUCCESS in 5m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f29dad0eb4b142e382218bc80a94f777 : SUCCESS in 14m 00s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7aba81bec4524a3d8fce579b75eab41c : SUCCESS in 5m 25s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/62b7059d62ba42b0ad74f6c7622aec49 : SUCCESS in 6m 13s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/b8ee60362a054b6193775125472d85f7 : SUCCESS in 7m 37s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/490e0c58c7b14d20a9cfac3d5c4d4975 : SUCCESS in 6m 33s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/e72988ab2bbb4ac18354e2ede4b3b7b6 : SUCCESS in 7m 34s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/c1731251a96e4f43bd47e44026a924a8 : SUCCESS in 9m 54s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/3dff811ceaae48bda49478e5d4568bcb : SUCCESS in 8m 02s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/4ebac6f2b9564bd5aac208c688e44821 : SUCCESS in 1h 25m 04s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e85640def8c6407e8f3128d8e9ab9acb : SUCCESS in 2h 17m 31s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/87d15f20684846bb85415843c591ab78 : SUCCESS in 51m 41s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/f3d7263b54354c7380ce35c7d1bf7ddc : SUCCESS in 41m 34s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/14235a05cba24b4a8f7337f393a983bd : SUCCESS in 1h 20m 02s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/87c753da4c654e4098a0be10d6737259 : SUCCESS in 35m 11s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/870c1be9a35045e09a513467ba5dec88 : SUCCESS in 1h 56m 51s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/3487f6d817134cda98556492e22e3059 : SUCCESS in 2h 18m 17s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/899197c05280436e94531ffade6d7fd4 : SUCCESS in 2h 00m 30s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/c42109504c77453abed4713ea9bd11a4 : SUCCESS in 1h 46m 02s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/08db1b2f272242d292a500da69a2b318 : SUCCESS in 1h 41m 07s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/e6d9fc1f195d4167822cee9a59ad3788 : SUCCESS in 2h 24m 38s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"059888a77551757066f677cf06ddd60065f39f5b","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-10-12 17:11:42.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7bc15a799cff3e4896af59e7447faa46d30bc81c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-12 17:12:28.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"621c3d550ae33a3f8434f8122fd52d43cbf18553","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-12 19:25:29.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/62c0b8532a644a7b8e1ba6bd2d59ca04\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7077bda7247e4bbebd382a682d16e104 : SUCCESS in 7m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fa7bd961638b4337bdcb1ff94c546634 : SUCCESS in 4m 51s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1d15c1623f4648fc98b2cf9b6cd5c7d5 : SUCCESS in 5m 14s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/66442761bca94f18b3f0d519c779faa8 : SUCCESS in 12m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/971ce0d773d14a64a2044775f62f969a : SUCCESS in 5m 07s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/9eebc5eb66ea41fb861aa2d6bac72328 : SUCCESS in 4m 40s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/4a592a03bf5849ccbb8b51e250915bb4 : SUCCESS in 8m 24s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/5bce7029727849929a78d81f4aed9e74 : SUCCESS in 1h 03m 34s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/164df2a800d94249b4bffe921b678c03 : SUCCESS in 2h 04m 05s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/5d58efd59e8d4b84a368a8edba10c632 : SUCCESS in 35m 01s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/1183da8de3c3450fa4f446d340e0d350 : SUCCESS in 37m 59s","accounts_in_message":[],"_revision_number":2},{"id":"db51f4323f61db435974e51638aed22bbad0f6c4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-12 19:25:29.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"f2ec6c3822fb48318da7f5a171f5f84e826dec2d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-12 19:26:53.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1e7e6e194295491db512e250330c98a8\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9a099f3312754f4da047a3fc8f407432 : SUCCESS in 58s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5be630f15c204265803f67f1b44390ba : SUCCESS in 43s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"27060603db697c64dd9d3a30910d67fed7b5906e","revisions":{"9f7062b75dd2e237ce82f77f20992bffea55f569":{"kind":"REWORK","_number":1,"created":"2023-10-10 09:36:34.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/78/897778/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/78/897778/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/78/897778/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/78/897778/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/78/897778/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/78/897778/1"}}},"commit":{"parents":[{"commit":"4c97b585ce6e519a5ff090c2cd97fab0e801be0c","subject":"Fix race condition in members batch update API call","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4c97b585ce6e519a5ff090c2cd97fab0e801be0c"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-09-25 11:48:01.000000000","tz":-240},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-10-10 09:36:22.000000000","tz":120},"subject":"Retry to set loadbalancer prov status on failures","message":"Retry to set loadbalancer prov status on failures\n\nIn case of DB outages when a flow is running, an exception is caught and\nthe flow is reverted. In most of the flows, the revert function of the\nfirst task\u0027s (the last to be reverted) unlocks the load balancer by\nsetting its provisioning status (to ERROR or ACTIVE, depending on the\nflow), but it fails if the DB is not reachable, leaving the LB in\na PENDING_* state.\nThis commit adds tenacity.retry to those functions, Octavia retries to\nset the status during ~2h45 (2000 attempts, 1 sec initial delay, 5 sec\nmax delay).\n\nConflicts:\n\toctavia/common/config.py\n\toctavia/tests/unit/controller/worker/test_task_utils.py\n\nCloses-Bug: #2036952\nChange-Id: I458dd6d6f5383edc24116ea0fa27e3a593044146\n(cherry picked from commit be91493332786365b8e997fcf88779a12d1ae130)\n(cherry picked from commit 96782e2c543bfb488c7629e9ba2cf009b2a6b033)\n(cherry picked from commit b78cc9c3dd832e0988e9bae4a4974686bcc4af7e)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9f7062b75dd2e237ce82f77f20992bffea55f569"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9f7062b75dd2e237ce82f77f20992bffea55f569"}]},"branch":"refs/heads/stable/zed"},"27060603db697c64dd9d3a30910d67fed7b5906e":{"kind":"REWORK","_number":2,"created":"2023-10-11 08:06:57.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/78/897778/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/78/897778/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/78/897778/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/78/897778/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/78/897778/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/78/897778/2"}}},"commit":{"parents":[{"commit":"4c97b585ce6e519a5ff090c2cd97fab0e801be0c","subject":"Fix race condition in members batch update API call","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4c97b585ce6e519a5ff090c2cd97fab0e801be0c"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-09-25 11:48:01.000000000","tz":-240},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-10-11 08:06:45.000000000","tz":120},"subject":"Retry to set loadbalancer prov status on failures","message":"Retry to set loadbalancer prov status on failures\n\nIn case of DB outages when a flow is running, an exception is caught and\nthe flow is reverted. In most of the flows, the revert function of the\nfirst task\u0027s (the last to be reverted) unlocks the load balancer by\nsetting its provisioning status (to ERROR or ACTIVE, depending on the\nflow), but it fails if the DB is not reachable, leaving the LB in\na PENDING_* state.\nThis commit adds tenacity.retry to those functions, Octavia retries to\nset the status during ~2h45 (2000 attempts, 1 sec initial delay, 5 sec\nmax delay).\n\nNote: stable/2023.1 and older, the patch also includes modifications for\nv1/tasks/lifecycle_tasks.py\n\nConflicts:\n\toctavia/common/config.py\n\toctavia/tests/unit/controller/worker/test_task_utils.py\n\toctavia/controller/worker/v1/tasks/lifecycle_tasks.py\n\nCloses-Bug: #2036952\nChange-Id: I458dd6d6f5383edc24116ea0fa27e3a593044146\n(cherry picked from commit be91493332786365b8e997fcf88779a12d1ae130)\n(cherry picked from commit 96782e2c543bfb488c7629e9ba2cf009b2a6b033)\n(cherry picked from commit 57833dbdad964f8a7861a0ab1d2847159f483577)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/27060603db697c64dd9d3a30910d67fed7b5906e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/27060603db697c64dd9d3a30910d67fed7b5906e"}]},"branch":"refs/heads/stable/zed"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
