)]}'
{"id":"openstack%2Foctavia~897783","triplet_id":"openstack%2Foctavia~stable%2F2023.2~I497f2d8587e25eda0d169aa919cb6fbede7e9fc6","project":"openstack/octavia","branch":"stable/2023.2","attention_set":{},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2023-10-16 08:31:44.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I497f2d8587e25eda0d169aa919cb6fbede7e9fc6","subject":"Fix timeout duration in start_vrrp_service during failovers","status":"MERGED","created":"2023-10-10 09:57:44.000000000","updated":"2023-10-16 08:32:53.000000000","submitted":"2023-10-16 08:31:44.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"897783","meta_rev_id":"a44dc5e68c768c317a9f80878ee04dda4fd1786e","_number":897783,"virtual_id_number":897783,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-10-16 08:31:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"recommended":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"all":[{"value":1,"date":"2023-10-13 14:32:06.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":2,"date":"2023-10-12 17:12:06.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-10-16 06:48:46.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-10-16 06:48:46.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-10-10 12:51:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-10-10 15:55:05.000000000","updated_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"reviewer":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"state":"CC"},{"updated":"2023-10-12 17:12:06.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2023-10-13 14:32:06.000000000","updated_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"reviewer":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"state":"REVIEWER"},{"updated":"2023-10-16 06:48:46.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"}],"messages":[{"id":"dac9ea3e40d4d6823371c91389cd62893255ad34","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-10-10 09:57:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f6442b5b87e235f3fbef398a599fc6f72e5cdea9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-10 12:51:01.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9ec21d49d515484db7d4fd4d406f2404\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/26e58b7e043a475e8329badfacc05569 : SUCCESS in 10m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3725e0cd68044674923e33f071028d57 : SUCCESS in 5m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/11736628bcbb4a7fbe41a67cd3f4b855 : SUCCESS in 5m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c5a27739b00f463ea7d2f77ca51f34ab : SUCCESS in 5m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8ad3c990817e495a823d96c5c11dadee : SUCCESS in 5m 34s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6c7876c883fe441dae08f3b53f4c4aa0 : SUCCESS in 5m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0527d2fd2fbe4b5ea75e1a9b73483924 : SUCCESS in 13m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6fbc197019904cf096948eb9a4461c71 : SUCCESS in 6m 01s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/3a26a01b23d142feae692c944a8990bf : SUCCESS in 6m 00s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/4d731376b8434e82a84719079f172e1b : SUCCESS in 8m 39s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/7f1f4d919e7844c59ecb278ffb435476 : SUCCESS in 4m 26s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/9eb8cfdb71b544b1ba90e491311eb2d1 : SUCCESS in 7m 02s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/65661766c2834ed3aa7b0dd881d22af9 : SUCCESS in 8m 50s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/a1dc351e71d8403b8967c467587b42cb : SUCCESS in 8m 08s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/07ceb71fe8274ecf8a0fa267164260c8 : SUCCESS in 1h 18m 08s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/569c1450ef81422fbc8e0c82dfff01bb : SUCCESS in 1h 22m 16s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/3c7c1bc19f364bfaad132cf52329e3e8 : SUCCESS in 1h 02m 11s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/35de4fb86ce54f9fb4b76d59482d0731 : SUCCESS in 1h 27m 35s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/f02bcbfad21d40a3a42d437a672a7743 : SUCCESS in 1h 06m 32s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/f42f8dbc4fe14f3099501f792628dec7 : SUCCESS in 40m 40s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/d264327527bf4f299a907e0eb8579b29 : SUCCESS in 41m 34s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/25c74b6f4ada4f69a3b2c14286d6c594 : FAILURE in 22m 36s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/9309739a78b0432da611d26feb53d406 : FAILURE in 10m 14s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8f8e50955d8c462791ff911568ef6a7d : SUCCESS in 43m 40s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/f75b9064eefd4339b2401d98aebe3f59 : SUCCESS in 2h 30m 39s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/9f04740ccd1348bd80a7b269b397c022 : SUCCESS in 2h 50m 44s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/ea74093d5a8c4dbeaff4d00152ead52f : SUCCESS in 2h 12m 03s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/dc57449e7e614905a2cc8bf628edf6c8 : FAILURE in 8m 25s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"92c2fecbe203d6780d33f1957c64aface39817ad","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2023-10-10 15:55:05.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"b45759fc11252d7f299e884df73eb0ec276c7653","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-10 18:36:52.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ca30ec6be8c0425292abbab3d2e6b9d9\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/099c25d995324f5cbe83215d5eccff1e : SUCCESS in 9m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c9afedfd299543c2b641e200771ec8e8 : SUCCESS in 5m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/70c47420d071461b9e0d5e13711ecac0 : SUCCESS in 5m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/0283dd2bf5a44caaa5cb359f3d7b05aa : SUCCESS in 4m 47s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/78c81378e0dd4bc58d61358c885a6c05 : SUCCESS in 4m 50s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1e79800821ab4da0b164834b433335ef : SUCCESS in 5m 10s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/14894fa1c61e40ea90cd74c8113e3c57 : SUCCESS in 11m 59s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d9536289363a494fad471ec8662f6f25 : SUCCESS in 5m 02s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/bbf00d4ee6ec4bfb80ac12eb18b0b16e : SUCCESS in 6m 05s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/f07e32a21e184d6bae949a34aba45bd0 : SUCCESS in 9m 29s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/8ba5cbf353d24d00934ba137cf7733eb : SUCCESS in 4m 40s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/7c9be3e5d26a45749d3c3b401a5a97a5 : SUCCESS in 6m 50s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/3097a34bad4e44bbb761197018dc26b7 : SUCCESS in 7m 58s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/868d0204b8a6412a89e9f7ac8b8c18cb : SUCCESS in 8m 25s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/8f150d03bfbc4789b72c581cd527014f : SUCCESS in 1h 38m 43s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/e3711fa000324368a6b73dbbeef99b1a : SUCCESS in 1h 39m 19s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/0b702592e4f4483d86bb836842b83fa6 : SUCCESS in 1h 14m 47s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/cf3f14d7174541108308b478c635bbc2 : SUCCESS in 1h 56m 22s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/d12e4b0029cf43639a89bac192a96c10 : SUCCESS in 1h 21m 57s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/a3fc26b5541142c7aed979934d689220 : SUCCESS in 54m 46s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/05be8cace9434563a00fb25dc02a3f67 : SUCCESS in 40m 55s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/1e8d2592444c4f31b35c7a8b112787d8 : SUCCESS in 1h 13m 01s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8616a6512494421baaa03a064d26f03f : SUCCESS in 31m 27s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/8b0953777eb7493fa9fb89a9b490930d : SUCCESS in 1h 52m 39s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/f5ab730229e948f9bea1d0b246361d9f : SUCCESS in 2h 38m 51s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/1d439c2f773540cf8bf2d4f2cf5e77b1 : SUCCESS in 1h 33m 15s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/7c195e3545014ec6b47d8894fcd5f34b : SUCCESS in 2h 20m 48s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"6b7b87af33620248c5879913d95df9ecd2125e75","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-10-12 17:12:06.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5363876bb29ba3ea07b6d3436b28fed3445e2f10","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2023-10-13 14:32:06.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7b9e6a64319059cc5c3c6f40cbdbcdcdb14b720b","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-10-16 06:48:46.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d22916d67d45d8f7b66ddf9aadde85257262d4e6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-16 06:49:21.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"24afbcd185fc1f943b52f1eb07a5eb6b03ca3184","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-16 08:31:44.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/04ec48310b98472987c836c6f473b1a3\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/24aab553382c45548cfeebf0191b06c7 : SUCCESS in 4m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/163419a8552e42f4bf0fdfd35ece048e : SUCCESS in 4m 58s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/46df9e7f8dd641d7a4580ff853641eec : SUCCESS in 4m 56s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/786a269b89064b5e9e78cccafe232dc2 : SUCCESS in 5m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b0187180e31b4ce4ad3825586ef53cff : SUCCESS in 12m 05s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a2bd0fa9405948fea1eef798cdebdaac : SUCCESS in 5m 42s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/22ac5b91392d4c709bc6523a21be204f : SUCCESS in 5m 11s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/53744e87af2f4a2282ee6e475f948f41 : SUCCESS in 7m 23s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/0b536d02621e4a56a94eb22a6987594e : SUCCESS in 1h 37m 34s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/bbba827652f144e9acee9b01cd88323f : SUCCESS in 1h 36m 59s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/fac6601bf5d44eaa8144131ed842ef08 : SUCCESS in 1h 12m 32s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/f89ddcc480344768b9466f39d75a63f8 : SUCCESS in 36m 09s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/9b8007de5bc94e27b729f38f269e347e : SUCCESS in 39m 30s","accounts_in_message":[],"_revision_number":1},{"id":"8ac713af2433bde859dae7058fde7a99cfd2cda7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-16 08:31:44.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"a44dc5e68c768c317a9f80878ee04dda4fd1786e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-16 08:32:53.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d8c0d07fd9bc496b8601b4e87286fca2\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f0803f8c827547f38fa3d7658020138f : SUCCESS in 49s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f03ace73bb4d4f29aee497cc43e8e848 : SUCCESS in 44s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"54eb1624f5762e9b848f10f712152faf02f8b84d","revisions":{"54eb1624f5762e9b848f10f712152faf02f8b84d":{"kind":"REWORK","_number":1,"created":"2023-10-10 09:57:44.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/83/897783/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/83/897783/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/83/897783/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/83/897783/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/83/897783/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/83/897783/1"}}},"commit":{"parents":[{"commit":"853d0f69e43a16e675103b515a82ec90d83e222f","subject":"Update .gitreview for stable/2023.2","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/853d0f69e43a16e675103b515a82ec90d83e222f"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-09-01 15:35:34.000000000","tz":-240},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-10-10 09:54:20.000000000","tz":120},"subject":"Fix timeout duration in start_vrrp_service during failovers","message":"Fix timeout duration in start_vrrp_service during failovers\n\nThe timeout_dict argument was not passed to the\n_populate_amphora_api_version, it could have made the failover of broken\namphora really long.\n\nRelated-Bug: #2033894\n\nChange-Id: I497f2d8587e25eda0d169aa919cb6fbede7e9fc6\n(cherry picked from commit cce348040d43b92b59146adeb15a4f1486946a64)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/54eb1624f5762e9b848f10f712152faf02f8b84d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/54eb1624f5762e9b848f10f712152faf02f8b84d"}]},"branch":"refs/heads/stable/2023.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
