)]}'
{"id":"openstack%2Foctavia~897784","triplet_id":"openstack%2Foctavia~stable%2F2023.2~Ib33a0b8d2875e4ff97c65933fe9360bb06994d32","project":"openstack/octavia","branch":"stable/2023.2","attention_set":{},"removed_from_attention_set":{"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2023-10-16 08:31:46.000000000","reason":"Change was submitted"},"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2023-10-16 08:31:46.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ib33a0b8d2875e4ff97c65933fe9360bb06994d32","subject":"Reduce duration of failovers with amphora in ERROR","status":"MERGED","created":"2023-10-10 09:57:44.000000000","updated":"2023-10-16 08:33:49.000000000","submitted":"2023-10-16 08:31:46.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"897784","meta_rev_id":"9ddf3327178b55df853d5648474257ec3ee56459","_number":897784,"virtual_id_number":897784,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-10-16 08:31:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"recommended":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"all":[{"value":1,"date":"2023-10-10 13:21:34.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-10-16 06:38:59.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-10-16 06:38:59.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"reviewers":{"CC":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}],"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-10-10 12:34:27.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-10-10 13:21:34.000000000","updated_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"reviewer":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"state":"REVIEWER"},{"updated":"2023-10-12 17:07:48.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"CC"},{"updated":"2023-10-14 19:10:49.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"CC"},{"updated":"2023-10-14 19:10:49.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2023-10-16 06:38:59.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"}],"messages":[{"id":"a890bb1652db0c64afbfddeee582de44814df7fc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-10-10 09:57:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"6da06ecc1df4889106b4585f6b2296df4fe420c8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-10 12:34:27.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c747795ca74e41d58d658066d62952f4\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c85327441ce94af49cad27d343357266 : SUCCESS in 12m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4a47acdd80b14e5196ce875e6412ee7a : SUCCESS in 5m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e01ec2ac028e404d99e59e55eba5d112 : SUCCESS in 4m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/59e5911c8acc458b854fb3c2693d1ec7 : SUCCESS in 6m 13s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/402fb437ef8a468ebf1788997848d272 : SUCCESS in 5m 02s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ddb2ac2462cb4d4cab926bba64ac0895 : SUCCESS in 5m 11s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/25a57465b86d4b3483e6703922eb90dd : SUCCESS in 13m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b0abc6d19f554aeeb9ab7b4d9298cc2e : SUCCESS in 6m 01s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/54842614cb6745b69c4a22a670f53751 : SUCCESS in 4m 46s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/7b16a84d40cc4769bc95c99bcd850ddd : SUCCESS in 8m 18s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/99b583c5556e4a179fdcf0c1dd23d82a : SUCCESS in 4m 59s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/0e835adad9774078b17b5d4304fb2e89 : SUCCESS in 8m 09s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/35ff573f20d946049f8192aaabef7351 : SUCCESS in 7m 38s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/0789aeac3ab040a8bcbd326fb0c54db9 : SUCCESS in 8m 29s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/f5e2904b2f094e799e33138a076a590f : SUCCESS in 1h 19m 25s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/22e083cadd0d48a99bbdd9edc35b8739 : SUCCESS in 1h 13m 43s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/39cc17b38b7448fb95bf31fbf78a9870 : SUCCESS in 56m 16s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/854dbde41ca54961b2fc03247bcb3ff1 : SUCCESS in 1h 45m 27s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/d47d9002b8c04b6d8954700395d6b0d7 : SUCCESS in 1h 17m 19s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/3bc467af77c8478e9bf74b2523f526aa : SUCCESS in 48m 13s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/084d4d05c44e48b2aea4ddd0afd854fe : SUCCESS in 48m 21s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/74400fec1dda4edba38a5adf8ec09240 : FAILURE in 28m 53s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/64de7072049c40588b631d6d4d98751b : FAILURE in 12m 43s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/95bba01cb208419b9afefd2a00de1f62 : SUCCESS in 44m 37s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/ab96da1b1ccf4419a941145611d4bf94 : SUCCESS in 2h 18m 40s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/bf8c03b8ced44ad0beaa5ab824586962 : SUCCESS in 2h 29m 58s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/e1a8a21fed9a44b08e3cd1114a21ee5f : SUCCESS in 2h 01m 33s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/25eb138d2094403db085a92677b05939 : FAILURE in 8m 51s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5c64370952965e4f979408658174950c13f1fe20","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2023-10-10 13:21:34.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"5bb0ed0a8131f53c26dc17b51a3977d82086c8ab","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-10-12 13:41:56.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f43e5f6071bdbb3193d72702a6225a25f59bf253","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-12 15:57:20.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/74bb5210724241d08441311283ed4d8e\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e6470278cd334081bb9f8352f1395abf : SUCCESS in 9m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3dcea923ed3846189c029c64c7966ebc : SUCCESS in 4m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/01b2fe6343eb47489b2da499391adaa7 : SUCCESS in 4m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/28c80781539b4e3c83aa072f8b2bd03c : SUCCESS in 5m 20s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6b0b1194ec5a4840b631480bf6fb5f83 : SUCCESS in 5m 53s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/22b5d74569ba42fe8a81315296bf2a58 : SUCCESS in 5m 01s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0dda535efe9441d7878f22bcf43ee867 : SUCCESS in 12m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c27d6fd029a640418a8bfae19a34e5a7 : SUCCESS in 5m 28s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/c5671514d1684bc38128cca6750d7625 : SUCCESS in 5m 12s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/43c148e903fa4934ba6b3af0714d813c : SUCCESS in 8m 39s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/5246680deff74882b32a3d691d76b9bf : SUCCESS in 5m 26s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/f49a3987b25a461f8af0d2d9e30e0487 : SUCCESS in 8m 47s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/be6f583c620245b29789228559c84773 : SUCCESS in 8m 30s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/b1489eb763a64f77b3e99726d4affe09 : SUCCESS in 7m 52s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/dd92489487a1451f96b75882b08a8d77 : SUCCESS in 1h 15m 42s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/adb42c603694497f883a825298632036 : SUCCESS in 1h 21m 33s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/56cea4f6765c45f69eed73f47e240668 : FAILURE in 1h 18m 48s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/80f0dc1bc96146698e1b10cb5702e614 : SUCCESS in 1h 23m 33s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/d79a06a196fe4bb988d7bab811b08dd0 : SUCCESS in 1h 05m 48s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/dffe3a78039145bb86886f5168199e71 : SUCCESS in 39m 19s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/2cc102e9e73841739375f74e0e029a8f : SUCCESS in 37m 40s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/0a4e68b8405044468c161c20e38652fe : FAILURE in 15m 59s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/237784e756e84dff8610a37781c434b7 : SUCCESS in 35m 57s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/6108d7a0810c472bab4174992a8b4abc : SUCCESS in 1h 56m 18s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/0eafd5d3b5f84a6a85f5705741c5f3d5 : SUCCESS in 2h 12m 18s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/2b9c819a89304625a877bdf3aee25100 : SUCCESS in 1h 40m 31s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/59a850f2cd41485090890100561fbce4 : FAILURE in 12m 04s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e935600dc4f28b4db30f9393cd9d30650f2eae31","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-10-12 17:07:48.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"bae0a60758cd264312b8627fa555dbcfffdc7356","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-12 19:48:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d9660641ebae4ab7813876fab82c31ef\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/621bb390c60b460d821363f1a4593087 : SUCCESS in 8m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f17466ba814d48508ad89e0f9c55b72d : SUCCESS in 5m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1400f688815d4fe9a42c308a661be580 : SUCCESS in 5m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/548357cd4c694298848a1eaabb593b9a : SUCCESS in 5m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a68d4966d1b44f3db896438157d22948 : SUCCESS in 5m 02s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/70094e6fcf5e4751ab6f135749fe74c7 : SUCCESS in 5m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b0334e9cca444268b6f681458bcc6fb5 : SUCCESS in 13m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b0d53f4bcf5b428abe5ce05416e2e094 : SUCCESS in 5m 23s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/0ce34515784e43b2941ac4534dd23919 : SUCCESS in 5m 30s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/ee7f08ed37ff403b83193656a9dc3592 : SUCCESS in 9m 54s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/7cb61aae73094a9cb31dc136ab35767e : SUCCESS in 4m 47s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/b41a72fb936e4ee8ac211866074dee9b : SUCCESS in 8m 45s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/480b55ae2ab8467995b94c8b9f66dfd0 : SUCCESS in 6m 40s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/f5b8581f499f4465a4eb1dff623b89da : SUCCESS in 8m 20s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/e35c09597879499b88df82469f10c1a7 : SUCCESS in 55m 19s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/4c3eda1b92f34646aeef4a4be326a226 : SUCCESS in 1h 34m 49s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/f2aa0e68dc5b4fbf8dcc9c6177980c51 : SUCCESS in 1h 10m 28s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/590594e8d3ec45f0afa1ac7a6a0c3f38 : SUCCESS in 1h 29m 06s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/e58a62d5bd9941959647bf686b282d88 : SUCCESS in 1h 17m 29s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/daece6e69f53468d936b08d22502af2a : SUCCESS in 38m 34s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/ba41bce773754c1cb1c11ed1e9babada : SUCCESS in 45m 16s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/61990accd5db4ea686fec239c817497d : FAILURE in 12m 06s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/332a4562c3d848a1a6eadc012fc39648 : SUCCESS in 33m 03s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/6ad59ab8a6544645b6ababb58ef2eb16 : SUCCESS in 1h 40m 44s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/4c58a4b59b00479db8f3800686e8f283 : SUCCESS in 2h 38m 39s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/0d50ef543b53498184053d5d521c8d96 : SUCCESS in 1h 46m 26s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/3b546a9658bf4fb7b5e2637f6530d59f : FAILURE in 12m 22s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a0438cb2b0d54fd02c735854c3bfc8ff323fca4e","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-10-16 06:38:59.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"6a3440aaa3826711f652dfa67a9485accbd5966a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-16 06:49:25.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f4d97eed2a8e086c7152da12b8816c7f5e97207c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-16 08:31:46.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/aa85fd406f2a4ac098f0bc0404fa4e3a\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/18d4e5fae3b744a38828e2164d6e30fa : SUCCESS in 4m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9cb2f989392445bbb6cb7cfd3f02d5c5 : SUCCESS in 4m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/f08063150437444ca12d382707133627 : SUCCESS in 6m 07s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f8c10be9447e49679ab9b546a051441e : SUCCESS in 4m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f81ce55d1978429e997fd1525b9732d7 : SUCCESS in 14m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/69ad42c13d054899bf8757b014081f02 : SUCCESS in 5m 00s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/7e602c39af474f209b6705ad31c17c38 : SUCCESS in 4m 41s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/9a934f49709b4342b27a146b1bcc137e : SUCCESS in 6m 33s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/29eac6db6a32470596c4e2edc97eed84 : SUCCESS in 1h 13m 01s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/0e53b6028a314ad7a9e4dfc120c80def : SUCCESS in 1h 12m 44s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/d7400aca530f4f13add670f397fd145f : SUCCESS in 1h 02m 52s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/03aac0e2756d480794880a3511ac794f : SUCCESS in 33m 46s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/4636f67f14c343a98b30333a93e19a80 : SUCCESS in 46m 08s","accounts_in_message":[],"_revision_number":1},{"id":"6461b988c083b99c3e421290ddfef3700f098167","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-16 08:31:46.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"9ddf3327178b55df853d5648474257ec3ee56459","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-16 08:33:49.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c47a5eaf22b843b7b4a9cd25e570dab6\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2d27d8c717d1493b9e2410604eb0fee1 : SUCCESS in 48s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/202cd40a9b564372a79b39ae5d5ddd4c : SUCCESS in 41s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"6fdc2079da3392fdfbdf4a80df05bfe7d149ca95","revisions":{"6fdc2079da3392fdfbdf4a80df05bfe7d149ca95":{"kind":"REWORK","_number":1,"created":"2023-10-10 09:57:44.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/84/897784/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/84/897784/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/84/897784/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/84/897784/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/84/897784/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/84/897784/1"}}},"commit":{"parents":[{"commit":"54eb1624f5762e9b848f10f712152faf02f8b84d","subject":"Fix timeout duration in start_vrrp_service during failovers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/54eb1624f5762e9b848f10f712152faf02f8b84d"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-09-01 13:08:37.000000000","tz":-240},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-10-10 09:54:20.000000000","tz":120},"subject":"Reduce duration of failovers with amphora in ERROR","message":"Reduce duration of failovers with amphora in ERROR\n\nIn the failover flow, there are multiple tasks for the configuration of\nVRRP for the other amphorae of the load balancer, but during outage the\nother amps may not be available. To prevent the tasks from attempting\nconnections to unreachable amphorae, we can detect in the first task\nthat an amp is unreachable and pass this information to the other tasks.\n\nThose connection attempts could have taken a lot of time, between 15 min\nand 40 min depending on the configuration of Octavia and the provider\ndriver (amphorav1 or amphorav2)\n\nCloses-Bug: #2033894\n\nChange-Id: Ib33a0b8d2875e4ff97c65933fe9360bb06994d32\n(cherry picked from commit ca70587f0bcb4b2a05bd5c243f111140a5762114)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6fdc2079da3392fdfbdf4a80df05bfe7d149ca95"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6fdc2079da3392fdfbdf4a80df05bfe7d149ca95"}]},"branch":"refs/heads/stable/2023.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
