)]}'
{"id":"openstack%2Foctavia~897785","triplet_id":"openstack%2Foctavia~stable%2F2023.2~I4bd027346c61b93b537ab53810c2ecb6160b6be2","project":"openstack/octavia","branch":"stable/2023.2","attention_set":{},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2023-10-16 09:16:59.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I4bd027346c61b93b537ab53810c2ecb6160b6be2","subject":"Fix amphorae in ERROR during the failover","status":"MERGED","created":"2023-10-10 09:57:44.000000000","updated":"2023-10-16 09:18:24.000000000","submitted":"2023-10-16 09:16:59.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"897785","meta_rev_id":"1075f1f16788294f745ba23e7ac730e306d60359","_number":897785,"virtual_id_number":897785,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-10-16 09:16:58.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"recommended":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"all":[{"value":1,"date":"2023-10-10 10:34:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":2,"date":"2023-10-12 17:08:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-10-16 06:51:46.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-10-16 06:51:46.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-10-10 10:34:59.000000000","updated_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"reviewer":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"state":"REVIEWER"},{"updated":"2023-10-10 12:33:53.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-10-12 17:08:20.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2023-10-16 06:51:46.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"}],"messages":[{"id":"fcded52d6d60ad8c8343389828c7257c3fc5ab68","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-10-10 09:57:44.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4a177e066827d91c7255239eb06e0737e971a6c1","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2023-10-10 10:34:59.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6c8a3d74246ff75f298d54762353d3928648f174","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-10 12:33:53.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8f6bde4cbcd846ed9caa62d83db3ee1c\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/31197234f88447c3902a80b884ac6f7a : SUCCESS in 10m 58s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2596d4eda293465698a52dc31e3f7156 : SUCCESS in 5m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8c5d1851442541ff89de1cfb8d619e9d : SUCCESS in 5m 06s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c036c0933181442d984f369e3ae636af : SUCCESS in 4m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ed142e67e1044e5783d7c5f6b772217b : SUCCESS in 5m 51s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/9306a53cdad345b4925f3264fdf61327 : SUCCESS in 5m 21s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8b21bba1252b4c8aa2aa717338d5104b : SUCCESS in 12m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e8ba7bd0490046c3a786b6b71a796934 : SUCCESS in 6m 12s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/70ab3f8007184adca99e4bf478961dc7 : SUCCESS in 5m 26s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/e39e1880a6bd45099cb93cb9f17e205d : SUCCESS in 8m 08s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/50d457c2749f4034bc92d90a6e319af5 : SUCCESS in 5m 27s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/98c5a69b67a444bb9a58b5b2057ff129 : SUCCESS in 7m 23s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/b1f0bfc687a04920b0b84ef7d9a15288 : SUCCESS in 8m 28s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/4635e49b355d455c93371a01fdbdfec5 : SUCCESS in 7m 12s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/66e8600c61034e5daa7afa9a4b18d47c : SUCCESS in 1h 12m 22s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/b64a6b65a64d473b9896366aaadc4571 : SUCCESS in 1h 39m 52s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/87e4801c1b2c41728d99e0375ba1a3b9 : SUCCESS in 1h 15m 16s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/dc42c681bcc147c3b506c1a4fee6e60d : SUCCESS in 1h 39m 32s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/f5cce9428a494b0090d0162f43a6f936 : SUCCESS in 1h 22m 19s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/4257790b0a2743888ac7d7b277281c51 : SUCCESS in 42m 05s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/f547e3615a25451b8035f83171477118 : SUCCESS in 50m 35s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/6a0e8294f2cb484483f637e781ac5cf6 : FAILURE in 34m 32s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/fe0a2704d5924710b569eb80c657f42a : FAILURE in 10m 46s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3c7167e8162e41789c8dc592b972b972 : SUCCESS in 40m 58s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/9ec335fec1c54911a6ce6ec76fb39b68 : SUCCESS in 2h 22m 21s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/f458038766ed410585ea37422316d7f1 : SUCCESS in 2h 27m 37s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/75f441af8c4a444196137a525a4027f2 : SUCCESS in 1h 49m 34s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/a77dd94d53f8403f889589d9a7c3de7c : FAILURE in 9m 01s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"04d972c4e16b276db3a64ec3a751c70b9183f371","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-10-12 13:42:04.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ece5ac0ee73de102fcdd814230196bb30af57273","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-12 15:39:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/139d1820a1c8413e8b876f6d5ae668b6\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6d58affe4cb14edb9a0c894aa352d412 : SUCCESS in 12m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0e44ee6c92414dd9a532eb97d8cf7167 : SUCCESS in 5m 12s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/19d1d044ce3848adb53b6b48d6849e6b : SUCCESS in 5m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/71e4b8c9034c41799ced7f834588ea59 : SUCCESS in 5m 14s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/28ae55fdb1464802a92624478228eaa4 : SUCCESS in 5m 04s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/18745000e22849a9aab7aed7e9630e98 : SUCCESS in 5m 02s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f060f17a305b4b7b863e9e3cfd4b188e : SUCCESS in 12m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7c734ff0a6a8447fa5fbdf8482cbf512 : SUCCESS in 5m 23s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/2295999e035c4d35bc6e9ec374b12e09 : SUCCESS in 5m 08s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/a3235d1b882d44ed900f0a44bb7ee8dc : SUCCESS in 7m 39s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/2e8b57b1aa0841ca8296df9c811d9109 : SUCCESS in 4m 42s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/0b1737e1b52648f1856eb170f8dc38dc : SUCCESS in 7m 11s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/645b9b43bcd24b8398b098e33b472602 : SUCCESS in 8m 38s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/2121a0f50ae54b4ab5018b10dd3d6a91 : SUCCESS in 8m 23s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/2dd941c40d894a0c88a1e51e9be6dc5d : SUCCESS in 1h 12m 51s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/e774b1a6fbd1408da2d2f6472e6c6bb9 : SUCCESS in 1h 15m 06s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/665f5d8520294614870386c9599fbb9e : SUCCESS in 1h 04m 53s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/e22ead8a828d40168c385b112f7e9ef4 : SUCCESS in 1h 39m 32s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/776990e9272441b7aa7f37f750a30de8 : SUCCESS in 1h 15m 13s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/9e925eb58d19439dbfb7f14edfab5efb : SUCCESS in 41m 30s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/01dbff9e380249bcad925bd1d156b2d5 : SUCCESS in 47m 06s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/67186738efc641c98393e55b3fd0018a : FAILURE in 14m 59s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3cd20433807c4e47906c831cd06f3ca4 : SUCCESS in 30m 54s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/0550b8e672684d2bbe9d51a246553285 : SUCCESS in 1h 54m 52s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/d1482f2bcba44a50a3832e765acbb366 : SUCCESS in 1h 48m 32s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/67f81d9a6e6b419383498f1f1331e038 : SUCCESS in 1h 54m 09s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/93663417cedf43e2828914fed6c2dea0 : FAILURE in 12m 06s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a302ba236511d9cb45b91267c67eb902d2ac79f2","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-10-12 17:08:20.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c40d4e6ba296ebb0cd9ecba67f8376e2551fa7b0","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-10-16 06:51:46.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d581c7e795a11d3519d09b6a260a016117af3858","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-16 06:52:44.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"4c692bc4326c02dfcccbeb5bdd27d4b57d70462c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-16 09:16:58.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0e24dee0926440598bc92fed7c1239fe\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/31506ef2bfe74f228146253fa1dee8bd : SUCCESS in 5m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9fe6d65a0ba14864a6e344697f020a1c : SUCCESS in 6m 05s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a884e5c1dc46464a83e8a42373c5d5cf : SUCCESS in 5m 41s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f6cdcfe602b0418db7ef9cd5ca25f304 : SUCCESS in 5m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b8cb0fb345f441dab841d4a2c7441499 : SUCCESS in 11m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/da85c7898ac5402fbf2efbb41c07010d : SUCCESS in 5m 05s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/e21d9de9485445cbb9fb177363038aca : SUCCESS in 4m 58s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/e21f086a11fc47c4a5fe40b1055042fb : SUCCESS in 7m 20s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/c0805daacbae4e2da70d3ec44e10aca8 : SUCCESS in 1h 07m 22s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/e18d3e02841149bdbb0aecd01a98f23e : SUCCESS in 1h 21m 56s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/b571e9564d74459a80211379a9e195e9 : SUCCESS in 1h 02m 07s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/4653b4adca9f4f21b811ef0d5f2f4138 : SUCCESS in 34m 42s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/770f8f746ea14b888d8ce5a8c6bca2e9 : SUCCESS in 38m 00s","accounts_in_message":[],"_revision_number":1},{"id":"4f02e578be53a415b35b043f6c76fc0f21947ee5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-16 09:16:59.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"1075f1f16788294f745ba23e7ac730e306d60359","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-16 09:18:24.000000000","message":"Patch Set 1:\n\nBuild failed (promote pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f589a7df80714d36bcfbf4d41e5adb64\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d8680dea20b74932b9a46d006e92163a : SUCCESS in 57s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2596ba3623474d2b9b62afa65a37ae69 : FAILURE in 41s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5622431534d3bf9b8b71ad8fdb2d744f94625ec4","revisions":{"5622431534d3bf9b8b71ad8fdb2d744f94625ec4":{"kind":"REWORK","_number":1,"created":"2023-10-10 09:57:44.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/85/897785/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/85/897785/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/85/897785/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/85/897785/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/85/897785/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/85/897785/1"}}},"commit":{"parents":[{"commit":"6fdc2079da3392fdfbdf4a80df05bfe7d149ca95","subject":"Reduce duration of failovers with amphora in ERROR","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6fdc2079da3392fdfbdf4a80df05bfe7d149ca95"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-09-04 09:56:25.000000000","tz":-240},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-10-10 09:54:20.000000000","tz":120},"subject":"Fix amphorae in ERROR during the failover","message":"Fix amphorae in ERROR during the failover\n\nWhen 2 amps were down, the failover flow created the first one and\nneeded to update both amp to configure VRRP, but as the 2nd was missing,\nit was set to ERROR. Then the health-manager could not trigger a\nfailover becasue amphorae in ERROR are excluded from the automated\nfailover process.\n\nThis commit changes the tasks that must be run on both amphorae during a\nfailover of one amphora, it doesn\u0027t mark the secondary amphora in ERROR\nif it is not reachable.\n\nCloses-Bug: #2033734\n\nChange-Id: I4bd027346c61b93b537ab53810c2ecb6160b6be2\n(cherry picked from commit 248cf2893ea913a74d870a763802a51586e0f63b)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5622431534d3bf9b8b71ad8fdb2d744f94625ec4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5622431534d3bf9b8b71ad8fdb2d744f94625ec4"}]},"branch":"refs/heads/stable/2023.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
