)]}'
{"id":"openstack%2Foctavia~897786","triplet_id":"openstack%2Foctavia~stable%2Fzed~I497f2d8587e25eda0d169aa919cb6fbede7e9fc6","project":"openstack/octavia","branch":"stable/zed","attention_set":{},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2023-10-16 09:16:45.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I497f2d8587e25eda0d169aa919cb6fbede7e9fc6","subject":"Fix timeout duration in start_vrrp_service during failovers","status":"MERGED","created":"2023-10-10 10:20:05.000000000","updated":"2023-10-16 09:17:52.000000000","submitted":"2023-10-16 09:16:45.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"897786","meta_rev_id":"9dd7fe2b09688ab3c00991b0e50ef4f196004293","_number":897786,"virtual_id_number":897786,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-10-16 09:16:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"recommended":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"all":[{"value":1,"date":"2023-10-12 09:23:52.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":2,"date":"2023-10-12 17:10:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-10-16 06:49:11.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-10-16 06:49:11.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-10-10 12:59:08.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-10-10 15:56:53.000000000","updated_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"reviewer":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"state":"REVIEWER"},{"updated":"2023-10-12 17:10:34.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2023-10-16 06:49:11.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"}],"messages":[{"id":"a0818517c5e8dfd624d056c44886694a048a4a58","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-10-10 10:20:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"a523f4249920a3b31aeb50a6370b5e737531d6d7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-10 12:59:08.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8c58f12cbabb43f7af2b875fecb014c2\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/976c7431787e4d96aee9060eb90ae073 : SUCCESS in 11m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/03385939e25a4948884adbbdf682650d : SUCCESS in 6m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b004eaacf6b24ef08d6fb1ea9b79b808 : SUCCESS in 4m 57s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/bd35d95f41b34e93a5c223eb16b8e8e9 : SUCCESS in 5m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/aa62660ee9054018986ff64c67845ab0 : SUCCESS in 4m 59s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/073447de0b1842f19c6384ca949bad5a : SUCCESS in 12m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1665de4768414eb283ce208e4fc0071d : SUCCESS in 6m 00s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/b3c23adaca8b44f8a6d968b8e48dec29 : SUCCESS in 5m 33s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/734fc1d85ba0474ea8cac8d2fe2299b5 : SUCCESS in 8m 02s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/1fb540ef3eb940ebaf90dd377a2273df : SUCCESS in 5m 02s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/67c5745af8474003a34bbbb97be4745b : SUCCESS in 7m 14s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/3210bdaa5b754bb2b396524279a23f2f : SUCCESS in 7m 59s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/e2cf2f11ff0f40c8a26246be2e2b031d : SUCCESS in 9m 39s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/deb105093ea74a2da1b58773216e5a42 : SUCCESS in 1h 30m 41s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/db31ef122cc84b7abecb8b906922d843 : SUCCESS in 1h 55m 00s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/7ae43045f17148d1af1cd3698f9841cf : SUCCESS in 35m 00s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/bdbaa9533fe24b3bb2bed3f9b5da3c18 : SUCCESS in 38m 49s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/01a4b26f80fa47ba91a8f35c940db45f : FAILURE in 11m 01s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c3a3aaa374bf4e639f3e6faac2957239 : SUCCESS in 35m 57s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/6ee134587c0b4558a634d9e539b8770f : SUCCESS in 2h 14m 31s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/01d7efb5b7684b1ca5c42664475ba849 : SUCCESS in 2h 35m 20s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/55bc253ae84a47d5b090631e8f8409f4 : SUCCESS in 2h 02m 36s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/eabc4a9eeb4b44399a1ff90fff179e26 : SUCCESS in 2h 05m 33s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/1b85a07d549b40b09c23dde129e70a56 : SUCCESS in 2h 10m 32s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/86bb41c15863422f97f23c306f8c437a : FAILURE in 7m 54s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d2cf4912ad1d22b47265daca3b1e2f3c9043766c","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2023-10-10 15:56:53.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f65348fa270d5c0433a3879d22875d8ef6052dbc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-10-12 09:23:52.000000000","message":"Uploaded patch set 2: New patch set was added with same tree, parent tree, and commit message as Patch Set 1.\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"ce7144a603370141b09f4a98765068470703c967","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-12 11:45:51.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8949f887826246acae09eaf46215a1e7\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f44c765f98484587817bd15dc6fcc189 : SUCCESS in 11m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/73a8544b758340c098b6ebaa88f4cd60 : SUCCESS in 7m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9c63c74363554d109ae093d764f75529 : SUCCESS in 5m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ab9774b7435b44308b82fa195dcf9794 : SUCCESS in 5m 26s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c7228e79438c41b29934676d093f51ee : SUCCESS in 6m 29s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2ecda9ee2a5e411095b983563b473a52 : SUCCESS in 11m 50s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4c4eaaf551bc4b0fa2ea54995e1874f7 : SUCCESS in 5m 11s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/21d009621bb741939a423ada25730f7c : SUCCESS in 5m 08s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/0a2a38d81fc84b90858d51a4ceb5171e : SUCCESS in 9m 30s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/ebe6b9d052864700a8e0ee9939540d2c : SUCCESS in 5m 10s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/e8a7c450710b4dc8b065ec21d08543c4 : SUCCESS in 8m 06s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/0c31e01e65e24198be91dc86d75adcad : SUCCESS in 8m 41s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/6cd8edebb2f640339578b383e38c3b7e : SUCCESS in 7m 21s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/e6d3d0d322ab48fa801c38f20ac36057 : SUCCESS in 1h 28m 33s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/314a6b58807547478cee9af6befcf76d : SUCCESS in 1h 48m 25s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/55566f879b0c427ba10333735239f8df : SUCCESS in 37m 18s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/8db3589a8ef74ff1aaad198e1689a4f6 : SUCCESS in 39m 53s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/58e278b192524bb38ebd74abde454150 : SUCCESS in 1h 04m 45s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/da02bfb2153e4469a0454e3ab6fe0b57 : SUCCESS in 33m 35s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/2aa4175a9277466f8c22b87e6bdebe9c : SUCCESS in 1h 50m 09s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/b7585285def54405a20f4f3556f8ab8a : SUCCESS in 2h 20m 35s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/52a10bbf7f0f43ef8b44518942f0916e : SUCCESS in 1h 56m 18s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/b6529946102041b191ef569026634dac : SUCCESS in 1h 40m 19s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/e41d631704c140e49b939d92729af479 : SUCCESS in 1h 38m 18s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/f0988b39a7094c74931353133b3cf4a1 : FAILURE in 2h 14m 31s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"311acd558b27c7b91203f4776c48bdd6e6a505a8","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-10-12 17:10:34.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"61d3a5ad187883f255d1ac64275efae88a3a5ff8","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-10-16 06:49:11.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"96df36224b32fda91f7af660afbe9ed3fc702a73","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-16 06:49:56.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"4828b4eb981f4be65462a9f755778d4a4974cde2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-16 09:16:45.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7006bcc334b145f5b1f9d009fde46833\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f4867a644eba4ec58a69a9df317762b9 : SUCCESS in 7m 02s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6880a8b7f29e41818ede0e29070b6e73 : SUCCESS in 4m 45s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2eca8bf0c0174dc7880365729e6d6010 : SUCCESS in 5m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/285ad3f218ac46dbb6db75bc7f19ec5a : SUCCESS in 10m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4201672b35f44d37bb4cdc8a4fe5ae06 : SUCCESS in 5m 15s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/7ed7dbefc2324283b318fa981fd77017 : SUCCESS in 4m 30s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/c7de0f64cefa4f2c922ce6d12155a262 : SUCCESS in 8m 14s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/1d9ef7be698446d1ae3f21d6e6b05bb1 : SUCCESS in 1h 21m 05s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d169a7d54d6548b3bfec187b6bc0d6ac : SUCCESS in 2h 11m 38s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/e969b66e7ffd4e029d62b0a33ff433fb : SUCCESS in 44m 13s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/7ce421a4288143f1b58504c4e3650973 : SUCCESS in 50m 48s","accounts_in_message":[],"_revision_number":2},{"id":"477e5ce6d379221f00908a317bb805ca953afa90","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-16 09:16:45.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"9dd7fe2b09688ab3c00991b0e50ef4f196004293","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-16 09:17:52.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d16a44623c8c479598c567f5b7733691\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/115fa5c901cc42d0a8e88c075d120e19 : SUCCESS in 44s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3e1c250246b44d18a2c0e080d7af17ec : SUCCESS in 43s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"1dadaff530e252e8e96ea8b4c6dcec902a797516","revisions":{"e61f80e773fc404edb4e50977fd7c4fc70bf2271":{"kind":"REWORK","_number":1,"created":"2023-10-10 10:20:05.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/86/897786/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/86/897786/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/86/897786/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/86/897786/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/86/897786/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/86/897786/1"}}},"commit":{"parents":[{"commit":"4c97b585ce6e519a5ff090c2cd97fab0e801be0c","subject":"Fix race condition in members batch update API call","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4c97b585ce6e519a5ff090c2cd97fab0e801be0c"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-09-01 15:35:34.000000000","tz":-240},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-10-10 10:19:13.000000000","tz":120},"subject":"Fix timeout duration in start_vrrp_service during failovers","message":"Fix timeout duration in start_vrrp_service during failovers\n\nThe timeout_dict argument was not passed to the\n_populate_amphora_api_version, it could have made the failover of broken\namphora really long.\n\nRelated-Bug: #2033894\n\nChange-Id: I497f2d8587e25eda0d169aa919cb6fbede7e9fc6\n(cherry picked from commit cce348040d43b92b59146adeb15a4f1486946a64)\n(cherry picked from commit 54eb1624f5762e9b848f10f712152faf02f8b84d)\n(cherry picked from commit 5c98d901337cb0d6e4c575e3280cb77839720cfe)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e61f80e773fc404edb4e50977fd7c4fc70bf2271"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e61f80e773fc404edb4e50977fd7c4fc70bf2271"}]},"branch":"refs/heads/stable/zed"},"1dadaff530e252e8e96ea8b4c6dcec902a797516":{"kind":"NO_CHANGE","_number":2,"created":"2023-10-12 09:23:52.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/86/897786/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/86/897786/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/86/897786/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/86/897786/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/86/897786/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/86/897786/2"}}},"commit":{"parents":[{"commit":"4c97b585ce6e519a5ff090c2cd97fab0e801be0c","subject":"Fix race condition in members batch update API call","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4c97b585ce6e519a5ff090c2cd97fab0e801be0c"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-09-01 15:35:34.000000000","tz":-240},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-10-12 09:22:02.000000000","tz":-240},"subject":"Fix timeout duration in start_vrrp_service during failovers","message":"Fix timeout duration in start_vrrp_service during failovers\n\nThe timeout_dict argument was not passed to the\n_populate_amphora_api_version, it could have made the failover of broken\namphora really long.\n\nRelated-Bug: #2033894\n\nChange-Id: I497f2d8587e25eda0d169aa919cb6fbede7e9fc6\n(cherry picked from commit cce348040d43b92b59146adeb15a4f1486946a64)\n(cherry picked from commit 54eb1624f5762e9b848f10f712152faf02f8b84d)\n(cherry picked from commit 5c98d901337cb0d6e4c575e3280cb77839720cfe)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1dadaff530e252e8e96ea8b4c6dcec902a797516"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1dadaff530e252e8e96ea8b4c6dcec902a797516"}]},"branch":"refs/heads/stable/zed"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
