)]}'
{"id":"openstack%2Foctavia~897787","triplet_id":"openstack%2Foctavia~stable%2Fzed~Ib33a0b8d2875e4ff97c65933fe9360bb06994d32","project":"openstack/octavia","branch":"stable/zed","attention_set":{},"removed_from_attention_set":{"31664":{"account":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"last_update":"2023-10-16 09:16:47.000000000","reason":"Change was submitted"},"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2023-10-16 09:16:47.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ib33a0b8d2875e4ff97c65933fe9360bb06994d32","subject":"Reduce duration of failovers with amphora in ERROR","status":"MERGED","created":"2023-10-10 10:20:05.000000000","updated":"2023-10-16 09:16:47.000000000","submitted":"2023-10-16 09:16:47.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"897787","meta_rev_id":"922210459aa113c000fe1676005508de14da6ed5","_number":897787,"virtual_id_number":897787,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-10-16 09:16:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":2,"date":"2023-10-12 17:08:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":2,"date":"2023-10-16 06:46:43.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":1,"date":"2023-10-16 06:46:43.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-10-10 13:27:07.000000000","updated_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"reviewer":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"state":"CC"},{"updated":"2023-10-10 13:47:32.000000000","updated_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"reviewer":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"state":"REVIEWER"},{"updated":"2023-10-10 14:21:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-10-12 17:08:35.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2023-10-16 06:46:43.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"}],"messages":[{"id":"c4af96214f65aae608b37fc0ce68dcc79ee20287","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-10-10 10:20:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"521d543b6cd004f2c9c91dcb33810f623d51cdc5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-10-10 11:32:02.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"7f3bc07ac39bf95465ddbc89f560bc963f6ebeb7","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2023-10-10 13:27:07.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d4d8ea7552b71772cef1139da7bfe8ebeec7eb85","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2023-10-10 13:47:32.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ade9e56810f1e33f40b709fc5522a90efa8bf73c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-10 14:21:42.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/004eb71a2e634140b3554238ea8d172f\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3b6e11ff024948168cd6f014f899c961 : SUCCESS in 9m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/217de311932e45d0adaa8bc48bd44546 : SUCCESS in 6m 20s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ec7a3a03a16f4c70b5147f9b716d0796 : SUCCESS in 4m 38s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/146188ab1a924fcfa9abdd728aac2481 : SUCCESS in 4m 59s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/23c3b618f147472aac03cc600b948a36 : SUCCESS in 5m 08s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ba69092ac824420a0abb91b9861d584 : SUCCESS in 11m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2def18fd50714f80867d17cd1501c16a : SUCCESS in 6m 24s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/597494206d0f475483e6179cef4335c5 : SUCCESS in 5m 08s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/280b69c7ed1b4fcea56d5f95ecf1fad6 : SUCCESS in 7m 10s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/ea6c81084c0247d1930a65155f090385 : SUCCESS in 4m 55s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/5ba9ee06127d4617953e4c93f9c8cd2e : SUCCESS in 7m 36s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/2116f3fdcb2c4da5b9ea297417e30555 : SUCCESS in 9m 29s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/a6dddec95d9c4d77bfd210dd13296754 : SUCCESS in 6m 59s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/d2547cfa27f24d3f9b4f9e0987d47f69 : SUCCESS in 1h 05m 54s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ee689354b2ed4656ae250d2258017ed5 : SUCCESS in 1h 56m 09s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/1709efbc15f24d88aec28c99617423c4 : SUCCESS in 37m 22s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/db918532becf47b0a8612d6d845acc68 : SUCCESS in 38m 59s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/c7a785b832704a5aa3cc7d1c69d29db3 : FAILURE in 10m 44s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/07ddf95ba42f4eb7aad20842354fe59a : SUCCESS in 32m 46s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/230e02af79c442e3a300793b26a64379 : SUCCESS in 2h 27m 24s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/d297a7e86c6a44e1b6149f8d4b4e910a : FAILURE in 2h 45m 13s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/f7c56f89450c49eb97dc27e1d0e6a492 : TIMED_OUT in 2h 34m 14s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/e083884b702c43fcad2f376c0ff9bead : SUCCESS in 2h 23m 15s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/60d2064977d6491bad081dca4a59f61b : TIMED_OUT in 2h 34m 28s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/0dfba532d4eb4228b4415c21734f930b : FAILURE in 7m 40s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"76fa54dab16ee8639d7bb481d19cfd74a5fdb09b","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-10-11 08:10:29.000000000","message":"Patch Set 2: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"8ecfc34bd1095f7bebeabdd4fb3d6fece1f74ea3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-10-12 09:23:52.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":3},{"id":"94600d436c342f99b815f95157156f2713c8ba29","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-10-12 09:37:08.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"791aa66debf844eef1bad4f74e2b6578ac8855e7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-12 12:33:29.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/287a8c5d6d6144c3a3561f38cb04c41b\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3acad8cf9bd143eba4b016ccb50e9675 : SUCCESS in 10m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/46bfb38c53d24ab59ee913c1b09b1924 : SUCCESS in 6m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cbaaa84e14164d3487c5c157ba52a1a5 : SUCCESS in 5m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fe1e1f41d0be48f384c9e4033bb7aae0 : SUCCESS in 5m 06s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d15be4a93a864690a7cf6b4adcbd105f : SUCCESS in 5m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0b0ef33e55a64c39a35a1b8cd29ab416 : SUCCESS in 11m 12s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9646e3a71f334a96a78592f1cabcf0ca : SUCCESS in 5m 56s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/352bee1c186147019490e633f2c6f549 : SUCCESS in 5m 40s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/dbd546e465e143ad875a23dce8d73a3b : SUCCESS in 8m 07s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/9b4973fce5d6437aaa32b33e816245e6 : SUCCESS in 5m 17s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/c0bcf2abe8b3461cb5572149a835cd75 : SUCCESS in 7m 20s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/ae7022fa8b7946289ea3826bc84ab620 : SUCCESS in 8m 09s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/003b9748c85f4cf78d265423eacd671b : SUCCESS in 8m 38s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/16994b3116fd4ad283984dcbf813950c : SUCCESS in 1h 33m 58s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f96a005ec2684972b0e0c25543a724d7 : SUCCESS in 2h 15m 26s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/0e2ff916ad184463bbddcd08c008904c : SUCCESS in 47m 03s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/bd7efbd8612d4342a959f415be6940bb : SUCCESS in 56m 30s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/87539e16e33d4967b4f8b29aa780f3b7 : SUCCESS in 1h 06m 45s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6cd1d399148c4d67bcb10609fa1556bf : SUCCESS in 43m 44s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/42acc33549d7488da5d12f1703bff702 : SUCCESS in 2h 05m 38s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/6e64848bdd9340efae54cf58c562cef3 : SUCCESS in 2h 51m 59s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/2cbf77a2de1e4fda94974feff0e02d82 : SUCCESS in 2h 10m 54s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/a6f4a09e56b14f8bbe371e8a7118d599 : SUCCESS in 1h 52m 58s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/9015c617a9ce43c097d12522a340ffbd : SUCCESS in 1h 57m 51s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/daf3e74e457e413ca56ddf28863d6fef : SUCCESS in 2h 10m 20s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"e09972aefd3df14d9b22f898e83833975bdd188b","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-10-12 17:08:35.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"38604f088e9d8365bfb3d9b07f60ed347f268937","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-10-16 06:46:43.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"fe10d45b823a2638a53ce2fd64229be45244a322","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-16 06:50:01.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"995c1ea7eb8d36edde5f69893df5a33e8494511a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-16 09:16:47.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d32534877bcf4cf58af9ca68c90090f9\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/44f6e855b1664d3d8028dc0b30c74a41 : SUCCESS in 6m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/511d0a60a09543128a4b414b6a98d6cd : SUCCESS in 5m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ea467f63d20a4425bfb4c74f6a6622a0 : SUCCESS in 5m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3d120910dccd40128cc05c368907c646 : SUCCESS in 14m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/facca6f0d4de40ea834a55a3f0c30e74 : SUCCESS in 5m 38s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/e51d52e5fe8f4ad98adf04f1c9b9e006 : SUCCESS in 4m 37s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/bee5ab19ab1742e18f63380dca661bd4 : SUCCESS in 6m 47s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/0c649aa85f6b468fa95c4ed14d549ba3 : SUCCESS in 1h 12m 53s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/59c527bd3353490fa9dd460befd03140 : SUCCESS in 1h 31m 37s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/9ecfe8f6f0bc4bfdb2de39837b385315 : SUCCESS in 34m 29s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/e7661777b3344fdfb18123cd74059e89 : SUCCESS in 37m 29s","accounts_in_message":[],"_revision_number":4},{"id":"922210459aa113c000fe1676005508de14da6ed5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-16 09:16:47.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"dd849b4c5c1a1d33e4fd4a9afa2f35903d4d9aac","revisions":{"61e27ffec304beee6325f915d78ada199f575ef7":{"kind":"REWORK","_number":1,"created":"2023-10-10 10:20:05.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/87/897787/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/87/897787/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/87/897787/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/87/897787/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/87/897787/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/87/897787/1"}}},"commit":{"parents":[{"commit":"e61f80e773fc404edb4e50977fd7c4fc70bf2271","subject":"Fix timeout duration in start_vrrp_service during failovers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e61f80e773fc404edb4e50977fd7c4fc70bf2271"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-09-01 13:08:37.000000000","tz":-240},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-10-10 10:19:17.000000000","tz":120},"subject":"Reduce duration of failovers with amphora in ERROR","message":"Reduce duration of failovers with amphora in ERROR\n\nIn the failover flow, there are multiple tasks for the configuration of\nVRRP for the other amphorae of the load balancer, but during outage the\nother amps may not be available. To prevent the tasks from attempting\nconnections to unreachable amphorae, we can detect in the first task\nthat an amp is unreachable and pass this information to the other tasks.\n\nThose connection attempts could have taken a lot of time, between 15 min\nand 40 min depending on the configuration of Octavia and the provider\ndriver (amphorav1 or amphorav2)\n\nCloses-Bug: #2033894\n\nConflicts:\n\toctavia/controller/worker/v2/tasks/amphora_driver_tasks.py\n\nChange-Id: Ib33a0b8d2875e4ff97c65933fe9360bb06994d32\n(cherry picked from commit ca70587f0bcb4b2a05bd5c243f111140a5762114)\n(cherry picked from commit 6fdc2079da3392fdfbdf4a80df05bfe7d149ca95)\n(cherry picked from commit 1c0b702f887cb4fe48a44d6a6972d337a9a3f187)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/61e27ffec304beee6325f915d78ada199f575ef7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/61e27ffec304beee6325f915d78ada199f575ef7"}]},"branch":"refs/heads/stable/zed"},"4c8c3360273f4c1a02f7401bb65ac578c018938d":{"kind":"REWORK","_number":2,"created":"2023-10-10 11:32:02.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/87/897787/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/87/897787/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/87/897787/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/87/897787/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/87/897787/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/87/897787/2"}}},"commit":{"parents":[{"commit":"e61f80e773fc404edb4e50977fd7c4fc70bf2271","subject":"Fix timeout duration in start_vrrp_service during failovers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e61f80e773fc404edb4e50977fd7c4fc70bf2271"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-09-01 13:08:37.000000000","tz":-240},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-10-10 11:30:58.000000000","tz":120},"subject":"Reduce duration of failovers with amphora in ERROR","message":"Reduce duration of failovers with amphora in ERROR\n\nIn the failover flow, there are multiple tasks for the configuration of\nVRRP for the other amphorae of the load balancer, but during outage the\nother amps may not be available. To prevent the tasks from attempting\nconnections to unreachable amphorae, we can detect in the first task\nthat an amp is unreachable and pass this information to the other tasks.\n\nThose connection attempts could have taken a lot of time, between 15 min\nand 40 min depending on the configuration of Octavia and the provider\ndriver (amphorav1 or amphorav2)\n\nCloses-Bug: #2033894\n\nConflicts:\n\toctavia/controller/worker/v2/tasks/amphora_driver_tasks.py\n\nChange-Id: Ib33a0b8d2875e4ff97c65933fe9360bb06994d32\n(cherry picked from commit ca70587f0bcb4b2a05bd5c243f111140a5762114)\n(cherry picked from commit 6fdc2079da3392fdfbdf4a80df05bfe7d149ca95)\n(cherry picked from commit 1c0b702f887cb4fe48a44d6a6972d337a9a3f187)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4c8c3360273f4c1a02f7401bb65ac578c018938d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4c8c3360273f4c1a02f7401bb65ac578c018938d"}]},"branch":"refs/heads/stable/zed"},"90717d4a1d4854f7eff865940598e013a5b973d2":{"kind":"REWORK","_number":3,"created":"2023-10-12 09:23:52.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/87/897787/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/87/897787/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/87/897787/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/87/897787/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/87/897787/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/87/897787/3"}}},"commit":{"parents":[{"commit":"1dadaff530e252e8e96ea8b4c6dcec902a797516","subject":"Fix timeout duration in start_vrrp_service during failovers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1dadaff530e252e8e96ea8b4c6dcec902a797516"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-09-01 13:08:37.000000000","tz":-240},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-10-12 09:22:08.000000000","tz":-240},"subject":"Reduce duration of failovers with amphora in ERROR","message":"Reduce duration of failovers with amphora in ERROR\n\nIn the failover flow, there are multiple tasks for the configuration of\nVRRP for the other amphorae of the load balancer, but during outage the\nother amps may not be available. To prevent the tasks from attempting\nconnections to unreachable amphorae, we can detect in the first task\nthat an amp is unreachable and pass this information to the other tasks.\n\nThose connection attempts could have taken a lot of time, between 15 min\nand 40 min depending on the configuration of Octavia and the provider\ndriver (amphorav1 or amphorav2)\n\nCloses-Bug: #2033894\n\nNote: stable/2023.1 and older, the patch also includes modifications in\n      octavia/controller/worker/v1/\n\nConflicts:\n\toctavia/controller/worker/v2/tasks/amphora_driver_tasks.py\n\nChange-Id: Ib33a0b8d2875e4ff97c65933fe9360bb06994d32\n(cherry picked from commit ca70587f0bcb4b2a05bd5c243f111140a5762114)\n(cherry picked from commit 6fdc2079da3392fdfbdf4a80df05bfe7d149ca95)\n(cherry picked from commit 97b1b8387e42f8361b612a0b81b1c60b6e14db37)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/90717d4a1d4854f7eff865940598e013a5b973d2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/90717d4a1d4854f7eff865940598e013a5b973d2"}]},"branch":"refs/heads/stable/zed"},"dd849b4c5c1a1d33e4fd4a9afa2f35903d4d9aac":{"kind":"REWORK","_number":4,"created":"2023-10-12 09:37:08.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/87/897787/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/87/897787/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/87/897787/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/87/897787/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/87/897787/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/87/897787/4"}}},"commit":{"parents":[{"commit":"1dadaff530e252e8e96ea8b4c6dcec902a797516","subject":"Fix timeout duration in start_vrrp_service during failovers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1dadaff530e252e8e96ea8b4c6dcec902a797516"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-09-01 13:08:37.000000000","tz":-240},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-10-12 09:35:42.000000000","tz":-240},"subject":"Reduce duration of failovers with amphora in ERROR","message":"Reduce duration of failovers with amphora in ERROR\n\nIn the failover flow, there are multiple tasks for the configuration of\nVRRP for the other amphorae of the load balancer, but during outage the\nother amps may not be available. To prevent the tasks from attempting\nconnections to unreachable amphorae, we can detect in the first task\nthat an amp is unreachable and pass this information to the other tasks.\n\nThose connection attempts could have taken a lot of time, between 15 min\nand 40 min depending on the configuration of Octavia and the provider\ndriver (amphorav1 or amphorav2)\n\nCloses-Bug: #2033894\n\nNote: stable/2023.1 and older, the patch also includes modifications in\n      octavia/controller/worker/v1/\n\nConflicts:\n\toctavia/controller/worker/v2/tasks/amphora_driver_tasks.py\n\nChange-Id: Ib33a0b8d2875e4ff97c65933fe9360bb06994d32\n(cherry picked from commit ca70587f0bcb4b2a05bd5c243f111140a5762114)\n(cherry picked from commit 6fdc2079da3392fdfbdf4a80df05bfe7d149ca95)\n(cherry picked from commit 97b1b8387e42f8361b612a0b81b1c60b6e14db37)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/dd849b4c5c1a1d33e4fd4a9afa2f35903d4d9aac"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/dd849b4c5c1a1d33e4fd4a9afa2f35903d4d9aac"}]},"branch":"refs/heads/stable/zed"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
