)]}'
{"id":"openstack%2Foctavia~897788","triplet_id":"openstack%2Foctavia~stable%2Fzed~I4bd027346c61b93b537ab53810c2ecb6160b6be2","project":"openstack/octavia","branch":"stable/zed","attention_set":{},"removed_from_attention_set":{"31664":{"account":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"last_update":"2023-10-16 09:17:02.000000000","reason":"Change was submitted"},"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2023-10-16 09:17:02.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I4bd027346c61b93b537ab53810c2ecb6160b6be2","subject":"Fix amphorae in ERROR during the failover","status":"MERGED","created":"2023-10-10 10:20:05.000000000","updated":"2023-10-16 09:18:42.000000000","submitted":"2023-10-16 09:17:02.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"897788","meta_rev_id":"f688ae88622866ec7f14f8004f0527fa236f292d","_number":897788,"virtual_id_number":897788,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-10-16 09:17:02.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-10-12 17:10:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":2,"date":"2023-10-16 06:54:01.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":1,"date":"2023-10-16 06:54:01.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-10-10 10:35:07.000000000","updated_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"reviewer":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"state":"CC"},{"updated":"2023-10-10 10:35:46.000000000","updated_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"reviewer":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"state":"REVIEWER"},{"updated":"2023-10-10 14:21:31.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-10-12 17:10:26.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2023-10-16 06:54:01.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"}],"messages":[{"id":"01e120a90011a1296a10e188d44ecfef4ee65292","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-10-10 10:20:05.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"3879a9757b6b9603960ee145b93de1f9650a63f8","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2023-10-10 10:35:07.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"879914c76dbf0350b096168dca5761b2e9a1e0fb","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2023-10-10 10:35:46.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"cd39c42fc90f2453872a63313e2ba3299019e118","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-10-10 11:32:02.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nCopied Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n","accounts_in_message":[],"_revision_number":2},{"id":"c9b3a7b2dbdd1b9d45ed39ed2bc3edb6631def42","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-10 14:21:31.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bc3b0a7ff5de412787c1af869d9096d4\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/52513ddcf9fc4869b7f480d85cc2f167 : SUCCESS in 12m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9318bff5690a4ac2a25808e147498d87 : SUCCESS in 7m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cc5cde80ec08418e809fd6365484aeb4 : SUCCESS in 5m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d28386b308914394821196c12f5bc7d7 : SUCCESS in 5m 19s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/4c1095e50a654183b4663491ea254650 : SUCCESS in 5m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8e9e69a8435045999d6c557e7a35c9d1 : SUCCESS in 13m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e571173756e24f939a766108058a0b21 : SUCCESS in 7m 08s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/0b628b56dc0a47fe82a96138ef8da67b : SUCCESS in 5m 50s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/e5377009a11349229466a8b096eeb2b7 : SUCCESS in 9m 00s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/a779ae91175e41808289153ec598d87e : SUCCESS in 4m 51s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/16be5cf959b34186be7c3243d65a9f53 : SUCCESS in 7m 44s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/085f1dd8982e42c1b48300b12dba1e6a : SUCCESS in 8m 18s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/6574bbe6a6644fffa38f86a900d83803 : SUCCESS in 8m 59s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/c0c42f72c3f445c1949e7b5ff3ffbff7 : SUCCESS in 1h 11m 09s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3979b69f3e934970be2fe050e01e33c7 : SUCCESS in 2h 14m 55s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/2a5cb4cc41a04e689668778ecbb80c49 : SUCCESS in 43m 38s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/c160c0706f414a7eb0612791b2b47fd0 : SUCCESS in 41m 39s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/d5248d35a9ab450997f8c76f928755ec : FAILURE in 13m 33s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/638ae52ddcc24f8d94fd2c73f73f70a2 : SUCCESS in 38m 58s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/9b7b8a9ee4e544c4a12b729e8cf55591 : SUCCESS in 1h 43m 15s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/cef7aeaeb586480696c7ad5d839eebd7 : SUCCESS in 2h 46m 26s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/db511e535b9242c687b2b3b6e53aaeb7 : SUCCESS in 1h 55m 16s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/37081ccdd348465894f9725b9b581256 : SUCCESS in 1h 54m 11s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/5955ed4a07314fe5bc2b11efed381af7 : SUCCESS in 2h 00m 18s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/a3eb9b4d17c04387bc34415d3dd6c0f9 : FAILURE in 7m 56s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"bfa897893463dc70304b643cb15a3ed50ac9fee9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-10-12 09:23:52.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"27969feabdd194d850ed74e6cc51d6305d970d37","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-10-12 09:37:08.000000000","message":"Uploaded patch set 4: Patch Set 3 was rebased.","accounts_in_message":[],"_revision_number":4},{"id":"680130d80b107a7d8e004b04ae079c1e0b14a67b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-12 11:58:02.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2409967106c14a15b1888a6bf8e82036\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/158539d57ae44f51abb35d114076285b : SUCCESS in 9m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4130e804e1c24ab5a7bf74a22bc3b658 : SUCCESS in 7m 17s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1c318d86fb974e73be643db67495cf68 : SUCCESS in 5m 02s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a3ac6901644c4e24ac11c471b6e306d1 : SUCCESS in 5m 09s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/6f79c003e0454939b2ab7fe0fdfbf466 : SUCCESS in 5m 26s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c1eacae7448744719670ab518abfaebd : SUCCESS in 13m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c00b2882c06441d18c65ac11371b5ff6 : SUCCESS in 5m 16s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/613c4adfa2d74c1b96c1e1e9a87caeaf : SUCCESS in 5m 15s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/2c7f148108e84c44aaac9b038e88082f : SUCCESS in 7m 53s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/0c162bfd0bd344e29bb4c3a18295f3d6 : SUCCESS in 5m 03s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/cc20e11954d14b398711f519ea3f8ef4 : SUCCESS in 7m 08s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/e5294adfad4e43108ad5ef2c60004b4b : SUCCESS in 8m 19s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/5d0cdf281dd9488e94098360aa158503 : SUCCESS in 8m 36s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/0960c5b59a634de793a1812976f95126 : SUCCESS in 1h 30m 41s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/49a3020943e1476d997099fcf9712820 : SUCCESS in 1h 49m 39s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/5c25b5c3d6c345268f97404ad7efc841 : SUCCESS in 44m 04s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/d47d43ec17e2409eaea94171035db8ba : SUCCESS in 39m 08s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/4436bcd019f54415a3f4c7f59016b1b1 : SUCCESS in 1h 04m 12s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7cb935595d3149188afe8168d2ff70a5 : SUCCESS in 31m 52s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/356843fc45334b3d8906e22a7ef54af5 : FAILURE in 1h 38m 18s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/d15292354d3e4698bf9ccbb3fd8139d0 : SUCCESS in 1h 43m 42s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/fd92f20a013f4222b24bf965fdb8bc88 : SUCCESS in 2h 11m 02s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/3d7721e263de476884183842d6021eb0 : FAILURE in 1h 35m 01s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/6291e68f59f94695ae63cb6c377460e4 : SUCCESS in 1h 29m 34s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/74677d66a8d94a0e802c915289d2f757 : SUCCESS in 2h 10m 57s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"9c5485d48367420874ad99f3f526b5090a46fb89","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-10-12 17:10:26.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"4544008e8bd306178e68a3a93c36effae12583f1","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-10-16 06:54:01.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":4},{"id":"d533cd9f56498fb25d6deb8047234dbe7c954d32","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-16 06:55:04.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"228f23f3eae9f5d369bc65ba50ffd350c104071e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-16 09:17:02.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7b3d5483c21b441891f59c27cb61aeb0\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/01ebc096c5a140ff8507a6139d975682 : SUCCESS in 6m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9892ee6cf107478ca06c2d1a245f6c8d : SUCCESS in 5m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/20fdcd1c95a043459a5f6e100f46e4c0 : SUCCESS in 5m 13s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f7187cbc7b3d4247ba4558379ecef748 : SUCCESS in 9m 38s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2647eebb3e2b402eb39b07017b82e50a : SUCCESS in 5m 38s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/adb2f08e1db84d74a30d8129847aeaeb : SUCCESS in 5m 21s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/9ec5270126f64465b21447e199a6e71b : SUCCESS in 6m 38s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/0709d09c1b224c15861fa199244cd695 : SUCCESS in 1h 16m 11s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/220493e2f3b64f51b57e664cf01ee975 : SUCCESS in 1h 40m 03s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/df22ce386b50496ca3d2b7a46eda651c : SUCCESS in 37m 29s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/c03323feb6814107827eceb4b7c7651f : SUCCESS in 39m 30s","accounts_in_message":[],"_revision_number":4},{"id":"eb12fa20f8bca29f85167d802920b87328ba2d17","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-16 09:17:02.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"f688ae88622866ec7f14f8004f0527fa236f292d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-16 09:18:42.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8873627a68664655b67eb1bb90ab8bc2\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4592a50b24fe45f2a74cdbfca9ac662c : SUCCESS in 43s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5c9c2e678c084cf4943d84cc6e8badfb : SUCCESS in 39s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"d09065c4f6c08b07930fa2f8ef6794de5417c2c0","revisions":{"63952f5268f692d2c0228688540759e6015a11d2":{"kind":"REWORK","_number":1,"created":"2023-10-10 10:20:05.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/88/897788/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/88/897788/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/88/897788/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/88/897788/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/88/897788/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/88/897788/1"}}},"commit":{"parents":[{"commit":"61e27ffec304beee6325f915d78ada199f575ef7","subject":"Reduce duration of failovers with amphora in ERROR","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/61e27ffec304beee6325f915d78ada199f575ef7"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-09-04 09:56:25.000000000","tz":-240},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-10-10 10:19:21.000000000","tz":120},"subject":"Fix amphorae in ERROR during the failover","message":"Fix amphorae in ERROR during the failover\n\nWhen 2 amps were down, the failover flow created the first one and\nneeded to update both amp to configure VRRP, but as the 2nd was missing,\nit was set to ERROR. Then the health-manager could not trigger a\nfailover becasue amphorae in ERROR are excluded from the automated\nfailover process.\n\nThis commit changes the tasks that must be run on both amphorae during a\nfailover of one amphora, it doesn\u0027t mark the secondary amphora in ERROR\nif it is not reachable.\n\nCloses-Bug: #2033734\n\nConflicts:\n\toctavia/controller/worker/v2/tasks/amphora_driver_tasks.py\n\nChange-Id: I4bd027346c61b93b537ab53810c2ecb6160b6be2\n(cherry picked from commit 248cf2893ea913a74d870a763802a51586e0f63b)\n(cherry picked from commit 5622431534d3bf9b8b71ad8fdb2d744f94625ec4)\n(cherry picked from commit f7583e43b39805f724460651458efef7b9d173d1)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/63952f5268f692d2c0228688540759e6015a11d2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/63952f5268f692d2c0228688540759e6015a11d2"}]},"branch":"refs/heads/stable/zed"},"70ce96ff985615afcc2b4a3fd03d9b3437fcb81d":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2023-10-10 11:32:02.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/88/897788/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/88/897788/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/88/897788/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/88/897788/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/88/897788/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/88/897788/2"}}},"commit":{"parents":[{"commit":"4c8c3360273f4c1a02f7401bb65ac578c018938d","subject":"Reduce duration of failovers with amphora in ERROR","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4c8c3360273f4c1a02f7401bb65ac578c018938d"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-09-04 09:56:25.000000000","tz":-240},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-10-10 11:30:58.000000000","tz":120},"subject":"Fix amphorae in ERROR during the failover","message":"Fix amphorae in ERROR during the failover\n\nWhen 2 amps were down, the failover flow created the first one and\nneeded to update both amp to configure VRRP, but as the 2nd was missing,\nit was set to ERROR. Then the health-manager could not trigger a\nfailover becasue amphorae in ERROR are excluded from the automated\nfailover process.\n\nThis commit changes the tasks that must be run on both amphorae during a\nfailover of one amphora, it doesn\u0027t mark the secondary amphora in ERROR\nif it is not reachable.\n\nCloses-Bug: #2033734\n\nConflicts:\n\toctavia/controller/worker/v2/tasks/amphora_driver_tasks.py\n\nChange-Id: I4bd027346c61b93b537ab53810c2ecb6160b6be2\n(cherry picked from commit 248cf2893ea913a74d870a763802a51586e0f63b)\n(cherry picked from commit 5622431534d3bf9b8b71ad8fdb2d744f94625ec4)\n(cherry picked from commit f7583e43b39805f724460651458efef7b9d173d1)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/70ce96ff985615afcc2b4a3fd03d9b3437fcb81d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/70ce96ff985615afcc2b4a3fd03d9b3437fcb81d"}]},"branch":"refs/heads/stable/zed"},"019fd7ec055a5893d92c7dae5dd06b65d81521ca":{"kind":"REWORK","_number":3,"created":"2023-10-12 09:23:52.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/88/897788/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/88/897788/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/88/897788/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/88/897788/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/88/897788/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/88/897788/3"}}},"commit":{"parents":[{"commit":"90717d4a1d4854f7eff865940598e013a5b973d2","subject":"Reduce duration of failovers with amphora in ERROR","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/90717d4a1d4854f7eff865940598e013a5b973d2"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-09-04 09:56:25.000000000","tz":-240},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-10-12 09:22:14.000000000","tz":-240},"subject":"Fix amphorae in ERROR during the failover","message":"Fix amphorae in ERROR during the failover\n\nWhen 2 amps were down, the failover flow created the first one and\nneeded to update both amp to configure VRRP, but as the 2nd was missing,\nit was set to ERROR. Then the health-manager could not trigger a\nfailover becasue amphorae in ERROR are excluded from the automated\nfailover process.\n\nThis commit changes the tasks that must be run on both amphorae during a\nfailover of one amphora, it doesn\u0027t mark the secondary amphora in ERROR\nif it is not reachable.\n\nCloses-Bug: #2033734\n\nNote: stable/2023.1 and older, the patch also includes modifications in\n      octavia/controller/worker/v1/\n\nConflicts:\n\toctavia/controller/worker/v2/tasks/amphora_driver_tasks.py\n\nChange-Id: I4bd027346c61b93b537ab53810c2ecb6160b6be2\n(cherry picked from commit 248cf2893ea913a74d870a763802a51586e0f63b)\n(cherry picked from commit 5622431534d3bf9b8b71ad8fdb2d744f94625ec4)\n(cherry picked from commit 10cfd8c7605e7789ec1ae3de511b3a84b8324e04)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/019fd7ec055a5893d92c7dae5dd06b65d81521ca"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/019fd7ec055a5893d92c7dae5dd06b65d81521ca"}]},"branch":"refs/heads/stable/zed"},"d09065c4f6c08b07930fa2f8ef6794de5417c2c0":{"kind":"TRIVIAL_REBASE","_number":4,"created":"2023-10-12 09:37:08.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/88/897788/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/88/897788/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/88/897788/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/88/897788/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/88/897788/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/88/897788/4"}}},"commit":{"parents":[{"commit":"dd849b4c5c1a1d33e4fd4a9afa2f35903d4d9aac","subject":"Reduce duration of failovers with amphora in ERROR","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/dd849b4c5c1a1d33e4fd4a9afa2f35903d4d9aac"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-09-04 09:56:25.000000000","tz":-240},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-10-12 09:35:42.000000000","tz":-240},"subject":"Fix amphorae in ERROR during the failover","message":"Fix amphorae in ERROR during the failover\n\nWhen 2 amps were down, the failover flow created the first one and\nneeded to update both amp to configure VRRP, but as the 2nd was missing,\nit was set to ERROR. Then the health-manager could not trigger a\nfailover becasue amphorae in ERROR are excluded from the automated\nfailover process.\n\nThis commit changes the tasks that must be run on both amphorae during a\nfailover of one amphora, it doesn\u0027t mark the secondary amphora in ERROR\nif it is not reachable.\n\nCloses-Bug: #2033734\n\nNote: stable/2023.1 and older, the patch also includes modifications in\n      octavia/controller/worker/v1/\n\nConflicts:\n\toctavia/controller/worker/v2/tasks/amphora_driver_tasks.py\n\nChange-Id: I4bd027346c61b93b537ab53810c2ecb6160b6be2\n(cherry picked from commit 248cf2893ea913a74d870a763802a51586e0f63b)\n(cherry picked from commit 5622431534d3bf9b8b71ad8fdb2d744f94625ec4)\n(cherry picked from commit 10cfd8c7605e7789ec1ae3de511b3a84b8324e04)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d09065c4f6c08b07930fa2f8ef6794de5417c2c0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d09065c4f6c08b07930fa2f8ef6794de5417c2c0"}]},"branch":"refs/heads/stable/zed"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
