)]}'
{"id":"openstack%2Foctavia~897789","triplet_id":"openstack%2Foctavia~stable%2F2023.1~I497f2d8587e25eda0d169aa919cb6fbede7e9fc6","project":"openstack/octavia","branch":"stable/2023.1","attention_set":{},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2023-10-16 08:31:48.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I497f2d8587e25eda0d169aa919cb6fbede7e9fc6","subject":"Fix timeout duration in start_vrrp_service during failovers","status":"MERGED","created":"2023-10-10 10:20:48.000000000","updated":"2023-10-16 08:33:07.000000000","submitted":"2023-10-16 08:31:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"897789","meta_rev_id":"2cda50fe6ee31960c3c03a0675622554d0c36031","_number":897789,"virtual_id_number":897789,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-10-16 08:31:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"recommended":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"all":[{"value":1,"date":"2023-10-10 15:56:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":2,"date":"2023-10-12 17:12:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-10-16 06:49:01.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-10-16 06:49:01.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-10-10 13:03:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-10-10 15:56:48.000000000","updated_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"reviewer":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"state":"REVIEWER"},{"updated":"2023-10-12 17:12:16.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2023-10-16 06:49:01.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"}],"messages":[{"id":"d20a0ca489672ebdeaa208be606daaeb12312e78","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-10-10 10:20:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b85d0e9d88940532f452ea11bc2ccf0a97e706e3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-10 13:03:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4088132637224dd081bd5bf4adc21a9d\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e4f39620d22749628d121603429ac0c9 : SUCCESS in 10m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f77c8c19648b4deeb22b95796ffeb5fc : SUCCESS in 5m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7196817a69e04b96bee1f7b08f113288 : SUCCESS in 4m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/f0863926f30b4ed0b185e1257a2cbfbb : SUCCESS in 5m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/71c07e27372d496a83ffe69757a4aa1a : SUCCESS in 12m 10s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4966756409e0403c8b8a2c937fc99f4d : SUCCESS in 5m 35s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/3672836ea06d49b6ac4e16fedcbe698e : SUCCESS in 5m 42s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/b7f87c926b044e73bad900673d99849e : SUCCESS in 8m 19s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/4707c00a831d4850b3c53954621a2a19 : SUCCESS in 5m 09s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/a193ec73fce24b79b16cd0dcbaee5c62 : SUCCESS in 7m 38s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/2c363fa539ea46aea220f5a1b53ac105 : SUCCESS in 8m 20s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/c7154373760548ee836a6b3580cde831 : SUCCESS in 9m 24s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/c573a1f7aca54e71b64cd41c3b6dcb76 : SUCCESS in 1h 27m 00s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/8b5ed30342cb42e3b697daac0d168ab4 : SUCCESS in 1h 29m 41s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/ba5b217e366c4120aaf4e884d91b982b : SUCCESS in 1h 19m 24s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/759cc935c3ba48b28877cf250fea9564 : SUCCESS in 47m 24s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/715cba7d54fb4187945414cf4d856018 : SUCCESS in 52m 10s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/8d072c2e143744ae9f467b0e5b783a96 : SUCCESS in 55m 57s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/d58ea6d14c8c4664aee837b1327233d3 : FAILURE in 10m 07s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d282223e02054b2caae2640bb699c166 : SUCCESS in 32m 35s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/0aceee24f87f49ecbf8776526f03b86e : SUCCESS in 2h 31m 48s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/bc5e23b9c17847459938c57a1bf83d9e : SUCCESS in 2h 33m 44s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/811a00d0ca9d4b83b5c52963b9b025e3 : SUCCESS in 2h 00m 31s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/0c21150c8a5d4ab9b59fe483839594ac : SUCCESS in 2h 18m 39s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/8276c22d4e1446609a02a23bd1db96ee : SUCCESS in 2h 20m 44s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/0b896b9686fe4847bcfd4ab2eabdabc1 : FAILURE in 8m 06s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2b4c79f8477acf16c41112fb39a191ad31e94d54","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2023-10-10 15:56:48.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4e4c3601b977795fcd9cdf5cc73166da3a654468","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-10-12 17:12:16.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7788a4b1c1e1f854e179474b7945a115b3a57c65","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-10-16 06:49:01.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"d35bff7857598f4fab80c841dd16033098396785","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-16 06:49:39.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"d804445419fdc6b68b9d6254096b7ee128c878a4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-16 08:31:48.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0115ffc62bcd4f79962fb772024b54b3\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c1933bb2eab44ee7a7e712fc601a238d : SUCCESS in 4m 41s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d246461cb3b44735bd2ea2a71296f6ec : SUCCESS in 5m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/d93a1993e377447fae49f75db6666f21 : SUCCESS in 4m 52s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6daa87dd3d624496afdf0a99fd6ec210 : SUCCESS in 11m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4326a98f1a7a469eaf41886e82dca4ab : SUCCESS in 5m 45s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/3838a3022b2445b39d466b05b0882f65 : SUCCESS in 4m 49s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/9e9fb00f96a04c84b583b5883c3d5fbd : SUCCESS in 6m 24s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/e2c2689cee3b4275b561209d2f94661f : SUCCESS in 1h 25m 23s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/1734e1bb946342aa86482e9bb7223f7b : SUCCESS in 1h 34m 43s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/2cafe78a59214c33b96afb0d6f4e3dca : SUCCESS in 1h 06m 52s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/45da069e81624e02800c8fc9380ac459 : SUCCESS in 36m 29s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/311bac6e09d24070a80e612acf89c8ba : SUCCESS in 38m 38s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/a8d7a617db1045328ba4c1e648ca3a5a : SUCCESS in 40m 43s","accounts_in_message":[],"_revision_number":1},{"id":"599cae751d67914c4405607ffce079754aebdd16","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-16 08:31:48.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"2cda50fe6ee31960c3c03a0675622554d0c36031","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-16 08:33:07.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e5a461f6ec864d77ad7af88dc4122651\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/403e410e2af043459f0ba9119b490a87 : SUCCESS in 46s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/414e6b89247f4537964dfa945bdd25b0 : SUCCESS in 43s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"5c98d901337cb0d6e4c575e3280cb77839720cfe","revisions":{"5c98d901337cb0d6e4c575e3280cb77839720cfe":{"kind":"REWORK","_number":1,"created":"2023-10-10 10:20:48.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/89/897789/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/89/897789/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/89/897789/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/89/897789/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/89/897789/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/89/897789/1"}}},"commit":{"parents":[{"commit":"64d9bdb97556652aa05a34276778260b541ad9c6","subject":"Fix race condition in members batch update API call","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/64d9bdb97556652aa05a34276778260b541ad9c6"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-09-01 15:35:34.000000000","tz":-240},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-10-10 10:06:09.000000000","tz":120},"subject":"Fix timeout duration in start_vrrp_service during failovers","message":"Fix timeout duration in start_vrrp_service during failovers\n\nThe timeout_dict argument was not passed to the\n_populate_amphora_api_version, it could have made the failover of broken\namphora really long.\n\nRelated-Bug: #2033894\n\nChange-Id: I497f2d8587e25eda0d169aa919cb6fbede7e9fc6\n(cherry picked from commit cce348040d43b92b59146adeb15a4f1486946a64)\n(cherry picked from commit 54eb1624f5762e9b848f10f712152faf02f8b84d)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5c98d901337cb0d6e4c575e3280cb77839720cfe"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5c98d901337cb0d6e4c575e3280cb77839720cfe"}]},"branch":"refs/heads/stable/2023.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
