)]}'
{"id":"openstack%2Foctavia~897790","triplet_id":"openstack%2Foctavia~stable%2F2023.1~Ib33a0b8d2875e4ff97c65933fe9360bb06994d32","project":"openstack/octavia","branch":"stable/2023.1","attention_set":{},"removed_from_attention_set":{"31664":{"account":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"last_update":"2023-10-13 14:33:16.000000000","reason":"\u003cGERRIT_ACCOUNT_31664\u003e replied on the change","reason_account":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}},"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2023-10-16 08:41:28.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ib33a0b8d2875e4ff97c65933fe9360bb06994d32","subject":"Reduce duration of failovers with amphora in ERROR","status":"MERGED","created":"2023-10-10 10:20:48.000000000","updated":"2023-10-16 08:42:34.000000000","submitted":"2023-10-16 08:41:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"897790","meta_rev_id":"7fe664e94e2b4d8a966de3e8cb17694dfd055f49","_number":897790,"virtual_id_number":897790,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-10-16 08:41:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"recommended":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"all":[{"value":1,"date":"2023-10-13 14:33:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":2,"date":"2023-10-12 17:10:54.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":2,"date":"2023-10-16 06:46:00.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":1,"date":"2023-10-16 06:46:00.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-10-10 13:14:36.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-10-10 13:25:12.000000000","updated_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"reviewer":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"state":"CC"},{"updated":"2023-10-10 13:48:25.000000000","updated_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"reviewer":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"state":"REVIEWER"},{"updated":"2023-10-12 17:10:54.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2023-10-16 06:46:00.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"}],"messages":[{"id":"9149f318671144d9cbb4dc44d3622af42292d984","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-10-10 10:20:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"52585689a4a80b0a0ada3a47f0c25cb426683135","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-10 13:14:36.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f4d2259ddc314e45874816b11c3d060d\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e22c484121db48cfb2226669858370be : SUCCESS in 11m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/262a0e8b17994d0b9050672b0b01f110 : SUCCESS in 5m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5d9f3581db944f48a1544e32ced2485d : SUCCESS in 5m 22s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7252155240ff497f8bec2d9152596de1 : SUCCESS in 5m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7df54b28b08f4fe8a068f2fb93072499 : SUCCESS in 14m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b223bfac2a524a1f8eb087ff12920560 : SUCCESS in 5m 30s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/eae52ffefa0142c18bd4b7c9209c76ce : SUCCESS in 5m 33s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/a33ff54dd0ea4e9dbfa63f399e4f1a6a : SUCCESS in 8m 15s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/ba4f9ed867e84cc79dcd42233f4b0108 : SUCCESS in 5m 34s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/0c9a3cb974fa46f58761cd830ba14ba4 : SUCCESS in 8m 03s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/98de5c0c18bd48ceaba07f4881487ecb : SUCCESS in 8m 06s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/de61c380f50e44ad8e74f740994c2c2b : SUCCESS in 8m 31s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/793c7587ea5d4667b26b9e2a318ac07b : SUCCESS in 1h 25m 47s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/ddeb2b298b734bfbbe09619ce3df31a0 : SUCCESS in 1h 42m 16s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/077d2228e80b422391a99bd835fdb7e3 : SUCCESS in 1h 22m 31s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/f6a78e85f518455892890ee60d1923ec : SUCCESS in 44m 28s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/58c86ba18807499fa15050a6acf8bd5a : SUCCESS in 57m 27s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/224cd3dff6094502b5bdc3537619b6f4 : SUCCESS in 51m 53s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/edb4afdfb0af469f83fea9eb3aaec7f8 : FAILURE in 11m 05s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b0e97206bbe44844b239b19a82921b03 : SUCCESS in 38m 58s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/255a8c02a42d442098c6a097701dc5fc : SUCCESS in 2h 28m 20s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/e469cf6ed42848d194cc85669f334ef4 : SUCCESS in 2h 40m 26s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/8ccc000c91524cee87e7b00c96ecc756 : SUCCESS in 1h 48m 38s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/54c5a2671d204efead077c9ff6e51213 : TIMED_OUT in 2h 34m 05s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/7096a51a0be743bea7a771c23838fb5c : SUCCESS in 2h 14m 28s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/b4f47b9753aa41fe8858526da84c5a6b : FAILURE in 10m 32s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1cad2f938273f7160854b28bf40fca95b3a1a798","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2023-10-10 13:25:12.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"890ecc2246d12d229e814a9547495ea51e1a9a21","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2023-10-10 13:48:25.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"33b1fbfa388897ae2da927c1fda5274f5ff8dc46","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-10-11 08:10:18.000000000","message":"Patch Set 1: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ab180afadcc51114a7d16336feaa2554354fb136","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-10-12 08:09:54.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":2},{"id":"100e3a8ef7688242f46ac1e3c752e3ecf6b7b7ff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-12 10:55:03.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/da48f7f685be47c7870a7eb0e57d4c61\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/eef75368fb3f4015b2c2413501c83773 : SUCCESS in 9m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c43b38c0af71422c868b7b9202a29add : SUCCESS in 4m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4ca1fa1b6f0045258358bca03b22e054 : SUCCESS in 5m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/590c86bf232f4054a5aa04adc3015f6e : SUCCESS in 4m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b441f7b4cce44eb38ee1256c381c3916 : SUCCESS in 12m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/69a42bfda93145148ec612091ad124ad : SUCCESS in 4m 47s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/e53777184b3548d5a82dda211b012e9f : SUCCESS in 5m 30s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/a1270a84d55f4368b063a29eaad68de2 : SUCCESS in 8m 08s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/7e776fdb445347089cb7852562b3da29 : SUCCESS in 5m 03s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/4539f0ebfced4c159c891d81b6022916 : SUCCESS in 6m 29s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/7ef4735b54174ca89e4ccb2ac2ea2129 : SUCCESS in 8m 03s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/97d9110160b4449d9963888ad0b25d22 : SUCCESS in 7m 31s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/442ba17421164b04acf6ec63846160af : SUCCESS in 1h 21m 54s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/3ce403b37deb4da6a519eb95493f257c : SUCCESS in 1h 32m 16s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/63ae08c8d8484c6aa1898d008367086b : SUCCESS in 1h 11m 19s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/ccf914ed02d0481d82c186819f2a4f52 : SUCCESS in 42m 39s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/f2988845374143e8af8740147d503702 : SUCCESS in 47m 48s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/60c3885f151c4b2298d772fcdded993b : SUCCESS in 58m 55s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/c3d39babc21b4ab2b84fa5d1f653ff50 : SUCCESS in 1h 17m 02s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b9d3963fa0dd4c9082e5678d49b8a6f0 : SUCCESS in 44m 58s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/71240b1f9bb34487947ca12ce2ea0266 : SUCCESS in 2h 16m 35s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/d6565a02bc194ee38bff46c9c60c5839 : SUCCESS in 2h 18m 04s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/6fe6b93330ca4df98726b396ff31d32d : SUCCESS in 1h 54m 43s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/ddbcfafecf2141fca83b7aea3b9cdf15 : SUCCESS in 2h 03m 53s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/5224b7d8867342f8bb9d08272c2832c2 : SUCCESS in 2h 13m 52s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/dd456fc6b7e542d8929a8411343b4c15 : SUCCESS in 2h 29m 01s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"e60d802823c3d05fcafa3b5fc43b08ee8b391781","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-10-12 17:10:54.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"a728cd4cffdbda412da445dd3ad9e61823ccf664","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2023-10-13 14:33:16.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"1c3b7d98bef52f411506ef2a473bee6196a68c83","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-10-16 06:46:00.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"5e736130573fd229e14178dc7387c782ebedf64f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-16 06:49:43.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"40c97d95b7a3b19c2a5d418133e97027d82197d3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-16 08:41:28.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8f159e2c63ea42df80ab9c220b5e1560\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/430e608755d349aea0742fcdc1b5a43d : SUCCESS in 4m 56s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fe5636dbfd6d4ef79e6676b8239c891b : SUCCESS in 6m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a0a70fc9b1b74c0cb0b9f94c3d533067 : SUCCESS in 4m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/05da66b1f4c14086abbd82754dd40423 : SUCCESS in 14m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c3702c6e8fa94ca6abb61f33a9b5dde0 : SUCCESS in 5m 31s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/abe10a5087e94410b86bcac4c800f1a5 : SUCCESS in 4m 29s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/9e74ab3f1b134a2789a12f0c08f11594 : SUCCESS in 6m 28s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/4783a66e5b014e9fa5ce3cfa0d4dff86 : SUCCESS in 1h 27m 33s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/70d2d5943f7c421b8c18357a7209c8c6 : SUCCESS in 1h 18m 32s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/c36a1948c9d04f9a88557169a961bc15 : SUCCESS in 1h 17m 23s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/175067e83b3241b38f8d62d0c61eea3e : SUCCESS in 36m 46s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/ce0dae9876004c8dba5891410fe61f3d : SUCCESS in 50m 25s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/9267b654966f4abab1e22275cb08c64e : SUCCESS in 50m 56s","accounts_in_message":[],"_revision_number":2},{"id":"25a42a063d6a7edf13e3737ee722868b31c803e5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-16 08:41:28.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"7fe664e94e2b4d8a966de3e8cb17694dfd055f49","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-16 08:42:34.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b80197db6ee54503bbe4e1273c72fb10\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a790de68ded845cca80dc39f08c98b1b : SUCCESS in 47s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/ec0b78f4d7304773977307a9422c0e92 : SUCCESS in 37s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"97b1b8387e42f8361b612a0b81b1c60b6e14db37","revisions":{"1c0b702f887cb4fe48a44d6a6972d337a9a3f187":{"kind":"REWORK","_number":1,"created":"2023-10-10 10:20:48.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/90/897790/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/90/897790/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/90/897790/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/90/897790/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/90/897790/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/90/897790/1"}}},"commit":{"parents":[{"commit":"5c98d901337cb0d6e4c575e3280cb77839720cfe","subject":"Fix timeout duration in start_vrrp_service during failovers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5c98d901337cb0d6e4c575e3280cb77839720cfe"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-09-01 13:08:37.000000000","tz":-240},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-10-10 10:07:52.000000000","tz":120},"subject":"Reduce duration of failovers with amphora in ERROR","message":"Reduce duration of failovers with amphora in ERROR\n\nIn the failover flow, there are multiple tasks for the configuration of\nVRRP for the other amphorae of the load balancer, but during outage the\nother amps may not be available. To prevent the tasks from attempting\nconnections to unreachable amphorae, we can detect in the first task\nthat an amp is unreachable and pass this information to the other tasks.\n\nThose connection attempts could have taken a lot of time, between 15 min\nand 40 min depending on the configuration of Octavia and the provider\ndriver (amphorav1 or amphorav2)\n\nCloses-Bug: #2033894\n\nConflicts:\n\toctavia/controller/worker/v2/tasks/amphora_driver_tasks.py\n\nChange-Id: Ib33a0b8d2875e4ff97c65933fe9360bb06994d32\n(cherry picked from commit ca70587f0bcb4b2a05bd5c243f111140a5762114)\n(cherry picked from commit 6fdc2079da3392fdfbdf4a80df05bfe7d149ca95)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1c0b702f887cb4fe48a44d6a6972d337a9a3f187"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1c0b702f887cb4fe48a44d6a6972d337a9a3f187"}]},"branch":"refs/heads/stable/2023.1"},"97b1b8387e42f8361b612a0b81b1c60b6e14db37":{"kind":"REWORK","_number":2,"created":"2023-10-12 08:09:54.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/90/897790/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/90/897790/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/90/897790/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/90/897790/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/90/897790/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/90/897790/2"}}},"commit":{"parents":[{"commit":"5c98d901337cb0d6e4c575e3280cb77839720cfe","subject":"Fix timeout duration in start_vrrp_service during failovers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5c98d901337cb0d6e4c575e3280cb77839720cfe"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-09-01 13:08:37.000000000","tz":-240},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-10-12 08:04:44.000000000","tz":-240},"subject":"Reduce duration of failovers with amphora in ERROR","message":"Reduce duration of failovers with amphora in ERROR\n\nIn the failover flow, there are multiple tasks for the configuration of\nVRRP for the other amphorae of the load balancer, but during outage the\nother amps may not be available. To prevent the tasks from attempting\nconnections to unreachable amphorae, we can detect in the first task\nthat an amp is unreachable and pass this information to the other tasks.\n\nThose connection attempts could have taken a lot of time, between 15 min\nand 40 min depending on the configuration of Octavia and the provider\ndriver (amphorav1 or amphorav2)\n\nCloses-Bug: #2033894\n\nNote: stable/2023.1 and older, the patch also includes modifications in\n      octavia/controller/worker/v1/\n\nConflicts:\n\toctavia/controller/worker/v2/tasks/amphora_driver_tasks.py\n\nChange-Id: Ib33a0b8d2875e4ff97c65933fe9360bb06994d32\n(cherry picked from commit ca70587f0bcb4b2a05bd5c243f111140a5762114)\n(cherry picked from commit 6fdc2079da3392fdfbdf4a80df05bfe7d149ca95)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/97b1b8387e42f8361b612a0b81b1c60b6e14db37"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/97b1b8387e42f8361b612a0b81b1c60b6e14db37"}]},"branch":"refs/heads/stable/2023.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
