)]}'
{"id":"openstack%2Foctavia~898101","triplet_id":"openstack%2Foctavia~stable%2Fyoga~I497f2d8587e25eda0d169aa919cb6fbede7e9fc6","project":"openstack/octavia","branch":"stable/yoga","attention_set":{},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2023-10-16 09:16:49.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I497f2d8587e25eda0d169aa919cb6fbede7e9fc6","subject":"Fix timeout duration in start_vrrp_service during failovers","status":"MERGED","created":"2023-10-12 09:44:11.000000000","updated":"2023-10-16 09:18:15.000000000","submitted":"2023-10-16 09:16:49.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"898101","meta_rev_id":"812633c16e22d43c2f7ba1b206edbf053decbe8e","_number":898101,"virtual_id_number":898101,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-10-16 09:16:48.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":2,"date":"2023-10-12 17:10:04.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-10-16 06:49:18.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-10-16 06:49:18.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-10-12 12:14:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-10-12 17:10:04.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2023-10-16 06:49:18.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"}],"messages":[{"id":"40537b2ee800539ece97296af664d32db0010ccb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-10-12 09:44:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c4c8074e4a6b3611c4a918506c883931e778ede7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-12 12:14:29.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/29e28857bc284d2a8300218c1d339587\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a171c70d9f9143c8889567800165a3c2 : SUCCESS in 8m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/23362a4daecf404587a820956c56fa7f : SUCCESS in 6m 12s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/2d393473340340ee83404cedbb1c54ea : SUCCESS in 5m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/ef15bc4338324ac1a668b5ca5bd3d230 : SUCCESS in 5m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f09e83659ad443809f8508bf868ab22f : SUCCESS in 13m 03s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/082d273fc38348948b688d4f01d8d2c8 : SUCCESS in 4m 42s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/372799a147fc46a8be0ac357fdaa9dc8 : SUCCESS in 5m 11s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/402b0aee868840efbfb4311e7f0a139c : SUCCESS in 5m 27s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/6314f0e2d9124115a3e356df652ad21a : SUCCESS in 6m 14s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/d0e5a2c9769245c7923913bd51d834d4 : SUCCESS in 10m 41s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/480c7d2b186e4d29a9e013a3ac855967 : SUCCESS in 1h 20m 21s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b7624fa6d994493f87601cdd73e05980 : SUCCESS in 2h 07m 32s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/9c9774b783754d04bc7554ee9b7dbe1d : SUCCESS in 48m 31s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/fd4a56aa263d46d79717f712b6a3b614 : SUCCESS in 49m 52s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/724274c06ab84be888172c1bcbd59141 : FAILURE in 58m 02s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1e62bde081cc4f8c8ab22701678ee6bb : SUCCESS in 39m 32s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/ea99866fbcf24d68915e3d985f3f9118 : SUCCESS in 2h 14m 02s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/cbeb19ad4bab4a4194b34334a9077534 : FAILURE in 9m 54s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/ab43d9be220e4041ac54819b1eaf6207 : FAILURE in 6m 04s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/9a26b63473a145159cefbb021795b1e3 : SUCCESS in 2h 25m 44s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/ad7bda091b624aeda61215a473e0c710 : SUCCESS in 1h 41m 42s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/74edfde5933445d888dc9d4aacf7e27d : FAILURE in 36m 14s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"dd0f991a9303110b6d012f0c112c1c31f82d0852","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-10-12 17:10:04.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"8d4ef7634357af9b14dd394d272e16fa6149ff24","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-10-16 06:49:18.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"218476f2392c5c74c13f59ccf4d29e7ade37ce94","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-16 06:50:04.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"de09a824ba9f3c622c56c6e952ff224e4f6199ce","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-16 09:16:48.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/7b9e79b9529f44edae09de6e3e095d16\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f68456778d8d4f25b835e018e68c9635 : SUCCESS in 7m 05s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/8cfa95df44644ce985f6fe115d2365e8 : SUCCESS in 4m 40s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/591e821566fa43369dc30c7d7bc13213 : SUCCESS in 5m 15s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f8d53fa8d80c4b48b9ad96a90a4bf650 : SUCCESS in 12m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/23f0af9440ca4b81b7e0c8087479542e : SUCCESS in 6m 02s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/bc496537e25448b09b91acc553f0ffa3 : SUCCESS in 5m 22s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/772166f06e8a4ce793d1b844e2cf44a1 : SUCCESS in 7m 28s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/22d58cf90e734eb491ff4e03c82004e9 : SUCCESS in 1h 05m 35s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7e6d6364b654422183f1a7dff230ba7b : SUCCESS in 1h 44m 17s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/7f7e65bf1de141e7b4e074236906833f : SUCCESS in 33m 10s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/116014104ad74afab0de6cbaa085b719 : SUCCESS in 44m 34s","accounts_in_message":[],"_revision_number":1},{"id":"4744349e2b28be6d7c7c3089be7e24b5c3515319","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-16 09:16:49.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"812633c16e22d43c2f7ba1b206edbf053decbe8e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-16 09:18:15.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/529c99cba2514a78b4718fc544ef9221\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7cfe62e96b6945fd8bfd95b004f81f05 : SUCCESS in 1m 00s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/96f6004007ee4bd4806e5efd4cab88fd : SUCCESS in 38s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"9e5a6340156113ec3a014e64b3bad6617cd1c082","revisions":{"9e5a6340156113ec3a014e64b3bad6617cd1c082":{"kind":"REWORK","_number":1,"created":"2023-10-12 09:44:11.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/01/898101/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/01/898101/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/01/898101/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/01/898101/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/01/898101/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/01/898101/1"}}},"commit":{"parents":[{"commit":"231398ca5fee7b825646eeed84523dbfcb414438","subject":"Fix race condition in members batch update API call","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/231398ca5fee7b825646eeed84523dbfcb414438"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-09-01 15:35:34.000000000","tz":-240},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-10-12 09:36:40.000000000","tz":-240},"subject":"Fix timeout duration in start_vrrp_service during failovers","message":"Fix timeout duration in start_vrrp_service during failovers\n\nThe timeout_dict argument was not passed to the\n_populate_amphora_api_version, it could have made the failover of broken\namphora really long.\n\nRelated-Bug: #2033894\n\nChange-Id: I497f2d8587e25eda0d169aa919cb6fbede7e9fc6\n(cherry picked from commit cce348040d43b92b59146adeb15a4f1486946a64)\n(cherry picked from commit 54eb1624f5762e9b848f10f712152faf02f8b84d)\n(cherry picked from commit 5c98d901337cb0d6e4c575e3280cb77839720cfe)\n(cherry picked from commit 1dadaff530e252e8e96ea8b4c6dcec902a797516)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9e5a6340156113ec3a014e64b3bad6617cd1c082"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9e5a6340156113ec3a014e64b3bad6617cd1c082"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
