)]}'
{"id":"openstack%2Foctavia~898102","triplet_id":"openstack%2Foctavia~stable%2Fyoga~Ib33a0b8d2875e4ff97c65933fe9360bb06994d32","project":"openstack/octavia","branch":"stable/yoga","attention_set":{},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2023-10-16 09:16:50.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ib33a0b8d2875e4ff97c65933fe9360bb06994d32","subject":"Reduce duration of failovers with amphora in ERROR","status":"MERGED","created":"2023-10-12 09:44:11.000000000","updated":"2023-10-16 09:16:50.000000000","submitted":"2023-10-16 09:16:50.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"898102","meta_rev_id":"8971a3e012620e2f5891bd477f74005955a1ba38","_number":898102,"virtual_id_number":898102,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-10-16 09:16:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"recommended":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"all":[{"value":1,"date":"2023-10-13 14:36:24.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":2,"date":"2023-10-12 17:10:13.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-10-16 06:43:30.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-10-16 06:43:30.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-10-12 12:06:01.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-10-12 17:10:13.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2023-10-13 14:36:24.000000000","updated_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"reviewer":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"state":"REVIEWER"},{"updated":"2023-10-16 06:43:30.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"}],"messages":[{"id":"c6f81242d10f79daf079da6271477f7df69a8b1b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-10-12 09:44:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"bef64cb04426deb26b878cda2345bf535836e067","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-12 12:06:01.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9c8902a61d8648c0bfc95748a3c305bd\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/86883c4b2c5c48999c07e213f13b8bee : SUCCESS in 10m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1e07615ec5ba45b4b7a878e0b299e03a : SUCCESS in 6m 49s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/59eb5197a2734fcab4028a07873003f2 : SUCCESS in 4m 55s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9aee58cb16e34e95bee2c42fe4710263 : SUCCESS in 5m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/54d2d22459bb4ba19337a3031419e01b : SUCCESS in 12m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/40aa1aeb341b4a0d9d843d28bd085a7b : SUCCESS in 5m 10s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/3960df9791584c3486c5208c685dcdd5 : SUCCESS in 5m 26s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/b2b19e2827084ad3906bbc769ff048f6 : SUCCESS in 5m 53s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/0f6682879f54401ebf6f05f81b757dd5 : SUCCESS in 5m 42s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/20a918dfc3c84accab73118d3d0b799f : SUCCESS in 9m 29s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/e8fbf97ab80f4e0d9eed7f14e450ef0c : SUCCESS in 1h 27m 42s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3f29f653d2984fe9973a6f996e29029e : SUCCESS in 1h 38m 24s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/919602a215f04427af27e52f3f271c21 : SUCCESS in 41m 57s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/165a24f619104e2ab6e286306af3e4cd : SUCCESS in 35m 04s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/f41c9d5567f0414d882f3f6d16548399 : FAILURE in 58m 02s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4fd4af5d4775470cbcb0e29572b0245a : SUCCESS in 44m 56s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/051bb37b99d74e7691d327bf38104d37 : SUCCESS in 2h 19m 37s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/94f3516cae454bd19fefb8ae9605140c : FAILURE in 15m 06s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/848a97a194ec43e8beaab1cd06f98068 : FAILURE in 11m 36s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/10babcca73ef4b10bfc3bbee2c81b128 : SUCCESS in 2h 06m 48s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/44fa2dd57c0b4912b6de86f5e13301e1 : SUCCESS in 1h 52m 57s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/32eb0774e42c4b92a5288b03d8557fd6 : FAILURE in 40m 48s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a791da6cccbeb58e05e0d6c7d568f84984a685ac","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-10-12 17:10:13.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3e8b799794cf4f82afbf015f22299f2e9f28da64","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2023-10-13 14:36:24.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1b00f7613336d10b3c050e968c313cf12baba5d3","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-10-16 06:43:30.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"96335b7bc7c0670abcd64daf16d289ebc1ab0e6a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-16 06:50:07.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"06dfe57c96bcc349ee2779c34b205c43b6cf4479","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-16 09:16:50.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/91c505014e4446a5a84c34f8f87c20f7\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/afabab8ddd994183b0ddd2057624ff51 : SUCCESS in 6m 37s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/0467efefdbb946d587c19e83483ab36b : SUCCESS in 5m 08s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e8359ad69c1b4c41a385aedf857c412c : SUCCESS in 5m 19s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0d1f3e1671144b8bb544efb614d89646 : SUCCESS in 11m 20s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/29954ecf8f254872a4a861fbf7878185 : SUCCESS in 5m 42s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/3c379f069e3b4968a95551156e7c3755 : SUCCESS in 4m 33s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c37e4c088b644c0e9f3f4890c3d3357b : SUCCESS in 8m 32s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/352e2546b32a4423a591e8638a9417f8 : SUCCESS in 1h 14m 26s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a5b8e23027a9486281a7a6ede03b1967 : SUCCESS in 1h 48m 55s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/7b510fa619514c708fc72c562cad489e : SUCCESS in 35m 45s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/63b2e2c4fe04481e9f05539db18afcb6 : SUCCESS in 35m 50s","accounts_in_message":[],"_revision_number":1},{"id":"8971a3e012620e2f5891bd477f74005955a1ba38","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-16 09:16:50.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"68cb6e38800d03746236d0ed3e910658afe54323","revisions":{"68cb6e38800d03746236d0ed3e910658afe54323":{"kind":"REWORK","_number":1,"created":"2023-10-12 09:44:11.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/02/898102/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/02/898102/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/898102/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/898102/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/02/898102/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/02/898102/1"}}},"commit":{"parents":[{"commit":"9e5a6340156113ec3a014e64b3bad6617cd1c082","subject":"Fix timeout duration in start_vrrp_service during failovers","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9e5a6340156113ec3a014e64b3bad6617cd1c082"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-09-01 13:08:37.000000000","tz":-240},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-10-12 09:37:19.000000000","tz":-240},"subject":"Reduce duration of failovers with amphora in ERROR","message":"Reduce duration of failovers with amphora in ERROR\n\nIn the failover flow, there are multiple tasks for the configuration of\nVRRP for the other amphorae of the load balancer, but during outage the\nother amps may not be available. To prevent the tasks from attempting\nconnections to unreachable amphorae, we can detect in the first task\nthat an amp is unreachable and pass this information to the other tasks.\n\nThose connection attempts could have taken a lot of time, between 15 min\nand 40 min depending on the configuration of Octavia and the provider\ndriver (amphorav1 or amphorav2)\n\nCloses-Bug: #2033894\n\nNote: stable/2023.1 and older, the patch also includes modifications in\n      octavia/controller/worker/v1/\n\nConflicts:\n\toctavia/controller/worker/v2/tasks/amphora_driver_tasks.py\n\nChange-Id: Ib33a0b8d2875e4ff97c65933fe9360bb06994d32\n(cherry picked from commit ca70587f0bcb4b2a05bd5c243f111140a5762114)\n(cherry picked from commit 6fdc2079da3392fdfbdf4a80df05bfe7d149ca95)\n(cherry picked from commit 97b1b8387e42f8361b612a0b81b1c60b6e14db37)\n(cherry picked from commit dd849b4c5c1a1d33e4fd4a9afa2f35903d4d9aac)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/68cb6e38800d03746236d0ed3e910658afe54323"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/68cb6e38800d03746236d0ed3e910658afe54323"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
