)]}'
{"id":"openstack%2Foctavia~898212","triplet_id":"openstack%2Foctavia~master~Id34bf69a3e9f78ad32f8e16d2817d0fecca6ac8f","project":"openstack/octavia","branch":"master","attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2023-10-13 12:24:03.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"Id34bf69a3e9f78ad32f8e16d2817d0fecca6ac8f","subject":"Fix failover of A/S LB when server-group quota is full","status":"NEW","created":"2023-10-13 09:18:45.000000000","updated":"2023-10-13 17:12:40.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"a03ad4a622e38da4ea53cc88487dec16990ab62b","_number":898212,"virtual_id_number":898212,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2023-10-13 17:12:40.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-10-13 12:24:03.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"b2d227f4d0a0da0a7123d31f5fcf128b4183d928","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-10-13 09:18:45.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"7652917f745b5065a8c366a240a535059a48f7f7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-13 12:24:03.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/6825162e4605404da9a0558d1d98331c\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/46efc444326c4aecac6852746b8aa5d3 : SUCCESS in 11m 01s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2fa280d264f44293941cff537f050ee2 : SUCCESS in 4m 54s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ab9c9a84ec774b83bee6a490c774e135 : SUCCESS in 5m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/3d854e9cd5fd4080aac7c8b89dc0316a : SUCCESS in 4m 58s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/482865a297b64c2a91122b1c3a4768a2 : SUCCESS in 4m 23s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ee395318036b496d9a753842da56e3e6 : SUCCESS in 5m 20s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c9eb58c0fb8444a8ba369cae703bb43b : SUCCESS in 12m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bd14bad13e3f4cf699ac89f243ff9666 : SUCCESS in 4m 51s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/5fd4e58126a54852981858d6969aac25 : SUCCESS in 5m 23s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/f77c35307f104cae8c4fb7806108d65f : SUCCESS in 8m 42s\n- octavia-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/6aa515c06bcb490a8f7786a97cfa3c99 : SUCCESS in 7m 48s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/6afc23d624b145a7bfe19e6d1400dcee : SUCCESS in 5m 26s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/66e401276e1b48a38f39e1ab2fe7ceea : SUCCESS in 7m 07s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/951aeb61dba247cba78a3bbbc2cd753a : SUCCESS in 6m 05s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/7d93c5f4deec4d9faf45fe718880e37c : SUCCESS in 7m 22s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/3bcdf31aa5d94cb5808d7a47f62b60da : FAILURE in 59m 43s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/f914cbc8595a47618725415c0248da82 : SUCCESS in 1h 38m 35s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/6d1f80e7a05e42e9978dd4256fa290ac : SUCCESS in 1h 07m 42s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/8041fafe97e54f04b6a3e2294d8f3a4d : SUCCESS in 1h 40m 19s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/5f1577a37e6e4e17abc60de325936795 : SUCCESS in 59m 54s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/762030726c4641448e0f530068a409d9 : SUCCESS in 36m 07s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/1ad23743dc0947beb2250f36a241ece4 : SUCCESS in 42m 20s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/13e8bd3487f44d33b17c2edb7be3b884 : SUCCESS in 40m 02s (non-voting)\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/dfaa21b8e6104737a22cbbe0b0cfcd62 : SUCCESS in 1h 04m 14s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f10c313c4b354f0eb24253f4148734eb : SUCCESS in 34m 56s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/be549f11af1046a5a7788cd6102c45c0 : SUCCESS in 2h 05m 12s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/a440d9e7fb3642fcb5ad8fc198914606 : TIMED_OUT in 3h 03m 17s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/5f57eee758564d928201f4d92c762f13 : TIMED_OUT in 2h 35m 49s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/cb9d124ea8cf42f8bdb3e0093e1400d9 : TIMED_OUT in 2h 35m 21s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"33ffd61ce5f5544e22ffa1d7663f7a68ee111bad","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-10-13 14:09:13.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"a03ad4a622e38da4ea53cc88487dec16990ab62b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-13 17:12:40.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6892599d15a549f697fba6702fcce6b4\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/060d3ae784e84b469f12ab210fb53896 : SUCCESS in 11m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e10c48b51980437ca05b344276b4c0a9 : SUCCESS in 5m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/246b7740cdb54f1095310658db0f8ae0 : SUCCESS in 5m 43s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/498e5eedd92d451aa812b2c1cdbb648a : SUCCESS in 6m 04s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/ff413a6ee62c451eaee0001875d83f11 : SUCCESS in 4m 46s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/50b62217d3fa4baaa1dd47ee78faf4c2 : SUCCESS in 5m 15s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e73f0d3b6a154265a29db34e9b5e848d : SUCCESS in 14m 15s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c3995fd1701640988be05b80ee69027d : SUCCESS in 5m 34s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/c807e725d34b46839023ccd9f794f0c9 : SUCCESS in 5m 32s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/f10ae3a67b3847f58a555bd855f795f0 : SUCCESS in 8m 31s\n- octavia-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/66358ca64ab446b7baea4cb013a802db : SUCCESS in 12m 35s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/f41a82243d0b40159eea5c793970acab : SUCCESS in 4m 02s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/d8e50789b3d4450ebb7e910468ac81ac : SUCCESS in 10m 02s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/487d877f38414d03be3c5a8e90ef4f16 : SUCCESS in 7m 50s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/acf0ea4218ba4af99ca9d0de0d60718e : SUCCESS in 7m 42s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/ec6c6c760a754167a404d9168aa633b0 : SUCCESS in 1h 28m 47s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/31c0913f723948f4bc593d6624dc5b75 : SUCCESS in 1h 44m 21s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/20bf2fe02ed24a9ca08d377cc7be6708 : SUCCESS in 1h 10m 54s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/835c9447f1b141a4b93f95e40fa40008 : SUCCESS in 1h 41m 13s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/7657e67e356842dca59a72b16fd607fa : SUCCESS in 1h 24m 21s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/6cb3f44f33a3495cb451b247028081b7 : SUCCESS in 36m 34s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/6b3aa76faced4cae85764486ed48e751 : SUCCESS in 39m 13s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/0cde7a419dd4495b99362c476f3faf8e : SUCCESS in 55m 32s (non-voting)\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/1f5755d9b34e40dda2e20dc07fbf9a8e : SUCCESS in 1h 06m 07s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b2e0b21a46ea4eeaa4dc08129a07e10b : SUCCESS in 46m 35s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/af8cd01273d04684aa966c1e46e8e26a : TIMED_OUT in 2h 34m 50s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/be4df6083bc04ecfa68fb84944fa2a8d : SUCCESS in 2h 56m 00s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/6dbc4d71bcff4652b893d4a0d1447b3c : TIMED_OUT in 2h 34m 10s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/bae2b11388d14ff69749e1a3d6b02feb : TIMED_OUT in 2h 34m 31s (non-voting)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"bfd2feb1b28fa84c84709abe54c7b0b17e75ba5c","revisions":{"b60af0165e036997677f0804f37234a5d5df8d43":{"kind":"REWORK","_number":1,"created":"2023-10-13 09:18:45.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/12/898212/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/898212/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/898212/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/898212/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/898212/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/898212/1"}}},"commit":{"parents":[{"commit":"8b275c9703fa610e6c9bb47645c4befbc848d6d4","subject":"Merge \"Fix Zuul warnings with regexps\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8b275c9703fa610e6c9bb47645c4befbc848d6d4"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-10-13 09:09:22.000000000","tz":-240},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-10-13 09:17:04.000000000","tz":-240},"subject":"Fix failover of A/S LB when server-group quota is full","message":"Fix failover of A/S LB when server-group quota is full\n\nWhen the server-group quota was full, the creation of the A/S LB failed\nbut the failover of the LB recreated the amphorae succesfully but\nwithout setting the server-group.\nThis commit ensures that the server-group is created in Nova and set in\nthe Octavia DB when creating or performing a failover of a load\nbalancer.\n\nCloses-Bug: #2037300\n\nChange-Id: Id34bf69a3e9f78ad32f8e16d2817d0fecca6ac8f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b60af0165e036997677f0804f37234a5d5df8d43"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b60af0165e036997677f0804f37234a5d5df8d43"}]},"branch":"refs/heads/master"},"bfd2feb1b28fa84c84709abe54c7b0b17e75ba5c":{"kind":"REWORK","_number":2,"created":"2023-10-13 14:09:13.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/12/898212/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/898212/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/898212/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/898212/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/898212/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/898212/2"}}},"commit":{"parents":[{"commit":"8b275c9703fa610e6c9bb47645c4befbc848d6d4","subject":"Merge \"Fix Zuul warnings with regexps\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8b275c9703fa610e6c9bb47645c4befbc848d6d4"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-10-13 09:09:22.000000000","tz":-240},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-10-13 14:05:53.000000000","tz":-240},"subject":"Fix failover of A/S LB when server-group quota is full","message":"Fix failover of A/S LB when server-group quota is full\n\nWhen the server-group quota was full, the creation of the A/S LB failed\nbut the failover of the LB recreated the amphorae succesfully but\nwithout setting the server-group.\nThis commit ensures that the server-group is created in Nova and set in\nthe Octavia DB when creating or performing a failover of a load\nbalancer.\n\nCloses-Bug: #2037300\n\nChange-Id: Id34bf69a3e9f78ad32f8e16d2817d0fecca6ac8f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bfd2feb1b28fa84c84709abe54c7b0b17e75ba5c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bfd2feb1b28fa84c84709abe54c7b0b17e75ba5c"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
