)]}'
{"id":"openstack%2Foctavia~898803","triplet_id":"openstack%2Foctavia~master~I085ffc78e768f2323ff4765375d3df9165b64d2a","project":"openstack/octavia","branch":"master","topic":"bug/2039430","attention_set":{"27442":{"account":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"last_update":"2023-10-19 11:50:24.000000000","reason":"A robot voted negatively on a label"}},"removed_from_attention_set":{},"hashtags":[],"change_id":"I085ffc78e768f2323ff4765375d3df9165b64d2a","subject":"Add validation for minimum number of ips when the subnet is specified","status":"NEW","created":"2023-10-19 09:08:01.000000000","updated":"2026-02-20 19:54:21.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"fca1cc2234d451e9b7c9c755a01ec0c9da90b359","_number":898803,"virtual_id_number":898803,"owner":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":-1,"date":"2026-02-20 19:54:21.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-10-19 11:50:24.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"}],"messages":[{"id":"a84587235c34e59bba9130252edb28ffb53c7c0b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"date":"2023-10-19 09:08:01.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4da7521028ef20c397022e11aec051622d8bb112","author":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"date":"2023-10-19 09:42:28.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"29d4f929a4b22f55a162c53c455be49ecd84e963","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-19 11:50:24.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b76011ed3b7f4731b29af2a168b6d7b3\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f54cca87573b453c8fe86b8e7ff4cd14 : FAILURE in 6m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/93ee5346c08343498690850969a08d3b : SUCCESS in 4m 45s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7f5e4bf39e90474b9bf58d48b71686e4 : SUCCESS in 4m 43s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ade88349bfa14ec691c74fe14015bdaa : SUCCESS in 7m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/49cb623b4d0f4205b11b731783ef1ad9 : SUCCESS in 11m 49s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/19cefbe4e3bc449fb244539bc48da713 : SUCCESS in 5m 23s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/9dca56e87e6d4d04b329da0ea41a24c3 : FAILURE in 5m 12s\n- octavia-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/cb190e69f0b64e22a29fe8695c1d9a65 : FAILURE in 7m 10s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/4b7f286932c345d9bd1d0353a35beab6 : SUCCESS in 4m 20s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/f1eb487a00bc47699b32f1a223068abc : FAILURE in 4m 35s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/7c0560859a9b4ec7be5827300d9a2be3 : FAILURE in 4m 26s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/ea2be0ef117840cc8f5c6f53c133c7ff : FAILURE in 4m 24s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/1eb85e7fb01a4eb58a789111e33b84f8 : SUCCESS in 1h 25m 03s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/bed3313c86604efaaf502902c461af30 : SUCCESS in 1h 36m 14s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/c6caf4a970f64408b455d6973e46c94a : SUCCESS in 56m 35s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/66c9afefae724998b1fe926aefe68c5a : SUCCESS in 1h 41m 26s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/612580189e4d4613adf35a52bb62a5c5 : SUCCESS in 1h 02m 45s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/616f602fd6574360af0afd64a02ac540 : SUCCESS in 37m 02s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/933dc5d8c2c849c28217d261f1204c7b : SUCCESS in 38m 41s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/2b330dcb7c814d5da6679f32174b4860 : SUCCESS in 36m 48s (non-voting)\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/c5368b644b644de780d07ba4f22846c4 : SUCCESS in 1h 11m 05s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/84224f6405cc40c1a8e8bdbbd0becdb4 : SUCCESS in 31m 54s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/f4dd02dced544154911e6fb62e463eba : SUCCESS in 2h 09m 50s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/9483b66a62f54124a2a03d9a91afad3a : SUCCESS in 2h 40m 40s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/0aa0b73a5d1c492f967e0a2d8dd0da11 : SUCCESS in 2h 30m 21s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/e14cf0d1c271487ba83678a295932dcc : TIMED_OUT in 2h 35m 11s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"08afe1671459e040a8b39dfe80cc9d34f8f15953","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"date":"2023-11-10 16:27:25.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"a6ba5acc8e834ce25fd9310ac77a0af43452734c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-10 19:19:49.000000000","message":"Patch Set 2: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ffddd47be9b142d993d2700c1c0402c0\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bc0b0fd885c24de7bd4d2a092d747c9b : FAILURE in 9m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ea8ed2c8d2324c98b976bdbe7641adfd : FAILURE in 3m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7562906cbbf541feade0ddbb04f3ccfd : SUCCESS in 4m 44s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1ebfb4fef18e48c1af99e5e041f3bffb : SUCCESS in 5m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/72e89b905af54a6289eed73142a09e95 : SUCCESS in 11m 32s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/e913f1b76f31492f90d50a89c7715916 : SUCCESS in 4m 40s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/4f2eced19ca142b2a724f570af37a2eb : FAILURE in 7m 37s\n- octavia-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/632288542fe742cc9e1975c9849a6739 : FAILURE in 8m 33s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/26cd500ce11b4cec89d010886e801daa : SUCCESS in 4m 52s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/e7e27ff8071140e5a142a5eadd4eee6f : FAILURE in 7m 38s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/95a5a970e494414ca3eb8a8a80832bca : FAILURE in 7m 47s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/b05bc8c22dec46f19da07b9049095925 : FAILURE in 7m 19s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/4ab34767fef34382aade64b6c4d4616c : FAILURE in 41m 14s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/c62b8d918bbc4a6bb4b9cfb3e2d98191 : SUCCESS in 1h 37m 34s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/998bc7d132454f1c81d4cf407cf0e1cc : SUCCESS in 1h 13m 34s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/2a2b2575d12d47be9797d641e1ebcb5b : SUCCESS in 2h 11m 23s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/e89bce157d0a455790959150f172329f : SUCCESS in 1h 18m 56s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/6396342c723f4efc84af2c87713b05e3 : SUCCESS in 44m 16s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/74cb9e29931c4b93a81880ad99ee36a7 : SUCCESS in 54m 58s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/f82b37039cec44dbb98c9898ff86fad2 : SUCCESS in 55m 35s (non-voting)\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/8bb3a21a75ff43c88d6f4ac54aed5438 : SUCCESS in 1h 07m 27s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f25e35c5669140acac83b33783ad9826 : FAILURE in 46m 03s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/3d878afc1b6c45c7920081fb2b86d3f4 : SUCCESS in 2h 26m 39s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/648cbb137e254e1ea743ad2150d4b1b7 : SUCCESS in 2h 41m 14s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/8b41b9fb744c458d9bc3e00be2a83b4b : SUCCESS in 1h 56m 37s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/f0cd0628bf5a475382120035cec69307 : TIMED_OUT in 2h 33m 45s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"1cf61457ec65b6cc694b3de8bbc8d433df908c89","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"date":"2023-11-16 18:10:42.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"90dcc5ad387132380ea93b6ca1ff87502ed3ffff","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-16 21:18:14.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7f9f98faf8bd4e3792901c5842f194db\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2e983788f9bc47aa84edb85e51e22b5d : FAILURE in 10m 17s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/326ac416963f41478a9571cf59ce9a53 : SUCCESS in 5m 16s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cb6260c2fbce42febf952f4b77ae4e2f : SUCCESS in 4m 11s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1709a67766b04cfda12094779485da5f : SUCCESS in 5m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ae0abc5ff1dd4e4ea08844746699cca3 : SUCCESS in 11m 36s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/0a7f8a37fb8e42dea9aa6ae94210c719 : SUCCESS in 5m 19s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/fcd077d4f9784ac8a7ec24d114db0adf : FAILURE in 6m 36s\n- octavia-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/223f86db21784f408f71b1168a947cab : FAILURE in 10m 50s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/04a53c38b59347e2ae2b75b8b21d120b : SUCCESS in 4m 26s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/f9bb5143e00d43cea3a48adda8dac54c : FAILURE in 4m 11s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/02acad97177c413296fe130857735a33 : FAILURE in 8m 51s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/97620acff95b43ab9b9e68639a9777d3 : FAILURE in 8m 29s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/bf75c5c2e0e3428a8f72b85e16e25556 : FAILURE in 22m 32s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/2d1150e2cbcb4d18b03f3deb5dcda981 : SUCCESS in 1h 38m 53s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/57dd2c5fda72411f981edd3ff7fc586e : SUCCESS in 58m 48s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/e5e57225860e49648480bfe18c7c359a : SUCCESS in 1h 43m 07s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/2acb120e300442c7a13c0e38912057c4 : SUCCESS in 1h 02m 07s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/f5429e0d748e4a4e9317697a929fdd55 : SUCCESS in 34m 38s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/5693201f47e443a5833b7fab9f719c21 : SUCCESS in 43m 08s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/682b757539a647948f73009fc3ee03e0 : SUCCESS in 54m 29s (non-voting)\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/6885c37695c04867830be71e993e3ac1 : SUCCESS in 59m 29s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/bfa453fbaea64067a8563fd190a218f0 : SUCCESS in 34m 43s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/b79463a797a44c45ab129002d0c67c68 : SUCCESS in 2h 03m 23s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/f8e0594c083145a98908b04d02fc0fdf : TIMED_OUT in 3h 04m 24s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/c0469efa7ad949beae5e791e122f04ce : SUCCESS in 2h 14m 57s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/b8edc504c4ec4e08b4aa2f296240d98a : TIMED_OUT in 2h 33m 31s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"bd340871e2c6baaee37bbf5ba58345054a48dbf8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"date":"2023-11-20 16:05:33.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"37fa9b50cad3116756d165469a238961561725f3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-20 19:00:33.000000000","message":"Patch Set 4: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/38d5f6605e9344269e8ae768bb38c6bd\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4255bf19d3ff4996bf9921afde1c30c4 : SUCCESS in 10m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4fdf40d6aa5a410494f433bd6475c038 : SUCCESS in 4m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0984b1bcd28e4718aecea98b8b25a6c3 : SUCCESS in 4m 31s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0cf185a10c2f4c7e85bdd90d3efbb243 : SUCCESS in 5m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9e7b2878fc70437d926709c6b359bde9 : SUCCESS in 11m 44s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/c5668db47c7f437193a589ea8a4c59d9 : SUCCESS in 4m 59s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/1d3aa78cfd8d4fe6a54b4e4e077777f3 : SUCCESS in 7m 13s\n- octavia-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/fb362b647b6d4cb1b94141e8e8f9d25a : SUCCESS in 9m 40s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/8f7c444d55694ba7b61fc5911146e164 : SUCCESS in 4m 58s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/8bcd473f35a94a10b25dad0617c9fd22 : FAILURE in 3m 57s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/342e1f095d6f4afd972ea21d7a84e3d2 : SUCCESS in 7m 35s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/e459663d80e647219ee927da012e0b9d : SUCCESS in 7m 00s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/47b3ed90041f4766954879de5dbd0b8b : FAILURE in 37m 40s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/1664b6fd52ed47489f2bc1021dfe4700 : SUCCESS in 1h 36m 42s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/574a69cdc7e247c3a5a392f1a0f6264f : SUCCESS in 1h 16m 07s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/594ae105b3a44e439db575b21947feb0 : SUCCESS in 1h 42m 27s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/5abba8841f994502b790eacde059fd25 : SUCCESS in 1h 16m 58s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/95231b505c9f441e9e15b9fcba51ba21 : SUCCESS in 44m 33s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/97f9b96b93c047f0b86b6a1996ec0451 : SUCCESS in 45m 00s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/80ad1d028df64f1986d734f127c43246 : SUCCESS in 36m 37s (non-voting)\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/92f2f280c1894343909962c3b9a523c4 : SUCCESS in 1h 06m 22s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5bc68e720c8f4ef09d53c35b94fa8611 : SUCCESS in 38m 38s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/c30db2daa54947cb8461aed7322f5c95 : SUCCESS in 2h 26m 42s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/a26f196bab8a4b218cf237ef4ee98d8b : SUCCESS in 2h 51m 25s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/b2282bc27a3843bb9cf20b5dd47ba874 : SUCCESS in 2h 19m 26s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/c1151480527945e9b250b15efd8bea40 : TIMED_OUT in 2h 33m 43s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"72f4d81d7ca47f9cbee12f66757c1e4e13fbfd4d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"date":"2023-11-21 18:46:39.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":5},{"id":"30eebcf703e6bfca8198f0c34eb3528a9678872d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-21 21:34:09.000000000","message":"Patch Set 5: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/432b36d8b47747f0aaf49d7e5a49e41e\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5a162d92ac9f4513955da24b57b43769 : SUCCESS in 9m 16s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0697abbd2fa74600acda7da735225b15 : FAILURE in 4m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9f29d52cb09148d7b19512beab7bb578 : SUCCESS in 4m 28s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/15db718e7fef452a94ba5cf6950a8813 : SUCCESS in 7m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9e4cc0a91e9944f88bbe52994677b8b9 : SUCCESS in 10m 50s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/95ebd3b44af34b8b8c8548260e282085 : SUCCESS in 5m 38s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/234ce7ee4d5e4f189f70540d1cda17a3 : SUCCESS in 7m 44s\n- octavia-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/897365d3edb14de79e32f14205923c49 : SUCCESS in 9m 16s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/cba6ee86f0764b9085195ba53e4084d9 : SUCCESS in 4m 27s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/c0ef6c6ee65647a6b9baf5a9cd042bef : SUCCESS in 7m 13s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/42bd1e6b0f674769913c174a67c9f95b : SUCCESS in 7m 27s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/9a5dfd8b71494c4a8257edb0eccca466 : SUCCESS in 7m 38s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/d48c2349c506492a963b666ca494ea0b : FAILURE in 36m 10s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/6a9873ef85e54a19aec77e755bb5605b : SUCCESS in 1h 32m 31s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/0c714008012c46efb82c5eb201802319 : SUCCESS in 1h 12m 43s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/c539a53868544f4f8be0f21ebd88ec2a : SUCCESS in 1h 50m 38s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/dde34640497d46a4862bd8f261d3e11e : SUCCESS in 1h 00m 11s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/83974d549bc744099a5215b93063ed10 : SUCCESS in 45m 06s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/5d0821769b4b4a67af226906943cbc53 : SUCCESS in 38m 19s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/d00c6ca77ea74ce0983918f0e82dcc80 : SUCCESS in 37m 15s (non-voting)\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/da77c9fbbc744d14bb3f6ab0a71bddbb : SUCCESS in 1h 07m 17s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/fa25360114ca4d42bd4e97d3f70de4de : SUCCESS in 34m 39s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/c8ad98a6d09b499db1f1be725e89d09f : TIMED_OUT in 2h 33m 44s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/e97f6e27cb9e49b995392e8453572dea : SUCCESS in 2h 26m 46s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/69565979d65b4c2381f6f92c65f1ee2d : SUCCESS in 2h 33m 29s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/60ddc8f686a2437eaa0cb522071cc1e3 : SUCCESS in 2h 25m 54s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"65efaa7fa444f21b3e79d815d4c72639a5b1fd42","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"date":"2023-11-22 09:33:19.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":6},{"id":"65e5eac4cce9e8b30fdefbaa6e691e8d77548258","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-22 12:16:11.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9452b420736d4f648e6d85a25020cb00\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3f385b6bc4f940b3b6660026db62c027 : SUCCESS in 12m 38s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2c85ecfec70b4ac99994fa7c7f647965 : SUCCESS in 4m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f38e717c6ec145208a105d3e818f5266 : SUCCESS in 4m 28s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/631883a39f6e4972a9523bf4b96adf78 : SUCCESS in 6m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d912cce912594edcb841818ce3f2610f : SUCCESS in 12m 17s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/0d4c353bddb447789c0a02fd8fdc90f6 : SUCCESS in 5m 12s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/851624ec59f74d45af68f447a9a5dd1b : SUCCESS in 8m 33s\n- octavia-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/146dfc5f7a66423b9c169676eabe33c1 : SUCCESS in 10m 07s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/2dc1777505df43bdb1998a444b89014e : SUCCESS in 5m 06s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/de8d30fba73d4107ae3f7e844ddd77b8 : SUCCESS in 6m 54s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/b27d4ba1b259434ab40c7646b8dfbc5f : SUCCESS in 7m 42s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/45b9d307884c4c2ab253f9087ada5820 : SUCCESS in 7m 26s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/51409f3fc72c4039bfc520653e988b6f : FAILURE in 36m 37s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/667be435f0b848e4bfba74d195a27023 : SUCCESS in 1h 34m 14s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/9ffddf6dc2f041d19f185d580768661b : SUCCESS in 57m 08s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/58d878c825ba4763ba914a29f2654e19 : SUCCESS in 1h 37m 00s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/43f2f93f8d3645089f64aed6b660770c : SUCCESS in 1h 31m 29s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/0c2e54d600ce414a910e9753a0a9cc40 : SUCCESS in 50m 39s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/04a7a56e42c242a9aa2c22942188c103 : SUCCESS in 38m 16s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/0264035f5cbd442f9833b44a12a3a1d2 : SUCCESS in 38m 17s (non-voting)\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/9044419e8eb74af0a99686651791ba79 : SUCCESS in 59m 54s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/092660a6f6634ca5badf1f2941a1b8c4 : SUCCESS in 44m 49s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/056de36b938744009648c8aaa7658115 : TIMED_OUT in 2h 33m 37s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/c17257beed0e41eb99e23f0b82a8feec : SUCCESS in 2h 41m 26s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/b32677028284438d84ce7c39baa690c4 : TIMED_OUT in 2h 33m 35s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/a5c21c7408d7464595de7733ee938789 : TIMED_OUT in 2h 34m 49s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"96f56617b9c282ddb282a250f1d85727bc87aa87","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"date":"2026-02-20 17:07:16.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"9374b4f249ec12047d0ef3cd653f48ec16a91585","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"date":"2026-02-20 17:15:25.000000000","message":"Uploaded patch set 8: Patch Set 7 was rebased.","accounts_in_message":[],"_revision_number":8},{"id":"fca1cc2234d451e9b7c9c755a01ec0c9da90b359","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2026-02-20 19:54:21.000000000","message":"Patch Set 8: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/3aa9dfc217db4f689eff5ac6e1989243\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/e5af32d60d2b44e7bedf8b10fce72717 : SUCCESS in 8m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0893b0e19bd14c65a720394050a60e07 : SUCCESS in 2m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/3e9e4e6f26284754978e4196ce25e475 : SUCCESS in 3m 40s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/efe8dd7e436640c79d32a9d4e3f77652 : SUCCESS in 4m 33s\n- openstack-tox-py314 https://zuul.opendev.org/t/openstack/build/46176fdb81a647e6a8d10278b1b817ee : SUCCESS in 7m 55s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b90519a74c1a4459a929c2a534c65397 : SUCCESS in 12m 21s\n- octavia-tox-py313-tips https://zuul.opendev.org/t/openstack/build/128a41e06c554a1489f7279aa75be45c : SUCCESS in 4m 34s\n- octavia-tox-functional-py313-tips https://zuul.opendev.org/t/openstack/build/ee9972ff3750490dba6105208663dbef : SUCCESS in 5m 19s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/9b9a0ecd8ee04894b20c24ccda0470d7 : SUCCESS in 3m 03s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/28e74171d0fd4d33b0f0b2a770c8dc26 : SUCCESS in 7m 49s\n- openstack-tox-functional-py313 https://zuul.opendev.org/t/openstack/build/022c20dfcd704d3a8a76999f20acdb67 : SUCCESS in 4m 13s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/1dd6829f46704bdbb43a146f584ce8bf : FAILURE in 19m 13s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/b9e1b3bcebdd403286556f397ef4f3b5 : SUCCESS in 1h 38m 43s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/1b8b36acbae845b69f2c5b1439c2e576 : SUCCESS in 1h 07m 07s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/92341fc8ae6c4aa5b02fd6a1f6647a29 : SUCCESS in 1h 38m 46s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/8dec26845dd4492396cb160f2f04fbfd : SUCCESS in 1h 11m 00s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/4f3306bc30744ee4928d1bee0fd7b383 : SUCCESS in 35m 52s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/c08150192562415fa51ef901a61e2e7a : SUCCESS in 43m 16s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/698e5777403d499db61c1cd47feb0ca0 : SUCCESS in 49m 22s\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5f21ae8c4d804dc983e1651147d62924 : SUCCESS in 33m 10s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/c508ac72e3474656beefc023f3987dc5 : SUCCESS in 34m 54s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/d2438c63776140e2bf6797c20fb0909f : SUCCESS in 2h 37m 04s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/b594735a5c594f8f84bb06b2ed110390 : SUCCESS in 1h 37m 32s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-traffic-ops https://zuul.opendev.org/t/openstack/build/08cefb03cda0453abf5bdfdd17817f27 : SUCCESS in 1h 26m 01s (non-voting)\n- octavia-v2-dsvm-scenario-centos-10-stream-non-traffic-ops https://zuul.opendev.org/t/openstack/build/317d8d1c69dd4bfdada23d9f1e2e2d19 : SUCCESS in 1h 09m 32s (non-voting)","accounts_in_message":[],"_revision_number":8}],"current_revision_number":8,"current_revision":"4c0f0fb30ce20afc7a682ae759ab81cb2a8bc7a0","revisions":{"dbb8819d1af8e58e66e048005ef810482d32660e":{"kind":"REWORK","_number":1,"created":"2023-10-19 09:08:01.000000000","uploader":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"ref":"refs/changes/03/898803/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/03/898803/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/03/898803/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/03/898803/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/03/898803/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/03/898803/1"}}},"commit":{"parents":[{"commit":"860d2f6aadfba9cef2c695d072dd48e29a7e85c8","subject":"Merge \"Fix incorrect masquerade rules in multivip LBs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/860d2f6aadfba9cef2c695d072dd48e29a7e85c8"}]}],"author":{"name":"Quentin GROLLEAU","email":"quentin.grolleau@ovhcloud.com","date":"2023-10-19 08:35:43.000000000","tz":120},"committer":{"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","date":"2023-10-19 09:05:40.000000000","tz":120},"subject":"Add validation for minimum number of ips when the subnet is specified","message":"Add validation for minimum number of ips when the subnet is specified\n\nThe _validate_vip_request_object is now calling _validate_subnet which\nwill verified is there is enought ip available in the subnet\n\nCloses-Bug: #2039430\nChange-Id: I085ffc78e768f2323ff4765375d3df9165b64d2a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/dbb8819d1af8e58e66e048005ef810482d32660e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/dbb8819d1af8e58e66e048005ef810482d32660e"}]},"branch":"refs/heads/master"},"a16c46e12178712e5270b40f402302eb895f9f60":{"kind":"REWORK","_number":2,"created":"2023-11-10 16:27:25.000000000","uploader":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"ref":"refs/changes/03/898803/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/03/898803/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/03/898803/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/03/898803/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/03/898803/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/03/898803/2"}}},"commit":{"parents":[{"commit":"860d2f6aadfba9cef2c695d072dd48e29a7e85c8","subject":"Merge \"Fix incorrect masquerade rules in multivip LBs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/860d2f6aadfba9cef2c695d072dd48e29a7e85c8"}]}],"author":{"name":"Quentin GROLLEAU","email":"quentin.grolleau@ovhcloud.com","date":"2023-10-19 08:35:43.000000000","tz":120},"committer":{"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","date":"2023-11-10 16:27:18.000000000","tz":60},"subject":"Add validation for minimum number of ips when the subnet is specified","message":"Add validation for minimum number of ips when the subnet is specified\n\nThe _validate_vip_request_object is now calling _validate_subnet which\nwill verified is there is enought ip available in the subnet\n\nCloses-Bug: #2039430\nChange-Id: I085ffc78e768f2323ff4765375d3df9165b64d2a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a16c46e12178712e5270b40f402302eb895f9f60"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a16c46e12178712e5270b40f402302eb895f9f60"}]},"branch":"refs/heads/master"},"3654fd25d86ad7d19ccd8879670801b00b55ae53":{"kind":"REWORK","_number":3,"created":"2023-11-16 18:10:42.000000000","uploader":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"ref":"refs/changes/03/898803/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/03/898803/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/03/898803/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/03/898803/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/03/898803/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/03/898803/3"}}},"commit":{"parents":[{"commit":"7310986de9bf68ed86de90de0501a1bc46945526","subject":"Add Noop Certificate Manager","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7310986de9bf68ed86de90de0501a1bc46945526"}]}],"author":{"name":"Quentin GROLLEAU","email":"quentin.grolleau@ovhcloud.com","date":"2023-10-19 08:35:43.000000000","tz":120},"committer":{"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","date":"2023-11-16 18:09:52.000000000","tz":60},"subject":"Add validation for minimum number of ips when the subnet is specified","message":"Add validation for minimum number of ips when the subnet is specified\n\nThe _validate_vip_request_object is now calling _validate_subnet which\nwill verified is there is enought ip available in the subnet\n\nCloses-Bug: #2039430\nChange-Id: I085ffc78e768f2323ff4765375d3df9165b64d2a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3654fd25d86ad7d19ccd8879670801b00b55ae53"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3654fd25d86ad7d19ccd8879670801b00b55ae53"}]},"branch":"refs/heads/master"},"931b22161769c1252fb14001ac13a164a6858538":{"kind":"REWORK","_number":4,"created":"2023-11-20 16:05:33.000000000","uploader":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"ref":"refs/changes/03/898803/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/03/898803/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/03/898803/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/03/898803/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/03/898803/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/03/898803/4"}}},"commit":{"parents":[{"commit":"7310986de9bf68ed86de90de0501a1bc46945526","subject":"Add Noop Certificate Manager","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7310986de9bf68ed86de90de0501a1bc46945526"}]}],"author":{"name":"Quentin GROLLEAU","email":"quentin.grolleau@ovhcloud.com","date":"2023-10-19 08:35:43.000000000","tz":120},"committer":{"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","date":"2023-11-20 15:55:19.000000000","tz":60},"subject":"Add validation for minimum number of ips when the subnet is specified","message":"Add validation for minimum number of ips when the subnet is specified\n\nThe _validate_vip_request_object is now calling _validate_subnet which\nwill verified is there is enought ip available in the subnet\n\nCloses-Bug: #2039430\nChange-Id: I085ffc78e768f2323ff4765375d3df9165b64d2a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/931b22161769c1252fb14001ac13a164a6858538"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/931b22161769c1252fb14001ac13a164a6858538"}]},"branch":"refs/heads/master"},"4e1f00ebe5efa7c07a67e8825afabc6a4effe449":{"kind":"REWORK","_number":5,"created":"2023-11-21 18:46:39.000000000","uploader":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"ref":"refs/changes/03/898803/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/03/898803/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/03/898803/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/03/898803/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/03/898803/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/03/898803/5"}}},"commit":{"parents":[{"commit":"7310986de9bf68ed86de90de0501a1bc46945526","subject":"Add Noop Certificate Manager","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7310986de9bf68ed86de90de0501a1bc46945526"}]}],"author":{"name":"Quentin GROLLEAU","email":"quentin.grolleau@ovhcloud.com","date":"2023-10-19 08:35:43.000000000","tz":120},"committer":{"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","date":"2023-11-21 18:46:31.000000000","tz":60},"subject":"Add validation for minimum number of ips when the subnet is specified","message":"Add validation for minimum number of ips when the subnet is specified\n\nThe _validate_vip_request_object is now calling _validate_subnet which\nwill verified is there is enought ip available in the subnet\n\nCloses-Bug: #2039430\nChange-Id: I085ffc78e768f2323ff4765375d3df9165b64d2a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4e1f00ebe5efa7c07a67e8825afabc6a4effe449"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4e1f00ebe5efa7c07a67e8825afabc6a4effe449"}]},"branch":"refs/heads/master"},"e1cce5c157ec955509e05ac733346fa1eaa001c7":{"kind":"REWORK","_number":6,"created":"2023-11-22 09:33:19.000000000","uploader":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"ref":"refs/changes/03/898803/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/03/898803/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/03/898803/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/03/898803/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/03/898803/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/03/898803/6"}}},"commit":{"parents":[{"commit":"7310986de9bf68ed86de90de0501a1bc46945526","subject":"Add Noop Certificate Manager","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7310986de9bf68ed86de90de0501a1bc46945526"}]}],"author":{"name":"Quentin GROLLEAU","email":"quentin.grolleau@ovhcloud.com","date":"2023-10-19 08:35:43.000000000","tz":120},"committer":{"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","date":"2023-11-22 09:33:08.000000000","tz":60},"subject":"Add validation for minimum number of ips when the subnet is specified","message":"Add validation for minimum number of ips when the subnet is specified\n\nThe _validate_vip_request_object is now calling _validate_subnet which\nwill verified is there is enought ip available in the subnet\n\nCloses-Bug: #2039430\nChange-Id: I085ffc78e768f2323ff4765375d3df9165b64d2a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e1cce5c157ec955509e05ac733346fa1eaa001c7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e1cce5c157ec955509e05ac733346fa1eaa001c7"}]},"branch":"refs/heads/master"},"5c08e3ad6d852ccc4d9e97d69f5de542d7fc7032":{"kind":"REWORK","_number":7,"created":"2026-02-20 17:07:16.000000000","uploader":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"ref":"refs/changes/03/898803/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/03/898803/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/03/898803/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/03/898803/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/03/898803/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/03/898803/7"}}},"commit":{"parents":[{"commit":"f42860a7302b521dd2895a22535dbd1de1f912f0","subject":"Remove url tags from README","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f42860a7302b521dd2895a22535dbd1de1f912f0"}]}],"author":{"name":"Quentin GROLLEAU","email":"quentin.grolleau@ovhcloud.com","date":"2023-10-19 08:35:43.000000000","tz":120},"committer":{"name":"Quentin GROLLEAU","email":"quentin.grolleau@ovhcloud.com","date":"2026-02-20 17:07:03.000000000","tz":60},"subject":"Add validation for minimum number of ips when the subnet is specified","message":"Add validation for minimum number of ips when the subnet is specified\n\nThe _validate_vip_request_object is now calling _validate_subnet which\nwill verified is there is enought ip available in the subnet\n\nCloses-Bug: #2039430\nChange-Id: I085ffc78e768f2323ff4765375d3df9165b64d2a\nSigned-off-by: Quentin GROLLEAU \u003cquentin.grolleau@ovhcloud.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5c08e3ad6d852ccc4d9e97d69f5de542d7fc7032"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5c08e3ad6d852ccc4d9e97d69f5de542d7fc7032"}]},"branch":"refs/heads/master"},"4c0f0fb30ce20afc7a682ae759ab81cb2a8bc7a0":{"kind":"TRIVIAL_REBASE","_number":8,"created":"2026-02-20 17:15:25.000000000","uploader":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"ref":"refs/changes/03/898803/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/03/898803/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/03/898803/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/03/898803/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/03/898803/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/03/898803/8"}}},"commit":{"parents":[{"commit":"a753d57186923cc8824464e644784519e0317010","subject":"Merge \"Add c10s jobs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a753d57186923cc8824464e644784519e0317010"}]}],"author":{"name":"Quentin GROLLEAU","email":"quentin.grolleau@ovhcloud.com","date":"2023-10-19 08:35:43.000000000","tz":120},"committer":{"name":"Quentin GROLLEAU","email":"quentin.grolleau@ovhcloud.com","date":"2026-02-20 17:15:02.000000000","tz":60},"subject":"Add validation for minimum number of ips when the subnet is specified","message":"Add validation for minimum number of ips when the subnet is specified\n\nThe _validate_vip_request_object is now calling _validate_subnet which\nwill verified is there is enought ip available in the subnet\n\nCloses-Bug: #2039430\nChange-Id: I085ffc78e768f2323ff4765375d3df9165b64d2a\nSigned-off-by: Quentin GROLLEAU \u003cquentin.grolleau@ovhcloud.com\u003e\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4c0f0fb30ce20afc7a682ae759ab81cb2a8bc7a0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4c0f0fb30ce20afc7a682ae759ab81cb2a8bc7a0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
