)]}'
{"id":"openstack%2Foctavia~898863","triplet_id":"openstack%2Foctavia~master~Ic7813d3a9073e4b3c1bc6a7839242df34a16d348","project":"openstack/octavia","branch":"master","attention_set":{},"removed_from_attention_set":{"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2024-02-27 11:48:43.000000000","reason":"Change was submitted"},"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2024-02-27 07:08:50.000000000","reason":"\u003cGERRIT_ACCOUNT_29244\u003e replied on the change","reason_account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}}},"hashtags":[],"change_id":"Ic7813d3a9073e4b3c1bc6a7839242df34a16d348","subject":"Remove some unused code","status":"MERGED","created":"2023-10-19 22:54:36.000000000","updated":"2024-02-27 11:49:45.000000000","submitted":"2024-02-27 11:48:43.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"898863","meta_rev_id":"a0fe97c0f66bc2292bc19fed991da5e3f980315e","_number":898863,"virtual_id_number":898863,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-02-27 11:48:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":2,"date":"2024-02-27 07:08:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-02-27 03:50:47.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":1,"date":"2024-02-27 07:08:50.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"CC":[{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-10-20 01:57:22.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-02-07 16:52:24.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"},{"updated":"2024-02-16 10:30:46.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2024-02-26 18:04:29.000000000","updated_by":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"reviewer":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"state":"CC"}],"messages":[{"id":"96692e2f71880c37d466de8b9229ddc95f74a0ec","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-10-19 22:54:36.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"db7cfb959e7c2c63f61ab95553f6cd2d1f629840","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-20 01:57:22.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d1e9d0598f9e4284afd45357a4c9fe56\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a410d8ef9965430e863b70b26417346b : SUCCESS in 8m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3ecc29157a8745c8a256c519be1c9872 : SUCCESS in 4m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a8cbf06229bd4949977a59b7ab6f15e3 : SUCCESS in 4m 43s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5f775012df354cfb802844549ce7b57d : SUCCESS in 5m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a484cdf95c2a4dd68270bcac01128848 : SUCCESS in 13m 00s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/6b29c2ca7f2a4f78830bbc4272855187 : SUCCESS in 5m 23s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/d0b333a691a346ee8eaa58c7315cafe3 : SUCCESS in 7m 20s\n- octavia-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/36bf81bd0786498eb1ab78d4905a488a : SUCCESS in 10m 10s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/f9599198d86b40f2bdbac26cd648100e : SUCCESS in 4m 26s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/2ee569f2cb1b438495292436a2d4e610 : SUCCESS in 8m 31s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/6aef084a14344d84b3a797d47227496b : SUCCESS in 7m 54s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/9b997457164d4cf6b87c7d4b7a4a6ce9 : SUCCESS in 7m 34s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/bcf92a4f897346a7a8ad2bde8468f208 : SUCCESS in 1h 39m 14s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/9f4d3fcfe80342d5a7d3b494cd00025d : SUCCESS in 1h 34m 56s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/ad45f7e4abd34959934c68292ab4ef26 : SUCCESS in 58m 48s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/3eda8ceb5e03416c94273685e92b4b41 : SUCCESS in 2h 09m 03s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/8408701fd8eb4413a41cd292f8bc6614 : SUCCESS in 1h 06m 11s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/ec5b609db44c4766ae91fbd82c4284f7 : SUCCESS in 42m 04s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/8935460483894a658cb6bd8de546bea9 : SUCCESS in 38m 53s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/4487d9f8c7bc434eb356526e608246c4 : SUCCESS in 41m 06s (non-voting)\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/83f84fec4ea34441a9ba335065c56076 : SUCCESS in 1h 06m 54s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b352682d2c20431e903b172aac3226ce : SUCCESS in 34m 28s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/108bfd74f57a4b069d97e0dd803dddaf : TIMED_OUT in 2h 33m 44s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/66a6318fcd6c4c70805a56d4a1713237 : SUCCESS in 2h 54m 48s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/4a878ee6a1794db3b182b0f688361d2c : SUCCESS in 1h 53m 14s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/fe7af4f4e24c42c282b569633c4f7567 : TIMED_OUT in 2h 33m 07s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"a907123deb42ec49e5bf9394f4b3fb5698a46908","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-02-07 16:52:24.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"b9444fec2a59d668486248411914fb64e2be0078","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2024-02-16 10:30:46.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"009d64cdf3b29e2ad7e5557d6bda6c9c05677b26","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-16 10:31:21.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"91ece533118d5e0ec9bf7c1f79d9db44d0f66d1a","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-16 12:12:18.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8d0b957bd1ab40ae81b7c9fea6fdca00\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/06ceb67662394acbb08c7f72947a7ccd : SUCCESS in 5m 04s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a9399f5e81d6432bb5e5a8ef957761d9 : SUCCESS in 4m 59s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7e91c3a0eb434e6aafd49b43e5771927 : SUCCESS in 4m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0419204a10d642f1963452d418726230 : SUCCESS in 11m 27s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/1da246c63c234248bda656316006ed02 : SUCCESS in 4m 40s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/ba9b72b2c6f44bde9e629958418f8932 : SUCCESS in 6m 47s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/db35d68d0b994624b4bee0047b79d6bc : SUCCESS in 1h 26m 14s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/a83c588d3ebe4cc0ba1a5064e2c4ca77 : CANCELED\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/169b18db969549f6b36903f3e30d4813 : FAILURE in 1h 39m 18s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/2eeff6be588e4229834c3967fd8d171b : SUCCESS in 36m 10s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/3b0cf604a3ce4720b8a8bc6ea3455c76 : SUCCESS in 49m 58s","accounts_in_message":[],"_revision_number":1},{"id":"c387c596868b733994d21a9e0689474743fb9766","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2024-02-16 16:29:23.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"578258a6c200330e8a56687d48e43bbc4a46e95d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-16 19:20:25.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/54766ed2e46841f694c7f26ba9f635c7\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/76561c1179b641feb17355f832294c60 : SUCCESS in 10m 28s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/81b734e7d3404b769b443fa2e919d7ab : SUCCESS in 5m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/85c172f3de584975b92202e795c6f3a5 : SUCCESS in 4m 32s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a24c686a3ebf4063be278a6c50d59e74 : SUCCESS in 4m 25s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c59ea0bba1934e35a143beb75dce84cf : SUCCESS in 10m 37s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/8cd30ce059ec47d59d43652620294787 : SUCCESS in 4m 57s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/10dd3a7f6dc24b8fb37663c1d556d985 : SUCCESS in 7m 36s\n- octavia-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/5aea2b6ac17f4da69eee6f7e1b3abaf4 : SUCCESS in 12m 06s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/400445ced47248de8027cfc725c0f05c : SUCCESS in 4m 30s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/71511370e3ad47c1ae36d682967829f9 : SUCCESS in 8m 40s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/4ad595beb78a474785df6f054a5cf96d : SUCCESS in 7m 29s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/3f1b4d9d2ea1459ea32a2856227063bd : SUCCESS in 7m 16s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/3905693e27a04e62a3834e301ebe8c95 : SUCCESS in 1h 36m 46s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/3bb7fcf3bbb843a78cac8f8bb32bf956 : SUCCESS in 1h 43m 27s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/97bc16f0e00046b1a5272d01714170ca : SUCCESS in 1h 09m 59s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/092e4aac51ab4a02943c124bccd74ec3 : TIMED_OUT in 2h 34m 15s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/1169428722d24fe1afc278a7350a81c9 : TIMED_OUT in 2h 34m 09s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/3ba744999131424a8a77bb4d3f1d7c8a : SUCCESS in 59m 03s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/e20cc8d1e24d489d86fc05491efb0ae0 : SUCCESS in 1h 03m 54s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/b0dfe6a271e64b85bf7df366791e3d9b : SUCCESS in 1h 04m 04s (non-voting)\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/ae4946f3b87e4e4b9f39e76576abe126 : SUCCESS in 1h 16m 35s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0144ae8c8fea4e4da8f13ebed6673fff : SUCCESS in 51m 53s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/e768ef0cb9434622bdaedacd44676595 : FAILURE in 2h 31m 13s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/4b7b655b91f041f5a671f12c920fc314 : SUCCESS in 2h 47m 25s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/dea6ba33a990475091c2b813fc4a1548 : FAILURE in 32m 24s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/b17b08dd9cae4807845a1ce1dd729a88 : TIMED_OUT in 2h 34m 39s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"43763b9291d30316144ef3a6b0fe25034819a05a","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2024-02-16 19:53:04.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"58e31ee65ac1668789c9d646384e4980e674a20f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-16 22:43:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3c84ca7353f844ee90b7cdecbc882e96\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/358479a04f504255984e12fc391044bc : SUCCESS in 10m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cd462c1e19fe48e0a41598e2be4962fd : SUCCESS in 4m 59s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/29c602f70a084a54b64d043f10925559 : SUCCESS in 4m 25s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/68ca56ed280642c380a2c5bdfb50413f : SUCCESS in 6m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6f87401789a141919cbad061a02976cd : SUCCESS in 11m 57s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/72523fff2b1d4d7bad04c6af8078ad3d : SUCCESS in 5m 15s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/908c7df4b73b46629611a9bd6cb11207 : SUCCESS in 7m 36s\n- octavia-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/b044ca65563c43ab81aa1a8fc149191d : SUCCESS in 10m 50s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/ed365a1abfa1441f836b59f8647a7f48 : SUCCESS in 5m 50s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/e8f72535eeab493e9d1708183e9b7612 : SUCCESS in 8m 13s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/d6636685556e4d2182aadc71b848c621 : SUCCESS in 6m 49s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/8f51ce60c8f34bd18c1de57650c8f5ab : SUCCESS in 9m 47s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/528b0712090f487bb531a95e077bcbf3 : SUCCESS in 1h 38m 03s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/66cd036aab624863a30e3d0e210eed62 : SUCCESS in 1h 41m 03s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/4c00f8a635dd472694388fd0e8022d2c : SUCCESS in 1h 15m 44s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/5cce1e2ec1f6491d93f9bb6e675a2dcb : SUCCESS in 1h 44m 13s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/24485e1e549f46efb69cf98ad5d66186 : SUCCESS in 1h 07m 12s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/5b912769445a4b49863b46d7db4a334a : SUCCESS in 41m 55s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/4e97d2730af84257a75890ea7d59d241 : SUCCESS in 46m 19s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/a9e9310a06e14235b9765e813a872a40 : SUCCESS in 36m 42s (non-voting)\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/e1c1110154d04fa29154b3528b3b22d8 : SUCCESS in 1h 13m 42s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0bee003acdf44b70a24cd149269fd9ed : FAILURE in 1h 02m 59s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/2cda05d819134b2f91ad6000739f70ea : FAILURE in 2h 30m 52s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/f33a726829bf483f92e913e329ed81e8 : SUCCESS in 2h 47m 26s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/7be9b0d52d0d4728b22505d269edffd9 : FAILURE in 28m 58s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/dbee432a0bd74329ad309b0636b8814f : TIMED_OUT in 2h 33m 15s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"dc1b3dc9a0d4e5dd301092da773925e65550bb7b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-16 22:44:30.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"94968fd978df82f99314482a55e40c5e2ab5d0d4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-16 23:58:57.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/877130d863854706bd0b450afe2af629\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bddb985d37264996ad98c6a14f9cef49 : SUCCESS in 4m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f9163ec0f22343a7930c3f37d55251a6 : SUCCESS in 4m 51s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/91da5ac75a8c4ee38b03fc2c05c65e02 : SUCCESS in 5m 17s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/30f76ee51e474749925bd626f6430e3b : SUCCESS in 11m 31s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/fadc7d047f6b4021b0d376c3ee5c9e42 : SUCCESS in 5m 11s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/62b63415db5f467b88127b9e6f80a0f0 : SUCCESS in 6m 52s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/e19967c82761460885409e74e0fd5efc : SUCCESS in 1h 10m 41s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/e205540726764627842077362ff7bf0e : CANCELED\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/b288281e62744dc6ac49a5f1b8a8f40c : FAILURE in 1h 14m 07s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/836f789cc7834334be9270ae8d0a08cf : SUCCESS in 57m 25s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/357e430fab1e454cad03524a230a6985 : CANCELED","accounts_in_message":[],"_revision_number":1},{"id":"70caaebae507e1d9a61c9dbb1ed769e925338592","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2024-02-17 01:17:57.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e48a4d8545460af327ee9ca60d23cdb023f30840","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-17 03:59:27.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6e9a7b5695af416a90c063592490a039\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d7251083b20348928c701698e28d707a : SUCCESS in 8m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/527e4be6a99846f4bf7f5aa0d379706a : SUCCESS in 4m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/540169a099cb4b6a8cf09db7653a1765 : SUCCESS in 4m 33s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3a449c12badb443f9fca2bcc1b94dd3d : SUCCESS in 3m 53s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/96b375832c9a41a490da3a95314136d1 : SUCCESS in 11m 26s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/a6c69b48c6564cd48913c1212c3a72fb : SUCCESS in 4m 38s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/6cc8a1451d1940e8a1eb93bcbc3ae02c : SUCCESS in 6m 45s\n- octavia-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/726901e5ea3d4fb581a32760490f110d : SUCCESS in 10m 15s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/b3fe1e212c7e4103a3bf818161cd03e3 : SUCCESS in 3m 58s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/a83acba1821a4242b8d07bfb91d3afc2 : SUCCESS in 6m 33s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/d18596a26e834608aeccc1665b99971b : SUCCESS in 6m 04s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/feb2309bb9824eb282354d0053f37cdc : SUCCESS in 7m 21s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/aee1684bf8ca42a4afcdd01d2e7513e6 : SUCCESS in 1h 33m 44s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/831c6c35afb745e280726ec040c90fe0 : SUCCESS in 1h 38m 19s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/5db51faefcd646cf8b935649c4c7cc7a : SUCCESS in 57m 56s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/04be16e1aa434586a2087bee163fb25d : SUCCESS in 2h 02m 37s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/ca1e1523519f4ee9acb31dde680cfb31 : SUCCESS in 1h 16m 06s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/8c292fcff3bc45289153247320ba4c56 : SUCCESS in 43m 23s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/a4f3ee8d7bc842e08d88b7b09d294458 : SUCCESS in 49m 06s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/2de728a9f6844aad83df44279b3eb787 : SUCCESS in 47m 27s (non-voting)\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/073784a35ff64c399b37157ed2df214c : SUCCESS in 1h 04m 11s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f23f5306c7914b0587ab3170cc5ac9bb : SUCCESS in 34m 12s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/fdbaf174dbaf4913ab98d48c3923b4e2 : SUCCESS in 2h 28m 09s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/74ea0ec449a9441db2bbc894a863c6c6 : SUCCESS in 2h 08m 38s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/238108bc149b4f9c96d9ee9c39f55a91 : FAILURE in 29m 22s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/033517a3fee8488989824babff0efce4 : TIMED_OUT in 2h 33m 01s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e0f07249a54dfb278598d175dddd1b30e785956c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-17 04:00:03.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"a2a20bda11a5da8305e8e9cbe4a6959389d1b8ba","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-17 05:34:10.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/2a02136f79d74ebf9ecbc513ee9e17d8\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fc34b815238548b1aeab70fedfd5c83c : SUCCESS in 4m 34s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/777bee1dda4c409b8b889f09f35cbbd2 : SUCCESS in 4m 07s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/60c30262e38249128e3a2407a591d3ff : SUCCESS in 5m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/49b007c0760f48f496931476a77bbaac : SUCCESS in 9m 53s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/de2e1390e0664cd9afe8bc773144eb40 : SUCCESS in 3m 54s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/cacb0b2e175d4fa98a7e183432309447 : SUCCESS in 7m 04s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/5a765ace91044a8d9c2e7d0457bbafaf : SUCCESS in 1h 10m 08s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/f6cd0ea6b5394e78aeabb856b7db0758 : CANCELED\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/6353f1119db54f009792078cc9735beb : FAILURE in 1h 32m 38s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/caf3a20463b44033a4efb587e43f1846 : SUCCESS in 52m 20s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/848bcb76d76149c7ae42fc58bb732599 : SUCCESS in 59m 26s","accounts_in_message":[],"_revision_number":1},{"id":"2446efb2bab3d648f5e447200f00c169a91ffd16","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2024-02-25 00:11:30.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-2\n* Workflow+1\n","accounts_in_message":[],"_revision_number":2},{"id":"95c1702fb7d81c8613e53ea119c8684149923a73","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-25 02:55:05.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e478926e94c74a3a98637fddcd2c7c78\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d56412a5ba9f4f0f82c68f11911d4be9 : SUCCESS in 7m 47s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bb8ec531dbeb4d7eb78d61e0db792d09 : SUCCESS in 3m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/fb176dca69e7446fa670abf4331f9a9b : SUCCESS in 4m 03s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/65a428ea7e394033ac43d0571c197423 : SUCCESS in 4m 01s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/36442702f3f1459c974066194605abe6 : SUCCESS in 9m 35s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/8aa0c6b6b62b44f68d96783cf166acb4 : SUCCESS in 4m 51s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/c26dae41b669406584750ae0835d92b5 : SUCCESS in 7m 22s\n- octavia-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/71a35f2b4c214613b08f44752ddbdeba : SUCCESS in 11m 51s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/00c8a1c8640c484ab3397db8f189779b : SUCCESS in 3m 53s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/31e080d978ac406fa04e32be829c5ead : SUCCESS in 5m 46s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/29ea5c156af142efae15293c5843a6aa : SUCCESS in 8m 01s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/7d682810097b4e6f97a58797782e9489 : SUCCESS in 7m 58s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/ee4515d8da9044b5b26d0cc91f5e5da3 : SUCCESS in 50m 53s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/023d3a29368840f4900e322cef22e3ef : SUCCESS in 2h 11m 20s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/d69b91fa2a2a463f92c84b373f5ee9d7 : FAILURE in 1h 13m 34s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/f46517cfd8fb482eab979d6855257b1e : SUCCESS in 1h 48m 09s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/cd73d33fe706459b8ece43239975f2c7 : FAILURE in 1h 08m 56s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/4a6cf45257504d7a82e087f4efc1f801 : SUCCESS in 49m 39s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/3c3ae94c7ce7422ea0bf0237a9adadd1 : SUCCESS in 41m 36s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/f532048698644fadb893b7b162bfd771 : SUCCESS in 48m 51s (non-voting)\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/d0afde817dab4e698883fd14a8ecf5c8 : FAILURE in 1h 00m 58s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3ac1c87c798049df8b51f7026a31bf9f : SUCCESS in 40m 58s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/a9205119b271404a8c63f62d812b7349 : SUCCESS in 2h 29m 36s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/7c99aa72fa8b49a7b7bb5e7d162ee648 : SUCCESS in 2h 26m 41s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/1bcb53a5aa4d42d381f920dfb76b5252 : FAILURE in 21m 29s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/cd62f957d6d848e4b8f07c263acb2b25 : TIMED_OUT in 2h 35m 15s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"3c3f4fe35dc3bb2a83265c06f81785e42cdda394","author":{"_account_id":13995,"name":"Nate Johnston","email":"nate.johnston@redhat.com","username":"natejohnston"},"date":"2024-02-26 18:04:29.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"9e530ebde863f7a760896e9e8dc7a760fee31b49","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-26 20:59:18.000000000","message":"Patch Set 2:\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/f7f7c3f2b46f42f3841acda36f9b7e6e\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0c61d932e8c549a69be74776907d7b31 : SUCCESS in 12m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/321e8ea2132a48dfbeb7ab72c4be871b : FAILURE in 5m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a2b540cc024b4b08bb8d37fa0bfd5113 : SUCCESS in 4m 47s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/aa4b93d9472c4624856752cf86f03540 : SUCCESS in 4m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9402d4d3460649238e88cd36a16d5d49 : SUCCESS in 11m 29s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/cabcb906fd7c4bab9f10646e3e38fcdf : SUCCESS in 5m 45s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/9f14ec8ce1e04ea0b8deeb578a3c445c : SUCCESS in 8m 44s\n- octavia-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/a3e0594092384727a237374119d15806 : SUCCESS in 10m 05s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/fe0e55dbcae74436b5a61e4fdd89f89d : SUCCESS in 5m 32s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/cba32b6180fb43cfbc753e2f022c2c0d : SUCCESS in 6m 57s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/e83017a7b6ab4ea0a39a99b8df2e82f8 : SUCCESS in 7m 49s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/670c67e920d6418f91d5fab9647dee7d : SUCCESS in 7m 44s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/0ae316d1bdc64326b9f4e1898c1b01bf : FAILURE in 1h 22m 18s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/46ff5fea74f4410ebb127bdc18752ccb : SUCCESS in 1h 43m 07s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/f912c0b9805d4d628ab2463d7a5d3535 : SUCCESS in 1h 04m 16s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/41f9d0d3ab1b413dba5c90ed21587818 : SUCCESS in 1h 47m 48s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/a062969fce7a43439ee6044f14390350 : FAILURE in 1h 15m 41s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/e00581b5eaac4e78aec2e0e0453aa304 : SUCCESS in 40m 52s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/26f32d8b6cac40219fef3586b2c303c5 : SUCCESS in 52m 18s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/dcd2d2e491ce4baf9a41102fc9ac654b : SUCCESS in 37m 25s (non-voting)\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/79f5a38e1e5c4787b94fdfa14d95ac67 : SUCCESS in 59m 13s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a277f696045640b08e83aa27c09aa068 : SUCCESS in 44m 07s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/0320973602a84988a450cd4d8ff635c1 : SUCCESS in 2h 27m 09s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/9bcc63d60a9840369da5ab33521f8892 : SUCCESS in 2h 53m 13s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/40956cde7a6a40fba294d16006c83568 : FAILURE in 30m 15s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/5de64d0107274799acb6bdd712413a73 : SUCCESS in 2h 23m 06s (non-voting)\n\nWarning:\n  Comments left for invalid file octavia/amphorae/backends/agent/api_server/osutils.py\n  Comments left for invalid file octavia/common/tls_utils/cert_parser.py","accounts_in_message":[],"_revision_number":2},{"id":"e4f46828591702b899caedd64df630f58e56d7f4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2024-02-27 03:50:47.000000000","message":"Patch Set 3: Patch Set 2 was rebased\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"3667c7ef4861a2343bd6dc01a3a1ed2ab26364f1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-27 06:35:14.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ed2adb11019b494a931fc8ab57441e98\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/45d94dc6ed38426eaa60e917697a0628 : SUCCESS in 8m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a2a9bccc097346ce977d9dea4b14ca54 : SUCCESS in 4m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2c128b381c90407782f649e46b7689e7 : SUCCESS in 4m 23s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6fe17d51a70a44b48056d832c3de0219 : SUCCESS in 5m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/016e72bb84b84f6dac4f366a9f0caf80 : SUCCESS in 11m 41s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/f5897f80beab41158395334a9f7b175a : SUCCESS in 5m 09s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/c377e69c7f3f4f66acca6164a9bae964 : SUCCESS in 8m 32s\n- octavia-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/c8742fee965e4ace87ef24b62222eb4b : SUCCESS in 10m 50s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/8f481c6c8dc74b73aae17bddc82eb220 : SUCCESS in 4m 37s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/457e3f7f0f8243dab8d5b137b8d192e9 : SUCCESS in 8m 46s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/6d5d1e4222cc440f85f3d02f373e4188 : SUCCESS in 7m 50s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/c2505f731655412494820249a66a58f4 : SUCCESS in 6m 34s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/f94d985b25cf45a48f9f9eee5f1c0a39 : SUCCESS in 1h 22m 12s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/72194c53766146cdafc314e0e6d95d0f : SUCCESS in 1h 56m 39s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/725739b5e7174d69945979b617ccee14 : SUCCESS in 1h 20m 50s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/7d819ec53c5b4fbe800c5e5c6343930e : SUCCESS in 2h 01m 13s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/e66d406161f949fbb5cde5085b9cac40 : FAILURE in 1h 23m 05s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/fd59c755b0de447aa597765ca7fe5846 : SUCCESS in 48m 33s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/18f787800d9846f39d5debf7846df266 : SUCCESS in 58m 19s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/48b5439f3e4346e3965da893711f5eba : SUCCESS in 50m 22s (non-voting)\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/67310cd95256426c8c82700b3a3f6e90 : FAILURE in 12m 53s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8b086866dae54616820d5a62bf7418a6 : SUCCESS in 42m 32s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/e8af389a68484eeda73aefb18fcc253c : TIMED_OUT in 2h 34m 21s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/87927e9a8ae0443284e8be2004a7a4ff : SUCCESS in 2h 42m 25s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/c046f3666f4948d799cbc2b666a43539 : FAILURE in 30m 13s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/1cb10d796d174c77bdd051df03ac28a4 : FAILURE in 10m 27s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"bc70bd5f8e23d355b5a0026778186de24d5aa28a","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2024-02-27 07:08:50.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"24858a753098be9c14b640b5c6ae2d962e2ed3a3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-27 10:09:40.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/793adef9b8f045e287236614c103d1cb\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/96cf5b2b474a44a3b823f41ff1e52b7d : SUCCESS in 10m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/791f6bd7161d4fa6aecee990ecf69419 : SUCCESS in 4m 19s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/45357a8e2e7648a691db3905dbae9018 : SUCCESS in 4m 04s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6ec514f77f1443da95944ad4bf585057 : SUCCESS in 5m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/39a3778c3d8a4e5da12fcfa7eb49f052 : SUCCESS in 12m 18s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/0f585d2025a0425085bb079b3d8fad5e : SUCCESS in 4m 35s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/82ca883b9ddc4380866a11df5c920a11 : SUCCESS in 7m 27s\n- octavia-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/51197e24a7b44e47b6645cd322ba75f1 : SUCCESS in 11m 05s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/988796ae5fd54a0a890aa828275fb1c6 : SUCCESS in 4m 07s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/6f485ff4edf045518e020f1147c72a34 : SUCCESS in 8m 11s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/d0413447862643f5be1088ac95073015 : SUCCESS in 9m 06s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/54d93b99b6e04a24a47ff2a71f9a394f : SUCCESS in 8m 15s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/0330f34f038a4cb58cece60cbdd7ba78 : SUCCESS in 1h 46m 53s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/d16c1142f3624adf9138d904ab1c16d5 : SUCCESS in 1h 38m 01s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/f415e636da384a5596d4e1f6aec71f1c : SUCCESS in 56m 28s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/a5af60494f234af2bc15c689d18a574c : SUCCESS in 1h 40m 46s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/aedd22183c264bc88946137f48690f6f : SUCCESS in 1h 16m 40s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/54012d94e88c4366909a998bd4dea8d6 : SUCCESS in 50m 01s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/f1f7f498e155409e9d2fd88d2ee07204 : SUCCESS in 57m 22s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/be80df2379c949dba7667e2291d3e3a5 : SUCCESS in 54m 28s (non-voting)\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/9f9f9256f3d24e2a81cc1742c4f5438c : SUCCESS in 1h 09m 44s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c0f0148467bb4c648a645c1ce0335e04 : SUCCESS in 33m 54s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/79220d3a642c439b95574ec859074294 : SUCCESS in 2h 04m 40s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/444cc10d56a74a8b85535599a6f4b05d : FAILURE in 2h 58m 36s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/8ef332c7a354464a9056c17438c6d58c : FAILURE in 23m 07s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/20d38027a91348898f590f363b3d8b87 : FAILURE in 10m 38s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"53a862a6bb2ba2db6ac1d1034aaedd542954ed7b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-27 10:10:14.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"f5f7e175e4985560df1844e9853fad2fd94f80e6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-27 11:48:43.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/432313e44c5d47c985c53c9d4a583e4f\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f25103303b94499a8111691ad6fa8db8 : SUCCESS in 4m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/59233f62c9f848798aff22540de61b15 : SUCCESS in 4m 16s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0d16acbd91a54156a5a549871949e77a : SUCCESS in 4m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e7922bfe95704b568e6c0c625ff70d5d : SUCCESS in 10m 09s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/ce7ba98b838140cd9abf4bbc3b6e02ca : SUCCESS in 4m 38s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/e6a8dacdfdd646d38909c34306ca5f55 : SUCCESS in 6m 34s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/76d82265a3664bbc83259662d9e95127 : SUCCESS in 1h 32m 22s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/9714162e98a740ac8fc16e30ee1c2b60 : SUCCESS in 1h 36m 04s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/df3e616b7f41491185d4b4c810260f11 : SUCCESS in 57m 59s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/f1ee1fa6ec674b8fb4ff023460fb55c2 : SUCCESS in 44m 44s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/d960d8b560ad4233b6ad107c8d873ab2 : SUCCESS in 59m 37s","accounts_in_message":[],"_revision_number":3},{"id":"90738ea99805529e17f2adbcad7159cd3b97c2d4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-27 11:48:43.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"a0fe97c0f66bc2292bc19fed991da5e3f980315e","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-27 11:49:45.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a758aa3a43144354be2092027264f296\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/18521093384e4387854fb1273b2a5c78 : SUCCESS in 45s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"ffc9d83197f103ff8f9f7ca00801dd6d65e8ea4d","revisions":{"c31b3fad9fa97fc8e835283c2fed186c21c0ca1a":{"kind":"REWORK","_number":1,"created":"2023-10-19 22:54:36.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/63/898863/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/63/898863/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/63/898863/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/63/898863/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/63/898863/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/63/898863/1"}}},"commit":{"parents":[{"commit":"c91ffe0e497a7333e283c4b6e90282c23053e2c9","subject":"Merge \"Fix Amphora Configure API call\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c91ffe0e497a7333e283c4b6e90282c23053e2c9"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-10-19 22:42:41.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-10-19 22:52:23.000000000","tz":0},"subject":"Remove some unused code","message":"Remove some unused code\n\nThese methods and tasks are no longer used in the code, so this patch proposes\nto remove them.\n\nChange-Id: Ic7813d3a9073e4b3c1bc6a7839242df34a16d348\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c31b3fad9fa97fc8e835283c2fed186c21c0ca1a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c31b3fad9fa97fc8e835283c2fed186c21c0ca1a"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"c91ffe0e497a7333e283c4b6e90282c23053e2c9","is_merged_in_target_branch":true}],"branch":"refs/heads/master"},"cef0f1bb00a5f8f10700fad9d9c700f4b936c4bc":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-02-25 00:11:30.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/63/898863/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/63/898863/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/63/898863/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/63/898863/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/63/898863/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/63/898863/2"}}},"commit":{"parents":[{"commit":"46f4020e7031a1032d3fb69ab3964a060f4da506","subject":"Merge \"Bump hacking in pre commit config\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/46f4020e7031a1032d3fb69ab3964a060f4da506"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-10-19 22:42:41.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2024-02-25 00:11:30.000000000","tz":0},"subject":"Remove some unused code","message":"Remove some unused code\n\nThese methods and tasks are no longer used in the code, so this patch proposes\nto remove them.\n\nChange-Id: Ic7813d3a9073e4b3c1bc6a7839242df34a16d348\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/cef0f1bb00a5f8f10700fad9d9c700f4b936c4bc"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/cef0f1bb00a5f8f10700fad9d9c700f4b936c4bc"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"46f4020e7031a1032d3fb69ab3964a060f4da506","is_merged_in_target_branch":true}],"branch":"refs/heads/master","description":"Rebase"},"ffc9d83197f103ff8f9f7ca00801dd6d65e8ea4d":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2024-02-27 03:50:47.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/63/898863/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/63/898863/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/63/898863/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/63/898863/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/63/898863/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/63/898863/3"}}},"commit":{"parents":[{"commit":"30ce858564904a6217c3a470985a72011adbcb9c","subject":"Merge \"Fix pylint error\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/30ce858564904a6217c3a470985a72011adbcb9c"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-10-19 22:42:41.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2024-02-27 03:50:47.000000000","tz":0},"subject":"Remove some unused code","message":"Remove some unused code\n\nThese methods and tasks are no longer used in the code, so this patch proposes\nto remove them.\n\nChange-Id: Ic7813d3a9073e4b3c1bc6a7839242df34a16d348\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ffc9d83197f103ff8f9f7ca00801dd6d65e8ea4d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ffc9d83197f103ff8f9f7ca00801dd6d65e8ea4d"}]},"parents_data":[{"branch_name":"refs/heads/master","commit_id":"30ce858564904a6217c3a470985a72011adbcb9c","is_merged_in_target_branch":true}],"branch":"refs/heads/master","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
