)]}'
{"id":"openstack%2Foctavia~899267","triplet_id":"openstack%2Foctavia~master~Ifdcfcaea679d0ddb6b7b0e3c88586a61a5d627cc","project":"openstack/octavia","branch":"master","topic":"codespell","attention_set":{},"removed_from_attention_set":{"34429":{"account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"last_update":"2023-11-01 19:45:32.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ifdcfcaea679d0ddb6b7b0e3c88586a61a5d627cc","subject":"Run codespell on doc directory","status":"MERGED","created":"2023-10-25 13:30:51.000000000","updated":"2023-11-01 19:47:32.000000000","submitted":"2023-11-01 19:45:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"899267-codespell","meta_rev_id":"bf06a158ad04086d5f4c9d19eb85fe39d98a9896","_number":899267,"virtual_id_number":899267,"owner":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-11-01 19:45:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":2,"date":"2023-11-01 17:54:16.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-10-30 10:08:46.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":1,"date":"2023-11-01 17:54:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"disliked":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":-1,"date":"2023-10-30 10:08:46.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","value":-1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-10-25 13:37:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-11-01 17:54:16.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"efc4a35b48293d896eab56644a76171f8aa75d46","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-10-25 13:30:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"039ed427faabf4285927609da68ac59f5c3b4001","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-10-25 13:33:50.000000000","message":"Patch Set 1: Workflow-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"eb22e29adbf4670b8b2507b799f9157087f09d63","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-25 13:37:29.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/19463fe5846f47c2896bc7c2bbc41072\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b1065f7dbad44bdca69700af9c1d195d : FAILURE in 4m 27s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e5a24e3878ce43e3adb462126732ba86 : FAILURE in 5m 32s","accounts_in_message":[],"_revision_number":1},{"id":"8fe5f6d3c12dd2b1883f5d0f1ba70bdf0dc0e41a","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-10-30 09:41:10.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n* Workflow-1\n","accounts_in_message":[],"_revision_number":2},{"id":"cfc5bbfa82202925f9a65731402d22e2457e7c61","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-30 09:52:30.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/68a49f8db1f24dbd8a857e253fd83bce\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8e584e42772f480c98418798bd7807d0 : SUCCESS in 4m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ad23e6a3b67e4381b34ee6347e69dd0b : SUCCESS in 10m 28s","accounts_in_message":[],"_revision_number":2},{"id":"4b015c9344effc253d8aa5b6a12fc05bf53b1670","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-10-30 10:08:46.000000000","message":"Patch Set 2: Backport-Candidate-1 Code-Review+2\n\n(2 comments)\n\nThis change is ready for review.","accounts_in_message":[],"_revision_number":2},{"id":"9e7c97e1e191fa67ed8961bac630edc0e6592818","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-11-01 17:54:16.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"4a5aa8de23c134e6638be517e3287d04280d1c42","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-01 17:57:48.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"e78803cec1a718d242f3a3f9eec3518c6d119b55","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-01 19:45:32.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6e643a08cafe43fe881ee63e2377c382\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1f6547a7ce4444bab28dffcab141125b : SUCCESS in 4m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/90f651c2f5534d7193ce2f266a0daede : SUCCESS in 12m 16s","accounts_in_message":[],"_revision_number":2},{"id":"dd8aecd8e1b997c5af97bcad85055799ec5c65dd","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-01 19:45:32.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"bf06a158ad04086d5f4c9d19eb85fe39d98a9896","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-01 19:47:32.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/503057f12d524b6193f8565f6e01089d\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/696cf65c80144d2fb488e77e1949a1f1 : SUCCESS in 47s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"deeb4acca5dc21c5b2cad27635d20fde0bb7e48a","revisions":{"d6228e7f9a0e6e0527f3c3bfaf8d0480513ddfb2":{"kind":"REWORK","_number":1,"created":"2023-10-25 13:30:51.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/67/899267/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/67/899267/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/67/899267/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/67/899267/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/67/899267/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/67/899267/1"}}},"commit":{"parents":[{"commit":"7ab75bfd959be202421bad72e271393cc2256518","subject":"Run codespell on octavia package directory","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7ab75bfd959be202421bad72e271393cc2256518"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2023-10-25 13:29:57.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2023-10-25 13:29:57.000000000","tz":120},"subject":"Run codespell on doc directory","message":"Run codespell on doc directory\n\nResult of running:\n\n$ codespell -w octavia\n\nChange-Id: Ifdcfcaea679d0ddb6b7b0e3c88586a61a5d627cc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d6228e7f9a0e6e0527f3c3bfaf8d0480513ddfb2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d6228e7f9a0e6e0527f3c3bfaf8d0480513ddfb2"}]},"branch":"refs/heads/master"},"deeb4acca5dc21c5b2cad27635d20fde0bb7e48a":{"kind":"REWORK","_number":2,"created":"2023-10-30 09:41:10.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/67/899267/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/67/899267/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/67/899267/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/67/899267/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/67/899267/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/67/899267/2"}}},"commit":{"parents":[{"commit":"7ab75bfd959be202421bad72e271393cc2256518","subject":"Run codespell on octavia package directory","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7ab75bfd959be202421bad72e271393cc2256518"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2023-10-25 13:29:57.000000000","tz":120},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2023-10-30 09:40:46.000000000","tz":60},"subject":"Run codespell on doc directory","message":"Run codespell on doc directory\n\nResult of running:\n\n$ codespell -w octavia\n\nChange-Id: Ifdcfcaea679d0ddb6b7b0e3c88586a61a5d627cc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/deeb4acca5dc21c5b2cad27635d20fde0bb7e48a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/deeb4acca5dc21c5b2cad27635d20fde0bb7e48a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
