)]}'
{"id":"openstack%2Foctavia~899444","triplet_id":"openstack%2Foctavia~master~I24e1a7059c7ed1bdfe0f287c81148db912ce52a9","project":"openstack/octavia","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2023-11-01 19:45:29.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I24e1a7059c7ed1bdfe0f287c81148db912ce52a9","subject":"diskimage-create: Fix tox.ini not compatible with tox 4.x","status":"MERGED","created":"2023-10-27 03:19:39.000000000","updated":"2023-11-01 19:45:29.000000000","submitted":"2023-11-01 19:45:29.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"899444","meta_rev_id":"548a6656fe4069b0f42ad884fafdffaff17f7bc8","_number":899444,"virtual_id_number":899444,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-11-01 19:45:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":2,"date":"2023-11-01 17:52:44.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-10-27 03:30:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"value":2,"date":"2023-10-30 11:08:41.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":1,"date":"2023-11-01 17:52:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-10-27 03:30:59.000000000","updated_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"reviewer":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"state":"REVIEWER"},{"updated":"2023-10-27 06:25:20.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-10-30 11:08:41.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"},{"updated":"2023-11-01 17:52:44.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"2788a2afd6158eedf706e7f142d97b145910f4c9","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-10-27 03:19:39.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"02618e0335b6a49eb1617dd8274da2a46fe90b28","author":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"},"date":"2023-10-27 03:30:59.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"2d7ab52f293156c01fc048fed3db863ba97e938c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-27 06:25:20.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/98cd9b0288ff4731a3fab7570027c6eb\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b1f6e2341329494a8c5fdd5bcaaf10e6 : SUCCESS in 10m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/38bda5aa7f24442c96b536d03452b63b : SUCCESS in 4m 50s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d9fb74bb01034a1aa652776a374c9dac : SUCCESS in 5m 29s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/0c990e49ac9f4147a088c001bd343790 : SUCCESS in 6m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/65f9e723c4e3455e9755aa658158030b : SUCCESS in 11m 49s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/4c7403a36d90497ab0a636b141ccb5cf : SUCCESS in 5m 23s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/f4b5ea93cc354e3984da61a985f6f158 : SUCCESS in 9m 00s\n- octavia-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/25040109bd214ee8a82317cefa2df27c : SUCCESS in 9m 35s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/6834f2d254e245808c62693626fd4664 : SUCCESS in 5m 16s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/0647c412960747ee8ded5320ec11d3a9 : SUCCESS in 7m 06s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/fea4cf0fb5f7477d970ce00d1a12538c : SUCCESS in 7m 42s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/d5b266b40512451ba0d52e83f969f2d4 : SUCCESS in 8m 08s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/2e1f3ae685e24fbabd480670e64e3534 : SUCCESS in 1h 28m 55s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/4bd9d54ed3074c668951baa5c2696d5d : SUCCESS in 1h 37m 28s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/ec65e555472a4ec5a7f9fa7cde6c8732 : SUCCESS in 57m 42s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/8cf3892f67184607ae3afecd6724cb18 : SUCCESS in 1h 40m 07s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/fd0a2abb73124179b266b43d40bbfc99 : SUCCESS in 1h 11m 51s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/0ed0b04c9a124ab6852bf33a9dd66a42 : SUCCESS in 35m 39s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/d282ec86844b435bac50983c125171e1 : SUCCESS in 41m 26s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/2a1b70900b6246eb9c6ef719a45ee079 : SUCCESS in 39m 44s (non-voting)\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/38429fc01aa44769ae79dbf626a7152e : SUCCESS in 1h 03m 35s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0aa7a70a67ce44089fb0e2cae40d0193 : SUCCESS in 35m 01s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/ce550345b8af44699fcfc071cb8968d4 : SUCCESS in 2h 00m 26s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/8a1150c606bb4dab8d2e98b1e5600c1e : TIMED_OUT in 3h 03m 05s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/fbb54d1b09a24537bb37ccc434421594 : SUCCESS in 1h 58m 02s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/cb509a50b69b400aa6bd9d58b16c0403 : SUCCESS in 2h 28m 45s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2e98874f5b4696a1823b087c308fe1680c9d49ae","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-10-30 11:08:41.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"afd822b6b5123516e9bea7b69771e0cb2448c961","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-11-01 17:52:44.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"54c6c9cd473d2b17120073c4fcc873d0223dc3b4","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-01 17:53:25.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"54e076868115dfc8dedec802d65c84ee4f5cee36","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-01 19:45:29.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ce4fb502fc3642ceb8562fbd9b8c1599\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3bc8f029b2af4d68bbaecb354d7fa228 : SUCCESS in 5m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/569b047edad048158bd7cb8f6adbcc98 : SUCCESS in 4m 49s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/91b61abfa7084beba18659ba678a95b5 : SUCCESS in 5m 18s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a20bf26099d24a0ba74c6b94fd129176 : SUCCESS in 9m 22s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/e3054084e4454fabb9a56974d62b3001 : SUCCESS in 4m 00s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/41cfa792bf0d423895545618e6989b99 : SUCCESS in 7m 31s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/2cb80a2dd83a4b9c8d11801def19a31a : SUCCESS in 1h 06m 04s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/d93d235b4c4d47ce858110f82eed4bc5 : SUCCESS in 1h 37m 01s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/69f5cebf1d12409db5a5f9ee0e7a6476 : SUCCESS in 1h 01m 21s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/93665e90c1b64a57b15138c563b58b04 : SUCCESS in 35m 54s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/8028d3d883a249a7b406b743079e6f80 : SUCCESS in 52m 26s","accounts_in_message":[],"_revision_number":1},{"id":"548a6656fe4069b0f42ad884fafdffaff17f7bc8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-01 19:45:29.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"753c30f1e0eb5def8aa3ec62513f83fe88a28953","revisions":{"753c30f1e0eb5def8aa3ec62513f83fe88a28953":{"kind":"REWORK","_number":1,"created":"2023-10-27 03:19:39.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/44/899444/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/44/899444/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/44/899444/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/44/899444/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/44/899444/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/44/899444/1"}}},"commit":{"parents":[{"commit":"05616ab7cc342c20ad7ed486d67b488c437bb0b9","subject":"Merge \"Update master for stable/2023.2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/05616ab7cc342c20ad7ed486d67b488c437bb0b9"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-10-27 03:19:17.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-10-27 03:19:17.000000000","tz":540},"subject":"diskimage-create: Fix tox.ini not compatible with tox 4.x","message":"diskimage-create: Fix tox.ini not compatible with tox 4.x\n\nChange-Id: I24e1a7059c7ed1bdfe0f287c81148db912ce52a9\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/753c30f1e0eb5def8aa3ec62513f83fe88a28953"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/753c30f1e0eb5def8aa3ec62513f83fe88a28953"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":26285,"name":"wu.chunyang","email":"wchy1001@gmail.com","username":"wu.chunyang"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
