)]}'
{"id":"openstack%2Foctavia~899662","triplet_id":"openstack%2Foctavia~master~I070d27df8acc069836a8c7d4b0cf492c5b6a6168","project":"openstack/octavia","branch":"master","attention_set":{},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2023-11-30 00:05:21.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I070d27df8acc069836a8c7d4b0cf492c5b6a6168","subject":"Cleanup duplicate DB sessions for quotas","status":"MERGED","created":"2023-10-31 10:02:51.000000000","updated":"2023-11-30 00:06:47.000000000","submitted":"2023-11-30 00:05:21.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"899662","meta_rev_id":"11515ca961e93cd0b857a8022c91d30135170708","_number":899662,"virtual_id_number":899662,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-11-30 00:05:21.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"recommended":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"all":[{"value":1,"date":"2023-11-23 13:52:16.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":1,"date":"2023-11-23 08:35:48.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"value":1,"date":"2023-11-29 06:06:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-11-29 22:27:43.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":2,"date":"2023-11-14 16:16:31.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-11-29 22:27:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-10-31 12:45:57.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-11-14 16:16:31.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"},{"updated":"2023-11-22 16:15:06.000000000","updated_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"reviewer":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"state":"CC"},{"updated":"2023-11-23 08:35:48.000000000","updated_by":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"reviewer":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"state":"REVIEWER"},{"updated":"2023-11-23 13:52:16.000000000","updated_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"reviewer":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"state":"REVIEWER"},{"updated":"2023-11-29 06:06:21.000000000","updated_by":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"reviewer":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"state":"REVIEWER"},{"updated":"2023-11-29 22:27:43.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"1f4b5a13d11431137ac6137a53c62e611b3b1f8b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-10-31 10:02:51.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"063ced344e4109b51a7fe27c9340640621ac9763","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-31 12:45:57.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ee963116b8b94ad4909899db7fd633c5\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0221470704ef47f8a00fb155ec2f01e4 : FAILURE in 9m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e4cadf0d00b3495d9be06800d12bf538 : SUCCESS in 5m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/08870e7cd06e4b48b80cb60a68a2fcbb : SUCCESS in 4m 40s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/85243c6dde5e4cb398411aef87e513a6 : SUCCESS in 4m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/80bb451040ab4484a7c9406fa2875056 : SUCCESS in 12m 14s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/c74169f210ad40bfb18fe9e11d12d76a : SUCCESS in 5m 39s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/bbe70fcfd56243e2ad6699d14bd451e0 : FAILURE in 7m 57s\n- octavia-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/a407d22e75ef45a0aac3e53948a0efd2 : FAILURE in 10m 25s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/d7fcdcc4a16e4c0a916c3316b6957990 : SUCCESS in 4m 59s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/5cc6cbd7913e40a79c8879c605361ea7 : FAILURE in 6m 32s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/ece67a81bca84526bc1581ea0bb33d6b : FAILURE in 8m 18s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/78178041d0ee491fbd731c2a5a67ce45 : FAILURE in 7m 29s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/e5820024dddb49458c48c3f02b6566ae : SUCCESS in 1h 17m 40s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/8b96e3dffc8342cb97d224fcc531f413 : SUCCESS in 1h 37m 27s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/2b1653de3ea44d9892a1054064689b20 : SUCCESS in 1h 05m 16s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/29c47fe9d596446391bd008aa821d304 : SUCCESS in 1h 45m 47s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/39db6d0a962d4955b0f3b988d46da7fd : SUCCESS in 59m 27s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/41e4a329737949b5ab903a4a7e0a4a29 : SUCCESS in 47m 05s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/6762f019ae584b22a8a4ff7ab19e7a98 : SUCCESS in 41m 47s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/6891c39017c644cca393b2b85ace66d0 : SUCCESS in 37m 12s (non-voting)\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/4cc7d455de4446d4878cfccdb0eb90e9 : SUCCESS in 1h 06m 58s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8fa84c9539374fec8be171b3d3d9d9a5 : SUCCESS in 32m 21s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/7d00e0efc3b34d30ae7b56eda3f60089 : SUCCESS in 2h 30m 35s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/657e079401664d14a08a8216cdbc9953 : SUCCESS in 2h 17m 30s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/b8493a31a6c0445ba441b416dc96a61d : SUCCESS in 2h 27m 44s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/593cfa0326d94ee5bf8a75310ab8a32f : TIMED_OUT in 2h 33m 44s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"59f17b1e911539a67a3a6188350d097c7f791cd2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-10-31 13:20:22.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"9fda451d1dcd0600cfa2b636de53fc97f3cac4ac","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-10-31 17:43:10.000000000","message":"Patch Set 2: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/722a391caff1436a91e4956092f68a09\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/536ad93a7e0442aeb5c4c3bd48ebaa6d : FAILURE in 11m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e154ec59700c418483499411cfb8b2b3 : SUCCESS in 5m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/09381facc642414798fd31d3ea98bc7a : SUCCESS in 5m 26s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6581357c1fdc49299cdbedeaac2ccc3b : SUCCESS in 4m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cf9fc8cecfba4ccb8eb15fbf57ec419e : SUCCESS in 14m 38s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/000a367d58dd4f338f4fc383f5e68ae2 : SUCCESS in 5m 43s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/ee383434b182488c9b70ae43bc939b13 : FAILURE in 7m 40s\n- octavia-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/b0d8e26438d84e458937efa15f599ee8 : FAILURE in 11m 58s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/37714040cdda49d1a2b65c63cccc4650 : SUCCESS in 4m 55s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/c0d2315d35af448ea8e91e25eb1199e2 : FAILURE in 7m 13s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/9dba532c62d44773921df83374630fed : FAILURE in 8m 41s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/df015e860efd4a41ac50f250a9541d74 : FAILURE in 8m 18s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/72abc7c2c79e4cb5af7d208df01a2518 : SUCCESS in 1h 06m 27s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/a6bf5cff04324331b1d1d989f70d55a2 : SUCCESS in 2h 04m 54s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/db6883f948aa4ac9b0f0f0e8323b8cef : SUCCESS in 1h 24m 36s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/d8d13ab7a920478c83601834b9d0cd6d : SUCCESS in 2h 14m 31s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/3dc0bc11b48748c2826b846500f2d3b3 : FAILURE in 1h 18m 06s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/7d3f12cef8874da5a92281e83bb7f280 : SUCCESS in 47m 28s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/d23c45e3a96d4f599357cfe796f855fc : SUCCESS in 51m 48s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/7a51eeadf2da48518f19eb026542b7bd : SUCCESS in 59m 20s (non-voting)\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/6270fc1bc00c420e8f6a7e8a62a2583b : SUCCESS in 1h 08m 33s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7864cba74e1d41f5bab08ee85c495606 : SUCCESS in 45m 06s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/7d43c28fdc244cd8a1280430e9b058ed : TIMED_OUT in 2h 33m 31s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/2bd1fa0afa6b44b085ea0bbad20dfc45 : TIMED_OUT in 3h 04m 37s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/b0ce1d1273d644a189603fdcf0b460ce : SUCCESS in 1h 57m 20s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/44a4e7a5aa9c44b897e7fed084fd8b53 : TIMED_OUT in 2h 34m 34s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"3388a4b1ff0a49c3939f3206ff7a7f08768dae5a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-11-02 13:48:57.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"4ae4f14a0fb0d5c238ffd25b88c3116e7e9bc9f2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-02 16:28:55.000000000","message":"Patch Set 3: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/cbdac6ececcd4cadb0cc221b66260f80\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/46dbd337a12b4c88be9f2fe2b776b413 : SUCCESS in 10m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e2da6dab884f49108ebc04880ca0b9c2 : SUCCESS in 5m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/720c8dd22b83455d9b33bc4f4caf2789 : SUCCESS in 4m 46s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3ef5d022073e4c72b9e9a4c1e3a91a32 : SUCCESS in 4m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ef70667e8dc4adaab208b7e244d7eb0 : SUCCESS in 11m 54s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/fb6276ece9ae48bbbd2eb2c23bb1a228 : SUCCESS in 4m 35s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/8adc2cad1d8140a58f26199778502651 : SUCCESS in 8m 39s\n- octavia-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/3ed628bfec22458dbf1a2d1c8775bfde : SUCCESS in 10m 36s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/5e7d12e00e2e4039a08af22e86844403 : SUCCESS in 4m 36s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/25b63676f3db42e1b10850974c6b9a00 : SUCCESS in 7m 40s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/f2a9103cb59e41b8b5ddf3a85ac0b4d8 : SUCCESS in 7m 45s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/98357f4692954cd7908f3a97669d08fe : SUCCESS in 7m 52s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/babe24c3491d4de3b66f56c82d9734df : SUCCESS in 2h 12m 13s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/a0d6a3b25a284bbc8a9305a02a63d30a : FAILURE in 1h 44m 18s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/002a55e3e41a4961a5a220be2ca7f413 : SUCCESS in 1h 01m 53s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/515c77f4f3514885b20f54dbaf2644ca : SUCCESS in 2h 02m 06s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/c1306c9713ff4863bf9cf570bd9cdc56 : SUCCESS in 1h 09m 13s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/2cb7f8e2329b45f29bbfc70bdc837863 : SUCCESS in 40m 38s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/c6cf15f7d0dc450c9c10ccc809a6786b : SUCCESS in 42m 15s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/f7dad28bd679452583fabfed67080a3b : SUCCESS in 41m 52s (non-voting)\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/251eeb48f8a0493fa2f63d9b0614833b : SUCCESS in 1h 02m 46s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/91e2ff4c944a445ca7497be75545098e : SUCCESS in 55m 16s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/459ab79d172c4619b7fc34b95ada5b47 : SUCCESS in 2h 20m 49s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/7c7d89a22f824f6fbb2e39d118d4b599 : SUCCESS in 2h 21m 39s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/f1fd5a8dde864fa4b1dd2a26b4348624 : TIMED_OUT in 2h 32m 44s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/b9a0d2dd0ee740a38c5136fd4e02cc89 : TIMED_OUT in 2h 33m 25s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"f62dc212c29d46e0d1b2acdc8f23c9548db3d084","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-11-10 07:08:18.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"a124209e7d2236ac4a862486e30c38d469621db5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-10 10:02:58.000000000","message":"Patch Set 3:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c5db22a8538e45a09ff2d08dc95e0065\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/31c7065404504353a0761f754c7c0fbd : SUCCESS in 12m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4a1dc7261f314b34b4e734414c441346 : SUCCESS in 7m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d69cca5c95224bdeada2e78e5ab70128 : SUCCESS in 4m 14s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a4b1c15b08334e369df8fd4fc032e820 : SUCCESS in 4m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/04403ab515654ed3baabd5438960cee1 : SUCCESS in 9m 02s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/71f202a6a46d4031b3166d63924560ba : SUCCESS in 5m 26s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/aeceb2f4f61e4b039159ec30e3dfe992 : SUCCESS in 8m 18s\n- octavia-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/5751a03b35ca4f00bbda3f1a17675bd9 : FAILURE in 6m 52s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/e2959d5d52a845d7b8b0f320922380ed : SUCCESS in 4m 14s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/04d75754055546f5916a12a86776d9bf : SUCCESS in 7m 16s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/25d7fe02be44417db942fe441adce39a : SUCCESS in 7m 57s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/e2f05034eed54aa7b8b724e7404becb0 : SUCCESS in 7m 15s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/98503fa1bf474d25a08660600be7c080 : SUCCESS in 1h 28m 51s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/f46a97c8b5614631af635235b141807e : SUCCESS in 1h 35m 45s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/719f78fb527d4eb59a4bcaf6e210536f : SUCCESS in 54m 11s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/41cd331e5d6f4287a152d248b2db002b : SUCCESS in 1h 37m 39s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/cadcb9cabb8c44f88d0b04f5ba39cec9 : SUCCESS in 1h 09m 46s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/80467b2fbb864a2cadde5025ed7de1ef : SUCCESS in 37m 07s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/2dfaef774efb40fcbcc431c601656fb3 : SUCCESS in 56m 53s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/79cf084222af41e69767de3a2086e090 : SUCCESS in 59m 22s (non-voting)\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/c0f62951267f45ffa92fb0369ed40e5d : SUCCESS in 1h 12m 08s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9ced6923a02542d5a6c3c3451ecbc4b0 : SUCCESS in 36m 01s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/ee5a0ebc015e4bde88799e280f6dfe22 : TIMED_OUT in 2h 34m 35s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/e9a5a8c499644080afa02201d57135dd : SUCCESS in 2h 47m 01s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/3cef639b20f641509f5ea09b6d512074 : SUCCESS in 2h 19m 29s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/5f37f2fb7919427c8aef996d6872e902 : SUCCESS in 2h 15m 23s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"e6bcd1307281493f0d211dce07ff91bbee370e14","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-11-13 10:40:40.000000000","message":"Uploaded patch set 4.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":4},{"id":"fffdcfd62bd42f9f104b4813f31d35d9b2c08fad","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-13 13:39:41.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/53bbdb98dbf44b8ebe45ac6eabfc592d\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8e0f7f7e61ab4b79a25e122151872b05 : SUCCESS in 11m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2940993b54b94ff79cfdad9264ff00a6 : SUCCESS in 4m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/47b3aa3180dd4982aff625e7bc8bf5bc : SUCCESS in 4m 48s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/56a4cdbab2244690bf618508ff9e982c : SUCCESS in 5m 22s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/01b27eef264b44f1a6763928d5fdc681 : SUCCESS in 11m 19s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/0b45c71b184b4dab835a42263290e908 : SUCCESS in 4m 54s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/39f77a87a5a044a5bedb40fc7b00d440 : SUCCESS in 7m 54s\n- octavia-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/ba84b0bffe954ad58fca9e3ea637cd58 : SUCCESS in 11m 52s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/c892315e48ba471f8c9f648f5a1055a1 : SUCCESS in 4m 26s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/283990d4e8a546deb059c58dffa72114 : SUCCESS in 7m 48s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/eb04a143fe444335b11dc35d4d1da3ca : SUCCESS in 7m 07s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/825216b1346342e180b10aa6fe6df091 : SUCCESS in 8m 26s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/89450c1d6a474f629308c6f84c671e91 : SUCCESS in 1h 24m 44s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/91d78ac80cdf40ad8bae8277dfe51009 : SUCCESS in 1h 34m 39s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/5ffbd9ac395e4a9391ca1da083f3d44d : SUCCESS in 1h 01m 29s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/ac45c26b52ab4a1cb0383a3999049dcc : SUCCESS in 1h 37m 28s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/f5a1027f934f41ccb8f7de5e66220934 : SUCCESS in 1h 17m 35s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/2f17feaa7a9e492e921942b3a6455670 : SUCCESS in 37m 04s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/d7ecf20efb2b4d258c8f16d228cb784b : SUCCESS in 41m 11s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/4148d5cb82184d39957e23af579ebcaa : SUCCESS in 56m 25s (non-voting)\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/47e7df8db1884194a2a6de1cb100a98a : SUCCESS in 1h 14m 54s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2ef6273b385848e3a53195b97e8febc1 : SUCCESS in 39m 40s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/43346cb145e14bc8adc74881b3e6e9ba : TIMED_OUT in 2h 33m 52s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/98d5d0f57d5441aea652adb6b41e91e2 : SUCCESS in 2h 54m 03s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/2415ba58f9004288932acb299186a805 : TIMED_OUT in 2h 33m 57s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/57a36fbb7c3b4b77861613b53036db83 : TIMED_OUT in 2h 33m 27s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"4a9deaee867bf55ae1875a3c2b345a6be80f4ae3","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-11-14 16:16:31.000000000","message":"Patch Set 4: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"b388a59dd270756bc87570efe57b7af52069b816","author":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"date":"2023-11-23 08:35:48.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"d0c07e235f2cd5d1ecbba1c2ccb815327275cc50","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2023-11-23 13:52:16.000000000","message":"Patch Set 4: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"3da9cd98e915678b064f27973adeac902ac84a88","author":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"date":"2023-11-29 06:06:21.000000000","message":"Patch Set 4: Code-Review+1","accounts_in_message":[],"_revision_number":4},{"id":"b217d381f6c9679436f0ae3e91c4b73abf85b4ef","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-11-29 22:27:43.000000000","message":"Patch Set 4: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":4},{"id":"603061616a8be32dc22eac80cf717ce7124e41bb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-29 22:28:20.000000000","message":"Patch Set 4: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":4},{"id":"f1359e58debb030ec39ab2e203914a088da49e5c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-30 00:05:21.000000000","message":"Patch Set 4: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/77e16ab86ffe4c398620898d3c457e99\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a4f51971c62541f2bf4d1cade9e9c3bb : SUCCESS in 5m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9ffea1bd12644f9ca9044f8974d1caaa : SUCCESS in 4m 13s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a296c480d1a045ae86b7d12a1a9b9c20 : SUCCESS in 5m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1913c7dea08f4a21b506ec81e4c888da : SUCCESS in 11m 07s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/11b15b302aef4e06bd84bab4b25ec21a : SUCCESS in 3m 41s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/44574c82e846485d87b81bf2e3ad31c8 : SUCCESS in 8m 10s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/b9466143213d4721aa5c18720f5ca277 : SUCCESS in 1h 19m 44s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/f56c3ff443014538afc749d2cec6d79d : SUCCESS in 1h 36m 37s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/ffc439d2e8d247f78674eb8d19b76174 : SUCCESS in 1h 18m 42s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/0f6db4776f624e48b728644721155e2d : SUCCESS in 39m 40s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/e92d993f52474aa7839d08a5ee39cfd6 : SUCCESS in 38m 03s","accounts_in_message":[],"_revision_number":4},{"id":"24a196df19fb57c2cb03aca66ba14f5a3a392c77","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-30 00:05:21.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":4},{"id":"11515ca961e93cd0b857a8022c91d30135170708","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-30 00:06:47.000000000","message":"Patch Set 4:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8ade23a8d15c4e519df6ad560f2eba02\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4494c6133f024c01991186d70c51a3f3 : SUCCESS in 1m 06s","accounts_in_message":[],"_revision_number":4}],"current_revision_number":4,"current_revision":"2a1c4709a0f4a14510ef48358f4063034c28027a","revisions":{"580479d4548444aa2a769a0b4bd30c5c632c08a1":{"kind":"REWORK","_number":1,"created":"2023-10-31 10:02:51.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/62/899662/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/62/899662/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/899662/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/899662/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/899662/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/62/899662/1"}}},"commit":{"parents":[{"commit":"c91ffe0e497a7333e283c4b6e90282c23053e2c9","subject":"Merge \"Fix Amphora Configure API call\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c91ffe0e497a7333e283c4b6e90282c23053e2c9"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-10-31 09:59:22.000000000","tz":-240},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-10-31 09:59:22.000000000","tz":-240},"subject":"WIP Cleanup duplicate DB sessions for quotas","message":"WIP Cleanup duplicate DB sessions for quotas\n\nChange-Id: I070d27df8acc069836a8c7d4b0cf492c5b6a6168\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/580479d4548444aa2a769a0b4bd30c5c632c08a1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/580479d4548444aa2a769a0b4bd30c5c632c08a1"}]},"branch":"refs/heads/master"},"ff106b5781a692a9db83250706c530637da8490a":{"kind":"REWORK","_number":2,"created":"2023-10-31 13:20:22.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/62/899662/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/62/899662/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/899662/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/899662/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/899662/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/62/899662/2"}}},"commit":{"parents":[{"commit":"c91ffe0e497a7333e283c4b6e90282c23053e2c9","subject":"Merge \"Fix Amphora Configure API call\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c91ffe0e497a7333e283c4b6e90282c23053e2c9"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-10-31 09:59:22.000000000","tz":-240},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-10-31 13:17:42.000000000","tz":-240},"subject":"WIP Cleanup duplicate DB sessions for quotas","message":"WIP Cleanup duplicate DB sessions for quotas\n\nChange-Id: I070d27df8acc069836a8c7d4b0cf492c5b6a6168\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ff106b5781a692a9db83250706c530637da8490a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ff106b5781a692a9db83250706c530637da8490a"}]},"branch":"refs/heads/master"},"d69df535411faeea29f32b18c7e883da33d88da8":{"kind":"REWORK","_number":3,"created":"2023-11-02 13:48:57.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/62/899662/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/62/899662/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/899662/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/899662/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/899662/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/62/899662/3"}}},"commit":{"parents":[{"commit":"c91ffe0e497a7333e283c4b6e90282c23053e2c9","subject":"Merge \"Fix Amphora Configure API call\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c91ffe0e497a7333e283c4b6e90282c23053e2c9"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-10-31 09:59:22.000000000","tz":-240},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-11-02 13:48:52.000000000","tz":60},"subject":"WIP Cleanup duplicate DB sessions for quotas","message":"WIP Cleanup duplicate DB sessions for quotas\n\nChange-Id: I070d27df8acc069836a8c7d4b0cf492c5b6a6168\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d69df535411faeea29f32b18c7e883da33d88da8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d69df535411faeea29f32b18c7e883da33d88da8"}]},"branch":"refs/heads/master"},"2a1c4709a0f4a14510ef48358f4063034c28027a":{"kind":"TRIVIAL_REBASE_WITH_MESSAGE_UPDATE","_number":4,"created":"2023-11-13 10:40:40.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/62/899662/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/62/899662/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/899662/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/899662/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/62/899662/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/62/899662/4"}}},"commit":{"parents":[{"commit":"7310986de9bf68ed86de90de0501a1bc46945526","subject":"Add Noop Certificate Manager","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7310986de9bf68ed86de90de0501a1bc46945526"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-10-31 09:59:22.000000000","tz":-240},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-11-13 10:39:17.000000000","tz":60},"subject":"Cleanup duplicate DB sessions for quotas","message":"Cleanup duplicate DB sessions for quotas\n\nSome functions in Octavia receive 2 DB session parameter, one is a\nsession in \"autocommit\" mode, the other without \"autocommit\" mode.\n\nNow that the autocommit feature is removed from sqlalchemy, we can clean\nup some leftovers of this migration, 2 session objects were passed to\nthese functions (and sometime the same object), only one is needed.\n\nRelated-Bug: #2038798\nChange-Id: I070d27df8acc069836a8c7d4b0cf492c5b6a6168\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2a1c4709a0f4a14510ef48358f4063034c28027a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2a1c4709a0f4a14510ef48358f4063034c28027a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
