)]}'
{"id":"openstack%2Foctavia~900142","triplet_id":"openstack%2Foctavia~master~Ic81e98c54c4bce100e3f44ff1a2fe6ce7b7f4256","project":"openstack/octavia","branch":"master","topic":"bug/2042787","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2023-12-14 12:56:28.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2023-12-19 15:03:19.000000000","reason":"Change was submitted"},"34429":{"account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"last_update":"2023-12-14 11:59:02.000000000","reason":"\u003cGERRIT_ACCOUNT_34429\u003e replied on the change","reason_account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}}},"hashtags":[],"change_id":"Ic81e98c54c4bce100e3f44ff1a2fe6ce7b7f4256","subject":"Use cryptography to load PKCS12 certificates","status":"MERGED","created":"2023-11-06 01:48:11.000000000","updated":"2023-12-19 15:04:32.000000000","submitted":"2023-12-19 15:03:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":4,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"900142-bug/2042787","meta_rev_id":"08a80f1c8c42ed740784af8cbc618de509534651","_number":900142,"virtual_id_number":900142,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-12-19 15:03:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-11-08 00:43:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":2,"date":"2023-12-14 11:59:02.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":1,"date":"2023-12-14 11:59:02.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"disliked":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":-1,"date":"2023-12-14 11:59:02.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","value":-1,"default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-11-06 03:17:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-11-08 00:43:24.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2023-12-14 10:36:50.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"},{"updated":"2023-12-14 10:36:50.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"}],"messages":[{"id":"39b68f3b617a62314af1912fc26101449bc32642","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-11-06 01:48:11.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"046eac4d62669fb65eb616a5d927ef1e601480eb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-11-06 01:49:06.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"3caab2b2b84fa55efa35795566fb3b989a7c5dd3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-06 03:17:37.000000000","message":"Patch Set 2: Verified-1\n\n(4 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0b11c526855349cd807c64efac418aab\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/d5ede0d761114d619b4bab7806f57b55 : SUCCESS in 1m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0a8bb3aa26c2410db5264f877e05823a : FAILURE in 4m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/feb44e263a124e48bf08d11be6e1c5c6 : FAILURE in 4m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/843ab63106174633acb6428ef1784b43 : FAILURE in 3m 57s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8d882e496024463fbee88034b2185239 : FAILURE in 4m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bbbf15d1a5e24948af9b1ca5f707e547 : FAILURE in 5m 42s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/bfb194ba49324682a84fb62947e6274e : FAILURE in 5m 03s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/50a14885ae9542b5b407d906ce0627dc : FAILURE in 6m 34s\n- octavia-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/c4ca5be2bb724798ac056e2da36788c9 : FAILURE in 7m 11s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/558ca8809f264370932245fe83e77d8f : FAILURE in 4m 37s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/32600fe0dea549988ca2317e4c5989f2 : FAILURE in 4m 19s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/a76c22804639466dafda59b88a946fec : FAILURE in 5m 01s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/35c14992b7ba477a8f15c7df2aff60c5 : FAILURE in 4m 51s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/2df9a3559ccf47b9b7371a996c08ae91 : SUCCESS in 1h 22m 50s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/33b06ff9d2964fa7851a41c7ff3d2323 : FAILURE in 25m 02s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/e986e0149206497a80a0310cbcc693fc : FAILURE in 24m 57s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/fd595b2320fc461281fa087d91b5dd83 : FAILURE in 24m 41s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/2ec8f589200444f7b85d3a418e11e869 : FAILURE in 25m 26s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/dbdff28379cc459082742e6ffdaf8bb7 : FAILURE in 29m 37s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/4c6bd2cc42df4a55ad1e7a97a8ada472 : FAILURE in 45m 03s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/560b7c86bea541af8c4e3edb073894f5 : FAILURE in 38m 01s (non-voting)\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/1bb223cc97564e4981233bf466af6b53 : FAILURE in 48m 17s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1d607e5193aa41269f63c6c3070d855e : FAILURE in 26m 01s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/def545c163894322acbcc00633e234b6 : FAILURE in 24m 42s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/dd560ab091714e74a9947051935b21b5 : FAILURE in 39m 17s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/d9c18d6e05c64b9e90dd93c3638e6cae : FAILURE in 26m 25s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/4e7d6a11b6574c4caa5b5667ef9a0357 : FAILURE in 24m 19s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"93d628654dc5c260ecd76dacede0addfc0dfbfa6","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-11-06 03:58:35.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":3},{"id":"a3e6aa5e641c4ea8f8dc88abfc255c80b8393de2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-11-06 03:59:43.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"643e85c7c57f3e2582a0b77df4bc35e8da387002","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-11-06 04:01:28.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"74bc8dc8617f24d651af01056ec635e7dc6813a8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-11-06 04:13:13.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"23c9496ce9b9116b31c73608a568522424d99a52","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-06 06:50:05.000000000","message":"Patch Set 6: Verified-1\n\n(2 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/bb4f501f0e5e44faba6e9bc54ac2706e\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/52491c9a89334cf484e76effc55cae6c : SUCCESS in 2m 54s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/d9b7809d585140ba8e0e19376a2c8b9b : FAILURE in 5m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ebabad0d1cff43a98f6c96ed47b81869 : FAILURE in 5m 08s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/22580a566b4e430c96241b784f476143 : FAILURE in 4m 46s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7739d9958024433bb2c5b7bf4cffc398 : FAILURE in 5m 40s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/017c1dd8400148d1bc79b3cf8b93bd0f : SUCCESS in 9m 51s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/6a24be24322e448a8bed39155fd9c587 : FAILURE in 6m 29s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/e2c001a0baf64374ad872b2d0a5d0b65 : SUCCESS in 6m 30s\n- octavia-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/946847fefbbe44eebf8430fb0307170e : SUCCESS in 10m 07s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/500f0803134d4bdead25859a88fd6428 : SUCCESS in 4m 13s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/f11ee4d5d90c4458a4c499b5019a85b5 : SUCCESS in 8m 28s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/a5e79d945d244ab0858b75cc90fbf4c9 : SUCCESS in 7m 26s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/08b663a23a7643a780bdc152f6e6a7f8 : SUCCESS in 6m 11s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/abc83041619040f994f667f7c9612b6d : SUCCESS in 56m 38s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/968e3645d92944b69c4821e1b32c4d4d : SUCCESS in 1h 49m 25s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/4011fc54cb7641c69ffc6a81ed7f56bb : SUCCESS in 1h 00m 56s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/f97f1dd9724e40ef84571d8fc165a635 : SUCCESS in 1h 50m 11s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/1a3b18da69f348f7897297d271a7c2de : SUCCESS in 1h 08m 57s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/82642761d76b4de896948049223c94ff : FAILURE in 34m 08s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/e89c7b6a0d494adb85d5b85be66a5e6a : SUCCESS in 51m 05s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/61c12aef280b44129b62c9d70e432672 : SUCCESS in 51m 43s (non-voting)\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/9092080687784772950a243ece0811fb : FAILURE in 1h 02m 58s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7dda9f4ad0b14268b86c94cfb1750991 : SUCCESS in 41m 55s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/5cc494de9280488d8a2253a40c84e110 : TIMED_OUT in 2h 32m 53s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/641a50a60fb24db09e0c265d26257f96 : SUCCESS in 2h 23m 15s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/f62802c23b054071b39ac2090bf50699 : SUCCESS in 2h 35m 00s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/905fec03afa44b469e5cb7e8a5914c2c : TIMED_OUT in 2h 34m 20s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"c1abf3c87e54fd80a2ace3a0927a4041a5f718ba","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-11-06 08:15:58.000000000","message":"Uploaded patch set 7.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":7},{"id":"7bd124bc6fab66863a4a8eb451c98e646b49a80a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-06 11:00:54.000000000","message":"Patch Set 7: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1232fe9c848645718407dfa6e5253f46\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/db348d41f8e54bf586f86542616437c6 : SUCCESS in 2m 07s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/311d61d0a6584e90930a80c638cb5c0d : FAILURE in 4m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0b8e6120415140d190b65314f1c8ac82 : SUCCESS in 4m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/98d48af0d9ef4715b2ee0c88e449b589 : FAILURE in 4m 41s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/cb08ddd302e14d65bab9da427e091bbc : FAILURE in 5m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f94530c574a94b3c925d6b994c6fbe34 : SUCCESS in 10m 51s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/b4c81c2e915747e7a6ec62744f62d8d6 : FAILURE in 5m 15s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/68dd470ca2c84e51bf86b04423aaf081 : SUCCESS in 8m 51s\n- octavia-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/c14f05d16106456da2dee079db4c199b : SUCCESS in 10m 25s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/0ef135add99a41eab4eea3792ad23832 : SUCCESS in 4m 14s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/30b9097e44184d51a5f86aaeca077fbe : SUCCESS in 8m 07s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/e659773190d84efebece6a42cf2df69e : SUCCESS in 5m 41s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/59f650b6b035414aa0b0859c98fac91b : SUCCESS in 6m 49s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/f2e8223cea5645d2b9a4d006b4899285 : SUCCESS in 1h 08m 49s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/891244ab56ae4633b627a740eb376893 : SUCCESS in 2h 00m 06s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/020b42b5f1e34059976e5fa01672a985 : SUCCESS in 1h 03m 22s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/3e352904095e48319c5b949bfccb81b6 : SUCCESS in 1h 47m 38s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/0dc557e8f05649719db59e4ab80bd96e : SUCCESS in 1h 04m 56s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/4693cd6f3679429ab3e282bd348e098d : FAILURE in 38m 44s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/623c9f12014c4376b100a10f586f6fb6 : SUCCESS in 38m 44s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/327c3b311ea14c4b9befc1a0ef7eb3ea : SUCCESS in 48m 20s (non-voting)\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/ed7e45dfd03a435e93603fa3bdfd8020 : FAILURE in 1h 10m 23s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/cb154e4c695b44938a95998299ce6b7a : SUCCESS in 39m 51s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/96947f4798b048f199bd23fecd09035c : SUCCESS in 1h 53m 39s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/aa6af4348a6b4833a7409ab94dfd740d : SUCCESS in 2h 43m 40s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/ce180d2c0a0f4ae69c7f14e494f7b004 : SUCCESS in 1h 53m 00s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/b6929f7c33a3473db01c774505062050 : SUCCESS in 2h 26m 26s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"049750f266077eabf71715637ca6d79baa6848e3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-11-06 12:37:04.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":8},{"id":"5be3928f5a9c33eb310e6647d3e391d009b7108d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-11-06 12:51:36.000000000","message":"Uploaded patch set 9.","accounts_in_message":[],"_revision_number":9},{"id":"dd8f4a30314b0ad4fb70415d306265292e058cc4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-11-06 12:52:27.000000000","message":"Uploaded patch set 10.","accounts_in_message":[],"_revision_number":10},{"id":"59ebb4997d919e519d4cc4058319fc4b021d5f68","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-11-06 12:58:14.000000000","message":"Uploaded patch set 11.","accounts_in_message":[],"_revision_number":11},{"id":"4c182f6cf98d14cbade62b405951dd9848ffa46e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-06 15:50:56.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e208dc27661b4a40b1c73f8f0d744177\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/14d6b6797789417f87b2dcbdbcdc3c65 : SUCCESS in 2m 27s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c9c471527aa849a28226e2f6ce5b7858 : SUCCESS in 11m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b06b5dc9d21d4901be1594bcf56eb160 : SUCCESS in 4m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0d52db1da0aa46e29ee249edbc8b799e : SUCCESS in 5m 11s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d781555580254b099a885e1e4b4d17ef : SUCCESS in 4m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b6689e4c1d464140a1c2984ae43222a3 : SUCCESS in 10m 51s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/96cb08b6d26b410eb6d625ceaf5fdf02 : SUCCESS in 5m 26s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/8457c16fd46a445fb0e8ea5197e84463 : SUCCESS in 7m 27s\n- octavia-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/d4562e54b0d248fa96f9fd5fbd04acac : SUCCESS in 12m 27s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/00c891a102dd4cb6b73b0932250d0ba6 : SUCCESS in 4m 59s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/d37383fb3ce14efbbbcba73697171745 : SUCCESS in 8m 25s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/9ca788de8bc04b528637d8a01c2045e1 : SUCCESS in 7m 42s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/ab9bd28a7723472aa055f62588575aa8 : SUCCESS in 6m 54s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/661957c98c3e40d48e5cbac75cfb41c5 : SUCCESS in 1h 23m 39s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/98315ed03d644d878f6a0ef8f2de4630 : SUCCESS in 1h 35m 34s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/642199aed5d745ed86e3a9d42f3c1c35 : SUCCESS in 1h 20m 18s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/7523aafeaf4b4532bcbb62547b4f3e91 : SUCCESS in 1h 44m 30s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/a07207aff0574c08a753b0f87ac0e77d : SUCCESS in 58m 54s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/699dfc1520cc44e4a91250e87994893a : SUCCESS in 45m 10s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/356e0cd1916645e5aacc4d5e719d93d1 : SUCCESS in 49m 30s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/d0ab40be9ec84077a42b1dade4cebd69 : SUCCESS in 45m 11s (non-voting)\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/4eb2a8f02ff341e797414faa98bc6002 : SUCCESS in 59m 33s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2c3760d8edfd4bc5b7b83ac5249b8c66 : SUCCESS in 37m 07s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/dad9e9975f3e45d6aba359c08d0c05b3 : SUCCESS in 2h 21m 22s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/6d74142e15564c3eaf754c249ea1f31c : FAILURE in 2h 40m 36s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/4b6573d315ae4c16aa5a93811ee95727 : TIMED_OUT in 2h 33m 50s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/be13ea819367447f8c4788af0beea2ef : SUCCESS in 2h 29m 15s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"e73f4d2c123317af30c1398092918ec131f8c5f8","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-11-08 00:43:24.000000000","message":"Patch Set 11: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"e7ff8c84fb50b9dda3e7c51becfd71dc6f3b3ea3","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-12-14 11:59:02.000000000","message":"Patch Set 11: Backport-Candidate-1 Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"8c9e55ab2d60f91e470d159a23126a9dbd63cdb1","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-14 11:59:38.000000000","message":"Patch Set 11: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":11},{"id":"62b7349666d3fa39d35038ed3a0423ed867acbfc","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-14 12:05:57.000000000","message":"Patch Set 11: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/811987c3ebc94426b10d587d345a21a2\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/18ee688142334d109f9c48ee7cec6719 : SUCCESS in 2m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6100d9f9698947f2ab57026aa445c186 : FAILURE in 4m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a6e01ef660204eeea262b1024e3756e5 : CANCELED\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a2369ec2d6f944988518d1358a79f534 : SUCCESS in 5m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/05461b19fbbf451191005c549d26704a : CANCELED\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/04b04e11cbba4fe8af059e37645db61e : CANCELED\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/3f0b92f4c5384f1595c359769c65656d : CANCELED\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/e08cf66a5ff342ae9da5602079baa0dc : CANCELED\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/50f91906d1bf41149f2d524d9275e11e : CANCELED\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/98c802ff884a48bf9e6c62c9f926c9f7 : CANCELED\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/f89937d229184d219934ebbb45e925cc : CANCELED\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/783e3c5d8b2f4c4ba7f1707707aa6b1d : CANCELED","accounts_in_message":[],"_revision_number":11},{"id":"30e4a21c7520e88a29335bb759222e12c1f54a8c","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-12-14 12:56:28.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"2c36fa8ad802f3c24365f2c97ad423bdedbcbac3","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-12-19 10:15:55.000000000","message":"Patch Set 11:\n\n(1 comment)","accounts_in_message":[],"_revision_number":11},{"id":"2f3247bf9c11721aec718538bdbd266c3ca60ad5","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-19 12:52:46.000000000","message":"Patch Set 11: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ca2c028257e942b5b81419d414f587ab\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/a0bda1b6989443378fa4c01b93e249df : SUCCESS in 2m 17s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1659066fa3ea421b8871d66cd9bdcade : SUCCESS in 10m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f93805b1b9284720b96a19c070250e2b : SUCCESS in 4m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d42cf2584b8d470d9fad271f7707d77a : SUCCESS in 5m 15s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/cdfea89fddd04de3b7cdbbb8eeec7aa8 : SUCCESS in 6m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c268342c6ca841d0805eca23ecd964d9 : SUCCESS in 13m 55s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/9c6631096e554200be393720011596b0 : SUCCESS in 4m 43s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/20d2eb6d50c04f099cd2f78cb9e4a3c4 : SUCCESS in 7m 07s\n- octavia-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/0f61d534390c45afb1ec551c865570f2 : SUCCESS in 10m 23s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/f18cc612719a4f89811159d0ff761345 : SUCCESS in 4m 07s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/ee325f19fff149c1a1e0202a2dc45769 : SUCCESS in 7m 39s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/a74496517b564c5a84bc5b669c6fc973 : SUCCESS in 7m 13s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/5ef75362388c48299cb8adaa6ab658cb : SUCCESS in 9m 11s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/0e7396de54854d40ba4d69afc5f3666d : SUCCESS in 1h 02m 03s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/492e1dfc9c7648daa32027f589ee272f : SUCCESS in 2h 11m 04s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/f47004e129fe4d96b50423735ee7a449 : SUCCESS in 1h 15m 31s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/151968d97e6a433e835e0a0f8606dae1 : SUCCESS in 2h 09m 45s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/00d99478458c4ca29cfb666d1e007aa6 : SUCCESS in 1h 18m 01s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/49df062212874a88aa15364fa442cae4 : SUCCESS in 55m 36s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/59e5a862054a4c76860e8fd863c59cc8 : SUCCESS in 38m 18s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/df14720fb737489da2be61de11d99fe8 : SUCCESS in 38m 50s (non-voting)\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/1832a4a4ba494723a52e307256099de3 : FAILURE in 1h 03m 32s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6694a4645bdd4f11bd4e4d0b395fffc3 : SUCCESS in 54m 31s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/081d6018142349eea8507de5214e2f78 : SUCCESS in 2h 22m 37s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/837fa5f9b980458cafd4e549d3b5ff9a : FAILURE in 2h 23m 18s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/93c3f3355c4f4ca2a0540c654ee0ccaa : TIMED_OUT in 2h 33m 06s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/147c1dca4d284342bdf570ba419061a8 : TIMED_OUT in 2h 34m 35s (non-voting)","accounts_in_message":[],"_revision_number":11},{"id":"b013017334b0cae0fa6bdcf530ff3700a0ab0fff","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-19 12:53:21.000000000","message":"Patch Set 11: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":11},{"id":"2901bc28accee0f6bea0c15546eeac916b6e08b0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-19 15:03:18.000000000","message":"Patch Set 11: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9b24aa6fe8184b919576cb6b27a8d196\n\n- requirements-check https://zuul.opendev.org/t/openstack/build/5d94f7ba1fe04408b01afab3bcdb7ad4 : SUCCESS in 2m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f668bf6eef8249da8a5138c3141cb438 : SUCCESS in 5m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/85858db528e64a99aed37f889d39359f : SUCCESS in 5m 11s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ea28007d047841aa83da22d12e8d0a87 : SUCCESS in 5m 05s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/183cca837e8f49f7b406ce4a6cf5a1dc : SUCCESS in 11m 51s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/b34643d4010e495499c3e22038319a5a : SUCCESS in 5m 46s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/9d5422ecb271433b846b6f1364763852 : SUCCESS in 8m 02s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/2e147b9a4a9f4190afc476d8034db74f : SUCCESS in 1h 31m 26s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/bd2f3455482941e8aafe255de1798884 : SUCCESS in 2h 06m 56s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/2606b796ac644c8db874d9fa437b6169 : SUCCESS in 1h 31m 29s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/c72954d0aedb42c6887ce0e62e837cc5 : SUCCESS in 49m 21s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/8d55e53a27aa4a8aa68ef91282f0093e : SUCCESS in 1h 02m 56s","accounts_in_message":[],"_revision_number":11},{"id":"e971f1148b65498f64f19b57719cb96e75c22201","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-19 15:03:19.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":11},{"id":"08a80f1c8c42ed740784af8cbc618de509534651","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-19 15:04:32.000000000","message":"Patch Set 11:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/75c9c536bf984e3b9fae738cde5d30b2\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2245028c023b4f8f85a8936e09d15e23 : SUCCESS in 52s","accounts_in_message":[],"_revision_number":11}],"current_revision_number":11,"current_revision":"d1d7fe7197dab64592f745424c469ddac0175b2d","revisions":{"e7d99b1671f53d1067950d75d29564f9e09f6d92":{"kind":"REWORK","_number":1,"created":"2023-11-06 01:48:11.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/42/900142/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/42/900142/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/42/900142/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/42/900142/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/42/900142/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/42/900142/1"}}},"commit":{"parents":[{"commit":"51ae11032e962c4bce957be8750d6bc6f9a9f1a5","subject":"Merge \"Run codespell on doc directory\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/51ae11032e962c4bce957be8750d6bc6f9a9f1a5"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-05 09:38:53.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-06 01:46:58.000000000","tz":540},"subject":"Use cryptography to load PKCS12 certificates","message":"Use cryptography to load PKCS12 certificates\n\n... because implementation in PyOpenSSL has been derprecated, according\nto the following warning.\n\n```\nDeprecationWarning: PKCS#12 support in pyOpenSSL is deprecated. You\nshould use the APIs in cryptography.\n```\n\nCloses-Bug: #2042787\nChange-Id: Ic81e98c54c4bce100e3f44ff1a2fe6ce7b7f4256\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e7d99b1671f53d1067950d75d29564f9e09f6d92"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e7d99b1671f53d1067950d75d29564f9e09f6d92"}]},"branch":"refs/heads/master"},"9a817ea6ba7aa327550dc503824aabb5fec55431":{"kind":"REWORK","_number":2,"created":"2023-11-06 01:49:06.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/42/900142/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/42/900142/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/42/900142/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/42/900142/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/42/900142/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/42/900142/2"}}},"commit":{"parents":[{"commit":"51ae11032e962c4bce957be8750d6bc6f9a9f1a5","subject":"Merge \"Run codespell on doc directory\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/51ae11032e962c4bce957be8750d6bc6f9a9f1a5"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-05 09:38:53.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-06 01:48:52.000000000","tz":540},"subject":"Use cryptography to load PKCS12 certificates","message":"Use cryptography to load PKCS12 certificates\n\n... because implementation in PyOpenSSL has been derprecated, according\nto the following warning.\n\n```\nDeprecationWarning: PKCS#12 support in pyOpenSSL is deprecated. You\nshould use the APIs in cryptography.\n```\n\nCloses-Bug: #2042787\nChange-Id: Ic81e98c54c4bce100e3f44ff1a2fe6ce7b7f4256\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9a817ea6ba7aa327550dc503824aabb5fec55431"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9a817ea6ba7aa327550dc503824aabb5fec55431"}]},"branch":"refs/heads/master"},"44cbf5eaced0a8691603b7e72175813654b2c28e":{"kind":"REWORK","_number":3,"created":"2023-11-06 03:58:35.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/42/900142/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/42/900142/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/42/900142/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/42/900142/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/42/900142/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/42/900142/3"}}},"commit":{"parents":[{"commit":"51ae11032e962c4bce957be8750d6bc6f9a9f1a5","subject":"Merge \"Run codespell on doc directory\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/51ae11032e962c4bce957be8750d6bc6f9a9f1a5"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-05 09:38:53.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-06 03:58:27.000000000","tz":540},"subject":"Use cryptography to load PKCS12 certificates","message":"Use cryptography to load PKCS12 certificates\n\n... because implementation in PyOpenSSL has been derprecated, according\nto the following warning.\n\n```\nDeprecationWarning: PKCS#12 support in pyOpenSSL is deprecated. You\nshould use the APIs in cryptography.\n```\n\nCloses-Bug: #2042787\nChange-Id: Ic81e98c54c4bce100e3f44ff1a2fe6ce7b7f4256\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/44cbf5eaced0a8691603b7e72175813654b2c28e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/44cbf5eaced0a8691603b7e72175813654b2c28e"}]},"branch":"refs/heads/master"},"317e7c73983d535f74f322a34349827072148f8c":{"kind":"REWORK","_number":4,"created":"2023-11-06 03:59:43.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/42/900142/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/42/900142/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/42/900142/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/42/900142/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/42/900142/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/42/900142/4"}}},"commit":{"parents":[{"commit":"51ae11032e962c4bce957be8750d6bc6f9a9f1a5","subject":"Merge \"Run codespell on doc directory\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/51ae11032e962c4bce957be8750d6bc6f9a9f1a5"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-05 09:38:53.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-06 03:59:33.000000000","tz":540},"subject":"Use cryptography to load PKCS12 certificates","message":"Use cryptography to load PKCS12 certificates\n\n... because implementation in PyOpenSSL has been derprecated, according\nto the following warning.\n\n```\nDeprecationWarning: PKCS#12 support in pyOpenSSL is deprecated. You\nshould use the APIs in cryptography.\n```\n\nCloses-Bug: #2042787\nChange-Id: Ic81e98c54c4bce100e3f44ff1a2fe6ce7b7f4256\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/317e7c73983d535f74f322a34349827072148f8c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/317e7c73983d535f74f322a34349827072148f8c"}]},"branch":"refs/heads/master"},"1d2d3953c92a39250b32de820a36aff0f2edb350":{"kind":"REWORK","_number":5,"created":"2023-11-06 04:01:28.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/42/900142/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/42/900142/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/42/900142/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/42/900142/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/42/900142/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/42/900142/5"}}},"commit":{"parents":[{"commit":"51ae11032e962c4bce957be8750d6bc6f9a9f1a5","subject":"Merge \"Run codespell on doc directory\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/51ae11032e962c4bce957be8750d6bc6f9a9f1a5"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-05 09:38:53.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-06 04:01:01.000000000","tz":540},"subject":"Use cryptography to load PKCS12 certificates","message":"Use cryptography to load PKCS12 certificates\n\n... because implementation in PyOpenSSL has been derprecated, according\nto the following warning.\n\n```\nDeprecationWarning: PKCS#12 support in pyOpenSSL is deprecated. You\nshould use the APIs in cryptography.\n```\n\nCloses-Bug: #2042787\nChange-Id: Ic81e98c54c4bce100e3f44ff1a2fe6ce7b7f4256\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1d2d3953c92a39250b32de820a36aff0f2edb350"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1d2d3953c92a39250b32de820a36aff0f2edb350"}]},"branch":"refs/heads/master"},"1f29359e368c6cfbf3e85e7eb13b34f5d938fb77":{"kind":"REWORK","_number":6,"created":"2023-11-06 04:13:13.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/42/900142/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/42/900142/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/42/900142/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/42/900142/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/42/900142/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/42/900142/6"}}},"commit":{"parents":[{"commit":"51ae11032e962c4bce957be8750d6bc6f9a9f1a5","subject":"Merge \"Run codespell on doc directory\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/51ae11032e962c4bce957be8750d6bc6f9a9f1a5"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-05 09:38:53.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-06 04:13:05.000000000","tz":540},"subject":"Use cryptography to load PKCS12 certificates","message":"Use cryptography to load PKCS12 certificates\n\n... because implementation in PyOpenSSL has been derprecated, according\nto the following warning.\n\n```\nDeprecationWarning: PKCS#12 support in pyOpenSSL is deprecated. You\nshould use the APIs in cryptography.\n```\n\nCloses-Bug: #2042787\nChange-Id: Ic81e98c54c4bce100e3f44ff1a2fe6ce7b7f4256\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1f29359e368c6cfbf3e85e7eb13b34f5d938fb77"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1f29359e368c6cfbf3e85e7eb13b34f5d938fb77"}]},"branch":"refs/heads/master"},"8efceaabd32559e4b98058ad8ff56d1ddff7b3f9":{"kind":"REWORK","_number":7,"created":"2023-11-06 08:15:58.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/42/900142/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/42/900142/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/42/900142/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/42/900142/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/42/900142/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/42/900142/7"}}},"commit":{"parents":[{"commit":"51ae11032e962c4bce957be8750d6bc6f9a9f1a5","subject":"Merge \"Run codespell on doc directory\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/51ae11032e962c4bce957be8750d6bc6f9a9f1a5"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-05 09:38:53.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-06 08:15:50.000000000","tz":540},"subject":"Use cryptography to load PKCS12 certificates","message":"Use cryptography to load PKCS12 certificates\n\n... because implementation in PyOpenSSL has been derprecated, according\nto the following warning.\n\n```\nDeprecationWarning: PKCS#12 support in pyOpenSSL is deprecated. You\nshould use the APIs in cryptography.\n```\n\nCloses-Bug: #2042787\nChange-Id: Ic81e98c54c4bce100e3f44ff1a2fe6ce7b7f4256\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8efceaabd32559e4b98058ad8ff56d1ddff7b3f9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8efceaabd32559e4b98058ad8ff56d1ddff7b3f9"}]},"branch":"refs/heads/master"},"b79b927ebf933cb8806baf3cf5fd6c726b1691c1":{"kind":"REWORK","_number":8,"created":"2023-11-06 12:37:04.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/42/900142/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/42/900142/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/42/900142/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/42/900142/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/42/900142/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/42/900142/8"}}},"commit":{"parents":[{"commit":"51ae11032e962c4bce957be8750d6bc6f9a9f1a5","subject":"Merge \"Run codespell on doc directory\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/51ae11032e962c4bce957be8750d6bc6f9a9f1a5"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-05 09:38:53.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-06 12:36:56.000000000","tz":540},"subject":"Use cryptography to load PKCS12 certificates","message":"Use cryptography to load PKCS12 certificates\n\n... because implementation in PyOpenSSL has been derprecated, according\nto the following warning.\n\n```\nDeprecationWarning: PKCS#12 support in pyOpenSSL is deprecated. You\nshould use the APIs in cryptography.\n```\n\nCloses-Bug: #2042787\nChange-Id: Ic81e98c54c4bce100e3f44ff1a2fe6ce7b7f4256\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b79b927ebf933cb8806baf3cf5fd6c726b1691c1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b79b927ebf933cb8806baf3cf5fd6c726b1691c1"}]},"branch":"refs/heads/master"},"083996ea57a0d7c29a6c15f94be8499dde47ed28":{"kind":"REWORK","_number":9,"created":"2023-11-06 12:51:36.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/42/900142/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/42/900142/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/42/900142/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/42/900142/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/42/900142/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/42/900142/9"}}},"commit":{"parents":[{"commit":"51ae11032e962c4bce957be8750d6bc6f9a9f1a5","subject":"Merge \"Run codespell on doc directory\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/51ae11032e962c4bce957be8750d6bc6f9a9f1a5"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-05 09:38:53.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-06 12:51:25.000000000","tz":540},"subject":"Use cryptography to load PKCS12 certificates","message":"Use cryptography to load PKCS12 certificates\n\n... because implementation in PyOpenSSL has been derprecated, according\nto the following warning.\n\n```\nDeprecationWarning: PKCS#12 support in pyOpenSSL is deprecated. You\nshould use the APIs in cryptography.\n```\n\nCloses-Bug: #2042787\nChange-Id: Ic81e98c54c4bce100e3f44ff1a2fe6ce7b7f4256\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/083996ea57a0d7c29a6c15f94be8499dde47ed28"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/083996ea57a0d7c29a6c15f94be8499dde47ed28"}]},"branch":"refs/heads/master"},"c82372c90adfa9c371fc97199ffc390e3de60e4a":{"kind":"REWORK","_number":10,"created":"2023-11-06 12:52:27.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/42/900142/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/42/900142/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/42/900142/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/42/900142/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/42/900142/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/42/900142/10"}}},"commit":{"parents":[{"commit":"51ae11032e962c4bce957be8750d6bc6f9a9f1a5","subject":"Merge \"Run codespell on doc directory\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/51ae11032e962c4bce957be8750d6bc6f9a9f1a5"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-05 09:38:53.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-06 12:52:19.000000000","tz":540},"subject":"Use cryptography to load PKCS12 certificates","message":"Use cryptography to load PKCS12 certificates\n\n... because implementation in PyOpenSSL has been derprecated, according\nto the following warning.\n\n```\nDeprecationWarning: PKCS#12 support in pyOpenSSL is deprecated. You\nshould use the APIs in cryptography.\n```\n\nCloses-Bug: #2042787\nChange-Id: Ic81e98c54c4bce100e3f44ff1a2fe6ce7b7f4256\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c82372c90adfa9c371fc97199ffc390e3de60e4a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c82372c90adfa9c371fc97199ffc390e3de60e4a"}]},"branch":"refs/heads/master"},"d1d7fe7197dab64592f745424c469ddac0175b2d":{"kind":"REWORK","_number":11,"created":"2023-11-06 12:58:14.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/42/900142/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/42/900142/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/42/900142/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/42/900142/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/42/900142/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/42/900142/11"}}},"commit":{"parents":[{"commit":"51ae11032e962c4bce957be8750d6bc6f9a9f1a5","subject":"Merge \"Run codespell on doc directory\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/51ae11032e962c4bce957be8750d6bc6f9a9f1a5"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-05 09:38:53.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-06 12:58:06.000000000","tz":540},"subject":"Use cryptography to load PKCS12 certificates","message":"Use cryptography to load PKCS12 certificates\n\n... because implementation in PyOpenSSL has been derprecated, according\nto the following warning.\n\n```\nDeprecationWarning: PKCS#12 support in pyOpenSSL is deprecated. You\nshould use the APIs in cryptography.\n```\n\nCloses-Bug: #2042787\nChange-Id: Ic81e98c54c4bce100e3f44ff1a2fe6ce7b7f4256\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d1d7fe7197dab64592f745424c469ddac0175b2d"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d1d7fe7197dab64592f745424c469ddac0175b2d"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
