)]}'
{"id":"openstack%2Foctavia~900334","triplet_id":"openstack%2Foctavia~master~Id0683e20d84a7e3b310e0b7ec88f166b842a1e80","project":"openstack/octavia","branch":"master","attention_set":{},"removed_from_attention_set":{"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2024-01-25 12:00:34.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Id0683e20d84a7e3b310e0b7ec88f166b842a1e80","subject":"Declare we use threads and processes for coverage","status":"MERGED","created":"2023-11-07 16:07:54.000000000","updated":"2024-01-25 12:02:29.000000000","submitted":"2024-01-25 12:00:34.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"900334","meta_rev_id":"386a7d33391843259d954f7e0f0d92d8438de431","_number":900334,"virtual_id_number":900334,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-01-25 12:00:34.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"recommended":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"all":[{"value":1,"date":"2024-01-05 19:35:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":1,"date":"2024-01-24 16:57:17.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"value":1,"date":"2023-11-29 06:09:44.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":2,"date":"2023-11-10 07:06:45.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-01-25 10:21:37.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-01-25 10:21:37.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-11-07 19:10:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-11-10 07:06:45.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2023-11-29 06:09:44.000000000","updated_by":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"reviewer":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"state":"REVIEWER"},{"updated":"2024-01-05 19:35:27.000000000","updated_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"reviewer":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"state":"REVIEWER"},{"updated":"2024-01-24 16:57:17.000000000","updated_by":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"reviewer":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"state":"REVIEWER"},{"updated":"2024-01-25 10:21:37.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"}],"messages":[{"id":"b79c3d07d23cf8de2a8c653fd993b9bb254799b2","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-11-07 16:07:54.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e442170232511acd455463b63e1d9ce52c01a780","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-07 19:10:15.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1333502ffbdc4326be93c6dd5da5bda7\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/60e9cb04ee5740f0826df2c62dd4c9b4 : SUCCESS in 10m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cd9bd46e0b874bcb8ddec3eefe5e03fc : SUCCESS in 5m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3532fe33d34344bca530d010f1666fb2 : SUCCESS in 4m 39s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/25ecd86414ee495d9fddf4b3a6c6f9c8 : SUCCESS in 4m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/364f75087ee647328a56f5543b629be9 : SUCCESS in 11m 43s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/69d317f2997b45ae8479956df58c1fb1 : SUCCESS in 5m 37s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/3a9ef477140a4f999de82ea81a58852e : SUCCESS in 7m 05s\n- octavia-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/55356eee7a6f4c0ba872e79a724c4e0a : FAILURE in 7m 30s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/4c672967b3d343b4961455fd5269feee : SUCCESS in 4m 31s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/05ebb54dc9b94c469e22fb8d63391d5c : SUCCESS in 6m 44s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/79ffa575a1db490a92f1b9dacc0c5cd3 : SUCCESS in 8m 32s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/fe6038554cea47f0af25c42163740686 : SUCCESS in 6m 50s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/50213e0168644b67920e1f5b95e67499 : FAILURE in 1h 22m 09s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/b9f9ada097c44f9ab5d89430b235b1ca : SUCCESS in 1h 36m 22s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/7064de78fc834cdca724deb243695329 : SUCCESS in 1h 03m 54s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/4f2fa08cc5d74cf8b3f82d7aefdb03a1 : SUCCESS in 1h 36m 59s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/edc4d77ec4f04500b10054d364bff40a : SUCCESS in 1h 11m 15s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/af29e38be8fb481ea24194e828a1be9d : SUCCESS in 40m 58s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/172f2bb19da243e38fd2e82565b8c527 : SUCCESS in 51m 01s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/bf200d73b4c14eb596aa83be9f80d4ad : SUCCESS in 46m 48s (non-voting)\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/9c595f6025684aa9b939708b2d741326 : SUCCESS in 1h 13m 08s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/eeb08e1d8d104ff4bd140a5abde14512 : SUCCESS in 45m 10s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/5d173688fa184b98a6ba315f2b3ffa71 : TIMED_OUT in 2h 34m 27s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/47acada228934cbeb9973ce4dd6877af : SUCCESS in 2h 56m 52s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/401d8887fa674b8fb0becd45415795f8 : SUCCESS in 2h 24m 48s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/09392b4a76b442a89e1137048ddfbf17 : TIMED_OUT in 2h 34m 21s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9b8fea0f6ce342643d4286b22e183a2af573e0c9","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-11-10 07:06:45.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"502918249604692ce0c6387b1ca826af53ce662a","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-10 09:48:50.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0e16a426e5d3403bb7833c3603698d14\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7da9d8d731264ae6bbcabca4c15ef01a : SUCCESS in 8m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f36f2a0ac5384c4f99b5910ce04a35eb : SUCCESS in 4m 05s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/577d30b8f8e8442493d595c652f529d9 : SUCCESS in 4m 29s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/ff62f5f50a364e94a56d3820d7fb9aed : SUCCESS in 4m 29s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/56e683ecb78546e3a34428aee27177f6 : SUCCESS in 9m 57s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/be5866818ff249f6bdac62387d5c5c13 : SUCCESS in 4m 38s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/404af69392484b58a5753f7d9c8f8802 : SUCCESS in 7m 59s\n- octavia-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/0bf3c6dd0f8b4112a8f6f5193d87a310 : FAILURE in 5m 55s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/a4ab949d6eb841739e8cdccc56d9f430 : SUCCESS in 4m 09s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/5d41b04e8d4c49978a0c931628e4a08d : SUCCESS in 8m 49s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/dba48691fdfd4b0f8a1c7d94edf2773c : SUCCESS in 7m 16s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/b901c4a2de834915b9ae94dcb4f17fc2 : SUCCESS in 6m 36s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/c64b19c7915f4b628cb68a7a4e85e136 : SUCCESS in 1h 31m 29s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/a25ee93338534de4ac490ebb5c9d9c93 : SUCCESS in 1h 49m 21s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/14014cc700ca457abd980bc964de3ca8 : SUCCESS in 1h 05m 39s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/c9eaf2c8b0c447a29c685751bf66feed : SUCCESS in 1h 38m 45s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/e301e34eb4504afda09f9e21672240e3 : SUCCESS in 1h 21m 12s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/38d1e088d6d149f1ad759398b169728c : SUCCESS in 48m 01s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/6d32b37bc7794a0d9fbf643744fe5ac2 : SUCCESS in 42m 26s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/a9efc9c43b8346b990dc4a70f6d3b71b : SUCCESS in 59m 23s (non-voting)\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/a4a582e664754136909009be56944e45 : SUCCESS in 1h 15m 08s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5400adaa45e3419bba2cabbcb2735772 : SUCCESS in 37m 05s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/de448728e3e048b3bc957cfa58b96601 : FAILURE in 2h 14m 42s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/57d75104983d4080a85cf5eead174459 : SUCCESS in 2h 40m 53s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/6d54cdbaf08c483bbd0943ce87de4349 : SUCCESS in 2h 19m 28s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/9eb5f066711a494e912b91342e6c1769 : SUCCESS in 2h 24m 33s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"20189c5a6378b62bb29f297d30f5a15919b5d3e8","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-11-21 22:02:04.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c0e9151d7e22b68639e5a9fccce9dad3f3bf59c4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-22 01:08:03.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c3e9a1924c55465e91d664d445abf986\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b72b1e03e29342fb843f1ec2e6e697f9 : SUCCESS in 10m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5ecf0340ada04f679c7842520c95f5ee : SUCCESS in 5m 35s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3360036a2e0246cbbcd2dbc1585832af : SUCCESS in 5m 14s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8c244ec67d184ee3acce72dfcac335be : SUCCESS in 4m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/3db56a676c7b4d3cadee0101057d0d79 : SUCCESS in 10m 47s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/75d03b88d6e746f7b8491a2705c3c1ed : SUCCESS in 4m 54s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/88ba39c118b041e39977fa004b706939 : SUCCESS in 7m 52s\n- octavia-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/8bdbd650114d4c67bd6cd46bbbcab438 : SUCCESS in 11m 05s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/7ca228a4e5114691b776222f9b58059d : SUCCESS in 4m 38s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/bf3946b1cf324547b7c28567cac9a8ee : SUCCESS in 10m 44s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/63ffaaadc1dd4760b79c0acea25ff7b2 : SUCCESS in 7m 31s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/b68fcf8bec5e432b9dc26855193ff693 : SUCCESS in 7m 28s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/65c87b07c9534e3cb67a0a701239f863 : SUCCESS in 1h 41m 12s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/ef9248e44d9848a793beba0a83a105b3 : SUCCESS in 1h 37m 27s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/b124fa4488b84afcaf492d6b2ac56a65 : SUCCESS in 1h 04m 59s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/0d62d134c0cf47b1923d693cd0d2e4f9 : SUCCESS in 1h 37m 28s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/d7322059c407475b8c09a27401e243ae : SUCCESS in 1h 06m 44s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/feecf4cbcaa140cda05c20fd12680fa5 : SUCCESS in 38m 35s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/c2fdec5a45a149268a1ba9de2919a139 : SUCCESS in 37m 52s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/68cefa9362aa4904a8d943537a655d4a : SUCCESS in 38m 04s (non-voting)\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/c68dce5ba4e14ab0aea68e5fc7a762c3 : SUCCESS in 1h 19m 36s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/6c1149f701e0473ea3b28c3d02fd0467 : SUCCESS in 37m 00s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/6335560db21644938a72bce7873fe03e : SUCCESS in 2h 02m 56s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/c2789bb8aac44af585e6b9195c167497 : TIMED_OUT in 3h 02m 47s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/ed7588e6c4a841d6a7e36126bc12a47b : TIMED_OUT in 2h 33m 36s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/604b614aee7e44ab920106dd533bf9ff : TIMED_OUT in 2h 34m 20s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"747d7a9891f79b0a160c5099c5ef4f94f34f007c","author":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"date":"2023-11-29 06:09:44.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"3bbd52aa3b45aca0e34a62835b4420a54093cc5e","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2024-01-05 19:35:27.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"342b542a2eb3d04e53913d94b30cafa2efdeddbd","author":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"date":"2024-01-24 16:57:17.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"b9ee0f445ff29824d77a8ee5dc22d67fe2c9bcb3","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-01-25 10:21:37.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"11a5885b0119383e5d59f98cdea0bd878271c97f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-25 10:22:19.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"fcfdfa20d774a68e5a17f44f8ec4d774ca407741","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-25 12:00:34.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/4c3948d493084c10bfb8f2f5451366be\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/193ceda0dcb0485cb23615464ef54812 : SUCCESS in 4m 57s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cf5e642e582a450bb7a1575f42357b0c : SUCCESS in 6m 36s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b380b5cd12d146ec89046a022cf44bb8 : SUCCESS in 5m 28s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2fab45ec5df646e68326eac89e28824b : SUCCESS in 13m 40s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/41bd903a474f4234bcf5bee900e56986 : SUCCESS in 5m 02s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/fe5cd6f910fa4f72afc46b1b3689ad48 : SUCCESS in 7m 43s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/bcc41104ed024ca0b7f51bf5edb5557a : SUCCESS in 1h 27m 08s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/50589943861f420d9ceb057dc48892f3 : SUCCESS in 1h 37m 27s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/4c3dbffcfa184f45b3dbb5c49ad538e0 : SUCCESS in 1h 02m 08s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/390ce91f8fed4864aa86ffcbc1ae9e00 : SUCCESS in 34m 39s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/b9b9e85a9f1c4ae7a93d0a37a8168e24 : SUCCESS in 47m 58s","accounts_in_message":[],"_revision_number":1},{"id":"f2772bdf4f79e2a9744d8832213cc088144019bb","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-25 12:00:34.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"386a7d33391843259d954f7e0f0d92d8438de431","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-25 12:02:29.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3061709272f842e391c617f9c655bb73\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/19896222a8f74670a665c21f080a7293 : SUCCESS in 44s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"657e7b8be49ec3fd5bb4a1f2195c8352e8a8c20a","revisions":{"657e7b8be49ec3fd5bb4a1f2195c8352e8a8c20a":{"kind":"REWORK","_number":1,"created":"2023-11-07 16:07:54.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/34/900334/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/34/900334/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/34/900334/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/34/900334/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/34/900334/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/34/900334/1"}}},"commit":{"parents":[{"commit":"7310986de9bf68ed86de90de0501a1bc46945526","subject":"Add Noop Certificate Manager","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7310986de9bf68ed86de90de0501a1bc46945526"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-11-07 16:06:56.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-11-07 16:07:53.000000000","tz":0},"subject":"Declare we use threads and processes for coverage","message":"Declare we use threads and processes for coverage\n\nWe should declare to coverage that we use the multiprocessing and thread\nlibraries in Octavia.\n\nChange-Id: Id0683e20d84a7e3b310e0b7ec88f166b842a1e80\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/657e7b8be49ec3fd5bb4a1f2195c8352e8a8c20a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/657e7b8be49ec3fd5bb4a1f2195c8352e8a8c20a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
