)]}'
{"id":"openstack%2Foctavia~900519","triplet_id":"openstack%2Foctavia~master~I02e5e7d8905d4a2a4cdb02eaba2af90e3e40a7cc","project":"openstack/octavia","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2023-11-13 12:53:13.000000000","reason":"Change was abandoned"},"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2023-11-13 12:53:08.000000000","reason":"\u003cGERRIT_ACCOUNT_29244\u003e replied on the change","reason_account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}}},"hashtags":[],"change_id":"I02e5e7d8905d4a2a4cdb02eaba2af90e3e40a7cc","subject":"Set sqlalchemy-tips job non-voting","status":"ABANDONED","created":"2023-11-09 13:11:04.000000000","updated":"2023-11-13 12:53:13.000000000","total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"1bca127f997f383d59c8a5dfae3b6ca7ff156060","_number":900519,"virtual_id_number":900519,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"disliked":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:check","value":-1,"date":"2023-11-09 16:20:14.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":-1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"disliked":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"all":[{"value":-1,"date":"2023-11-13 11:45:36.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-11-13 08:52:13.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":-1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-11-09 16:20:14.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-11-13 08:52:13.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"},{"updated":"2023-11-13 11:45:36.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"}],"messages":[{"id":"1a420430ae44f045da43f7b86ccd56e741a2fbd1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-11-09 13:11:04.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4f8b6172e0786cf2f9bc1ca1637e2cc4e7c6d17f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-09 16:20:14.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/47d09164f19d46348ca9cd7ba59eccaa\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/81cf02d793d04ae88ec96c0a95b56e42 : SUCCESS in 11m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6bbedd323253434a9947bbfbb2358cda : SUCCESS in 4m 37s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dbaa4e7d907249f9b000457a560f430a : SUCCESS in 4m 29s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e21f9e9e73174413a75619466292fc81 : SUCCESS in 4m 55s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dbf7edc5635440b39b10df03120f14b5 : SUCCESS in 12m 37s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/299a74633f2a4125967da8079cf78bcb : SUCCESS in 5m 01s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/bba3ef5977764f0784a95c7da9623144 : SUCCESS in 6m 56s\n- octavia-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/8e1b78bb9f9945ddbcf9b1dccee53eb4 : FAILURE in 5m 43s (non-voting)\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/306167c864ad4b5a9cd28ca0292e8fb8 : SUCCESS in 5m 06s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/a4e3aa681f084548bce19db9cead994f : SUCCESS in 7m 59s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/0f38018c2ed647739b285601e72e95e0 : SUCCESS in 7m 03s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/7e727d7a983f42c18a3800d67049b83a : SUCCESS in 6m 32s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/0d8ac583efcf4bfa84e75e3d6726ef92 : SUCCESS in 1h 13m 28s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/67f7f0456a3948abac378b7e23fd2265 : SUCCESS in 1h 53m 26s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/0e67318985e4430faefb4ee3756b5517 : FAILURE in 1h 13m 26s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/fe5413aadfaf42bfac7987225486b11e : SUCCESS in 1h 57m 05s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/aa0db8c60d4f435b9ceb1ff7886bbac9 : SUCCESS in 1h 11m 09s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/4d78a1b6b6fe4720a5fe84e06208d733 : SUCCESS in 41m 37s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/d30ccff7b2f4459cb49b92663535ca29 : SUCCESS in 39m 23s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/be47991b8a3d48b0bd05cbac0fd67144 : SUCCESS in 44m 12s (non-voting)\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/d403323d83394406bd938843c0a54b3d : SUCCESS in 1h 04m 11s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/297fff4e3a504995b760ba45942b3ef0 : SUCCESS in 34m 01s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/555c9d621e7a4c229cbc4d5e38413e9b : SUCCESS in 2h 09m 25s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/be68944dfd2c4b56b42480a3f5022562 : FAILURE in 2h 49m 14s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/86b922cdbddc4160a512bd406004abba : SUCCESS in 2h 05m 36s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/883c47cf28f74db69f1bcd8d589785b3 : TIMED_OUT in 2h 33m 41s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"5ac7a62f1b3c3e29cd3dfed25e9f4827e46489db","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-11-13 08:52:13.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e42407beacaa24aaadba2433f7e3c589f10d5b27","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-11-13 11:45:36.000000000","message":"Patch Set 1: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a86af67fbff08322c11616f435adc6d0aa1bee55","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-13 11:48:03.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9fece72ca5434c4bab36dc27951f42e3\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c8a6cc0d32d74f4794c324d3843cfecc : SUCCESS in 12m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c5f72c7929934676958a163e003607d0 : SUCCESS in 5m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/790c15f4fd21469cbde32b656c7f34e8 : SUCCESS in 4m 47s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d85107db813c4b739411d0b0394333af : SUCCESS in 4m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/28632817656d48a9ab5c58b2b73ade8a : SUCCESS in 9m 38s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/2754000c33d24d2ca84f054f45646ffd : SUCCESS in 4m 55s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/2f613ec05b564238a3035b28901feaec : SUCCESS in 9m 28s\n- octavia-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/b4e2bd15490347b0b4d6e106d913ed09 : SUCCESS in 9m 42s (non-voting)\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/64972d4e6cc3474b873c86730d875027 : SUCCESS in 4m 51s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/941d3813f5504d00ad07be988e0308dd : SUCCESS in 7m 24s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/c519c1e7b8114753928ddf9ee26a9b9a : SUCCESS in 8m 35s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/40bc660d6f8f48c6874dde99024ca093 : SUCCESS in 6m 52s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/37f943cb6f784c15b26c15b65fa4de00 : SUCCESS in 1h 18m 23s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/eb09c2cef3d84bf9830dba8cd580688d : FAILURE in 1h 44m 29s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/8092eaef0890438d99f0edf49a262289 : SUCCESS in 1h 12m 28s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/659032d2ca044561935b7ac23d4a8e82 : SUCCESS in 1h 47m 37s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/32063f4843554eb5a426a58559d78969 : SUCCESS in 1h 07m 10s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/6f82d39ab1d342a5999e606ed2c386a0 : SUCCESS in 43m 15s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/7985555f694b463397aed9371c8772f7 : SUCCESS in 39m 48s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/9648781f733e4fcf8185af98559a5cab : SUCCESS in 48m 21s (non-voting)\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/ec5d80a816d34c13aafaeab79eb620bd : NODE_FAILURE Node request 300-0022704886 failed in 0s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/83192237ec174ac19fe2eae594d331ba : SUCCESS in 41m 39s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/17d1bdc0b82641d2be868d0c9dc9184f : SUCCESS in 2h 14m 25s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/70a1375022ac47aabac816419d391a80 : SUCCESS in 2h 37m 50s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/25286b5799b34c269910f36546558090 : SUCCESS in 2h 13m 30s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/af54728cdf9c4fee8dafd99a711afc9d : TIMED_OUT in 2h 33m 51s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d22699f761350a1c95bbf1e143a6f80ccd500ca2","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-11-13 12:53:08.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1bca127f997f383d59c8a5dfae3b6ca7ff156060","tag":"autogenerated:gerrit:abandon","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-11-13 12:53:13.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"f71660eaba17efcf1f46e8ee08dbf31a25f7e6c4","revisions":{"f71660eaba17efcf1f46e8ee08dbf31a25f7e6c4":{"kind":"REWORK","_number":1,"created":"2023-11-09 13:11:04.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/19/900519/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/19/900519/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/19/900519/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/19/900519/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/19/900519/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/19/900519/1"}}},"commit":{"parents":[{"commit":"7310986de9bf68ed86de90de0501a1bc46945526","subject":"Add Noop Certificate Manager","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7310986de9bf68ed86de90de0501a1bc46945526"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-11-09 13:10:50.000000000","tz":60},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-11-09 13:10:50.000000000","tz":60},"subject":"Set sqlalchemy-tips job non-voting","message":"Set sqlalchemy-tips job non-voting\n\nsqlalchemy-tips job started to fail because of [0]\nMark it as non-voting until it\u0027s fixed\n\n[0] https://bugs.launchpad.net/oslo.utils/+bug/2042886\n\nChange-Id: I02e5e7d8905d4a2a4cdb02eaba2af90e3e40a7cc\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f71660eaba17efcf1f46e8ee08dbf31a25f7e6c4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/f71660eaba17efcf1f46e8ee08dbf31a25f7e6c4"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
