)]}'
{"id":"openstack%2Foctavia~900925","triplet_id":"openstack%2Foctavia~master~I1c50f232c4f0c5c10a3dd5a928466f7ef67a9763","project":"openstack/octavia","branch":"master","topic":"idle-timeout","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2024-02-28 01:09:03.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}},"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2024-02-28 05:27:01.000000000","reason":"Change was submitted"},"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2024-02-23 01:17:46.000000000","reason":"\u003cGERRIT_ACCOUNT_11628\u003e replied on the change","reason_account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}}},"hashtags":[],"change_id":"I1c50f232c4f0c5c10a3dd5a928466f7ef67a9763","subject":"Honor connection_recycle_timeout in MysqlPersistenceDriver","status":"MERGED","created":"2023-11-14 16:55:52.000000000","updated":"2024-02-28 05:28:19.000000000","submitted":"2024-02-28 05:27:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":9,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"900925-idle-timeout","meta_rev_id":"5c1ebf5224c479e7842f821c777f623e85d2d611","_number":900925,"virtual_id_number":900925,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-02-28 05:27:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"recommended":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"all":[{"value":1,"date":"2024-01-05 17:55:04.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":1,"date":"2023-11-14 17:05:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"value":2,"date":"2024-02-23 01:17:46.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-11-15 10:56:26.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},{"value":1,"date":"2024-02-23 01:24:09.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":1,"date":"2024-02-23 07:38:33.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"value":1,"date":"2024-02-23 01:17:46.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},{"value":0,"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-11-14 17:05:30.000000000","updated_by":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"reviewer":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"state":"REVIEWER"},{"updated":"2023-11-14 19:59:42.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-11-15 10:56:26.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"},{"updated":"2023-12-05 14:08:42.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2023-12-05 14:08:42.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2024-01-05 17:55:04.000000000","updated_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"reviewer":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"state":"REVIEWER"},{"updated":"2024-02-23 01:24:09.000000000","updated_by":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"reviewer":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"state":"REVIEWER"},{"updated":"2024-02-23 07:38:33.000000000","updated_by":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"reviewer":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"state":"REVIEWER"}],"messages":[{"id":"cb9e73107fdd6eea42b7f5928fc81456c27898cb","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-11-14 16:55:52.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d86a1e2752698add9f3658f41d9a6e064a103d36","author":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"date":"2023-11-14 17:05:30.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"38ebd2b04843539f360d3f920c4e7284485d5004","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-14 19:59:42.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/8a52afe1e305477aa204d8951d311a80\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/676c7bddb7c14cf3a9e36f5b3e771fca : SUCCESS in 15m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0ad153f502f54765b532cd58eb17dace : SUCCESS in 5m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/6caaf00a219540048098d9991dcb2b50 : SUCCESS in 5m 14s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d1bfd232916c4f27b543950be0e8e9ed : SUCCESS in 4m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/420fbefa63484b30a05ebd799f9f9f26 : SUCCESS in 9m 09s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/955431d54956405ea97f480b65011ff6 : SUCCESS in 4m 27s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/e158b13b1f8843628f709d992816d2b3 : SUCCESS in 4m 58s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/bded7c4d6a6f4d539bdc77a0094cbb43 : SUCCESS in 8m 16s\n- octavia-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/beecc7d6871b42e2a71698e74e4d621d : SUCCESS in 11m 23s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/3d29f85dee914b449f6c94330cfd985f : SUCCESS in 5m 40s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/334ed6b03c254419b26f8be0d806e282 : SUCCESS in 7m 16s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/6d8e44f4e81b4bd9bf4a0a72f4a6cd44 : SUCCESS in 6m 51s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/21a53c1ab6804c0191ac357307b5e432 : SUCCESS in 7m 39s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/ed36b9f01fac445da673bfadf64bc626 : SUCCESS in 1h 27m 58s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/3b6578f4ed6147f4935b0c7efb964429 : SUCCESS in 1h 36m 45s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/8aba76ae97fc4d2fa7674ae1108d0fd8 : SUCCESS in 1h 08m 44s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/ce643ea068444f09a946c8681edb592b : SUCCESS in 1h 56m 21s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/b322cd9f1b454423a081d1996c9401ef : SUCCESS in 1h 15m 40s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/167ab967f9284e8f94abd505d2b579f2 : SUCCESS in 40m 47s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/d2b0b1beedb94cd6a845c4dbdc2329ee : TIMED_OUT in 3h 02m 20s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/88f1683b078548ce9c1d45cfdc03bd3f : SUCCESS in 54m 05s (non-voting)\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/447b0ff11cbd4521aa16fb0b6d1cc8d6 : SUCCESS in 1h 08m 18s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/412bd478e83d46cbb69ed1540c268af9 : SUCCESS in 35m 02s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/825cbf92c60e4090b78d2bb96da96996 : SUCCESS in 2h 12m 24s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/f10df355db0b4e848972ae48d1ee3ec6 : SUCCESS in 2h 25m 22s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/87f05ff082a447adb11c6824b3a5d9cc : SUCCESS in 2h 23m 00s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/79fd663e40d34aed8975d40f2a20ee7b : TIMED_OUT in 2h 33m 14s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0c6732b821207de65106267ad7ebaef58a9a1929","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-11-15 01:08:20.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"fe9dcbd54dc16516b983f7376b57574a4ed687ea","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-15 03:59:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ab78adbb871b4a56bfb30996d448ca0d\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/09f8329bb639471b8fccb0263fcfcf5d : SUCCESS in 10m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b748cc873b8d4d99ba5aa79377360bbe : SUCCESS in 4m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/22eaa47304a84d14b35233d02b828636 : SUCCESS in 5m 01s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/55103f0649dd47f0895ae0ab9e988c3e : SUCCESS in 5m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ea39d3eb0c044cf397d2a2a3751a0303 : SUCCESS in 11m 53s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/de75f7fde9bf46f2be0465aedb6f9806 : SUCCESS in 4m 43s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/793fe7a60e4a45649d67409494c4b412 : SUCCESS in 4m 43s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/c40c2fe6aa3a4396a27c4ec31c0ec990 : SUCCESS in 8m 55s\n- octavia-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/ef231d71f8074de5be643375165b1867 : SUCCESS in 12m 05s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/771c4d84477b4f788c966a3f4a1443e1 : SUCCESS in 4m 52s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/d13456d2451343479f7806a75d5c0e0d : SUCCESS in 6m 49s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/38f0eede0caf48d2b175da3fb5e53c3a : SUCCESS in 8m 44s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/800d235c3bad4bd5af46443900c9de67 : SUCCESS in 7m 12s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/613797837e6d4ba8a4b18080ae7d2aae : SUCCESS in 1h 17m 09s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/b5b2d4396b60447a83a6aa0aeee4f33f : SUCCESS in 1h 49m 24s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/520b42ff2cea4bd8b56c55da9fd46147 : SUCCESS in 1h 18m 30s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/8b920efe43ff4eaf975b3efdef7f3111 : SUCCESS in 2h 13m 03s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/4cbeb487c9b84167afe3fc078af38518 : SUCCESS in 1h 18m 32s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/f2355a820ff742379285d856566140ef : SUCCESS in 48m 20s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/a39242f1eeea416ca46baaebd7c49845 : SUCCESS in 48m 14s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/cb27c82bfded4334897741d05d3cd9f8 : SUCCESS in 49m 30s (non-voting)\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/b173965e0aef46cc8ee866ebbdfbef77 : SUCCESS in 1h 09m 34s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b989e0a52b61488498119588e9938cec : SUCCESS in 43m 31s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/fbf9500230ad49a3a87a504a1ea8e47c : SUCCESS in 2h 19m 05s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/2776962aa44c426ca7eabb19044af7f2 : SUCCESS in 2h 49m 55s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/63e71f18a658482eb01cc9bc74e9cd94 : SUCCESS in 2h 11m 03s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/8acaf4e2f98a4f31a0e4f4bcf0529e7a : TIMED_OUT in 2h 34m 45s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"437dadcb30f5fe5313322c17d3209443650a18e1","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-11-15 10:56:26.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"19a4b4ca60d073f8fd7f754a64afc55f4b3dc3de","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2024-01-05 17:55:04.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"30237fec9e28ae1632a889d046bfc9db4077daa6","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2024-02-23 01:17:46.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"4d9257b650acdac99a9abe739d04e6e2ee361577","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-23 01:18:24.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"7b436a96451baa24060584c90cbd1ebbb5f21ce9","author":{"_account_id":32029,"name":"likui","email":"likui@yovole.com","username":"likui"},"date":"2024-02-23 01:24:09.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"778238756163c3b457375fb472e27eee1d88e2b5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-23 03:00:00.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/deba2565df19433a801dcfabe78c2c85\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4d2f1b1fbd274f748620273c059a2e2d : SUCCESS in 4m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ffbf6b29247446cd911282cfddfc5e6b : SUCCESS in 4m 37s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4bfa7464eeda4615bbb262b0659269ab : SUCCESS in 4m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9a7c6d6c22574e94a45c494fcce44b7d : SUCCESS in 12m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cf1d23b55bc24146b036110e0b8f64cf : SUCCESS in 4m 34s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/73842a9e13d146fb8db19a50890e1965 : SUCCESS in 4m 35s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/498c60fb7c8a48cea75ce2182a093829 : SUCCESS in 5m 50s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/50b6e23812c34c6a9d020513d924b88f : SUCCESS in 1h 16m 18s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/eb09ce82ec724ec1be2c1cec33852f14 : CANCELED\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/9b6fb3a70d9c46e38ea2d7fca8658063 : FAILURE in 1h 39m 56s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/bf1d34d8adaa4ccd8bcba7d5cf8d599f : SUCCESS in 52m 17s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/05099e9cf60d4f919a4a9f65ec05844e : SUCCESS in 1h 02m 22s","accounts_in_message":[],"_revision_number":1},{"id":"34cffd0a8ea13a17e8c360b48b387336868754c3","author":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"date":"2024-02-23 07:38:33.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"6df36e061392992c6138146543908789281a4b3e","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-24 15:20:55.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ebb8a16d03864e939408f808c2d2b2e98b5f8b4b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-24 18:07:12.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/13cfca3befdb45b9abb5876091b2ba5f\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3091af2377194c499253cacf29926986 : SUCCESS in 11m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8e2a251f42ff4d7196f2144db125d495 : SUCCESS in 5m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d129c04b310042cbac0775251b2a592f : SUCCESS in 4m 20s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d79549ee58fc4369925df293fb5f7688 : SUCCESS in 5m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c1d5cbd4a98948d6abd65b563a051d66 : SUCCESS in 10m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2790075f5c4441a792b61080b4fdb3fc : SUCCESS in 4m 39s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/b09fef28fb7e466b937cc33176f3a206 : SUCCESS in 4m 22s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/cee8cbb8437349c997ccc85c51f718e0 : SUCCESS in 6m 56s\n- octavia-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/84401d6508384a3a849ce4555370647d : SUCCESS in 8m 38s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/e6b0d4fc7a4a46cda7f32af60665ab90 : SUCCESS in 4m 42s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/839485483e45485aaff8585775a9ffe0 : SUCCESS in 6m 48s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/1cbd1ffa986040dc8f6a755d1caf1781 : SUCCESS in 8m 32s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/08590b2c03544cce9eb93697552bc6ee : SUCCESS in 7m 32s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/a058395c70a5419b9287ffb67e482fa7 : SUCCESS in 1h 23m 38s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/c1c299db1dfc40739c1700a38d949c22 : SUCCESS in 1h 47m 17s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/06deedcbe7824f1e8b2e94c7daaaa18d : FAILURE in 1h 42m 24s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/f09356f0b8154ab4a8289cce3ede3df3 : SUCCESS in 1h 50m 17s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/d618fb429cbf4b8692a06cb87cadfa82 : FAILURE in 1h 39m 16s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/fc53aef74bb242ccbdf835502007c99c : SUCCESS in 43m 32s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/6a6a57eb800844f2bd4d344cfe7baac6 : SUCCESS in 39m 40s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/938f02d83a1b4a8da2b5bc5c8f088010 : SUCCESS in 45m 21s (non-voting)\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/5356ad0e7c4a4fa8982e07346ace1e79 : SUCCESS in 1h 09m 36s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/80fc729ccff74537be4a398944cb2a7d : SUCCESS in 39m 17s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/c0e54c5f82ff48ce9770d03720a083ca : TIMED_OUT in 2h 33m 38s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/f02614d49ed34b74b81015d384757f94 : SUCCESS in 2h 42m 38s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/4b57579b13b94965aef65c557e8e8528 : FAILURE in 36m 10s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/3bee5139d9ed4f9f9156cbb62ef67c1e : TIMED_OUT in 2h 35m 13s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"34585ab20e11f2e3844c485e3c5c87847671c797","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-25 16:03:20.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"c1c29106832080695610dc576abd4ec9492d7f24","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-25 19:11:07.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3c548ad3819e4d24aed31c8762867715\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/5cd20fc7d8264060b965d76b706d03b2 : SUCCESS in 10m 54s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a5ba11d6aa1844608e3f01d5f72a8792 : SUCCESS in 3m 44s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4aaf583e213b47dc9024f02844137e37 : SUCCESS in 4m 59s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e6e884b9fd9c4510b25f361d1d048393 : SUCCESS in 7m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e6ccaaeba1c24037a4c80915cb96fe59 : SUCCESS in 11m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/39614dcd611b4419a0002ac31549af07 : SUCCESS in 4m 42s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/575c2b9637f14f1db67370e3a2d5daba : SUCCESS in 5m 18s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/df2ba83e69714f6cad135881317bc5e3 : SUCCESS in 7m 24s\n- octavia-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/4615700317a44d26b525eaab8f751bd7 : SUCCESS in 13m 58s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/05dc999819e641d38521714a9fd3f6c6 : SUCCESS in 4m 25s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/6517bdb4b0d045b38e9b3c203c5b2cb4 : SUCCESS in 7m 20s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/5a639acd8c7341fea05051ab2acbec8c : SUCCESS in 6m 00s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/b7b200b09b8e46c1b1848030c1cfbb4a : SUCCESS in 7m 55s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/0e6e26ccce89494d95dfce30160ca3b7 : SUCCESS in 1h 15m 38s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/32df78a5f4d146f5adb1d43d062a4543 : SUCCESS in 1h 50m 06s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/c10367d5e6754e56bb73f7dd4a9807f9 : SUCCESS in 1h 02m 49s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/47e82b9a4eae4f2a8169f197b0481bc5 : SUCCESS in 2h 00m 41s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/cd2af1df20fd4e918554fe20d7b6c71a : SUCCESS in 1h 01m 14s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/069f39dc42c3413590000d49b728e3ac : SUCCESS in 34m 26s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/e9525610c51f4a0f81d697567ad60d03 : SUCCESS in 42m 41s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/de28983291a3490d80867dab96d72897 : SUCCESS in 42m 44s (non-voting)\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/40055cdc448e463fa370063228878d4b : SUCCESS in 1h 02m 11s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e6c221c626b94a5c9fb134bb6fea33e4 : SUCCESS in 37m 26s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/59200e34cff04835b1d0f22192b6a0b0 : SUCCESS in 2h 15m 53s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/dc92f03c60e44bf59345bf3b9cf635b6 : SUCCESS in 2h 58m 17s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/a9d7eacd03524c63abe148e2ca5b088d : FAILURE in 28m 54s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/cbdf5b39f83248f8b4685b6d91546700 : TIMED_OUT in 2h 34m 23s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"d1e67cfcffc5c3efc0cda2c71af15c299b422c26","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-25 19:11:42.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"18f6b48cc03e23727d63bf5f4617f2bd63bb8131","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-25 19:23:07.000000000","message":"Patch Set 1: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9e79c3a962ed49f38b03aa4797768a60\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eba0e42a0b8f4b3199b4dec4d30d052c : FAILURE in 5m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1540c8df55284604b0c571d0d79be815 : SUCCESS in 3m 35s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b69dc0dea06540329d4032ef9f40631b : SUCCESS in 5m 20s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/04c7790dcc584ae69a05cc4aadfd3b03 : SUCCESS in 8m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/6730d9b168674a638df5f5ca5c29e5f3 : SUCCESS in 5m 45s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/4b3826181a1f4d41ae0e44aaf92704c0 : SUCCESS in 3m 27s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/1fa270affff040dfa9f89c2662fa98ff : CANCELED\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/dfa5eca1fc7e4b95a76595b941ef9534 : CANCELED\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/02e3da163efc46899a06a459e4335c83 : CANCELED\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/46d27111a3354b1b93812a2461107334 : CANCELED\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/3b6dbea96dac46749dd59e2e723d544f : CANCELED\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/dd78c168e3cb4f9ea417b3defa21e57d : CANCELED\n\nWarning:\n  Comments left for invalid file octavia/amphorae/backends/agent/api_server/osutils.py\n  Comments left for invalid file octavia/common/tls_utils/cert_parser.py\n  Comments left for invalid file octavia/network/drivers/noop_driver/driver.py","accounts_in_message":[],"_revision_number":1},{"id":"0ab697491c43a7d2469bfa74b7a986ac2cc9a10b","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-26 00:40:07.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7883fe12625f7eb4dd2dc647fe33f929652492e3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-26 03:46:55.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/26d7a461b4ee45d482ce2b98bbf8b1eb\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2b4cd3826cef4e7f9446bdb2fa9cd51f : SUCCESS in 11m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/042b90a99fd548cf997b791b1db2be92 : FAILURE in 6m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dc78ed99430f4fab8a9e26d68cdbb5e8 : SUCCESS in 8m 07s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c7bd9b22fe3a42efad40ff42da305474 : SUCCESS in 5m 06s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5690182136414d17933c8e253e6c0703 : SUCCESS in 12m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/57ff941c45e04dcdbbaf1d38225058e6 : SUCCESS in 4m 28s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/c3a52d1d722b480b8574e19460ab1823 : SUCCESS in 7m 10s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/61b9675cf22a49d380e03642a4455a6c : SUCCESS in 6m 28s\n- octavia-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/59f05dff2bd34a168062630e6160aa8d : SUCCESS in 8m 51s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/d9cc39f7727e450fb51eaada9f56ae0e : SUCCESS in 4m 13s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/ecf3fde3f13d4de88889a1b553de27a0 : SUCCESS in 7m 11s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/d647262404f94c23b6d706412ce59712 : SUCCESS in 6m 02s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/51add3d82f2041aebff9ebe72fc5d0f4 : SUCCESS in 6m 48s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/59b46a0f61634c86b7bfa4cba1bbc998 : SUCCESS in 1h 28m 49s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/73fdbfbd61ae40dfb5a5fcd89e0e33c7 : SUCCESS in 1h 42m 33s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/4959fee4b6ac4ca2a868cf484e9ed0d1 : SUCCESS in 1h 09m 36s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/23045b1a0a0243a5a4e0fe43aae8eee2 : SUCCESS in 1h 45m 13s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/4c9f5cc83ecb42a4965e9683f223d9ce : SUCCESS in 1h 05m 02s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/fc6d1c754f754f94876dd7ebd2163bb4 : SUCCESS in 36m 11s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/d5037ece522a4bf5a2d7435cf4f1fe1f : SUCCESS in 45m 27s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/2cd7cf40b39d493a98af11d5697c98ff : SUCCESS in 45m 12s (non-voting)\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/24640d8aea274e8d88bbde4f589b477d : SUCCESS in 1h 09m 46s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2aa30632c72b46d4b662668f9de59673 : SUCCESS in 41m 03s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/f7ca25814a9748df91f6df453809628e : SUCCESS in 2h 21m 46s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/2db5600fef004a4894a5fe88abae8d0a : SUCCESS in 2h 33m 42s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/af4195907e024236bffd2ca25873c5bf : FAILURE in 36m 23s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/e4e75b4d22244a66bbc14247afbf9fab : TIMED_OUT in 2h 34m 45s (non-voting)\n\nWarning:\n  Comments left for invalid file octavia/amphorae/backends/agent/api_server/osutils.py\n  Comments left for invalid file octavia/common/tls_utils/cert_parser.py\n  Comments left for invalid file octavia/network/drivers/noop_driver/driver.py","accounts_in_message":[],"_revision_number":1},{"id":"cd2e4f819f12fe9f2e4786dae1db516cf0a658d8","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-27 16:43:43.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"40ea4bb74b1199fc897c1a644733c24050bcf1a0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-27 19:29:03.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/cbcb7c0ad3ee488a8e6f4a917d1b2cc4\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4e55d7bbcaea456cb52052557ca7839f : SUCCESS in 11m 11s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/90e94a91da924898a9dc3dc34bae34f5 : SUCCESS in 4m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7f04bc8817d146569828d66ed994f7a6 : SUCCESS in 4m 27s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3f69740772254957bd2384891bb09244 : POST_FAILURE in 5m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/86b9154f824b409baad5764d5b60fc05 : SUCCESS in 13m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0a334f440cc748fbb4299ab339e49edf : SUCCESS in 5m 30s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/a5331cef75f546f0ae0c65ebdf4dad6d : SUCCESS in 6m 28s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/f6c07805cbe64465bcf1a0ff24a61bf1 : SUCCESS in 8m 25s\n- octavia-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/8527042d3a6644caa085d6828e8b1ae7 : SUCCESS in 10m 08s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/0c2f9790958e4b2e8f1ace020b6b4ac6 : SUCCESS in 5m 10s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/b7e047b150c040ce90bfa12fe2958e1d : SUCCESS in 6m 47s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/9aa8fd10af564be8b165cd5db7e72715 : SUCCESS in 7m 48s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/3bba5632fcb34b4f9d867524eba94471 : SUCCESS in 10m 43s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/2a1df3d2e8f04d61acba2091c2e453db : SUCCESS in 1h 42m 00s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/050d0b1e43bb4d18bec5cac9abf0d976 : TIMED_OUT in 2h 35m 45s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/515f903072384249899919022dba816c : FAILURE in 1h 28m 27s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/327382f607d5478e93de8ba9765680ac : TIMED_OUT in 2h 34m 54s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/602a61e28562425789bd5f9f48345368 : SUCCESS in 1h 16m 27s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/106e4db654c3433498324d68c012cbc5 : SUCCESS in 48m 56s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/f51456bec6084403bc33a992e36e138c : SUCCESS in 39m 16s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/c38c7cbe9f71428db8f794a43860aef3 : SUCCESS in 39m 13s (non-voting)\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/d40b55f21a814180b94c84bc7a81ab92 : SUCCESS in 1h 22m 13s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/dc848950ae9d49e2a6988946a4e426dd : SUCCESS in 48m 19s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/8232dc76f5c44b448ecaeca91475924d : TIMED_OUT in 2h 33m 39s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/8e24d32b0b004c6598ab0697923b8bbf : SUCCESS in 2h 36m 21s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/14fc2fbc59f94787b4ea61890af2bb2c : FAILURE in 34m 52s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/99bedd048c654cd1b010dbb769ee26dd : TIMED_OUT in 2h 33m 37s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"65c000730f700ee79d3d78a44eb989ea6128e578","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-02-28 01:09:03.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ee99b0c0fdffefb455f419aef65eda57a2c9e6c3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-28 03:47:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/fc34f805dd8f427ba1c47d4a104cece9\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f5c34d3cb22540379d670b356bb2a8b9 : SUCCESS in 11m 00s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f5d19ea3be384e34bd232c397fc52cf8 : SUCCESS in 5m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/94500c3fb9a84851bbc6f8b29016cfd5 : SUCCESS in 4m 42s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/cb5806b8a8b24bc79c149e6417fa1749 : SUCCESS in 7m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/cac08b54ee924414a90b9ecd0eeaf32e : SUCCESS in 12m 08s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5e181e270d08482fa2a1bbd164ff33f8 : SUCCESS in 5m 13s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/ccbfc79a95c34d4fa8d1412029de8181 : SUCCESS in 5m 26s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/7dc198d00e6049639da5080cec19d599 : SUCCESS in 9m 11s\n- octavia-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/f1c6efa4129341c3a4a5e361545aff7a : SUCCESS in 9m 43s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/67f12adf75464776977a4d653d1cc202 : SUCCESS in 4m 40s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/8d1a2c6f9d284146a3543a00a05931e4 : SUCCESS in 8m 24s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/da00d452ad4b483bb38370834db1914c : SUCCESS in 7m 08s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/746e5b41531945eea67bcce2190afd75 : SUCCESS in 9m 40s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/9682a916f2d143aeacd9f6a64b1b6958 : SUCCESS in 1h 20m 39s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/15f8699b4c584a51ac4667cbf74c3d24 : SUCCESS in 2h 30m 22s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/21929131504b40e3a7c129b63860dbc0 : SUCCESS in 1h 22m 05s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/2038f4cb3d834b839ef138b90e3384b5 : SUCCESS in 2h 32m 23s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/66325789a3484a44a4da57992a94fe0e : SUCCESS in 1h 22m 34s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/e079ecee70f74db0b2c5d09201d800c2 : SUCCESS in 37m 13s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/16c1d0a2e6f24ba0805c68a3ec51355d : SUCCESS in 40m 48s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/82d8aee83ebf4140947ca8ad888d493a : SUCCESS in 37m 34s (non-voting)\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/b1a999f3c3ff4582a3a3f79772f8c1c8 : SUCCESS in 1h 14m 47s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3c7ef8d4820e4571804bbbe584686fc8 : SUCCESS in 44m 04s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/4b5ae7f8803f4404b2b52697a18af81a : TIMED_OUT in 2h 35m 12s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/65de0895f3e44bf8a91017d80f53ee77 : SUCCESS in 2h 19m 55s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/b635f11d32f141909029cceca213a965 : FAILURE in 33m 39s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/2ac0843e911a44a2abda5140f649b81a : TIMED_OUT in 2h 35m 03s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e9bfac520c879580fa97950a7ddd0cadfcc253ef","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-28 03:47:52.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"45bafdf9bac68a9a52c583c968c18223cb7638ec","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-28 05:27:01.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/de61c414eb6f450d9fd1275c18d48aca\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/889671abe8cd4468976dfe4b2a00f948 : SUCCESS in 7m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5630daaad41c49be97845e3179f63d5b : SUCCESS in 6m 35s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/582d390239ad497ab3fb1c2b32e4c142 : SUCCESS in 5m 04s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4833c3fd49cc4aada84fa0d7a4a16e7a : SUCCESS in 11m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/dc3be64ec3f04c289e4ab03a5b999ddd : SUCCESS in 4m 59s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/a551e6c9a2a84fef875a2dab1ba63a9b : SUCCESS in 4m 07s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/4c930eb9c4eb42a2bb934afe61e4f55b : SUCCESS in 8m 27s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/3d569d51cb0044f5a6fb8c54c0f29ec1 : SUCCESS in 1h 29m 15s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/3978f68c55124a0ea0c10576d47b9890 : SUCCESS in 1h 38m 38s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/3a39b81eeadd4663a2523d2ee903a630 : SUCCESS in 58m 16s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/597b8e16eaf64fb78d44c216d5ca9ecb : SUCCESS in 37m 17s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/924693d4274a4a54ab6be22c8288eebf : SUCCESS in 39m 53s","accounts_in_message":[],"_revision_number":1},{"id":"792a2ef6a8805ecc856f79c78877702a1ab662c7","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-28 05:27:01.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"5c1ebf5224c479e7842f821c777f623e85d2d611","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-28 05:28:19.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5bc7c745b5f04083a7ff417de8fde7f2\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/61730c1347f347e8b84b601bd90be9fa : SUCCESS in 54s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3742a414b2c24c5f811643c0a7975967 : SUCCESS in 48s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ff1b201d586e4c15a7d0f6e2eff9449bf6bfa2eb","revisions":{"ff1b201d586e4c15a7d0f6e2eff9449bf6bfa2eb":{"kind":"REWORK","_number":1,"created":"2023-11-14 16:55:52.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/25/900925/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/25/900925/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/25/900925/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/25/900925/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/25/900925/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/25/900925/1"}}},"commit":{"parents":[{"commit":"51ae11032e962c4bce957be8750d6bc6f9a9f1a5","subject":"Merge \"Run codespell on doc directory\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/51ae11032e962c4bce957be8750d6bc6f9a9f1a5"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-14 16:51:49.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-14 16:55:14.000000000","tz":540},"subject":"Honor connection_recycle_timeout in MysqlPersistenceDriver","message":"Honor connection_recycle_timeout in MysqlPersistenceDriver\n\nThe taskflow library allows us to customize idle_timeout. This change\nmakes the option set according to the equivalent option in oslo.db\nsimilarly to the other options such as max_overflow.\n\nChange-Id: I1c50f232c4f0c5c10a3dd5a928466f7ef67a9763\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ff1b201d586e4c15a7d0f6e2eff9449bf6bfa2eb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ff1b201d586e4c15a7d0f6e2eff9449bf6bfa2eb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
