)]}'
{"id":"openstack%2Foctavia~900997","triplet_id":"openstack%2Foctavia~stable%2F2023.1~I700c65fb17bad28b2b922e03d9c94c4716de9cbe","project":"openstack/octavia","branch":"stable/2023.1","topic":"add_noop_cert_manager-stable/2023.1","attention_set":{},"removed_from_attention_set":{"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2023-12-01 12:35:19.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I700c65fb17bad28b2b922e03d9c94c4716de9cbe","subject":"Add Noop Certificate Manager","status":"MERGED","created":"2023-11-15 17:01:26.000000000","updated":"2023-12-01 12:36:38.000000000","submitted":"2023-12-01 12:35:19.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"900997-add_noop_cert_manager-stable/2023.1","cherry_pick_of_change":894196,"cherry_pick_of_patch_set":9,"meta_rev_id":"84652c8e77339d58c41bcddc9909bb69c9e65d50","_number":900997,"virtual_id_number":900997,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-12-01 12:35:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"recommended":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"all":[{"value":1,"date":"2023-11-20 09:26:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":2,"date":"2023-12-01 09:35:18.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-11-20 12:38:00.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":1,"date":"2023-12-01 09:35:18.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-11-15 17:01:26.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"state":"CC"},{"updated":"2023-11-15 19:50:13.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-11-20 09:26:26.000000000","updated_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"reviewer":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"state":"REVIEWER"},{"updated":"2023-11-20 12:38:00.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"},{"updated":"2023-12-01 09:35:18.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"}],"messages":[{"id":"9ea6d1c7750dc0e94b07e2fa150ef7a1adcc5ed1","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-11-15 17:01:26.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"99199bb50910d8d2bf4adbf8bb72b9fd862c996f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-15 19:50:13.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/574f603eed4643199e654a6da852984a\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/4bfe80923d6745f489c7d73e9e66d1cd : SUCCESS in 2m 12s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1c7d843bf5494875923fb298eb331033 : SUCCESS in 11m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/556a27473be740908ae6323e211a9236 : SUCCESS in 5m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/bf226068e20b4e76a5d282840c593308 : SUCCESS in 4m 39s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/874cb364cf8d4460af4065d5dec44cc2 : SUCCESS in 14m 23s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/84ced7430f054be5bfbef18df3264609 : SUCCESS in 5m 46s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/af7897ed4ed1441cb4c854b2cd68dfa5 : SUCCESS in 8m 26s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/a107f4b7ed1c457fbff79ccde2071acb : SUCCESS in 7m 30s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/ff5ca3a99bae4c74bc2b68cfaccc3cb4 : SUCCESS in 8m 30s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/c8eac401a6bc4c0bbc978efdd6d5f38b : SUCCESS in 8m 10s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/0195514970db4bee96f0d61490e94d12 : SUCCESS in 8m 00s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/e96d55cfaf864899850846c52afd3a34 : SUCCESS in 9m 56s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/ea13791194874dcaa065e45b9ab0b940 : SUCCESS in 1h 27m 24s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/9d10dccbe04d4982b3a4b8786e55e6e4 : SUCCESS in 1h 32m 30s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/a48dabfe4b7244c09eb44f9d226205dc : SUCCESS in 1h 09m 21s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/c626c06cb4c5478192972b5082581dd7 : SUCCESS in 48m 54s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/474b773b5dd74c5a96bb5a1514868e58 : SUCCESS in 53m 26s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/e1a92e882ab34561835cb544e3559175 : SUCCESS in 1h 02m 57s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/e393b0b940484e89acbf2bb4f4940058 : SUCCESS in 1h 12m 16s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/54c4943159814419b2d2385624b4f25d : SUCCESS in 42m 39s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/5f03cedfa6ef4d7cb42aaf365c52da54 : SUCCESS in 1h 53m 57s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/92d1aba847ee4eb682384d97ed743e51 : SUCCESS in 2h 45m 26s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/5fbd52b0860047b09a0a1fb471713fc4 : SUCCESS in 2h 09m 16s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/04eec204707142168a6d604f361b87d4 : SUCCESS in 2h 07m 36s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/dd3126411d3d455b9ab0a2d893def3fc : SUCCESS in 2h 27m 52s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/9cea4054616b48feb9aad15cab9bea3b : SUCCESS in 2h 25m 17s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"4757fe72c0a080516a6df86281136dc814c3e760","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2023-11-20 09:26:26.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"34ef0c540f0642faf27dac9c4f3d84d223148e8f","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-11-20 12:38:00.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"c859a91499cc7ed597c3020adae158646cc9a934","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-12-01 09:35:18.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"4e9ec0eb86beb7459ad322fe8b9f57be0cc4b33e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-01 09:36:01.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"f9aafd67ec3cf76f85ffbd5ff907da16fc447b48","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-01 12:35:18.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6ecd9814901c48878f6dc7a35af000bc\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/567fceeebbb148bfb5b81f7df3bfa950 : SUCCESS in 2m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/4a2cb60deff8470b95f0103bb65d0434 : SUCCESS in 6m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/2d8eb5a1619b49008704a79601db757e : SUCCESS in 4m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/814eab6533a549e49b7a78b879dcf3c0 : SUCCESS in 11m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a02e380e84c14b15b5ddc6e57457cd41 : SUCCESS in 5m 14s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/1572db09fb5e4c26b63768a2069b2f06 : SUCCESS in 4m 36s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/01edfac27a5747d68bc060c466336eb2 : SUCCESS in 7m 37s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/86d9f9d7f6664baea6497271dd163fb6 : SUCCESS in 1h 31m 03s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/584bd77bd11043799828668586c99046 : SUCCESS in 1h 29m 34s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/2b07495740804ea484b259fad82356af : SUCCESS in 1h 07m 21s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/9e722a5464614e92a9967266efcf3b4f : SUCCESS in 41m 41s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/6cb47b9b4c4740cfa1ec55dc1f11a6e7 : SUCCESS in 48m 33s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/58ba64db790147f3bfeb4ef9a692c5ff : SUCCESS in 49m 01s","accounts_in_message":[],"_revision_number":1},{"id":"479456eb4ad12f7d2f1673f2a2aae101dffeffa4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-01 12:35:19.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"84652c8e77339d58c41bcddc9909bb69c9e65d50","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-01 12:36:38.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b5073df25280404eb5beea166aaf932f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d405306e69694bc9a3cd2fe03233f550 : SUCCESS in 1m 01s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3387e91298094d7d8a05a93ba3fbff7d : SUCCESS in 40s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"ec36152a03bd27a183d189d70effe3f095825e91","revisions":{"ec36152a03bd27a183d189d70effe3f095825e91":{"kind":"REWORK","_number":1,"created":"2023-11-15 17:01:26.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/97/900997/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/97/900997/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/97/900997/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/97/900997/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/97/900997/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/97/900997/1"}}},"commit":{"parents":[{"commit":"74499a036bd30c5fb895656c4ae762337ea9172a","subject":"Merge \"Fix incorrect removal of IP rules in the amphora\" into stable/2023.1","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/74499a036bd30c5fb895656c4ae762337ea9172a"}]}],"author":{"name":"Omer","email":"oschwart@redhat.com","date":"2023-09-07 13:42:36.000000000","tz":120},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-11-15 17:01:26.000000000","tz":0},"subject":"Add Noop Certificate Manager","message":"Add Noop Certificate Manager\n\nSo far, Octavia noop drivers were using real certificate managers, which\nhave validated the cerifitates for every certificate required operation,\nsometimes without any need.\n\nOctavia should have a Noop Certificate Manager for faster testing\npurposes.\n\nThis patch adds it.\n\nCloses-Bug: #2034711\n\nChange-Id: I700c65fb17bad28b2b922e03d9c94c4716de9cbe\n(cherry picked from commit 7310986de9bf68ed86de90de0501a1bc46945526)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ec36152a03bd27a183d189d70effe3f095825e91"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ec36152a03bd27a183d189d70effe3f095825e91"}]},"branch":"refs/heads/stable/2023.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
