)]}'
{"id":"openstack%2Foctavia~901435","triplet_id":"openstack%2Foctavia~master~I5698558857cbaa585f8a3d7ac37aaa31c0189d46","project":"openstack/octavia","branch":"master","attention_set":{},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2023-11-30 00:35:32.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I5698558857cbaa585f8a3d7ac37aaa31c0189d46","subject":"Fix health-monitors with ALPN members","status":"MERGED","created":"2023-11-20 08:03:21.000000000","updated":"2023-11-30 00:37:37.000000000","submitted":"2023-11-30 00:35:32.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"901435","meta_rev_id":"c53816b315da211e7aa8a9f55c29aac5ab1cfb1d","_number":901435,"virtual_id_number":901435,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":33408,"name":"Pierre-Yves Jourel","email":"pierre-yves.jourel@ovhcloud.com","username":"PYJ"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-11-30 00:35:32.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"recommended":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"all":[{"value":1,"date":"2023-11-20 15:43:14.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":1,"date":"2023-11-22 16:09:55.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":33408,"name":"Pierre-Yves Jourel","email":"pierre-yves.jourel@ovhcloud.com","username":"PYJ"},{"value":1,"date":"2023-11-29 06:06:05.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":2,"date":"2023-11-29 22:36:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-11-20 12:12:48.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":33408,"name":"Pierre-Yves Jourel","email":"pierre-yves.jourel@ovhcloud.com","username":"PYJ"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":1,"date":"2023-11-29 22:36:40.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":33408,"name":"Pierre-Yves Jourel","email":"pierre-yves.jourel@ovhcloud.com","username":"PYJ"},{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":2,"date":"2023-11-20 08:13:10.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"_account_id":33408,"name":"Pierre-Yves Jourel","email":"pierre-yves.jourel@ovhcloud.com","username":"PYJ"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-11-20 11:08:30.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-11-20 12:12:48.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"},{"updated":"2023-11-20 15:43:14.000000000","updated_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"reviewer":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"state":"REVIEWER"},{"updated":"2023-11-22 16:09:55.000000000","updated_by":{"_account_id":33408,"name":"Pierre-Yves Jourel","email":"pierre-yves.jourel@ovhcloud.com","username":"PYJ"},"reviewer":{"_account_id":33408,"name":"Pierre-Yves Jourel","email":"pierre-yves.jourel@ovhcloud.com","username":"PYJ"},"state":"REVIEWER"},{"updated":"2023-11-29 06:06:05.000000000","updated_by":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"reviewer":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"state":"REVIEWER"},{"updated":"2023-11-29 22:36:40.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"ebaeffbf43c8b077b970818c57c4b0a1762bcec0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-11-20 08:03:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"ecac094a99e0232e473352d989ab0548153665a6","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-11-20 08:13:10.000000000","message":"Patch Set 1: Backport-Candidate+2","accounts_in_message":[],"_revision_number":1},{"id":"5fd7084c918cffea45f2ed765ddb1d123071cc47","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-20 11:08:30.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f1f32a6d971d47788566bba02b9d8dbb\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b626e782adf04d6cbdf8d2dfae6d54c3 : SUCCESS in 9m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a6d522c067564d248a1ff625b7d624cf : SUCCESS in 5m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8114f706f24147cc94037f34f7f04e42 : SUCCESS in 4m 18s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/b7d5b11f16f447c0b18c69865dc2a2df : SUCCESS in 5m 16s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/4e2709e889e34f5aac21cef2c194ee2d : SUCCESS in 9m 57s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a495e27d5f814474bd7b6f37ebf01a19 : SUCCESS in 4m 47s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/4ee9d9301e9946f2a4610153022663dc : SUCCESS in 4m 36s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/1d1c6250693b43dc8e8a1cafa555dbdc : SUCCESS in 7m 43s\n- octavia-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/1ec525af02034e37b43b97bca45a089a : SUCCESS in 9m 36s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/7e44a2898ca34cc08a2132cdf194b546 : SUCCESS in 3m 55s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/1832541e552541cdaf30da7612c2627f : SUCCESS in 6m 41s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/1aeb8190d0614e3aae452e609ed51300 : SUCCESS in 7m 58s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/602e0f4e421549fba98c2114e57e923f : SUCCESS in 7m 19s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/97faa63239024d6fb9790d2f0230e0e3 : SUCCESS in 1h 13m 35s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/a4b179269c864914980bafc6682da3cc : SUCCESS in 1h 58m 41s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/60c70645fec247cf8c486b20a8eae725 : SUCCESS in 1h 10m 28s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/1e396bb1531248c4aad84f7db15d5731 : SUCCESS in 2h 11m 15s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/ec7c5a35a27a41078b6e5948759c5cb3 : SUCCESS in 1h 18m 26s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/5412b560102f4330b32c1d92fdea129f : SUCCESS in 47m 00s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/a58a6148477848189637c0244e8ebe28 : SUCCESS in 45m 36s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/b8c9c3c884f6444dbd55c4c804107fcf : SUCCESS in 45m 32s (non-voting)\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/c57b4408e3984e29a01e04dcf2355b90 : SUCCESS in 1h 09m 31s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d44d088b79334f8690eb12be766a4939 : SUCCESS in 37m 38s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/8a9b881a8e3347f6b3d32d99a74ece6a : TIMED_OUT in 2h 32m 57s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/1ddc3f3d77fb49d983ba87eda5b9ad96 : SUCCESS in 3h 02m 30s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/be01a28bffe64a83808a478f5b2c7c79 : SUCCESS in 2h 02m 06s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/9babc1af8b0141fd938c8bbd653a21a8 : SUCCESS in 2h 12m 39s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8ec437cbc783f9033716fc3b5ece2169bfb72ae7","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-11-20 12:12:48.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"eea7531b7c25a2aab8f1d02f47013fcd01b2f0ea","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2023-11-20 15:43:14.000000000","message":"Patch Set 1: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"a10233aa3053141816ac6655bd5ea47db486eb12","author":{"_account_id":33408,"name":"Pierre-Yves Jourel","email":"pierre-yves.jourel@ovhcloud.com","username":"PYJ"},"date":"2023-11-22 16:09:55.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"bd6c21eb53890f2772f4af31d6c8412ee158c2c4","author":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"date":"2023-11-29 06:06:05.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"b7f3cf6e31349fb1bd84ddf2a793832b30ca7436","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-11-29 22:36:40.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"757f8db15afd510afdaf6c4e73a2107c562ae6b2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-29 22:37:19.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"07f547e86bb4694040eecb99c1866eedaf300df9","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-30 00:35:32.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/86648e861ab84e18846801b36e96fa36\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5d90522a21244307826f1733cfe70ba9 : SUCCESS in 5m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/5d2fbdb57b404cd899c7151d75782d2a : SUCCESS in 4m 33s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/791fc3752c024ad084112fb493a12c32 : SUCCESS in 7m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9c632621e99a4f1f943b2d88f87ced18 : SUCCESS in 10m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c3db06516a5243dcae7a5ba1c2d37daf : SUCCESS in 4m 48s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/611527fa42a94b4b895e857cc758fd3c : SUCCESS in 5m 04s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/1836db8dff4545cc9689f3f2d740f23a : SUCCESS in 7m 38s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/389ebc79de564ec7a7414c234377449d : SUCCESS in 1h 18m 16s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/aa2c8dba338044e7a87b18e516285b95 : SUCCESS in 1h 45m 59s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/7e9f10317e8c4af083a28abc5dc49650 : SUCCESS in 1h 05m 13s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/4f5cf1ced7734904a7aa521f3d8d5238 : SUCCESS in 39m 42s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/85a08f884aa44de5809c8c378c9721d3 : SUCCESS in 38m 25s","accounts_in_message":[],"_revision_number":1},{"id":"c9deece12070ee895962df4db06f9490710657a3","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-30 00:35:32.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"c53816b315da211e7aa8a9f55c29aac5ab1cfb1d","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-30 00:37:37.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/806ab71fdcca405591a0d2cf672fdd00\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c03225cb523047e88badafc8d385dcb9 : SUCCESS in 51s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/12f4cda60489414db0451dab45f0ed02 : SUCCESS in 49s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"c28eb84dcafcb066746b04e9e6ef7a51decb68cd","revisions":{"c28eb84dcafcb066746b04e9e6ef7a51decb68cd":{"kind":"REWORK","_number":1,"created":"2023-11-20 08:03:21.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/35/901435/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/35/901435/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/35/901435/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/35/901435/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/35/901435/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/35/901435/1"}}},"commit":{"parents":[{"commit":"7310986de9bf68ed86de90de0501a1bc46945526","subject":"Add Noop Certificate Manager","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7310986de9bf68ed86de90de0501a1bc46945526"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-11-20 07:16:37.000000000","tz":-300},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-11-20 07:58:52.000000000","tz":-300},"subject":"Fix health-monitors with ALPN members","message":"Fix health-monitors with ALPN members\n\nUsing HTTP or HTTPS health-monitor on a ALPN pool failed with SSL\nerrors.\nhaproxy doc mentions that when using \"check\" with ALPN servers, the\ncheck-alpn option must be enabled.\n\n[0] https://docs.haproxy.org/2.8/configuration.html#5.2-check\n\nCloses-Bug: #2043812\n\nChange-Id: I5698558857cbaa585f8a3d7ac37aaa31c0189d46\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c28eb84dcafcb066746b04e9e6ef7a51decb68cd"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c28eb84dcafcb066746b04e9e6ef7a51decb68cd"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
