)]}'
{"id":"openstack%2Foctavia~901436","triplet_id":"openstack%2Foctavia~master~Icf75d5ea0b408e2f3e8647edfc8f6b8e5cf6d7f7","project":"openstack/octavia","branch":"master","attention_set":{},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2023-11-21 07:42:11.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Icf75d5ea0b408e2f3e8647edfc8f6b8e5cf6d7f7","subject":"Deprecate TLS-HELLO HMs","status":"ABANDONED","created":"2023-11-20 08:03:21.000000000","updated":"2023-11-21 07:42:11.000000000","total_comment_count":2,"unresolved_comment_count":1,"has_review_started":true,"meta_rev_id":"7e7daf523e4a1ab9d50e1dfd225a81e513c4ef14","_number":901436,"virtual_id_number":901436,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:check","value":1,"date":"2023-11-20 13:55:14.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"rejected":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"recommended":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"all":[{"value":1,"date":"2023-11-20 13:44:44.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"value":-2,"date":"2023-11-20 17:02:13.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2023-11-20 11:31:06.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-11-20 11:06:35.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-11-20 11:31:06.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"},{"updated":"2023-11-20 13:44:44.000000000","updated_by":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"reviewer":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"state":"REVIEWER"},{"updated":"2023-11-20 17:02:13.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"6c9f4e8a56616886064f0ca747ab4f145e039d99","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-11-20 08:03:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"219a9c564cc05ee9e318f52d2e1a052c9b13ee78","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-20 11:06:35.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6fe25052f30741d2b02afaf612ed52e3\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/0278b1bc40e444568bdd010b53cd4a26 : SUCCESS in 5m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bb9c355f948745e69a8681d0d71c9e3b : SUCCESS in 11m 32s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1f547cb07d7947bc98f45df5603b88dc : SUCCESS in 4m 52s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/cd71a6ad2d574471b5de3d59d03ed50d : SUCCESS in 5m 13s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f9cf330dcbd246f3b987d9c3449e4f27 : SUCCESS in 5m 48s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1f0388ab4a2849a59c45b8fc5af9a5d5 : SUCCESS in 11m 32s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fc123ec29db342e99369f7fa17592342 : SUCCESS in 5m 46s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/e5e86854ee6640ae83aff84b2f802c48 : SUCCESS in 5m 13s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/7d1e2e40164a42ba83fb9c0b146f05c5 : SUCCESS in 6m 37s\n- octavia-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/6c83f10163b44150b58ddcd8d1d6b46a : SUCCESS in 11m 16s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/9d1b014a90c649cb972c7783f0cb1db1 : SUCCESS in 4m 49s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/fb9d4aca29354e99b765b03488c51175 : SUCCESS in 6m 46s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/cca145b845044b21a4cde3244811aff5 : SUCCESS in 7m 03s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/8bc0fa57289846ea8ba75ca37c95b863 : SUCCESS in 6m 31s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/cee810c11c884cd0b0fc6555b36a2821 : SUCCESS in 1h 12m 51s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/612bc153a91e4c18b4dbf9245a70cd90 : SUCCESS in 1h 35m 10s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/9c6cf50db1c04d0ea047812c9f9f5ca7 : SUCCESS in 58m 46s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/79b08df0418a4e82bbdde2630cc23d0d : SUCCESS in 1h 40m 08s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/cc13c8245e6d47229f9f1feab01b8f51 : SUCCESS in 58m 09s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/bc5bb3d6f73d46038a3a6f23b5f16805 : SUCCESS in 37m 51s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/f8177c116f094695988cd28b9761088b : SUCCESS in 43m 28s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/eea3d93e6fea4042ad2f656ed1f60709 : SUCCESS in 47m 18s (non-voting)\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/8346bc1864244cc98ed33717cd41dc74 : SUCCESS in 1h 04m 14s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d2996af02b734c97b3970c7111d04d02 : SUCCESS in 35m 43s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/375ac855d7be4af195d125692ca13c7b : SUCCESS in 2h 14m 05s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/fbf889c7837f41b69defca51cc81d462 : SUCCESS in 2h 59m 02s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/4c9ca4be4b494d4f91a527f086db919a : SUCCESS in 2h 29m 11s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/a923de501d0a46b795bdceeaf5a24684 : TIMED_OUT in 2h 34m 30s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"9eed97624cfd43cdc985337e923928cf2c5e1519","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-11-20 11:15:04.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"47b959dbc2564d4b79f684c4b6a933947fac40e6","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-11-20 11:31:06.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"61af2104c9e0f597e1599809616aab4ba8d36975","author":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"date":"2023-11-20 13:44:44.000000000","message":"Patch Set 2: Code-Review+1","accounts_in_message":[],"_revision_number":2},{"id":"621fd84463677c341c8eba63db8dd9af04b0cd49","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-20 13:55:14.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/91c3b924a47d4d8d86285669e4c8d813\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/5ba8e7700b134ca182c98cb9839990d7 : SUCCESS in 4m 58s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/2459f2bdbcb84344911724f01d60df8d : SUCCESS in 10m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/749751bc036e42188e7f5dc27996b875 : SUCCESS in 5m 24s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/7457caf0b07a47ff94e5f618a7b95553 : SUCCESS in 4m 31s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/5a37d993ca2b47edbd49a40764fd7eec : SUCCESS in 4m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/18d3cf4738814ea19447c4170ca6c00f : SUCCESS in 12m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/3343a68252154f58b4b4a8ea4d396030 : SUCCESS in 5m 20s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/2536d5e4a82045c9b088a9c42b02eb8b : SUCCESS in 5m 03s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/30d45c4baad54c7184cad4d3bc1e09ff : SUCCESS in 7m 08s\n- octavia-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/a0bb634dd83f4b26be2424bc6b800a53 : SUCCESS in 9m 23s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/e584ebea284c487096047f44d84f157c : SUCCESS in 4m 25s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/a80151e1166c4b22ade17145a2695c53 : SUCCESS in 6m 36s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/6ddcac1e9eaa430e8f9ce854a7e79409 : SUCCESS in 8m 08s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/94c692b7f3c746fd89a178939af692e5 : SUCCESS in 7m 07s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/bffd263d91194802b28b4984342763e8 : SUCCESS in 1h 23m 12s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/62e71ece00af431fbe787d60b3503288 : SUCCESS in 1h 33m 07s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/306797b300da409d9c231e009b2820eb : SUCCESS in 1h 07m 34s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/ca6d6c26163245ab8ba3e3574d7c4a4b : SUCCESS in 1h 43m 32s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/2138fc0a87d64d79bf537f37fb2e365f : SUCCESS in 1h 01m 57s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/c2f0b6c4faf24783b8a2a3841be31720 : SUCCESS in 36m 29s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/90670a5df1124433b1d7fb03fa78b81d : SUCCESS in 41m 06s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/3188027610db4849a8043932e0e9c819 : SUCCESS in 37m 38s (non-voting)\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/811470008a8d4ad4822e9e39a5581b80 : SUCCESS in 1h 00m 00s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/53e75e7bb87549429013ab3b4c225647 : SUCCESS in 36m 00s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/5f761a536463425b9a9622244f257d6d : SUCCESS in 2h 06m 51s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/c1ab36a5eb8d4e6eb53e81cb5e435f90 : SUCCESS in 2h 38m 48s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/ccde2a0a594d45e7b9be882ee6fce21c : SUCCESS in 2h 02m 27s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/c4d8cde097344116be501d3bc03c6998 : SUCCESS in 2h 09m 26s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"78258525b9b2984ec4f855047652ece0fbf179d4","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-11-20 17:02:13.000000000","message":"Patch Set 2: Code-Review-2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7e7daf523e4a1ab9d50e1dfd225a81e513c4ef14","tag":"autogenerated:gerrit:abandon","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-11-21 07:42:11.000000000","message":"Abandoned\n\nreplaced by https://review.opendev.org/c/openstack/octavia/+/901524","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"fc1c84ad9c81c5f33fa4aef20dc9e0fc5834b9c6","revisions":{"b09cefcb5ffa87d5edf6356b75cb906e6ba0ee53":{"kind":"REWORK","_number":1,"created":"2023-11-20 08:03:21.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/36/901436/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/36/901436/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/36/901436/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/36/901436/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/36/901436/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/36/901436/1"}}},"commit":{"parents":[{"commit":"c28eb84dcafcb066746b04e9e6ef7a51decb68cd","subject":"Fix health-monitors with ALPN members","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c28eb84dcafcb066746b04e9e6ef7a51decb68cd"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-11-20 07:44:35.000000000","tz":-300},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-11-20 07:59:13.000000000","tz":-300},"subject":"Deprecate TLS-HELLO HMs","message":"Deprecate TLS-HELLO HMs\n\nTLS-HELLO HMs are based on SSLv3 messages and don\u0027t work when SSLv3 is\ndisabled (and most distributions have disabled SSLv3)\n\nRelated-Bug: #2043812\nChange-Id: Icf75d5ea0b408e2f3e8647edfc8f6b8e5cf6d7f7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b09cefcb5ffa87d5edf6356b75cb906e6ba0ee53"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b09cefcb5ffa87d5edf6356b75cb906e6ba0ee53"}]},"branch":"refs/heads/master"},"fc1c84ad9c81c5f33fa4aef20dc9e0fc5834b9c6":{"kind":"REWORK","_number":2,"created":"2023-11-20 11:15:04.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/36/901436/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/36/901436/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/36/901436/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/36/901436/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/36/901436/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/36/901436/2"}}},"commit":{"parents":[{"commit":"c28eb84dcafcb066746b04e9e6ef7a51decb68cd","subject":"Fix health-monitors with ALPN members","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c28eb84dcafcb066746b04e9e6ef7a51decb68cd"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-11-20 07:44:35.000000000","tz":-300},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-11-20 11:11:11.000000000","tz":-300},"subject":"Deprecate TLS-HELLO HMs","message":"Deprecate TLS-HELLO HMs\n\nTLS-HELLO HMs are based on SSLv3 messages and don\u0027t work when SSLv3 is\ndisabled (and most distributions have disabled SSLv3)\n\nRelated-Bug: #2043812\nChange-Id: Icf75d5ea0b408e2f3e8647edfc8f6b8e5cf6d7f7\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fc1c84ad9c81c5f33fa4aef20dc9e0fc5834b9c6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fc1c84ad9c81c5f33fa4aef20dc9e0fc5834b9c6"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":["label:Code-Review\u003dMIN"],"failing_atoms":["label:Code-Review\u003dMAX"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
