)]}'
{"id":"openstack%2Foctavia~901595","triplet_id":"openstack%2Foctavia~master~Icb919f5f72a16713c1cee9d610a6a5fa90b2686b","project":"openstack/octavia","branch":"master","topic":"bug/2044283","attention_set":{"28691":{"account":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"last_update":"2023-11-27 13:47:07.000000000","reason":"\u003cGERRIT_ACCOUNT_35963\u003e replied on the change","reason_account":{"_account_id":35963,"name":"Lê Minh Thư","display_name":"Le Minh Thu","email":"thultm9@viettel.com.vn","username":"thultm9"}},"36379":{"account":{"_account_id":36379,"name":"Nguyễn Ngọc Hiếu","display_name":"ZiTechDev","email":"zitech.dev@gmail.com","username":"zitechdev201"},"last_update":"2023-11-27 13:47:42.000000000","reason":"\u003cGERRIT_ACCOUNT_35963\u003e replied on the change","reason_account":{"_account_id":35963,"name":"Lê Minh Thư","display_name":"Le Minh Thu","email":"thultm9@viettel.com.vn","username":"thultm9"}},"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2023-11-22 16:46:07.000000000","reason":"\u003cGERRIT_ACCOUNT_35963\u003e replied on the change","reason_account":{"_account_id":35963,"name":"Lê Minh Thư","display_name":"Le Minh Thu","email":"thultm9@viettel.com.vn","username":"thultm9"}},"1004":{"account":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"last_update":"2023-11-22 16:46:07.000000000","reason":"\u003cGERRIT_ACCOUNT_35963\u003e replied on the change","reason_account":{"_account_id":35963,"name":"Lê Minh Thư","display_name":"Le Minh Thu","email":"thultm9@viettel.com.vn","username":"thultm9"}},"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2023-11-22 16:46:07.000000000","reason":"\u003cGERRIT_ACCOUNT_35963\u003e replied on the change","reason_account":{"_account_id":35963,"name":"Lê Minh Thư","display_name":"Le Minh Thu","email":"thultm9@viettel.com.vn","username":"thultm9"}}},"removed_from_attention_set":{"35963":{"account":{"_account_id":35963,"name":"Lê Minh Thư","display_name":"Le Minh Thu","email":"thultm9@viettel.com.vn","username":"thultm9"},"last_update":"2023-11-27 13:47:07.000000000","reason":"\u003cGERRIT_ACCOUNT_35963\u003e replied on the change","reason_account":{"_account_id":35963,"name":"Lê Minh Thư","display_name":"Le Minh Thu","email":"thultm9@viettel.com.vn","username":"thultm9"}}},"hashtags":[],"change_id":"Icb919f5f72a16713c1cee9d610a6a5fa90b2686b","subject":"Fix duplicate specified VIP among load balancers","status":"NEW","created":"2023-11-22 16:44:15.000000000","updated":"2023-11-27 13:47:42.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":true,"submittable":false,"total_comment_count":8,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"ebd291eb5cb1107c5e8db2d6e5c0894062cde564","_number":901595,"virtual_id_number":901595,"owner":{"_account_id":35963,"name":"Lê Minh Thư","display_name":"Le Minh Thu","email":"thultm9@viettel.com.vn","username":"thultm9"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":36379,"name":"Nguyễn Ngọc Hiếu","display_name":"ZiTechDev","email":"zitech.dev@gmail.com","username":"zitechdev201"},{"tag":"autogenerated:zuul:check","value":1,"date":"2023-11-23 18:13:11.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"recommended":{"_account_id":36379,"name":"Nguyễn Ngọc Hiếu","display_name":"ZiTechDev","email":"zitech.dev@gmail.com","username":"zitechdev201"},"all":[{"value":1,"date":"2023-11-24 16:07:01.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":36379,"name":"Nguyễn Ngọc Hiếu","display_name":"ZiTechDev","email":"zitech.dev@gmail.com","username":"zitechdev201"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":36379,"name":"Nguyễn Ngọc Hiếu","display_name":"ZiTechDev","email":"zitech.dev@gmail.com","username":"zitechdev201"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":36379,"name":"Nguyễn Ngọc Hiếu","display_name":"ZiTechDev","email":"zitech.dev@gmail.com","username":"zitechdev201"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"}],"REVIEWER":[{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":36379,"name":"Nguyễn Ngọc Hiếu","display_name":"ZiTechDev","email":"zitech.dev@gmail.com","username":"zitechdev201"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-11-22 16:46:07.000000000","updated_by":{"_account_id":35963,"name":"Lê Minh Thư","display_name":"Le Minh Thu","email":"thultm9@viettel.com.vn","username":"thultm9"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2023-11-22 16:46:07.000000000","updated_by":{"_account_id":35963,"name":"Lê Minh Thư","display_name":"Le Minh Thu","email":"thultm9@viettel.com.vn","username":"thultm9"},"reviewer":{"_account_id":1004,"name":"Mohammed Naser","email":"mnaser@vexxhost.com","username":"mnaser"},"state":"REVIEWER"},{"updated":"2023-11-22 16:46:07.000000000","updated_by":{"_account_id":35963,"name":"Lê Minh Thư","display_name":"Le Minh Thu","email":"thultm9@viettel.com.vn","username":"thultm9"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2023-11-22 19:22:51.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-11-23 01:53:30.000000000","updated_by":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"reviewer":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"state":"CC"},{"updated":"2023-11-24 16:07:01.000000000","updated_by":{"_account_id":36379,"name":"Nguyễn Ngọc Hiếu","display_name":"ZiTechDev","email":"zitech.dev@gmail.com","username":"zitechdev201"},"reviewer":{"_account_id":36379,"name":"Nguyễn Ngọc Hiếu","display_name":"ZiTechDev","email":"zitech.dev@gmail.com","username":"zitechdev201"},"state":"REVIEWER"}],"messages":[{"id":"4cf15d3b59fe928ff6ed278a69bf25ae822cdca5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35963,"name":"Lê Minh Thư","display_name":"Le Minh Thu","email":"thultm9@viettel.com.vn","username":"thultm9"},"date":"2023-11-22 16:44:15.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"d389bd90f0c53d21a21e6e7a506e5e723450bd88","author":{"_account_id":35963,"name":"Lê Minh Thư","display_name":"Le Minh Thu","email":"thultm9@viettel.com.vn","username":"thultm9"},"date":"2023-11-22 16:46:07.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"20b4cd7993a75de101093b57f431028c2c94e687","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-22 19:22:51.000000000","message":"Patch Set 1: Verified-1\n\n(3 comments)\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/226f0de7d47a445db826897c16d1a908\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/bf8eff74a3764c6e8f99484067254c03 : SUCCESS in 12m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ed669350d31c4c0785b676267fb389f8 : FAILURE in 4m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/96689a6344b44fc0963443253458609a : SUCCESS in 4m 21s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4ae93faade9c432d8b21a8043cbba282 : SUCCESS in 5m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/6777bc6a91f745059294b69f5106ea81 : SUCCESS in 10m 37s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/5ef5f63885554453bbdbf909e58c4435 : SUCCESS in 4m 37s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/4956281a718a4b0aa00db2112a8ceb52 : SUCCESS in 7m 07s\n- octavia-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/59d5e3365b2a4bc4b782115530d87b71 : SUCCESS in 11m 50s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/3d49ff35b027402d91495af8ca99b928 : SUCCESS in 4m 45s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/c35885de1ab94ef9ba0939d0e41b37c7 : SUCCESS in 8m 37s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/1678da2dae6849a3b1c7fd879f57e174 : SUCCESS in 8m 30s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/773773e2c12a4eb5a438f4cfd02329c9 : SUCCESS in 6m 14s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/e23fb3e4351f4f90817c810d62400c9b : SUCCESS in 1h 50m 18s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/86bdef352ac84240af1289ead088c651 : SUCCESS in 1h 38m 18s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/d417597c98f5451ebeab0e7b31d5f49f : SUCCESS in 1h 01m 58s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/9a3a5850652948e6a56166cb4b18d32c : SUCCESS in 1h 42m 32s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/bf3664b269894c158780d56f6c7a5ff1 : SUCCESS in 1h 03m 33s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/2586740809154c02bf0ab9f3c3462dba : SUCCESS in 39m 28s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/fd081ef298584a11b54e4023d1c15fe7 : SUCCESS in 41m 50s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/357f42472d2c45b8891335e4195aecb6 : SUCCESS in 40m 51s (non-voting)\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/4ef9df79fade41f5a06555df0a62092c : SUCCESS in 54m 36s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b65a0d59bacb4901a57cdd3cfdc9a904 : SUCCESS in 32m 17s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/9bc15ac6797e4c5d8666681e2b72a32e : FAILURE in 22m 14s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/f9eb65677b3545d081a756a6bd6450d2 : SUCCESS in 2h 37m 15s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/a4ecaba9cd984f3c8720a8229e71946a : TIMED_OUT in 2h 33m 16s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/c04635f3047544a09c6005cfca003c9b : TIMED_OUT in 2h 34m 55s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"52a4a54ee8c39c414e1c63ad5ecb60b5747b1e6b","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2023-11-23 01:53:30.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6812fa2ada45ab1e1ef5a95a8de5a3a58db2bb39","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":35963,"name":"Lê Minh Thư","display_name":"Le Minh Thu","email":"thultm9@viettel.com.vn","username":"thultm9"},"date":"2023-11-23 14:53:40.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"dbca464cd2c127b289a9fcf2c4369281da5169bb","author":{"_account_id":35963,"name":"Lê Minh Thư","display_name":"Le Minh Thu","email":"thultm9@viettel.com.vn","username":"thultm9"},"date":"2023-11-23 15:24:01.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"336ea9b190be456e4dfbd97c2d1d72f0f3ee0053","author":{"_account_id":35963,"name":"Lê Minh Thư","display_name":"Le Minh Thu","email":"thultm9@viettel.com.vn","username":"thultm9"},"date":"2023-11-23 15:28:02.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"dfb228ace8d694e96ca4dd6504cd4a56c2d8c2a3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-23 18:13:11.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/613aca6434cf454cb35e29a1dba67b9b\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/aba38fe748f642bf90c5a08b49af320c : SUCCESS in 11m 22s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/847bf004efde4a2fb9efd1583f63c848 : SUCCESS in 5m 03s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d6a4b506af37471aaaf099349bdf2667 : SUCCESS in 6m 17s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/06f59a5d40f34b32a78c5a8f7b6c9d29 : SUCCESS in 5m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0b847df9df074f369ea5bc242b54e9fd : SUCCESS in 10m 38s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/b3e917e31ecf4ca0b04047963734e8ff : SUCCESS in 5m 29s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/21162cc12e584cd59ad4f9b389d2fe9e : SUCCESS in 6m 51s\n- octavia-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/2aa5b4b19cb044b9b46f5f9382f84436 : SUCCESS in 10m 23s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/56f631c14c074feda1c7fe5cac554306 : SUCCESS in 4m 18s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/28b70b58e5d7471ea8da7ea13125182a : SUCCESS in 7m 31s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/d741903288f6420082b616f7de9cd03c : SUCCESS in 6m 59s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/f8548be7b0f44d6d91fd3e26043a3009 : SUCCESS in 7m 20s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/db621ff3ede34c8880b2e3ac1a5fef91 : SUCCESS in 1h 02m 08s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/4070fabf9b1844859a5b811433f933b1 : SUCCESS in 1h 53m 53s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/7158a30030d14a3a976b3873f5527502 : SUCCESS in 1h 08m 35s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/7add5599966d420596df351bb226af2b : SUCCESS in 1h 49m 44s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/988af75e1b4b4b30b7ed7b4b59579ecc : SUCCESS in 1h 22m 24s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/adb0ad6de76f4b6c8f73e1cfd03cadc7 : SUCCESS in 39m 49s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/21b4756d77694c0bb61b6c79178c5cbc : SUCCESS in 39m 06s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/d079a8ed1f7245218778b8caeba027da : SUCCESS in 51m 09s (non-voting)\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/72797b97ea9d483788c84731749ee5b6 : SUCCESS in 53m 26s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/527aa357e6054eb4ac3955cb68f3c302 : SUCCESS in 34m 22s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/867b10736bd9460ca7366b470b8867df : SUCCESS in 2h 13m 22s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/378b64d518a14e73bb0cb4edba51656b : TIMED_OUT in 3h 02m 51s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/4cc1887c96e246078799096581c75991 : TIMED_OUT in 2h 34m 05s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/654a92ce36a544cb8a74bc2f2e61186e : TIMED_OUT in 2h 34m 14s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"8f067ff234d001701c062ddec53c504af02b1bb3","author":{"_account_id":28691,"name":"Bo Tran","email":"ministry.96.nd@gmail.com","username":"ministry"},"date":"2023-11-24 01:57:03.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"c2f8f8e598e5d6ca3ec5146a11bbb3321bb98e72","author":{"_account_id":36379,"name":"Nguyễn Ngọc Hiếu","display_name":"ZiTechDev","email":"zitech.dev@gmail.com","username":"zitechdev201"},"date":"2023-11-24 16:07:01.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"b6d0ac0c5346871a7b74def6d7d5e471107c299d","author":{"_account_id":35963,"name":"Lê Minh Thư","display_name":"Le Minh Thu","email":"thultm9@viettel.com.vn","username":"thultm9"},"date":"2023-11-27 13:47:07.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"ebd291eb5cb1107c5e8db2d6e5c0894062cde564","author":{"_account_id":35963,"name":"Lê Minh Thư","display_name":"Le Minh Thu","email":"thultm9@viettel.com.vn","username":"thultm9"},"date":"2023-11-27 13:47:42.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"64df65daa0cb23294f5497226c139499f050017b","revisions":{"b977aab08b1fd5f30feb459dcb7b5cc036e75c2f":{"kind":"REWORK","_number":1,"created":"2023-11-22 16:44:15.000000000","uploader":{"_account_id":35963,"name":"Lê Minh Thư","display_name":"Le Minh Thu","email":"thultm9@viettel.com.vn","username":"thultm9"},"ref":"refs/changes/95/901595/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/95/901595/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/95/901595/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/95/901595/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/95/901595/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/95/901595/1"}}},"commit":{"parents":[{"commit":"7310986de9bf68ed86de90de0501a1bc46945526","subject":"Add Noop Certificate Manager","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7310986de9bf68ed86de90de0501a1bc46945526"}]}],"author":{"name":"thultm9","email":"thultm9@viettel.com.vn","date":"2023-11-22 16:14:22.000000000","tz":420},"committer":{"name":"thultm9","email":"thultm9@viettel.com.vn","date":"2023-11-22 16:43:46.000000000","tz":420},"subject":"Fix duplicate specified VIP among load balancers","message":"Fix duplicate specified VIP among load balancers\n\nWhen create lb with specified vip port, update device_id of this port in order not to use it for creating other load balancer while it is attached.\n\nCloses-Bug: #2044283\n\nChange-Id: Icb919f5f72a16713c1cee9d610a6a5fa90b2686b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b977aab08b1fd5f30feb459dcb7b5cc036e75c2f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b977aab08b1fd5f30feb459dcb7b5cc036e75c2f"}]},"branch":"refs/heads/master"},"64df65daa0cb23294f5497226c139499f050017b":{"kind":"REWORK","_number":2,"created":"2023-11-23 14:53:40.000000000","uploader":{"_account_id":35963,"name":"Lê Minh Thư","display_name":"Le Minh Thu","email":"thultm9@viettel.com.vn","username":"thultm9"},"ref":"refs/changes/95/901595/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/95/901595/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/95/901595/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/95/901595/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/95/901595/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/95/901595/2"}}},"commit":{"parents":[{"commit":"7310986de9bf68ed86de90de0501a1bc46945526","subject":"Add Noop Certificate Manager","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7310986de9bf68ed86de90de0501a1bc46945526"}]}],"author":{"name":"thultm9","email":"thultm9@viettel.com.vn","date":"2023-11-22 16:14:22.000000000","tz":420},"committer":{"name":"thultm9","email":"thultm9@viettel.com.vn","date":"2023-11-23 14:51:28.000000000","tz":420},"subject":"Fix duplicate specified VIP among load balancers","message":"Fix duplicate specified VIP among load balancers\n\nWhen create lb with specified vip port, update device_id of this port in order not to use it for creating other load balancer while it is attached.\n\nCloses-Bug: #2044283\n\nChange-Id: Icb919f5f72a16713c1cee9d610a6a5fa90b2686b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/64df65daa0cb23294f5497226c139499f050017b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/64df65daa0cb23294f5497226c139499f050017b"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":36379,"name":"Nguyễn Ngọc Hiếu","display_name":"ZiTechDev","email":"zitech.dev@gmail.com","username":"zitechdev201"}},{"label":"Workflow","status":"MAY"},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
