)]}'
{"id":"openstack%2Foctavia~901689","triplet_id":"openstack%2Foctavia~master~I06911f42b9bf29cf9a5f2e76d8333d8a2f1bc60b","project":"openstack/octavia","branch":"master","attention_set":{},"removed_from_attention_set":{"6577":{"account":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"last_update":"2023-12-01 07:25:43.000000000","reason":"\u003cGERRIT_ACCOUNT_6577\u003e replied on the change","reason_account":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"}},"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2024-01-06 15:54:20.000000000","reason":"Change was submitted"},"34429":{"account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"last_update":"2024-01-06 14:03:50.000000000","reason":"\u003cGERRIT_ACCOUNT_34429\u003e replied on the change","reason_account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}}},"hashtags":[],"change_id":"I06911f42b9bf29cf9a5f2e76d8333d8a2f1bc60b","subject":"Fix issue with certificates with no subject or CN","status":"MERGED","created":"2023-11-22 21:50:17.000000000","updated":"2024-01-06 15:55:28.000000000","submitted":"2024-01-06 15:54:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":13,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"901689","meta_rev_id":"1534ed082cf2869aae8419a74e7c9ac8fbddbc63","_number":901689,"virtual_id_number":901689,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-01-06 15:54:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"recommended":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"all":[{"value":1,"date":"2024-01-05 19:33:57.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":1,"date":"2023-12-01 07:25:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-12-01 09:16:56.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":2,"date":"2024-01-06 14:03:50.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":1,"date":"2024-01-06 14:03:50.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"recommended":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":1,"date":"2023-12-01 07:25:43.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","value":1,"default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-11-23 01:54:52.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-11-23 06:18:50.000000000","updated_by":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"reviewer":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"state":"REVIEWER"},{"updated":"2023-11-29 17:07:43.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"CC"},{"updated":"2023-11-29 17:11:24.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"},{"updated":"2023-12-01 09:16:56.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2024-01-05 19:33:57.000000000","updated_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"reviewer":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"state":"REVIEWER"}],"messages":[{"id":"e2b78f7feb23769b210dbf254a21647dc8a79d17","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-11-22 21:50:17.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"4f4e5b0f211d54370c07232e2eae6875c1531256","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-11-22 21:50:31.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"bffc9209c110e3a21eb5ed80bd830b57774a9d4d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-11-22 22:49:21.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Workflow-1\n","accounts_in_message":[],"_revision_number":2},{"id":"d13ee3094f8d6053d0c1b6ae95184c3f9494ec30","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-23 01:54:52.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/023efa6559634bdc891c96f2b7989aaf\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b316d38c8caa4239b21bd041c68be34f : SUCCESS in 8m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/94cfd45bc0c34ab1a42563a70bb54aca : SUCCESS in 4m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/a342f0a5945f4e5099fc401185ee8479 : SUCCESS in 4m 16s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a3debcee1768429d9dad990c01876c0c : SUCCESS in 4m 51s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f9e3142ebf9546df9dc9a43acbf6b286 : SUCCESS in 11m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/7b576d48df16425486c4249a3d65048e : SUCCESS in 5m 17s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/9f0e4e87183049f1a8edc872115be0cd : SUCCESS in 5m 13s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/13fccc8ba7ac49c1a4e320a0d3e9954a : SUCCESS in 8m 13s\n- octavia-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/cd041bb80c76418b8b11de95f48ed121 : SUCCESS in 9m 32s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/0896a7b360974adf80ccf25e33edc41a : SUCCESS in 4m 14s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/fa217d9354184588b0f9288d52b87503 : SUCCESS in 6m 08s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/5da39e277f8f4d38a103251c635ed7e2 : SUCCESS in 6m 37s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/12769228b09542cba6ba6a35bfdc8b15 : SUCCESS in 6m 59s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/e5ee6a05e11548afb5c4af6c468d5552 : SUCCESS in 1h 08m 00s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/016e78da9cb642f2a1d2bc44659cdef2 : SUCCESS in 1h 40m 41s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/e025e2067b714097ac620672200d71a2 : SUCCESS in 56m 38s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/7a7777e917fa49b582e65c93f27fa5f3 : SUCCESS in 1h 44m 41s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/cdd5d3b5da5e4e65beb96bd61ed20575 : SUCCESS in 1h 00m 49s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/59c73f9560494c78a273c120d2bef6c6 : SUCCESS in 39m 50s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/cefbfecb72d94e9db33e3d5306a8ce11 : SUCCESS in 39m 12s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/fbe9864b46924d6fa231cb2086538ab1 : SUCCESS in 38m 35s (non-voting)\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/ac7923b1defe44dea65fb611e3cff0fc : SUCCESS in 1h 01m 28s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ff0782be7c284ffea3adb9f1af5c4228 : SUCCESS in 34m 29s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/2e886c3922a347b78c68582b0565b388 : SUCCESS in 2h 20m 41s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/765906dbb4d34511b6307607157d3a39 : TIMED_OUT in 3h 02m 42s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/7503d04f666047c6b3b43fb840ca788b : SUCCESS in 2h 28m 01s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/018d9a07803d47df9feecefd2c38b78f : SUCCESS in 2h 12m 03s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"918a12491a8eb2f3de98b41e4c82a11434e411db","author":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"date":"2023-11-23 06:18:50.000000000","message":"Patch Set 2: Backport-Candidate+1 Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"da98c9dfba59d5fe04fe57dc309aaadda903942a","author":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"date":"2023-11-23 06:20:07.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"647b0fa0f23dd05214fa39b11296bcdb6b986efa","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-11-28 22:25:17.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"cda121c7abc33e34f8ac717d09588195a28bb1ef","author":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"date":"2023-11-29 03:30:07.000000000","message":"Patch Set 2:\n\n(2 comments)","accounts_in_message":[],"_revision_number":2},{"id":"a1f09386ff0954f81699066502c8bf7c45da5c3a","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-11-29 17:07:43.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"857daac99e258ab5c0ab1b870776cb4cc8d59700","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-11-29 17:11:24.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"180d2c1a7d2d22cf9d4c42a720caef8c44879d87","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-12-01 00:36:09.000000000","message":"Uploaded patch set 3.\n\nOutdated Votes:\n* Backport-Candidate+1 (copy condition: \"changekind:NO_CODE_CHANGE OR changekind:TRIVIAL_REBASE\")\n* Code-Review+1, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"0f326d9c669e6d640d205b88ba830cae88792c94","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-12-01 00:49:35.000000000","message":"Patch Set 3:\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"1396ee4743e42bb4956c707791b18188bba1ad91","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-01 03:30:09.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c8fabe63293e478290ebbacc190cfb05\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/034ec79cde0f425cb22bd7ab31c1f636 : SUCCESS in 11m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e57615b0fd6544c3aab68f1f3c8ef196 : SUCCESS in 5m 01s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b105916daf2e4962874b4d5cd31a0ea0 : SUCCESS in 4m 59s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/42ab1098c0a24820af40d5edbd4ee606 : SUCCESS in 5m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/23ab1c764d154314b7172a917db2dc21 : SUCCESS in 11m 21s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/2401025f0e4046938d47dcca8e610da9 : SUCCESS in 4m 45s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/d5f6fb531b994c599327c666566dbc1e : SUCCESS in 5m 02s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/7ba90aa0cd954a28bdcbecd2446b6882 : SUCCESS in 6m 47s\n- octavia-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/cfd128e344004253a4168b407818271a : SUCCESS in 11m 16s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/19fbe78696e74c848d85dcabe5d74328 : SUCCESS in 4m 41s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/fad412a0d2b54b20bccd3c9247415c9d : SUCCESS in 7m 25s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/2a94459b27ba498c80eac021a963c476 : SUCCESS in 6m 26s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/c2342db197b04394b5309ea799c1a6bc : SUCCESS in 7m 41s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/fefffa3757dc461aa475ab40e480a6ac : SUCCESS in 1h 14m 12s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/85e3b175f93f4e45b3b00858b0435c4a : SUCCESS in 1h 40m 08s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/b9c38b754cfd4a0b99dc851868055305 : SUCCESS in 1h 02m 15s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/68657317a00f4c6ba58493e9d76c62e5 : SUCCESS in 1h 40m 38s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/a5a4e3a962c348fbb0447345cd724c23 : SUCCESS in 56m 00s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/ae1bf38a99d44af1b614039064952fdd : SUCCESS in 38m 38s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/be70ac087f7741639272a53b2febc4b3 : SUCCESS in 38m 07s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/aebf12f923f944b4931e4803a3595856 : SUCCESS in 48m 58s (non-voting)\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/5baaa69a7112476080f7e39b5bf94d42 : SUCCESS in 1h 04m 01s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/cc0f95700dbf4cf3a83c4ca5c17869a4 : SUCCESS in 32m 04s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/ad0f04e886e54c2386aa0ebab6f8afa5 : SUCCESS in 2h 07m 06s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/194818bb23374b6a9e0a28c7d4d6168a : FAILURE in 2h 51m 56s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/339184f5cacf4ac19e462d9823a827af : SUCCESS in 2h 09m 15s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/3d08b8e4b213448ab626aa148576e936 : TIMED_OUT in 2h 34m 50s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"df16c9594778a7d5711770c161f26227356a0d49","author":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"date":"2023-12-01 07:25:43.000000000","message":"Patch Set 3: Backport-Candidate+1 Code-Review+1","accounts_in_message":[],"_revision_number":3},{"id":"28b68b004626a85f1ee9679d84df4fa3db8e0690","author":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"date":"2023-12-01 07:27:27.000000000","message":"Patch Set 3:\n\n(2 comments)","accounts_in_message":[],"_revision_number":3},{"id":"fb8883df81b2f76460d09c584893b82e4550c54b","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-12-01 09:16:56.000000000","message":"Patch Set 3: Code-Review+2","accounts_in_message":[],"_revision_number":3},{"id":"daa1a49840a3fbf58e16cc06a3b252deef316536","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2024-01-05 19:33:57.000000000","message":"Patch Set 3: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":3},{"id":"74f8c3697789627e213ea455b0e58c06565eab62","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-01-06 14:03:50.000000000","message":"Patch Set 3: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":3},{"id":"8e44fe740c395cc2cfcaf11e9aef4165625eb53e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-06 14:04:30.000000000","message":"Patch Set 3: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":3},{"id":"19a71caca122c3fe4d8e0576f1e1b470b39751c6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-06 15:54:20.000000000","message":"Patch Set 3: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/70d8e4bbb31d4135af94e67ea8d3c999\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/56904dec587b4955bf655aa1d83125fe : SUCCESS in 5m 29s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f79f9b3c23774d08b49974d3f3e1bcb9 : SUCCESS in 4m 51s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/fe109ae2aae947b2b4757a4bbe62a800 : SUCCESS in 5m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c6118203f6c7419bb503095fe8259f66 : SUCCESS in 11m 31s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9f19c01441f64cc49e55ef506dd6132e : SUCCESS in 5m 20s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/02df43cdcb09488e81e6da02d7f1bc76 : SUCCESS in 4m 42s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/7de6616f8b05466f8a29fabe46f07aff : SUCCESS in 7m 54s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/64ecca4a10aa4de2bbf7db17fb2ba4a1 : SUCCESS in 1h 09m 02s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/785bd54d0bd54e16b62fa2c385a89295 : SUCCESS in 1h 42m 15s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/9b3f8e6f3c2b4ae6a01462412bb4f871 : SUCCESS in 1h 23m 18s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/3d671d9289594a3893ea1187e75c7569 : SUCCESS in 42m 13s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/719de85067954a5780d785180041b69a : SUCCESS in 58m 40s","accounts_in_message":[],"_revision_number":3},{"id":"3fdd36a9e8c4177cd2b6a8ae993007ca2f3549e4","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-06 15:54:20.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":3},{"id":"1534ed082cf2869aae8419a74e7c9ac8fbddbc63","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-06 15:55:28.000000000","message":"Patch Set 3:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0c6422b9783147a7b64ad1d04597f847\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/464dbbacc02e439aac29c6864d637737 : SUCCESS in 48s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/78dfa1e5d9ba4dde8337a903cc0f2684 : SUCCESS in 37s","accounts_in_message":[],"_revision_number":3}],"current_revision_number":3,"current_revision":"73cdee503ff1cb3419d8db0295b20b2b6ddf30c0","revisions":{"2e2b7ca9da10b6fc7a59a1ad7a783348230e1755":{"kind":"REWORK","_number":1,"created":"2023-11-22 21:50:17.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/89/901689/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/89/901689/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/89/901689/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/89/901689/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/89/901689/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/89/901689/1"}}},"commit":{"parents":[{"commit":"7310986de9bf68ed86de90de0501a1bc46945526","subject":"Add Noop Certificate Manager","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7310986de9bf68ed86de90de0501a1bc46945526"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-11-22 21:45:44.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-11-22 21:50:16.000000000","tz":0},"subject":"Fix issue with certificates with no subject or CN","message":"Fix issue with certificates with no subject or CN\n\nThis patch fixes an issue where if the user attempts to use a\ncertificate that does not have a subject or CN, we would fail to create\na listener using the certificate.\nPer the x.509 specification, a blank subject is allowed as long as the\nsubjectAltName extension is present in the certificate.\nOctavia will now check for the a valid subAltName if the subject CN can\nnot be retrieved. If both are missing an appropriate error is raised for\nthe user.\n\nCloses-Bug: #2043582\nChange-Id: I06911f42b9bf29cf9a5f2e76d8333d8a2f1bc60b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2e2b7ca9da10b6fc7a59a1ad7a783348230e1755"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2e2b7ca9da10b6fc7a59a1ad7a783348230e1755"}]},"branch":"refs/heads/master"},"2b9582f39267aff263364e2de7dfaefe4b31258b":{"kind":"REWORK","_number":2,"created":"2023-11-22 22:49:21.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/89/901689/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/89/901689/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/89/901689/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/89/901689/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/89/901689/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/89/901689/2"}}},"commit":{"parents":[{"commit":"7310986de9bf68ed86de90de0501a1bc46945526","subject":"Add Noop Certificate Manager","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7310986de9bf68ed86de90de0501a1bc46945526"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-11-22 21:45:44.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-11-22 22:47:54.000000000","tz":0},"subject":"Fix issue with certificates with no subject or CN","message":"Fix issue with certificates with no subject or CN\n\nThis patch fixes an issue where if the user attempts to use a\ncertificate that does not have a subject or CN, we would fail to create\na listener using the certificate.\nPer the x.509 specification, a blank subject is allowed as long as the\nsubjectAltName extension is present in the certificate.\nOctavia will now check for the a valid subAltName if the subject CN can\nnot be retrieved. If both are missing an appropriate error is raised for\nthe user.\n\nCloses-Bug: #2043582\nChange-Id: I06911f42b9bf29cf9a5f2e76d8333d8a2f1bc60b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2b9582f39267aff263364e2de7dfaefe4b31258b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2b9582f39267aff263364e2de7dfaefe4b31258b"}]},"branch":"refs/heads/master"},"73cdee503ff1cb3419d8db0295b20b2b6ddf30c0":{"kind":"REWORK","_number":3,"created":"2023-12-01 00:36:09.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/89/901689/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/89/901689/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/89/901689/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/89/901689/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/89/901689/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/89/901689/3"}}},"commit":{"parents":[{"commit":"7310986de9bf68ed86de90de0501a1bc46945526","subject":"Add Noop Certificate Manager","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7310986de9bf68ed86de90de0501a1bc46945526"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-11-22 21:45:44.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-12-01 00:36:03.000000000","tz":0},"subject":"Fix issue with certificates with no subject or CN","message":"Fix issue with certificates with no subject or CN\n\nThis patch fixes an issue where if the user attempts to use a\ncertificate that does not have a subject or CN, we would fail to create\na listener using the certificate.\nPer the x.509 specification, a blank subject is allowed as long as the\nsubjectAltName extension is present in the certificate.\nOctavia will now check for the a valid subAltName if the subject CN can\nnot be retrieved. If both are missing an appropriate error is raised for\nthe user.\n\nCloses-Bug: #2043582\nChange-Id: I06911f42b9bf29cf9a5f2e76d8333d8a2f1bc60b\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/73cdee503ff1cb3419d8db0295b20b2b6ddf30c0"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/73cdee503ff1cb3419d8db0295b20b2b6ddf30c0"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Backport-Candidate","status":"MAY","applied_by":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"}}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
