)]}'
{"id":"openstack%2Foctavia~901899","triplet_id":"openstack%2Foctavia~master~Ia51db61bf9d3effdad99532ea62f38ed2e8dd995","project":"openstack/octavia","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2023-11-28 01:15:04.000000000","reason":"\u003cGERRIT_ACCOUNT_9816\u003e replied on the change","reason_account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"}}},"hashtags":[],"change_id":"Ia51db61bf9d3effdad99532ea62f38ed2e8dd995","subject":"Add publish amphora image job for Jammy","status":"MERGED","created":"2023-11-27 07:54:49.000000000","updated":"2023-11-28 05:59:01.000000000","submitted":"2023-11-28 05:54:36.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":5,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"901899","meta_rev_id":"3901647759976858f08ca8fa47c76250826855f0","_number":901899,"virtual_id_number":901899,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-11-28 05:54:36.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":2,"date":"2023-11-27 14:46:27.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2023-11-27 13:18:40.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":1,"date":"2023-11-27 14:46:27.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"date":"2023-11-27 13:39:37.000000000","permitted_voting_range":{"min":0,"max":1},"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-11-27 10:50:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-11-27 13:18:40.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"},{"updated":"2023-11-27 14:46:27.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"}],"messages":[{"id":"e1ac59f3d6470d6b9119dc0864a105cbba718273","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-11-27 07:54:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"74e9f5f5e16eddb5f25ca2d1d63ab7f24c746a5b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-27 10:50:15.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0d7e4d2743a2495b989855e5cf23eb11\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/af6c2d93f60b4c13a492544e4a82f3ad : SUCCESS in 10m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0095edefcd1447489234a78b5bdeff51 : SUCCESS in 5m 10s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/9dbe57b1f2b14d2f88aaf577bd4f260f : SUCCESS in 4m 46s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/bad9e8b0095c42189d0d371e4570c44d : SUCCESS in 5m 23s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/36d612a9e0ef4c9c9597426ee12e8cb1 : SUCCESS in 11m 22s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/66e83cd45c9e481dababe1321449f810 : SUCCESS in 4m 35s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/bd58e7f248894d85b0ce8122c61667c7 : SUCCESS in 7m 54s\n- octavia-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/eb35a6351c3a4de6955280e86fb34f1c : SUCCESS in 8m 06s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/ae4c4f71ece4451f9780ebc6d1754711 : SUCCESS in 5m 16s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/97d216ab5f564b3ab775630e54262235 : SUCCESS in 7m 26s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/d59bbae26a484a248f2e44d027dfa2e9 : SUCCESS in 6m 54s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/d5108434a1664b80aa084e869f9ce392 : SUCCESS in 7m 17s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/395c135668974d9ba0f6facfd3c29c4e : SUCCESS in 1h 41m 57s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/5e5444314b6c4b77834c274699f87ad8 : SUCCESS in 1h 39m 20s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/db126e644fbf4feda86360a94f75e911 : SUCCESS in 59m 21s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/ba1c29246d4044269071426b7f3ba4fc : FAILURE in 56m 06s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/10fd30506cb7465d8b62c5978f814b15 : SUCCESS in 57m 01s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/a057b655e6494bc581ec434985782305 : SUCCESS in 37m 12s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/eab4a373721c4ebc9258b0c23cf6c47a : SUCCESS in 37m 19s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/3f035aa904ac4540ad2699e7fa146293 : SUCCESS in 37m 20s (non-voting)\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/3dcd3ae317ef47afa72001b62256f3d8 : SUCCESS in 1h 03m 14s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/40736d1cd11143c792539bfea70d0b3a : SUCCESS in 40m 35s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/7573bd599bd54425a4cc49f4946938d8 : SUCCESS in 2h 15m 41s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/fe6d78eaaf2c4168aa59f1dae63abf68 : SUCCESS in 2h 46m 53s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/721fd650ab654fbc8a413c638cb0e6df : SUCCESS in 2h 04m 56s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/506278b777af41a7b47e81a0b28e858a : TIMED_OUT in 2h 34m 26s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8dd6a97bd554001d8915733c10a2e996af883fbf","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-11-27 12:43:31.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"577e301819b04c496c87bfa2df4f948c10fc0cac","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-11-27 13:18:40.000000000","message":"Patch Set 1: Backport-Candidate+2 Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ea34ecf1eeb265d8368e39d664cdb942d1b35cb5","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-11-27 13:35:08.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"67ab6589ef15c2f969bfb347a424ba986b72abd7","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2023-11-27 13:39:37.000000000","message":"Patch Set 1: -Backport-Candidate\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"1450629a7bdfebc460f0376c7c0723513c26a2a1","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-11-27 14:46:27.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"3651a7b98a16c0be977db9ecabeb9652084d5b47","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-27 16:06:26.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/b61e97e4fa534bb497ff4a42109020de\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8ad03ac72274420584c6661258310ebb : SUCCESS in 10m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5723d461acdd4412b1e8ed42ead635f0 : SUCCESS in 4m 39s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c45780aba89941c4b95a1d89f0e72a94 : SUCCESS in 5m 41s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d6416e3bb5c244c8bc17c7ae345c4141 : SUCCESS in 6m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ff895b9b6584360ba33abb4f03b35d2 : SUCCESS in 13m 59s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/6b14139002a8494b932e1220d0188e76 : SUCCESS in 5m 54s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/387e90a5ad4b48e4871e36112099f92f : SUCCESS in 8m 05s\n- octavia-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/795924136b5140d5ae6d5594b93d848c : SUCCESS in 9m 33s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/52952c5609d844ae9be868f96b4454a7 : SUCCESS in 4m 15s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/4a6947fa788644b19ef2ba5f086322e9 : SUCCESS in 9m 20s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/1205bd3b69e94e7baa46b6f57aa2cc40 : SUCCESS in 6m 55s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/d5c28c0b667e4dbcaf756e9725adda9e : SUCCESS in 7m 47s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/c4e5254590714d4188bf69a225d41d7e : FAILURE in 1h 07m 49s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/1c569350f8154e9780e51b5c3a199e4f : SUCCESS in 1h 36m 17s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/cb100a83b127437d8ade7c0988290897 : SUCCESS in 1h 05m 09s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/852de17dea324b56af5ddcf970cb246e : SUCCESS in 1h 42m 34s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/08c61f1de44c446abe9d38e82d3e5297 : SUCCESS in 58m 48s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/2b80d6bcb4714733b08baa2df58bfd2c : SUCCESS in 34m 13s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/50d86e747d994748a8a082fbb777d054 : SUCCESS in 39m 55s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/7bfd8d3139bf46d6bfb0818cbe6c7fff : SUCCESS in 48m 31s (non-voting)\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/a0e36d31140c4068b8b9b0bc419eb1da : SUCCESS in 1h 09m 42s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/e03d6900451f4d59ae6d34125f800e86 : SUCCESS in 35m 27s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/7709b8f8a0c44d0f81ac4ed822158890 : SUCCESS in 2h 17m 21s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/7b47c967a11d4c749ee1929180e6c282 : SUCCESS in 2h 51m 17s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/3387f68b3a114f2ea3564e8e62b94073 : SUCCESS in 2h 20m 31s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/2ab148669e1f466f8058bacdfbea72a2 : TIMED_OUT in 2h 33m 18s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e4b6ff2a7e7a93ab281bed41483ebcffd56f1d9f","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-11-28 01:15:04.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"e0022a32d7071680a61ae8c762e318574e0c58b6","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-28 04:09:25.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8486e3cbf3b7454e83606171c71dbecb\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f2da758e224f44d0ac22dc604fdf6c17 : SUCCESS in 10m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7db743ef36e5409aba9980b5105f5844 : SUCCESS in 4m 25s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/799fe1479f1a413695653ebfd5fcccde : SUCCESS in 4m 42s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f95b210b1a0349c1a1a028074c80d766 : SUCCESS in 5m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/dacde07ad8224bcd9862e43915fcc6fa : SUCCESS in 11m 44s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/0ad5bae6b4d54316a20499dec1a842dc : SUCCESS in 6m 11s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/19b790b072314aea804a3f1afc40b016 : SUCCESS in 9m 08s\n- octavia-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/6b4253c137fa46e7bde8e36023127562 : SUCCESS in 10m 30s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/4fdd3f1b73df4219b24a5dd00950b8b2 : SUCCESS in 4m 31s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/6c131c83caa947d7aad741f31514a9d5 : SUCCESS in 7m 42s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/c33a964cb7b7492c906cff918422a2ef : SUCCESS in 7m 10s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/8a66e3e41c814d5a82d20419a6067bd0 : SUCCESS in 9m 02s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/bccd37d0e0f34af6a4e31cdd2a3e9c3a : SUCCESS in 59m 19s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/a386ed570153425da628a9d89d6e46df : SUCCESS in 1h 51m 02s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/3d2f20d8dd9d4292bd0cc7a200fd2b1c : SUCCESS in 1h 15m 58s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/35e9b50270584e89aa4aaa764e16009c : SUCCESS in 2h 08m 58s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/2a4f8581718e427e83bae34993cc5a8e : SUCCESS in 1h 17m 59s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/a5d911a3af6049319729489a3d1b14a2 : SUCCESS in 38m 43s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/3c2f0254d961449eb067eb4891312566 : SUCCESS in 40m 27s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/7402fed2081941cb89e82632c8156dd6 : SUCCESS in 43m 17s (non-voting)\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/9290f32597c44f7c8e3b33fa74a07bb1 : SUCCESS in 1h 05m 08s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1c4068f8938d496aac53960e1e37b7bf : SUCCESS in 34m 43s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/e6bdabbb3f18439fbe4668ef15e44758 : SUCCESS in 2h 29m 25s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/8cff154a47e1456ea4671e31040f6db6 : FAILURE in 2h 51m 58s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/526806b72004420390899355e176a6c4 : SUCCESS in 2h 26m 17s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/94ae53d20ac042539605c674f84c9ad1 : TIMED_OUT in 2h 34m 21s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"c4385849cb7058d393ac73a7ee3bc3f9c8e3514e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-28 04:10:09.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"cdc686b3fefa4ea0bd17b5cf987b3dbc367e8ad0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-28 05:54:36.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/37e5ed80167e4bb4a2f9c3c57a7bd184\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fd7f697544ee46cfa0c3949e8bafcf06 : SUCCESS in 4m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f05d2e0ffc3444db8d3378bc59604b92 : SUCCESS in 5m 39s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/88a300f93e4946f7bd136c27a58ee14f : SUCCESS in 4m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/520a22ff537b455eb7ca962467d20020 : SUCCESS in 11m 00s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/4dd9a34e50db41d7964fa16234b2747c : SUCCESS in 4m 50s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/4a99d6d15ce54832a65d2a001bb40d44 : SUCCESS in 7m 59s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/ea7820f6a37740c1a5182a31411d5314 : SUCCESS in 1h 23m 45s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/58b9936b483e486ba88a8497a5561070 : SUCCESS in 1h 43m 18s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/4bfdbf039ad44eb1b820d07ea89202fc : SUCCESS in 1h 12m 07s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/a2136db37ced4e9ea28d84d91f53b5ac : SUCCESS in 34m 13s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/e1dbdf6041f544168d8368eeb490a2d8 : SUCCESS in 53m 34s","accounts_in_message":[],"_revision_number":1},{"id":"c40ddc6012f2e8b1d77f5215679b0a852b1e498e","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-28 05:54:36.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"3901647759976858f08ca8fa47c76250826855f0","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-11-28 05:59:01.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1c57274951cb489cb97229e49b92b7c3\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/20b3d7d4220a43428a702d32a714b209 : SUCCESS in 48s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"7df86e12771a600aec01495589f212fb1b46d7b1","revisions":{"7df86e12771a600aec01495589f212fb1b46d7b1":{"kind":"REWORK","_number":1,"created":"2023-11-27 07:54:49.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/99/901899/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/99/901899/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/99/901899/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/99/901899/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/99/901899/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/99/901899/1"}}},"commit":{"parents":[{"commit":"7310986de9bf68ed86de90de0501a1bc46945526","subject":"Add Noop Certificate Manager","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7310986de9bf68ed86de90de0501a1bc46945526"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-27 07:53:57.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-11-27 07:54:25.000000000","tz":540},"subject":"Add publish amphora image job for Jammy","message":"Add publish amphora image job for Jammy\n\nThe job definition was added earlier by [1], but it was not actually\nenabled.\n\n[1] 37ddc6fba4c5d216a0088f7e8a8c0baf01f84b64\n\nChange-Id: Ia51db61bf9d3effdad99532ea62f38ed2e8dd995\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7df86e12771a600aec01495589f212fb1b46d7b1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7df86e12771a600aec01495589f212fb1b46d7b1"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
