)]}'
{"id":"openstack%2Foctavia~902812","triplet_id":"openstack%2Foctavia~master~I54f8144a3347c3c5bf2e2e99e9d500a0c6fb89eb","project":"openstack/octavia","branch":"master","topic":"remove-wsgi_scripts","attention_set":{},"removed_from_attention_set":{"31664":{"account":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"last_update":"2025-05-05 08:59:09.000000000","reason":"Change was submitted"},"27442":{"account":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"last_update":"2025-05-05 08:59:09.000000000","reason":"Change was submitted"},"15334":{"account":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"last_update":"2025-05-05 08:59:09.000000000","reason":"Change was submitted"},"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2025-05-05 08:59:09.000000000","reason":"Change was submitted"},"31291":{"account":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"last_update":"2025-05-05 08:59:09.000000000","reason":"Change was submitted"},"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2025-05-05 08:59:09.000000000","reason":"Change was submitted"},"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2025-05-05 07:11:30.000000000","reason":"\u003cGERRIT_ACCOUNT_29244\u003e replied on the change","reason_account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}}},"hashtags":[],"change_id":"I54f8144a3347c3c5bf2e2e99e9d500a0c6fb89eb","subject":"Add wsgi module to Octavia and remove wsgi script","status":"MERGED","created":"2023-12-06 21:07:49.000000000","updated":"2025-05-05 12:49:43.000000000","submitted":"2025-05-05 08:59:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":11,"unresolved_comment_count":1,"has_review_started":true,"submission_id":"902812-remove-wsgi_scripts","meta_rev_id":"58ee2d166356a66f9e63d91474d6ca03664c727a","_number":902812,"virtual_id_number":902812,"owner":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2025-05-05 08:59:08.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"recommended":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"value":1,"date":"2025-05-05 08:23:59.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2025-05-02 00:29:50.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":2,"date":"2025-05-05 07:11:30.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":1,"date":"2025-05-05 07:11:30.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"value":0,"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"value":0,"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"}],"REVIEWER":[{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-12-07 02:12:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-12-07 11:02:16.000000000","updated_by":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"reviewer":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"state":"CC"},{"updated":"2023-12-07 16:12:43.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2024-01-05 15:28:25.000000000","updated_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"reviewer":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"state":"REVIEWER"},{"updated":"2024-01-24 16:58:50.000000000","updated_by":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"reviewer":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"state":"REVIEWER"},{"updated":"2024-09-30 05:58:56.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"state":"REVIEWER"},{"updated":"2025-04-29 08:15:10.000000000","updated_by":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"reviewer":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"state":"CC"},{"updated":"2025-05-05 08:23:59.000000000","updated_by":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"reviewer":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"state":"REVIEWER"}],"messages":[{"id":"cdf7077bc79d96691209d56b9daf3adf668691c0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-12-06 21:07:49.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"68310f98bc7f72e3e4daf99ec63680a7dae798f5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-12-06 21:12:52.000000000","message":"Uploaded patch set 2.","accounts_in_message":[],"_revision_number":2},{"id":"9ce91846e9e00fac48d78b420c34cf3776ecc8f4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-12-06 21:14:52.000000000","message":"Uploaded patch set 3.","accounts_in_message":[],"_revision_number":3},{"id":"54034c13b89cbda45460aeea846be3a985ad3fe4","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-12-06 23:03:02.000000000","message":"Uploaded patch set 4.","accounts_in_message":[],"_revision_number":4},{"id":"46f505a56d2e786736fd9b9c26c88b169aea0c78","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-12-06 23:04:37.000000000","message":"Uploaded patch set 5.","accounts_in_message":[],"_revision_number":5},{"id":"76014e7753357d04deb7116c4258491d642f9b24","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-12-06 23:20:26.000000000","message":"Uploaded patch set 6.","accounts_in_message":[],"_revision_number":6},{"id":"b2e258349dfa7462a5cfd5cf6f2e33935e3fd9d3","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-07 02:12:15.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/46dde1c8fc744e1abeedc60170cd3a62\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/e0215b65246b41c29f881af85a088f67 : SUCCESS in 2m 18s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7cb9616a8ed24f74b955f136ff8bc9f3 : SUCCESS in 8m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2facc186a4704f6bad61f0f2ffcdf93c : SUCCESS in 4m 43s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/931c79308b384df082ab8b8d878a4998 : SUCCESS in 4m 52s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8538c3c7f1cc4f8c837619f9fe27ef56 : SUCCESS in 5m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ca6d94b1230748d0a1548554618274d5 : SUCCESS in 9m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/25ae930a39f74b6a8cfa8a2e426948c2 : SUCCESS in 5m 23s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/c494c31c0183453a86dccb5476428484 : SUCCESS in 5m 44s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/e488f02f753f456aa996af1805a9e7aa : SUCCESS in 8m 17s\n- octavia-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/756025a971fe4505a2819bbe33d4d6ab : SUCCESS in 10m 24s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/977b4052c85d435f9f6f76ed2bb229ac : SUCCESS in 5m 29s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/213d4db26e6a4651a3a5087da5c58236 : SUCCESS in 7m 59s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/efef510890124a04ab01fabe2d09cfab : SUCCESS in 6m 44s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/c77df9e77c764a6eb840299f488193e1 : SUCCESS in 10m 03s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/29579f4d925d4007a6a8081b3fbc4115 : SUCCESS in 1h 17m 53s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/e2422128e7b745bdb3fa953ca823d3a3 : SUCCESS in 1h 58m 14s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/14e55fd64d2f49f5a63ea00b82a8edd5 : SUCCESS in 1h 15m 49s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/536abf3c969048b1841edaa2cbb6465f : SUCCESS in 2h 06m 10s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/1fba7deaac6f4acabf8d0f4fae892e5e : SUCCESS in 1h 16m 39s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/2167dc702554423b90f1ab9dd2bb27ea : SUCCESS in 51m 34s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/42958a05e2734539b9d221407427ecfd : SUCCESS in 57m 42s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/4235dd0154924502972e47ab2b4beca3 : SUCCESS in 57m 03s (non-voting)\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/9ee5a29295a6425fbfe158bf19eda884 : SUCCESS in 1h 06m 54s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ac81715777a041bfb57784fd62e6b408 : SUCCESS in 48m 52s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/94e54f82abb94d2bbb87467365fa0c9f : SUCCESS in 2h 02m 26s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/928ff4825f504725ae777eecf7e13c59 : SUCCESS in 2h 42m 15s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/e7c5fa0a5ef24bf9832c0639e8d117cd : SUCCESS in 1h 44m 14s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/263dfd20e5b84cf4836755e4db4a002f : SUCCESS in 2h 18m 15s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"8a6b5d16dd1371f0ae8c3ae17b36c67dbd0429da","author":{"_account_id":15334,"name":"Stephen Finucane","display_name":"stephenfin","email":"stephenfin@redhat.com","username":"sfinucan"},"date":"2023-12-07 11:02:16.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"84a01e04104449210cc8a691b67315c2c1c8d77c","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-12-07 17:28:12.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"f415a74bc80d77478e53469f2e6d77e1894571c6","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2024-01-05 15:28:25.000000000","message":"Patch Set 6: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"0a0e1089ffe11100b8078997db4432683e1b5ab9","author":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"date":"2024-01-24 16:58:50.000000000","message":"Patch Set 6: Code-Review+1","accounts_in_message":[],"_revision_number":6},{"id":"4929c5e3a19b991e5f4a82d2643e14f8c48d0d0a","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2024-09-30 05:58:56.000000000","message":"Patch Set 6: Code-Review-1\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"9bbbcb068a58e4f529f144a6a5145de30c3f0b47","author":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"date":"2025-04-29 08:15:10.000000000","message":"Patch Set 6:\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"199c590dd10beb8ae15108f85c4169b0a032315e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-29 11:20:28.000000000","message":"Patch Set 6: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/c943bee301af4f389d09998e2cbe618c\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/07284977d19b439e8bee49ba63a0c0c7 : SUCCESS in 2m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1b455fac87e34c7381d3df84e752c42b : SUCCESS in 13m 18s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5c866e3bd27e41c285062ca2a8f71afb : SUCCESS in 4m 48s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/dd12e23ee1b24c7ca2ed6e13a9b6f48f : SUCCESS in 3m 57s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f8e8f0241165427cbff5cd8eaeb2d325 : SUCCESS in 4m 41s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9b846a1f0ec44b6380fdf08158ff598c : SUCCESS in 5m 14s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8f8e8aba05b44ac5ab1634e752ad5fd5 : SUCCESS in 12m 44s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f59aad97fc244f238894f9872db2c8fa : SUCCESS in 3m 21s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/bd4beae69488437fbccca00d6595f092 : SUCCESS in 4m 39s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/0e78a1920d33459f9020fc16140fdf07 : SUCCESS in 3m 38s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/5a0a79b273e3401cb087ebef232fc2f9 : SUCCESS in 3m 41s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/855ed6c481044184b497ba4cffdc86fd : FAILURE in 1m 32s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/7c4f7fc0046b40099006098a676135b6 : SUCCESS in 7m 21s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/a26b15c60fa8432f91d263e2ffb9bd38 : SUCCESS in 3m 36s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/67edfc7a88db4923a57c9968365000fa : SUCCESS in 38m 07s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/c08b8a48e092433a976ddc8339553b34 : SUCCESS in 1h 59m 28s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/9d8d0f93d58c45fe9b1eb4dd95e51b35 : SUCCESS in 1h 14m 26s\n- octavia-v2-dsvm-scenario-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b8b1685942b241b4b83fac7a5a3efc32 : SUCCESS in 2h 10m 15s\n- octavia-v2-dsvm-scenario-non-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/18e7ae573e1f4b379c74a757575e9ba2 : SUCCESS in 47m 33s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/3052fb4298ac4e138e67f9ef8d5096a9 : SUCCESS in 2h 01m 05s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/4a62dfa03aa9416fb3ad3f4babf597a5 : SUCCESS in 1h 03m 44s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/26081360c15e4cf59c358bbe06854a07 : FAILURE in 34m 40s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/35e7ff3b00ce4a81ab8ad9fa74411c32 : FAILURE in 30m 12s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/abdca693a26042beb737acb9efb0614f : FAILURE in 12m 37s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/12d06139ee10444786c4d9238b067904 : SUCCESS in 35m 52s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/ee605aea16b04c4aa1615a75f7319928 : SUCCESS in 2h 29m 30s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/f927aa0091ce400f9e0571153094ec3e : TIMED_OUT in 3h 03m 19s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/4720a3faa674484a98edc630fb650781 : TIMED_OUT in 2h 33m 01s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/a726721c1bb7452ead4825b62cc57ed4 : FAILURE in 7m 24s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"c6406a286ba6aaf7620a4f4a31503267ce2b7e8a","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"real_author":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"date":"2025-04-30 07:08:25.000000000","message":"Patch Set 7: Patch Set 6 was rebased on behalf of \u003cGERRIT_ACCOUNT_11628\u003e\n\nCopied Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"_revision_number":7},{"id":"31d9a77fb590e8d1ed9c03fe2c4fa95a25f720a7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-30 10:01:47.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6371757f9f0649439a6987d4e7dfaa2f\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/3a11baf210c144e8b29162f5235b02d1 : SUCCESS in 2m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/40a7998f97d14602b995cadc0189ca30 : SUCCESS in 14m 10s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/418daf4ad15b41f4a6ec9caf28b74f5a : SUCCESS in 2m 43s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/fc5f4d80f515414986faed102b312c06 : SUCCESS in 2m 44s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a2fcc64fd6b04801bbf68ba7f5745105 : SUCCESS in 2m 53s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/2fe71e364061421289309380a0dbdf89 : SUCCESS in 4m 06s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/29e00624f39f4b45b464e4f6458762fe : SUCCESS in 12m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/86734b5e303f4cb880958303dcfc5c42 : SUCCESS in 4m 33s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/a8a382e252a543859de4ba694b6042b5 : SUCCESS in 4m 28s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/e98fb42aee2647efaa2344b9d55669ef : SUCCESS in 6m 55s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/23648859ff3d45efadc8ff857f0994a8 : SUCCESS in 3m 31s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/eba4013e1eaf47459b0445005ea10337 : SUCCESS in 7m 22s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/8f3e4c4d988741a4a0a061c9391bb5a1 : SUCCESS in 6m 29s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/ed6b8201a51449d69a4255c9440692dd : SUCCESS in 1h 31m 06s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/a3d8139c60fb436eb73dab54de96d048 : SUCCESS in 2h 00m 30s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/9b6788c845b340a0aeb68c7f4250cc34 : SUCCESS in 1h 33m 30s\n- octavia-v2-dsvm-scenario-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e43209ebf9be49899ec1d154b11b0e1b : SUCCESS in 1h 55m 45s\n- octavia-v2-dsvm-scenario-non-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/505e09024fba40f49724933107ebf384 : SUCCESS in 1h 06m 10s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/92605692fa1f4ecb97979be767a898d0 : SUCCESS in 1h 56m 51s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/3bb550b8de024d69ace6b26224aed0aa : SUCCESS in 1h 10m 57s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/45748bc16fcb42b4a34be947a07843d6 : FAILURE in 25m 46s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/8345cc3882fb43f18bd32b34e20325a7 : FAILURE in 33m 42s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/52d0872a8d554ba395c911a3c3b45ab3 : SUCCESS in 34m 27s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/edb083ebed5847aa8514a1d3429b1481 : TIMED_OUT in 2h 33m 12s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/7f33daed59b74241a5ed238ef08341da : SUCCESS in 2h 51m 12s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/be32c8a28abb4e61b4fabda8a30ff2eb : SUCCESS in 2h 19m 31s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"6186045ba5f4dd17163b21d7461bd673ec6b14e7","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"date":"2025-04-30 10:08:40.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"2abd23b7e75f5546dc421bfa34f9fda1ca334c97","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-30 13:12:55.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/bab82b566629432da43487867066d15f\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/9cc3ed9abc754ce89fe7d4ebdb131965 : SUCCESS in 2m 19s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/9680bbda4b70452e8deab2dd91146061 : SUCCESS in 14m 40s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ccb230e1d37f4e8ea92b1d647d604ffa : SUCCESS in 4m 16s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/141fd40d1eff437dbe259d41b3f0ab04 : SUCCESS in 4m 19s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a01e4ab3011f4ccebcbe350d57c9c2e3 : SUCCESS in 4m 18s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/4854ae1c87734770979e71f2d5125a29 : SUCCESS in 7m 31s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/064557aa947f4613bd50d353c3498a45 : SUCCESS in 12m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a248f8757b2b46c6a5c87a4361f6a8b7 : SUCCESS in 4m 35s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/66082dcd179d480a987868f2a33a4dba : SUCCESS in 3m 49s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/171fc36f2e0640ea90b3eb9a0339ea05 : SUCCESS in 6m 45s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/8d0ed8fc4780434e97119bbbb1dde4fa : SUCCESS in 3m 41s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/2f4ac8cd0eca4cfcb60bd1bd8a5da4f2 : SUCCESS in 3m 30s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/578fdf27adfc4c928cae87f10723de23 : SUCCESS in 6m 14s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/cb271234960b417296b6d9cf060bbb71 : SUCCESS in 1h 01m 28s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/8542cdc994c04ea494dcb571357844d9 : SUCCESS in 1h 24m 39s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/c5b4b2ba828242fe9089c1bb70ce54e3 : SUCCESS in 1h 03m 01s\n- octavia-v2-dsvm-scenario-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/5e7c286d17db499a9d0495e0b8202345 : SUCCESS in 1h 59m 03s\n- octavia-v2-dsvm-scenario-non-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/663453648e2c48f4950d017f1a0195d1 : SUCCESS in 56m 32s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/41b09b95745f439185d2074049285252 : SUCCESS in 2h 04m 03s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/376dbe84ff4d494aaee53a12f63bd1aa : SUCCESS in 1h 09m 42s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/57f23886a6c64b5ca02ecfa39608786b : FAILURE in 37m 38s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/6d5e4ef493234175883a48b1a2fe2f7c : FAILURE in 32m 52s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a7acdf0d1f9a4fc4a8e687ca3640498d : SUCCESS in 35m 47s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/bec89e6c0ee44cf89eeaea28056fb213 : TIMED_OUT in 2h 32m 47s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/5053f953e8b143d8a9e5df462a0af973 : TIMED_OUT in 3h 02m 52s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/272425d51281443c8024ae2e60f5ec55 : SUCCESS in 2h 30m 44s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"4019eeea7e8e1a5fb8eb2d1785bdda504dedc217","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2025-04-30 21:58:58.000000000","message":"Patch Set 8: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"eb8d1b0994e373cd4c6153dc36c317fcad187bfd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-30 21:59:31.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"3d5827c521c07cdf8d6a6fd00a756e7f731e6efd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-04-30 22:28:20.000000000","message":"Patch Set 8: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/ba0cbf86e67746009f0656c95c18ab80\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/50a63ee55ede4d7f9681100f1c08125b : SUCCESS in 2m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7f1537d1fd2c47b984eb50df4f30eb5d : SUCCESS in 4m 46s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/bad99bf3aace4348a2e3a9c25d47bd63 : SUCCESS in 4m 08s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1cfff78138634d6c858ff4e1676e1d65 : SUCCESS in 4m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e139a1e6f1124b73a9ebba9cacc07a05 : SUCCESS in 5m 43s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/753c0dc3902e4e67bfd721ec97c4eebe : SUCCESS in 2m 36s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/4550b965ec804bfc9f8fa3423cc60432 : SUCCESS in 1m 51s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/1785c6c52f8c451998d5374844ed95b6 : SUCCESS in 3m 48s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/a3e417534e7647adab8ab2d7ce12bb81 : CANCELED\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/f58e21694e684088922450d9325ac884 : CANCELED\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/e78acbcd61f145e3b4abc96fc5ba4630 : CANCELED\n- octavia-v2-dsvm-scenario-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/b47a08c5f4454bf9a28ff7f347933b6d : CANCELED\n- octavia-v2-dsvm-scenario-non-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/c284add46625405495a4f0e98cc2a164 : CANCELED\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/964e77bce762468c85e0f8de3ca51518 : CANCELED\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/62b8b05eafda4cfab48c784ed9a35db8 : FAILURE in 27m 11s","accounts_in_message":[],"_revision_number":8},{"id":"ec7de62a0412353e6bf4994f9844a15d5f3abbaa","author":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"date":"2025-05-01 07:03:11.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"675a04e83740abfe1b8e07147b0a84a6aa390708","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-01 09:37:11.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f512e1b870e34653ba4e803bb78780fc\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/8c7749cb25f044118d50cafcc87cbdc7 : SUCCESS in 3m 25s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/08f9aca1b5ff4e7fb0e5f3f52e49874a : SUCCESS in 11m 02s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/03931f6736474594b62042cf8967e037 : SUCCESS in 6m 05s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/8de9ffac38954d84968f2d94b7f46267 : SUCCESS in 4m 43s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/36459ffb08f54dd28af840f5310b412a : SUCCESS in 6m 24s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/9c594fcdcde34518b936d37e29558775 : SUCCESS in 6m 24s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7034af94a84c4836a6f9183a0366ea96 : SUCCESS in 10m 51s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/21282d6f934745b0a02b8980e2b984d5 : SUCCESS in 4m 30s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/689d4229dba0459b8822c8b6c9b976b9 : SUCCESS in 4m 02s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/adad26c8864d4f7b8cd1c58a19e9821f : SUCCESS in 8m 03s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/bca493c5f4aa4b6388b919f30cd910e4 : SUCCESS in 4m 00s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/0c3a1b2fb51a4746a0db2ab94945295e : SUCCESS in 7m 37s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/8cf7f3e987b74f9fa5cc43b1be919bbe : SUCCESS in 6m 45s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/88833dfacd114482bf351c726fe8c4cb : SUCCESS in 1h 23m 02s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/c81da8544b3e4a73bea28255fd9a5619 : SUCCESS in 1h 57m 18s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/fd36a96bf9d84dc3937c8a7a91399ba6 : SUCCESS in 40m 59s\n- octavia-v2-dsvm-scenario-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/11a31a24479a4f25ae9db14803fc68a4 : SUCCESS in 2h 01m 26s\n- octavia-v2-dsvm-scenario-non-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/9c4e083f14f941a4a3481ef031b48d09 : SUCCESS in 1h 10m 43s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/b7c53a3b1cf94bd1a328dbbdf066964a : SUCCESS in 1h 58m 25s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/3ca51d119a3d49218cfe0091b3314fbe : SUCCESS in 1h 01m 08s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/95fa31b0b69b45e0af072796bd1f127b : FAILURE in 34m 23s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/e313bd665f2143989c2baa4d46c48cf5 : FAILURE in 26m 16s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/bc382478b8aa40c8a60ae22ff45304b5 : SUCCESS in 22m 13s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/04f1ade91d0a451aaad87fd1184f4ad7 : TIMED_OUT in 2h 32m 56s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/38218ab7b9564dcca5e182758b2c8868 : SUCCESS in 2h 24m 41s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/7c1753a8083f4ef2a9a37e419c56ff4f : SUCCESS in 2h 08m 34s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"25c24f16dadc896e6b00ed9235d99a26c2d98f95","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-01 09:37:45.000000000","message":"Patch Set 8: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":8},{"id":"4c5a6262b43fbdd029c56aa7d873029f566a77ed","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-01 09:53:54.000000000","message":"Patch Set 8: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/32e3710e1a3e4020be13d05f0a676904\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/9f11fcec37604657a5191ab0b7b14d7d : SUCCESS in 3m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c937e96905714b60ad6b2ecbc1310427 : SUCCESS in 4m 53s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/9885df1061254e9f8c0288c9b39cf05e : SUCCESS in 3m 55s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9a835410958947d18f4136c0b5b35918 : SUCCESS in 5m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8146a23ad194431aa695a738ca1597ce : SUCCESS in 9m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/27b515c544704cd482ddff7c396d6616 : SUCCESS in 3m 54s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/8e1d8b1338af426c9ba1c70d8c5a3582 : SUCCESS in 3m 24s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/834576a492ed4afd8fe2d0a4dced8e1a : SUCCESS in 5m 41s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/424c8df0c5a04ab2bef9a3a5bea3765b : CANCELED\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/cfd42f61fd6a44da822f773a5caa1f94 : CANCELED\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/eca5e3a0333949c1b5fe568c17808d55 : CANCELED\n- octavia-v2-dsvm-scenario-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/148e9748e2934cff83087ea09c4905ee : CANCELED\n- octavia-v2-dsvm-scenario-non-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/da1df9b41d274daf8807864b0fc9212f : CANCELED\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/312a18e257194b8b8afcf0db61efe6bd : CANCELED\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/681b8bec9d794618b9d8bf7a3158c613 : FAILURE in 15m 33s","accounts_in_message":[],"_revision_number":8},{"id":"d5c3abce943f89760214fe808ae5ded9b23ee5fd","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"date":"2025-05-01 15:29:59.000000000","message":"Patch Set 9: Published edit on patch set 8.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified-2\n* Workflow+1\n","accounts_in_message":[],"_revision_number":9},{"id":"7117ada973b1ffec357cb8ac07fc4f20afe25745","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-01 18:34:26.000000000","message":"Patch Set 9: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\nand https://docs.openstack.org/project-team-guide/testing.html#how-to-handle-test-failures\n\nhttps://zuul.opendev.org/t/openstack/buildset/4e6b4ea76a0344eea09d7a92b416cf4d\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/c47985a94b8d4b66809e9e42bb7f6f8f : SUCCESS in 2m 16s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7186945fa8aa417cb82ff480c160c701 : SUCCESS in 11m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ad24d31381b444a1a19027dba7505bcb : SUCCESS in 4m 29s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/7c2f7920d6844db5a5b16a5ada199249 : SUCCESS in 4m 44s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/f20e93f7fa3e4c4ab50a42404694158b : SUCCESS in 2m 52s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/1bc5f47d03c2439984798d24d6d86531 : SUCCESS in 7m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2d9a953c41454121bf9428f50c0a35af : SUCCESS in 13m 01s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/339f1e4ac8c148cfa2b5ad5b7ac7f6ed : SUCCESS in 2m 25s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/76031be909c6412f9959c3909071fe9c : SUCCESS in 4m 29s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/284691cc346d4906940e486ea9968981 : SUCCESS in 6m 40s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/64aacaf4426341fbb53bf93a52efa374 : SUCCESS in 3m 52s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/28946cabf52e4226a134a07c553af90e : SUCCESS in 6m 44s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/fa7c509f56fd454aabe9fd410f282ebc : SUCCESS in 8m 04s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/b09d764b3990423abf5313e2d7083ce2 : FAILURE in 1h 44m 52s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/eadf0a34e07c409aa5c1e37b8764d5b7 : SUCCESS in 1h 48m 58s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/ad36ad3ad9274a358fa037f992b737e2 : SUCCESS in 53m 56s\n- octavia-v2-dsvm-scenario-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/91ef8468505b4c01801abf1389ae420e : SUCCESS in 1h 45m 18s\n- octavia-v2-dsvm-scenario-non-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/7f3cb4985a3f48ceb89e0ee62865f790 : SUCCESS in 1h 07m 30s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/739b6987825944e49f11036865af262f : SUCCESS in 1h 50m 24s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/a778879c62d44a67bebfc7b496758be9 : SUCCESS in 1h 08m 24s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/a6d2d130c0134ddfbd34a9f811cc0985 : FAILURE in 27m 04s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/2c70d4dac9274e23bacbcd78109df16f : FAILURE in 27m 35s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3767a6d5ee0449208e9ebb66897cde3f : SUCCESS in 33m 38s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/434aea513dba46529db373e072396d63 : TIMED_OUT in 2h 32m 53s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/92fb2a7efe45450f9fc8796475880c90 : TIMED_OUT in 3h 03m 22s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/122655d1a2d04135ad68365c85753802 : SUCCESS in 2h 30m 34s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"8d30c92f8fa5b83348b7354860481e3176fefcda","author":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"date":"2025-05-01 20:33:35.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"392652694b90598cb4c0c2ce9e73b9f2d33b11b0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-01 23:07:44.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/57cf743954de4af3b9eb858c2ed69813\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/04b5c0a598a048d788ca7d9612c80e7d : SUCCESS in 2m 14s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/4f51ed817fd942bc9fc88aa464d21ee0 : SUCCESS in 14m 20s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/fac5cc7e4e524c14945c47ca986c3a39 : SUCCESS in 2m 30s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/1bfb707823094560befd8bd19835999d : SUCCESS in 4m 34s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/ab72a2d42582437fbfb6ceec25a199fa : SUCCESS in 4m 31s\n- openstack-tox-py313 https://zuul.opendev.org/t/openstack/build/af35ee3dec66443ebb9c595c09274e6c : SUCCESS in 5m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ea24b1bae1e74ed288d733f0364772e5 : SUCCESS in 10m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/80cf93d2938c4081a2da05b8dd0dcf3c : SUCCESS in 2m 41s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/67c537358c524596b66e7e190a5e683f : SUCCESS in 4m 02s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/ca34b4b8f9a04e90a99ed68595cf157c : SUCCESS in 5m 56s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/1de7febda99049238c988ddf32f7c367 : SUCCESS in 3m 13s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/b5a6111253664d72a6a768b18894518a : SUCCESS in 6m 53s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/3cc7d18c1b8745f08279416fc1a23adc : SUCCESS in 4m 01s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/b92b2808fba141c4bdf75cc3f25cf0b1 : SUCCESS in 1h 00m 15s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/14b680d9aed84bbba7e9f9d4bac59091 : SUCCESS in 1h 47m 01s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/9c2e019af39d4d3ca205e9ff5e267ac2 : SUCCESS in 1h 08m 03s\n- octavia-v2-dsvm-scenario-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/14625c1591ad498592b67e72e6be1102 : SUCCESS in 1h 23m 49s\n- octavia-v2-dsvm-scenario-non-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/f68f487048f74311a46927d43eb8fd78 : SUCCESS in 1h 04m 23s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/3b4cce3eb8aa4fe1be15fc220d4b8570 : SUCCESS in 1h 51m 37s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/cbd4e9f493844d01966bed063a43ecdd : SUCCESS in 46m 08s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/931888c041864d3eab2f19244c1ad692 : FAILURE in 25m 24s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/ee40ed71e549401a9dd1802055f7a54c : FAILURE in 27m 02s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c7e91cffbd2a43e2891408bc2d89ded1 : SUCCESS in 22m 02s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/2839506d2f2642449530df133916a26c : TIMED_OUT in 2h 32m 56s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/e119f4d3e160461183ee3e82675ad4e8 : SUCCESS in 2h 11m 02s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/4a2f9d5d882846348c1363f3c2cec77c : SUCCESS in 2h 21m 00s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"8e68d197de2706a2fc447368cc3a64c90696b361","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2025-05-02 00:29:50.000000000","message":"Patch Set 9: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"5a74755ab6e7002e5ab1708dfd31db250f9c487e","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2025-05-05 07:11:30.000000000","message":"Patch Set 9: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":9},{"id":"c00d2e3b32ea34f0e844293f91dbc8045887533c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-05 07:12:03.000000000","message":"Patch Set 9: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":9},{"id":"2df4539cdbe73966625c88c8e1d4480b23d12048","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2025-05-05 08:23:59.000000000","message":"Patch Set 9: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":9},{"id":"14383b5680c6ff90001e9ad55331fed0e9ef93a0","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-05 08:59:08.000000000","message":"Patch Set 9: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6614ec997ef541d9adb91b8c36e6a041\n\n- test-release-openstack https://zuul.opendev.org/t/openstack/build/0ca313fae2334c3bbca53d9b002b87a4 : SUCCESS in 2m 29s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1797978107084a0ba1194e0800e29cb7 : SUCCESS in 5m 12s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/0da2fc62027c48fa833b61c248d78073 : SUCCESS in 4m 11s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/5839626ff3ce40b9b27ba1fed77919ec : SUCCESS in 4m 34s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5c614a7c05084d81954939afa2d09ba9 : SUCCESS in 12m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/bd83d96f77da4f0d8f20ddccb4f14397 : SUCCESS in 4m 49s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/c0a9723fa64d4eb19bb57ac56eaa8848 : SUCCESS in 3m 08s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/1f49869903394e2a88b7cedc22c10e5e : SUCCESS in 8m 02s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/f23512083ede471bb359f0f8f09a9500 : SUCCESS in 1h 36m 10s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/0d7647fc90124b6cb542b54f8aef3444 : SUCCESS in 1h 46m 16s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/f0dc07e8721e4b1a9061c5eedfb6ac46 : SUCCESS in 56m 12s\n- octavia-v2-dsvm-scenario-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/0ce6bbb7938f41598f938bb6e50fefa7 : SUCCESS in 1h 44m 20s\n- octavia-v2-dsvm-scenario-non-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/3b404ab529034d2aaf37859dc475a9b1 : SUCCESS in 52m 49s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/11412ea76d7a4bb2a45ec25468045477 : FAILURE in 28m 05s (non-voting)\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/d52f77121d1b4c2683414b5197b9e121 : FAILURE in 26m 19s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"e1390ca80b497d05825dbeb156fe95f7a6542aba","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-05 08:59:09.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":9},{"id":"4851a9d06456e69d65e951d95b4f6f6ac8ba6c30","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-05-05 09:00:33.000000000","message":"Patch Set 9:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3f0a715ffe8e40e7ba73712a3a9ee219\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/904a75096cb6455c947c12d9a02f51ff : SUCCESS in 50s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f6a13161a7b148a2ba590c6a359d81b0 : SUCCESS in 42s","accounts_in_message":[],"_revision_number":9},{"id":"58ee2d166356a66f9e63d91474d6ca03664c727a","author":{"_account_id":13861,"name":"yatin","email":"ykarel@redhat.com","username":"yatinkarel"},"date":"2025-05-05 12:49:43.000000000","message":"Patch Set 9:\n\n(1 comment)","accounts_in_message":[],"_revision_number":9}],"current_revision_number":9,"current_revision":"8ebf8d27c4209932ea1e5dfb3ca397af6c92e14e","revisions":{"6a8247d05ad9cfaae0ded505e10eabdef792bdf4":{"kind":"REWORK","_number":1,"created":"2023-12-06 21:07:49.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/12/902812/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/902812/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/902812/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/902812/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/902812/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/902812/1"}}},"commit":{"parents":[{"commit":"261ee7dcffb1fe634152c4f2ccf79264c1a653b5","subject":"Merge \"Add spec for SR-IOV in Octavia Amphora\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/261ee7dcffb1fe634152c4f2ccf79264c1a653b5"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-12-06 21:01:56.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-12-06 21:07:47.000000000","tz":0},"subject":"Add wsgi module to Octavia and remove wsgi script","message":"Add wsgi module to Octavia and remove wsgi script\n\nChanges in python packaging tooling mean that the wsgi_scripts\nfunctionality via PBR may not longer function.\n\nThis patch switches Octavia from using the PBR wsgi_scripts method to\nusing a new wsgi module that provides the same behavior as the generated\nwsgi scripts provided.\n\nA related devstack patch enables devstack to setup uWSGI to use the new\nmodule path instead of the generated wsgi scripts.\n\nThis also aligns Octavia to a new proposed OpenStack goal[1].\n\n[1] https://review.opendev.org/c/openstack/governance/+/902807\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/902758\nChange-Id: I54f8144a3347c3c5bf2e2e99e9d500a0c6fb89eb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6a8247d05ad9cfaae0ded505e10eabdef792bdf4"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6a8247d05ad9cfaae0ded505e10eabdef792bdf4"}]},"branch":"refs/heads/master"},"146c40c1b34b8ceb5c2682a1a75ef5d26fac23b7":{"kind":"REWORK","_number":2,"created":"2023-12-06 21:12:52.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/12/902812/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/902812/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/902812/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/902812/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/902812/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/902812/2"}}},"commit":{"parents":[{"commit":"261ee7dcffb1fe634152c4f2ccf79264c1a653b5","subject":"Merge \"Add spec for SR-IOV in Octavia Amphora\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/261ee7dcffb1fe634152c4f2ccf79264c1a653b5"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-12-06 21:01:56.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-12-06 21:12:46.000000000","tz":0},"subject":"Add wsgi module to Octavia and remove wsgi script","message":"Add wsgi module to Octavia and remove wsgi script\n\nChanges in python packaging tooling mean that the wsgi_scripts\nfunctionality via PBR may not longer function.\n\nThis patch switches Octavia from using the PBR wsgi_scripts method to\nusing a new wsgi module that provides the same behavior as the generated\nwsgi scripts provided.\n\nA related devstack patch enables devstack to setup uWSGI to use the new\nmodule path instead of the generated wsgi scripts.\n\nThis also aligns Octavia to a new proposed OpenStack goal[1].\n\n[1] https://review.opendev.org/c/openstack/governance/+/902807\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/902758\nChange-Id: I54f8144a3347c3c5bf2e2e99e9d500a0c6fb89eb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/146c40c1b34b8ceb5c2682a1a75ef5d26fac23b7"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/146c40c1b34b8ceb5c2682a1a75ef5d26fac23b7"}]},"branch":"refs/heads/master"},"5732241abb5c884c36e8d1bd348bfa1b1d571d84":{"kind":"REWORK","_number":3,"created":"2023-12-06 21:14:52.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/12/902812/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/902812/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/902812/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/902812/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/902812/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/902812/3"}}},"commit":{"parents":[{"commit":"261ee7dcffb1fe634152c4f2ccf79264c1a653b5","subject":"Merge \"Add spec for SR-IOV in Octavia Amphora\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/261ee7dcffb1fe634152c4f2ccf79264c1a653b5"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-12-06 21:01:56.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-12-06 21:14:45.000000000","tz":0},"subject":"Add wsgi module to Octavia and remove wsgi script","message":"Add wsgi module to Octavia and remove wsgi script\n\nChanges in python packaging tooling mean that the wsgi_scripts\nfunctionality via PBR may not longer function.\n\nThis patch switches Octavia from using the PBR wsgi_scripts method to\nusing a new wsgi module that provides the same behavior as the generated\nwsgi scripts provided.\n\nA related devstack patch enables devstack to setup uWSGI to use the new\nmodule path instead of the generated wsgi scripts.\n\nThis also aligns Octavia to a new proposed OpenStack goal[1].\n\n[1] https://review.opendev.org/c/openstack/governance/+/902807\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/902758\nChange-Id: I54f8144a3347c3c5bf2e2e99e9d500a0c6fb89eb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5732241abb5c884c36e8d1bd348bfa1b1d571d84"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5732241abb5c884c36e8d1bd348bfa1b1d571d84"}]},"branch":"refs/heads/master"},"e5a0ca51a54ff51d7a97006a196e0c0c5a4ed965":{"kind":"REWORK","_number":4,"created":"2023-12-06 23:03:02.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/12/902812/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/902812/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/902812/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/902812/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/902812/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/902812/4"}}},"commit":{"parents":[{"commit":"261ee7dcffb1fe634152c4f2ccf79264c1a653b5","subject":"Merge \"Add spec for SR-IOV in Octavia Amphora\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/261ee7dcffb1fe634152c4f2ccf79264c1a653b5"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-12-06 21:01:56.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-12-06 23:02:35.000000000","tz":0},"subject":"Add wsgi module to Octavia and remove wsgi script","message":"Add wsgi module to Octavia and remove wsgi script\n\nChanges in python packaging tooling mean that the wsgi_scripts\nfunctionality via PBR may not longer function.\n\nThis patch switches Octavia from using the PBR wsgi_scripts method to\nusing a new wsgi module that provides the same behavior as the generated\nwsgi scripts provided.\n\nA related devstack patch enables devstack to setup uWSGI to use the new\nmodule path instead of the generated wsgi scripts.\n\nThis also aligns Octavia to a new proposed OpenStack goal[1].\n\n[1] https://review.opendev.org/c/openstack/governance/+/902807\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/902758\nChange-Id: I54f8144a3347c3c5bf2e2e99e9d500a0c6fb89eb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e5a0ca51a54ff51d7a97006a196e0c0c5a4ed965"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/e5a0ca51a54ff51d7a97006a196e0c0c5a4ed965"}]},"branch":"refs/heads/master"},"b459a2ef32bff65a1c13f0baba86c32d6ca0dc25":{"kind":"REWORK","_number":5,"created":"2023-12-06 23:04:37.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/12/902812/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/902812/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/902812/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/902812/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/902812/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/902812/5"}}},"commit":{"parents":[{"commit":"261ee7dcffb1fe634152c4f2ccf79264c1a653b5","subject":"Merge \"Add spec for SR-IOV in Octavia Amphora\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/261ee7dcffb1fe634152c4f2ccf79264c1a653b5"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-12-06 21:01:56.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-12-06 23:04:28.000000000","tz":0},"subject":"Add wsgi module to Octavia and remove wsgi script","message":"Add wsgi module to Octavia and remove wsgi script\n\nChanges in python packaging tooling mean that the wsgi_scripts\nfunctionality via PBR may not longer function.\n\nThis patch switches Octavia from using the PBR wsgi_scripts method to\nusing a new wsgi module that provides the same behavior as the generated\nwsgi scripts provided.\n\nA related devstack patch enables devstack to setup uWSGI to use the new\nmodule path instead of the generated wsgi scripts.\n\nThis also aligns Octavia to a new proposed OpenStack goal[1].\n\n[1] https://review.opendev.org/c/openstack/governance/+/902807\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/902758\nChange-Id: I54f8144a3347c3c5bf2e2e99e9d500a0c6fb89eb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b459a2ef32bff65a1c13f0baba86c32d6ca0dc25"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b459a2ef32bff65a1c13f0baba86c32d6ca0dc25"}]},"branch":"refs/heads/master"},"984b8abe25b16bf90d346c87b81b41d88cc80ed9":{"kind":"REWORK","_number":6,"created":"2023-12-06 23:20:26.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"ref":"refs/changes/12/902812/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/902812/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/902812/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/902812/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/902812/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/902812/6"}}},"commit":{"parents":[{"commit":"261ee7dcffb1fe634152c4f2ccf79264c1a653b5","subject":"Merge \"Add spec for SR-IOV in Octavia Amphora\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/261ee7dcffb1fe634152c4f2ccf79264c1a653b5"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-12-06 21:01:56.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-12-06 23:19:58.000000000","tz":0},"subject":"Add wsgi module to Octavia and remove wsgi script","message":"Add wsgi module to Octavia and remove wsgi script\n\nChanges in python packaging tooling mean that the wsgi_scripts\nfunctionality via PBR may not longer function.\n\nThis patch switches Octavia from using the PBR wsgi_scripts method to\nusing a new wsgi module that provides the same behavior as the generated\nwsgi scripts provided.\n\nA related devstack patch enables devstack to setup uWSGI to use the new\nmodule path instead of the generated wsgi scripts.\n\nThis also aligns Octavia to a new proposed OpenStack goal[1].\n\n[1] https://review.opendev.org/c/openstack/governance/+/902807\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/902758\nChange-Id: I54f8144a3347c3c5bf2e2e99e9d500a0c6fb89eb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/984b8abe25b16bf90d346c87b81b41d88cc80ed9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/984b8abe25b16bf90d346c87b81b41d88cc80ed9"}]},"branch":"refs/heads/master"},"784587b217cb1ed4f15a2bec9713f66f6a218512":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2025-04-30 07:08:25.000000000","uploader":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"real_uploader":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"ref":"refs/changes/12/902812/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/902812/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/902812/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/902812/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/902812/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/902812/7"}}},"commit":{"parents":[{"commit":"ba617006a611abcb343570ce09a7f5d32834e4a0","subject":"zuul: Drop centos9/py39 and make grenade and barbican non-voting","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ba617006a611abcb343570ce09a7f5d32834e4a0"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-12-06 21:01:56.000000000","tz":0},"committer":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2025-04-30 07:08:25.000000000","tz":0},"subject":"Add wsgi module to Octavia and remove wsgi script","message":"Add wsgi module to Octavia and remove wsgi script\n\nChanges in python packaging tooling mean that the wsgi_scripts\nfunctionality via PBR may not longer function.\n\nThis patch switches Octavia from using the PBR wsgi_scripts method to\nusing a new wsgi module that provides the same behavior as the generated\nwsgi scripts provided.\n\nA related devstack patch enables devstack to setup uWSGI to use the new\nmodule path instead of the generated wsgi scripts.\n\nThis also aligns Octavia to a new proposed OpenStack goal[1].\n\n[1] https://review.opendev.org/c/openstack/governance/+/902807\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/902758\nChange-Id: I54f8144a3347c3c5bf2e2e99e9d500a0c6fb89eb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/784587b217cb1ed4f15a2bec9713f66f6a218512"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/784587b217cb1ed4f15a2bec9713f66f6a218512"}]},"branch":"refs/heads/master","description":"Rebase"},"45ebc92121eca79d58189555ef5a46b3c3939e51":{"kind":"REWORK","_number":8,"created":"2025-04-30 10:08:40.000000000","uploader":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"ref":"refs/changes/12/902812/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/902812/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/902812/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/902812/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/902812/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/902812/8"}}},"commit":{"parents":[{"commit":"84f512687a6a1995378bfc5b13753ef4a145601c","subject":"Merge \"Ignore load_balancer graph in task results to avoid break write data in jobboard DB\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/84f512687a6a1995378bfc5b13753ef4a145601c"}]}],"author":{"name":"Eduardo Olivares","email":"eolivare@redhat.com","date":"2025-04-30 07:01:33.000000000","tz":120},"committer":{"name":"Eduardo Olivares","email":"eolivare@redhat.com","date":"2025-04-30 10:08:31.000000000","tz":120},"subject":"Add wsgi module to Octavia and remove wsgi script","message":"Add wsgi module to Octavia and remove wsgi script\n\nChanges in python packaging tooling mean that the wsgi_scripts\nfunctionality via PBR may not longer function.\n\nThis patch switches Octavia from using the PBR wsgi_scripts method to\nusing a new wsgi module that provides the same behavior as the generated\nwsgi scripts provided.\n\nA related devstack patch enables devstack to setup uWSGI to use the new\nmodule path instead of the generated wsgi scripts.\n\nThis also aligns Octavia to a new proposed OpenStack goal[1].\n\n[1] https://review.opendev.org/c/openstack/governance/+/902807\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/902758\n\nCloses-Bug: #2109665\n\nIn order to make the CI work, this patch was merged with:\n\nzuul: Drop centos9/py39 and make grenade and barbican non-voting\n\nThe requirements repo has dropped upper-constraints for python3.9, so\nthe centos9 and py39 based jobs are no longer working on master, let\u0027s\ndrop them.\n\nThe grenade jobs are broken due to the #2109665, make them\ntemporarily non-voting, so that we can merge the fix[0] on master first,\nbackport it and then re-enable them.\n\nFinally, octavia-v2-dsvm-tls-barbican fails due to #2109584, so a fix\nsimilar to [0] will be needed for barbican.\n\n[0] https://review.opendev.org/c/openstack/octavia/+/902812\n\nRelated-Bug: #2109665\nRelated-Bug: #2109584\n\nChange-Id: I54f8144a3347c3c5bf2e2e99e9d500a0c6fb89eb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/45ebc92121eca79d58189555ef5a46b3c3939e51"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/45ebc92121eca79d58189555ef5a46b3c3939e51"}]},"branch":"refs/heads/master"},"8ebf8d27c4209932ea1e5dfb3ca397af6c92e14e":{"kind":"REWORK","_number":9,"created":"2025-05-01 15:29:59.000000000","uploader":{"_account_id":31291,"name":"Eduardo Olivares","email":"eolivare@redhat.com","username":"eolivare"},"ref":"refs/changes/12/902812/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/12/902812/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/902812/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/902812/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/12/902812/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/12/902812/9"}}},"commit":{"parents":[{"commit":"84f512687a6a1995378bfc5b13753ef4a145601c","subject":"Merge \"Ignore load_balancer graph in task results to avoid break write data in jobboard DB\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/84f512687a6a1995378bfc5b13753ef4a145601c"}]}],"author":{"name":"Eduardo Olivares","email":"eolivare@redhat.com","date":"2025-04-30 07:01:33.000000000","tz":120},"committer":{"name":"Eduardo Olivares","email":"eolivare@redhat.com","date":"2025-05-01 15:29:59.000000000","tz":0},"subject":"Add wsgi module to Octavia and remove wsgi script","message":"Add wsgi module to Octavia and remove wsgi script\n\nChanges in python packaging tooling mean that the wsgi_scripts\nfunctionality via PBR may not longer function.\n\nThis patch switches Octavia from using the PBR wsgi_scripts method to\nusing a new wsgi module that provides the same behavior as the generated\nwsgi scripts provided.\n\nA related devstack patch enables devstack to setup uWSGI to use the new\nmodule path instead of the generated wsgi scripts.\n\nThis also aligns Octavia to a new proposed OpenStack goal[1].\n\n[1] https://review.opendev.org/c/openstack/governance/+/902807\n\nDepends-On: https://review.opendev.org/c/openstack/devstack/+/902758\n\nCloses-Bug: #2109665\n\nIn order to make the CI work, this patch was merged with:\n\nzuul: Drop centos9/py39 and make grenade and barbican non-voting\n\nThe requirements repo has dropped upper-constraints for python3.9, so\nthe centos9 and py39 based jobs are no longer working on master, let\u0027s\ndrop them.\n\nThe grenade jobs are broken due to the #2109665, make them\ntemporarily non-voting, so that we can merge the fix[0] on master first,\nbackport it and then re-enable them.\n\nFinally, octavia-v2-dsvm-tls-barbican fails due to #2109584, so a fix\nsimilar to [0] will be needed for barbican.\n\n[0] https://review.opendev.org/c/openstack/octavia/+/902812\n\nRelated-Bug: #2109665\nRelated-Bug: #2109584\n\nChange-Id: I54f8144a3347c3c5bf2e2e99e9d500a0c6fb89eb\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8ebf8d27c4209932ea1e5dfb3ca397af6c92e14e"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/8ebf8d27c4209932ea1e5dfb3ca397af6c92e14e"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
