)]}'
{"id":"openstack%2Foctavia~903081","triplet_id":"openstack%2Foctavia~master~I9aea55371d3f3d054eba5673d09eead23d5c95a8","project":"openstack/octavia","branch":"master","attention_set":{},"removed_from_attention_set":{"27442":{"account":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"last_update":"2024-05-22 20:28:24.000000000","reason":"Change was submitted"},"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2024-05-22 20:28:24.000000000","reason":"Change was submitted"},"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2024-05-03 14:46:17.000000000","reason":"\u003cGERRIT_ACCOUNT_11628\u003e replied on the change","reason_account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},"34429":{"account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"last_update":"2024-05-14 13:23:32.000000000","reason":"\u003cGERRIT_ACCOUNT_34429\u003e replied on the change","reason_account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}}},"hashtags":[],"change_id":"I9aea55371d3f3d054eba5673d09eead23d5c95a8","subject":"Fix incorrect filtering when using bool attrs","status":"MERGED","created":"2023-12-07 12:48:56.000000000","updated":"2024-05-22 20:29:33.000000000","submitted":"2024-05-22 20:28:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":7,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"903081","meta_rev_id":"1484f3f49e29bc6ddfbacfd9de0a57528905330c","_number":903081,"virtual_id_number":903081,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-05-22 20:28:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"recommended":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"all":[{"value":1,"date":"2024-05-03 15:28:07.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":2,"date":"2024-05-03 14:46:17.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":2,"date":"2024-05-14 13:23:32.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":1,"date":"2024-05-14 13:23:32.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"value":0,"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-12-07 15:54:59.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-01-02 10:53:56.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"CC"},{"updated":"2024-01-24 16:58:38.000000000","updated_by":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"reviewer":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"state":"REVIEWER"},{"updated":"2024-02-23 00:36:44.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2024-05-03 15:28:07.000000000","updated_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"reviewer":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"state":"REVIEWER"},{"updated":"2024-05-14 13:23:32.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"}],"messages":[{"id":"7e01ee7635372d5c495cfcdf8987a10d55f26b71","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-12-07 12:48:56.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"158134397fc5136e5ccd930f005bdbcdcab11437","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-07 15:54:59.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/31effff71e5348d68aac459629b56ae5\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/90ce5aa0c33d47dbb9cc403615349457 : SUCCESS in 13m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/c17c484348e64f3a9a628e023f117810 : SUCCESS in 4m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/ee5b060385c54fed9af934f18e97f4d6 : SUCCESS in 4m 22s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f7ee7f1b7b3947e48661423cae5a1042 : SUCCESS in 6m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/91fb2c91e18040e5ad3ce1ffdb49a03b : SUCCESS in 12m 27s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/4fba7fc8b29348da80f62eb701f8cfb1 : SUCCESS in 5m 18s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/8d3527ec4d8c4c239a2a5ef9a37875fd : SUCCESS in 5m 29s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/a2ca5fb5e9a1488297c715974552fb33 : SUCCESS in 7m 04s\n- octavia-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/a8d8883150c54ac3a3275152ad360427 : SUCCESS in 12m 28s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/e6ebf78c63314bb5990e86a8975a0828 : SUCCESS in 4m 24s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/9325e7830df540f1ac65d8adbbfd9c52 : SUCCESS in 7m 26s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/8ad3bbd76ccf40869f4c2d03c7942887 : SUCCESS in 10m 12s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/489e8c1284dd4e1c8f9b589658b2c4b4 : SUCCESS in 7m 23s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/f42b99d44eff4cc0a23bd53206f8cc3a : SUCCESS in 1h 33m 17s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/df7a7159d84646998db20b2c96faf7d2 : SUCCESS in 1h 36m 48s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/4df54d958e874e7cbaaa5d36b48251fa : SUCCESS in 1h 03m 36s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/34c873ac82a7433b91a87084551fbc8c : SUCCESS in 1h 43m 30s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/6a4b1e29314e4becb5a0006a87095bc5 : SUCCESS in 1h 01m 54s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/de50a7e9fb5c4374a7eebea50450261a : SUCCESS in 35m 05s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/0eeaf367441545faa1d3c7ca2c138776 : SUCCESS in 39m 27s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/39542da015674b72874bbdd3ad11d178 : SUCCESS in 56m 15s (non-voting)\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/6db1ec9163584a7f8eb7e761d075548b : SUCCESS in 1h 13m 48s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/2b9562d9b48241649a16ce8795864b8a : SUCCESS in 33m 32s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/6e47595a5fef4698b6199d0a02c3c2ad : SUCCESS in 2h 00m 13s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/7a545ab5551c42829f9d97f33cb2d238 : TIMED_OUT in 3h 02m 55s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/03b657ab34d5495f8d3a2c498b90c6f9 : SUCCESS in 2h 09m 06s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/4b73cedd9e694c9483216aa8cab1c72c : SUCCESS in 2h 34m 00s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"7e12060eb74b7330e7c52e1842b151b61935f506","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-12-08 08:18:41.000000000","message":"Patch Set 1: Workflow-1","accounts_in_message":[],"_revision_number":1},{"id":"df52a06547caaac195b79079d29d6d6cee7f296d","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-12-08 13:14:50.000000000","message":"Patch Set 1: -Workflow","accounts_in_message":[],"_revision_number":1},{"id":"9d134daeeb1d20818714028ab9f42f59a20568e0","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-01-02 10:53:56.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f09917500164522b0204014408c6429dad0a3b70","author":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"date":"2024-01-24 16:58:38.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"ca65b7b8758ec9d00df734ca6838dc13fe5506b5","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2024-02-23 00:36:44.000000000","message":"Patch Set 1: Code-Review-1\n\n(2 comments)","accounts_in_message":[],"_revision_number":1},{"id":"85811fc2c77aa482523d882d04e9529612d0821b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2024-05-03 13:42:08.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+1, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"f265b9280b5b340c55d0202f9da8edd9ae1a9eb8","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2024-05-03 14:46:17.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"850c672addd0602d00f4d62d36ad7851f0938757","author":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"},"date":"2024-05-03 15:28:07.000000000","message":"Patch Set 2: Code-Review+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"552001a28e661c80671e833e27d3365883dbe4d7","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-03 16:49:02.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1c75bd3a574444e2a29f8ae290b06f40\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c7851f1a07c143249e50bbe772b9efe2 : SUCCESS in 10m 49s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/5b17a8228962454e8f34acbe8bc8996e : SUCCESS in 4m 54s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/5bf12456fad74c7d94fde3dc257518af : SUCCESS in 6m 02s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/10ec92923e194991aeea430b1583523c : SUCCESS in 5m 46s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/8c1a862669954ef090247274b6fb0032 : SUCCESS in 7m 58s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/06f5cb98142a4bd2bbeb172ea202218f : SUCCESS in 12m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9844ed75891c4afc88109d517955a2fe : SUCCESS in 4m 49s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/28f42bda731844faaadc0e88e9064f10 : SUCCESS in 5m 31s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/b29f7e5fb44d4b02958727ef67898196 : SUCCESS in 6m 46s\n- octavia-tox-functional-py311-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/37c39b64089f47ec9d7d5283286a4564 : SUCCESS in 7m 49s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/33b63053a519409da3e235acdb07c3af : SUCCESS in 4m 45s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/fb979daff2cc45f3a8cc2515fef495a2 : SUCCESS in 8m 49s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/9e6b73fd6db5456bb4e9ce1f9f73e820 : SUCCESS in 6m 40s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/c1ca5ec7394e40a5b5ca56dc8f3a2f30 : SUCCESS in 5m 19s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/ff9dcbff0cec4014b09f3bdd73c2c217 : SUCCESS in 1h 35m 42s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/0dfa224a3cb54ad490195cf05267be37 : SUCCESS in 1h 43m 44s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/b0244007bdc14fec9af58f971d925b91 : SUCCESS in 1h 00m 57s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/2ec1e4ec0b6b4cf39129c2464fdca96d : SUCCESS in 2h 11m 23s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/1c2412f12b26427c8e6ffc0509c32a61 : SUCCESS in 1h 04m 32s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/a0b1f63e4a7b40f4878db6b6f300fc82 : SUCCESS in 44m 31s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/0f3be0a37b1e47f39774fa1b4553098c : SUCCESS in 38m 54s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/5c8ec5cc09a544e7983c3137252e22ad : SUCCESS in 1h 02m 30s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/fe38772418f3425daa9eb5556575124a : SUCCESS in 37m 41s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/e413fa75a3bc419aa0ec9f70c4168531 : SUCCESS in 2h 06m 51s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/2037c845f3eb4100a65318044d84f529 : TIMED_OUT in 3h 03m 19s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/57d0554284804e40bc42c7951aaf65fc : FAILURE in 27m 04s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/f45c6d1319fa45f096f329470fc810fe : TIMED_OUT in 2h 33m 35s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"015ad21f2d3b4851cd8b51e02d91a1c911edc4a4","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-05-14 13:23:32.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"7cdd0e57532af8bdb7d8c50c62de4e8336a86ae3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-14 13:24:20.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"ebcdf68f13af04b536e668030eb281abde808006","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-14 15:06:32.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/67adc45bb25c4e5b8f9f6448e2d1e96d\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9544dd32ede042cf8a7234217b77143d : FAILURE in 5m 24s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b83e9086ce484edbb3ab8c95798db360 : SUCCESS in 5m 00s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e567b92802e0437eb9b68f8361d0f211 : SUCCESS in 5m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2d04c51d5edc4eacbded5acdca99664f : CANCELED in 4m 29s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9ae2f75a14434d5eb5581f9ae89e4d68 : CANCELED in 6m 02s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/5f2b0621c4184f3c9009404f16f471ad : SUCCESS in 4m 03s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/2c2306b125014e2e8bf73b1de4c7d032 : SUCCESS in 5m 24s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/0bde9c920d704035aa597e1a040611e4 : CANCELED in 22s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/6f0b242bafd64948a657fb9ed6755567 : CANCELED in 10m 34s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/5fd491cb42584bb78e849107a700561d : CANCELED in 10m 31s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/125deeaa8a2f45f39647812c7f3af6a1 : CANCELED in 10m 33s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/86e2cd9a12884d03949698c502e5b249 : CANCELED in 10m 00s\n\nWarning:\n  Comments left for invalid file octavia/amphorae/backends/agent/api_server/keepalived.py\n  Comments left for invalid file octavia/amphorae/backends/agent/api_server/keepalivedlvs.py\n  Comments left for invalid file octavia/amphorae/backends/agent/api_server/loadbalancer.py\n  Comments left for invalid file octavia/common/tls_utils/cert_parser.py\n  Comments left for invalid file octavia/controller/worker/v2/tasks/compute_tasks.py","accounts_in_message":[],"_revision_number":2},{"id":"4d2c0baa710a69ce9264cb176e39aef7ab0c4928","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-05-22 15:27:30.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"2af9830fe85bd0a9bf47d085db16c8b86703101c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-22 18:29:45.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8a7b402404cb4c4e90d192741c1d3ed9\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1026689f93934f8097b8207599d8528f : SUCCESS in 10m 03s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e3184f6555954e1e96291ffb59406783 : SUCCESS in 4m 22s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/93c726d5df9a423ba633455520d258f5 : SUCCESS in 5m 20s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c5982c4932fb4077a66cdcf3642fe09c : SUCCESS in 5m 08s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/eb0181ab53e4455b9aed0bee31b88725 : SUCCESS in 9m 19s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/054a6c9289b540a1a37d847f9e7922a0 : SUCCESS in 10m 19s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/212bb1972ee1408482e7cdbd6e3b8b2a : SUCCESS in 5m 28s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/6fd545278a364ff7a96e9f96ba9f4db5 : SUCCESS in 5m 29s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/d0472dfb304f4796ac51d0e6a864046b : SUCCESS in 6m 37s\n- octavia-tox-functional-py311-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/dfb278ae08bf47a19da74e76c66b6228 : SUCCESS in 6m 56s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/8b3d6e5372854a7bba4ca7cd06567be7 : SUCCESS in 5m 28s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/73e70fc469d04dd79f7877af1568bac3 : SUCCESS in 7m 25s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/f71b3d6ec6cc4d7ab05ba2e590a958ac : SUCCESS in 8m 59s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/d7b024bcc4ba4ed283d900fdf5d57517 : SUCCESS in 4m 44s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/e65a858636f34273a56ade008e160586 : SUCCESS in 1h 42m 33s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/32f2f1fa8aa1462aa9f8cce1c0ca30b6 : SUCCESS in 1h 43m 31s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/09a4ea5bb08945f6a99956f52bfb8c8d : SUCCESS in 1h 11m 52s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/34a8395b40274e57a89c91893e7dde04 : SUCCESS in 2h 01m 34s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/08d1011edcd548e89a42ed77b231a790 : SUCCESS in 1h 09m 30s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/1c7fa858b0634560b1429648b38f076d : SUCCESS in 38m 32s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/7e07b484a0c0475ca0166b975093d453 : SUCCESS in 39m 54s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/63b4bcb548784a3ea7d4a6dff6ba9672 : SUCCESS in 1h 12m 38s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b17946c2a33e46e39f70d1259e5ffdf5 : SUCCESS in 42m 13s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/2c3b316d2c8e4208a45fe0cb273e482b : TIMED_OUT in 2h 33m 15s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/6a1c0f1e8adc4094926c7a70f7f88e17 : SUCCESS in 2h 59m 52s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/f74dfda0d7e9490aba154695bfbd28b0 : FAILURE in 31m 42s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/afb4c94e447b4661ba8041f6949f58eb : TIMED_OUT in 2h 33m 22s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"0c44b7728e14fa4e11707160e918d871b34f0ea5","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-22 18:30:29.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"8d6457d0921847c890e6327d45eb2009409c55fb","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-22 20:28:24.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d0c408adb5304b44b08ef0c754c9a84c\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8d6d1c95573e4f26955a79c3b1aeb4da : SUCCESS in 4m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/065bbb0d03ea497eb1b96d914c5d79e0 : SUCCESS in 5m 19s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/44af8a49edb94bbc9040d714b6d0ce6a : SUCCESS in 5m 24s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1896969577964e8086fa433bd5d9915f : SUCCESS in 11m 46s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/71be63427d524da9ac5cba0e60b1ea78 : SUCCESS in 5m 28s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/4e3d04e1b83740b684990d4b7c56fefd : SUCCESS in 4m 36s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/f2feaedc6bc8411aba0cad3480a5f82e : SUCCESS in 4m 41s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/f4985b4892064a84b1502382355dc96f : SUCCESS in 1h 53m 00s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/179b88e855b24ae6949cb779e470bb28 : SUCCESS in 1h 45m 07s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/1d61b0d810ab47d1ac14f2d26888cd1d : SUCCESS in 1h 07m 35s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/8fec83a34c2c49c8adaa5bae5600ae50 : SUCCESS in 40m 43s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/5d9f2e7e79a6493e8797a0a2928adaa9 : SUCCESS in 43m 41s","accounts_in_message":[],"_revision_number":2},{"id":"fade72f88e13c5cef83e8a5c1edc7b7845c465a8","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-22 20:28:24.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"1484f3f49e29bc6ddfbacfd9de0a57528905330c","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-22 20:29:33.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/3c62f716be1d4a239a178a5853642607\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0ec73dc7eb48450fb0b7c2dd79fa9f4b : SUCCESS in 49s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/22f2b5f05828429390de209c368da404 : SUCCESS in 40s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"cdbcc3fd777d641e3b30b6577470213d30d4269a","revisions":{"49068dc96a8fef220ea1901aa03e9bb08a14e57b":{"kind":"REWORK","_number":1,"created":"2023-12-07 12:48:56.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/81/903081/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/81/903081/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/81/903081/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/81/903081/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/81/903081/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/81/903081/1"}}},"commit":{"parents":[{"commit":"261ee7dcffb1fe634152c4f2ccf79264c1a653b5","subject":"Merge \"Add spec for SR-IOV in Octavia Amphora\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/261ee7dcffb1fe634152c4f2ccf79264c1a653b5"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-12-07 09:14:47.000000000","tz":-300},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-12-07 12:43:47.000000000","tz":-300},"subject":"Fix incorrect filtering when using bool attrs","message":"Fix incorrect filtering when using bool attrs\n\nWhen filtering the GET calls with a boolean attribute, the value of the\nattribute was a string and not a boolean, it made the queries incorrect.\nNow, every boolean attribute is converted from a string to a boolean\nbefore querying the DB.\n\nCloses-Bug: #2045803\nChange-Id: I9aea55371d3f3d054eba5673d09eead23d5c95a8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/49068dc96a8fef220ea1901aa03e9bb08a14e57b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/49068dc96a8fef220ea1901aa03e9bb08a14e57b"}]},"branch":"refs/heads/master"},"cdbcc3fd777d641e3b30b6577470213d30d4269a":{"kind":"REWORK","_number":2,"created":"2024-05-03 13:42:08.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/81/903081/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/81/903081/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/81/903081/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/81/903081/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/81/903081/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/81/903081/2"}}},"commit":{"parents":[{"commit":"33c755d753c6a7502aede5901b3e7ea75fa0203c","subject":"Merge \"Updates for new Python runtimes in 2024.2\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/33c755d753c6a7502aede5901b3e7ea75fa0203c"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-12-07 09:14:47.000000000","tz":-300},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2024-05-03 13:28:58.000000000","tz":-240},"subject":"Fix incorrect filtering when using bool attrs","message":"Fix incorrect filtering when using bool attrs\n\nWhen filtering the GET calls with a boolean attribute, the value of the\nattribute was a string and not a boolean, it made the queries incorrect.\nNow, every boolean attribute is converted from a string to a boolean\nbefore querying the DB.\n\nCloses-Bug: #2045803\nChange-Id: I9aea55371d3f3d054eba5673d09eead23d5c95a8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/cdbcc3fd777d641e3b30b6577470213d30d4269a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/cdbcc3fd777d641e3b30b6577470213d30d4269a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":31664,"name":"Omer Schwartz","email":"oschwart@redhat.com","username":"oschwart"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
