)]}'
{"id":"openstack%2Foctavia~903674","triplet_id":"openstack%2Foctavia~master~I68880b83c8a555ed30c3ce18b73ae27b8db34dc8","project":"openstack/octavia","branch":"master","attention_set":{},"removed_from_attention_set":{"9816":{"account":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"last_update":"2023-12-19 10:02:48.000000000","reason":"Change was submitted"},"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2023-12-19 07:52:11.000000000","reason":"\u003cGERRIT_ACCOUNT_29244\u003e replied on the change","reason_account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},"34429":{"account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"last_update":"2023-12-19 10:02:48.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I68880b83c8a555ed30c3ce18b73ae27b8db34dc8","subject":"Unpin flake8-import-order","status":"MERGED","created":"2023-12-14 12:49:21.000000000","updated":"2023-12-19 10:04:04.000000000","submitted":"2023-12-19 10:02:48.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"903674","meta_rev_id":"1e04b49da4e819c8a828f845e5d8b624d352b3bb","_number":903674,"virtual_id_number":903674,"owner":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2023-12-19 10:02:47.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"recommended":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"all":[{"value":1,"date":"2023-12-18 10:25:20.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"value":2,"date":"2023-12-14 16:41:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":2,"date":"2023-12-19 07:52:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":1,"date":"2023-12-19 07:52:11.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2023-12-14 15:58:49.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2023-12-14 16:41:01.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2023-12-18 10:25:20.000000000","updated_by":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"reviewer":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"state":"REVIEWER"},{"updated":"2023-12-19 01:01:23.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2023-12-19 01:01:23.000000000","updated_by":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"}],"messages":[{"id":"eaa5bd2f675aa76aba5210e1eba4a3f8f40e5bb8","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"date":"2023-12-14 12:49:21.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"33369ebf7cddaaaccf6c727c26d29d365b85ea17","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-14 15:58:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/92efc5d83b234ff2b77dfcc20d0cf799\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/ebbf41c17014403ea855e381fe6085eb : SUCCESS in 7m 30s\n- requirements-check https://zuul.opendev.org/t/openstack/build/3f0fa8777b0a464899e49d025c415ee0 : SUCCESS in 4m 23s\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/87f9e8d6e23f4fdc8015eb84f165f3c4 : SUCCESS in 11m 21s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6f1a749752d740499a443d2ae25fe220 : SUCCESS in 4m 46s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c332bde970f74d59ae54f41fb1891a1d : SUCCESS in 5m 03s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/57b69f6d38ef44a29aa6fdef2c47a646 : SUCCESS in 4m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1ecc7475373d4641bde33e0e410ed004 : SUCCESS in 10m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8711eeb7c62842c99579c05823debb19 : SUCCESS in 5m 18s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/d5c93e5eb0a147bda1cdb112f03de00d : SUCCESS in 5m 18s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/b079bcb6b97a44cca247bb07ca6dd388 : SUCCESS in 7m 51s\n- octavia-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/83d3a17c5c294756b380834bba99107f : SUCCESS in 9m 28s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/3f8bf4f33aa64a568ff95af02ee36d97 : SUCCESS in 4m 35s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/f808382031504bfc90392728705129e6 : SUCCESS in 7m 18s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/0f4be9b4a10c4e179cf8dc21a293f381 : SUCCESS in 7m 21s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/d122cf32cf17430aa20cf8132b0849e2 : SUCCESS in 8m 03s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/8a1db97d6d7c414c92533466771c8217 : SUCCESS in 1h 11m 49s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/abba9ec98a44447d869992706c427db6 : SUCCESS in 1h 36m 09s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/cf970d79ff1c4ca9b5e0fc3767ad1c38 : SUCCESS in 55m 50s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/5201e55a54024042acc64ed60596a5fa : SUCCESS in 1h 39m 40s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/9a164a10dc134625b64f180926bfa08d : SUCCESS in 1h 30m 58s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/88d157efde6e4b62ac8d5b018bcfdc45 : SUCCESS in 35m 34s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/1f0a58b258af40fbaa8893fae1098c1d : SUCCESS in 38m 22s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/70d6a6bfc2b946ad9abfdafb72cdd692 : SUCCESS in 37m 54s (non-voting)\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/04a8546a249349dcb1789ee6a0932d98 : SUCCESS in 1h 02m 32s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a958a416c6944f0e99ebced1efb24183 : SUCCESS in 32m 58s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/af44a36fbf6f4040917713702e75e4b2 : TIMED_OUT in 2h 34m 23s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/9f5996616c954f7689ca8263aa9df632 : TIMED_OUT in 3h 04m 59s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/8ab85ec239474281be914bb96dd7c5b5 : TIMED_OUT in 2h 34m 06s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/ee2594dc2fcc404ba704b8883cea6759 : TIMED_OUT in 2h 34m 42s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"07318c430de0a531cc8858ac0f3fa11550e1b48d","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2023-12-14 16:41:01.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7f4c75e51496b5a72a0bd55e8426ec226e3df902","author":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"date":"2023-12-18 10:25:20.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"f36bfc54c5b6ea95954f78871efc52446ecc89e0","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2023-12-19 07:52:11.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":1},{"id":"77de5ca37c737b8a19f68021d2546a8997f76f9f","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-19 07:52:46.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"67e775d69e4a761561a32f2b18c9a4acdf51c8b3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-19 10:02:47.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/88a2a5f92c6e4d9b99807a1d11a565d5\n\n- build-openstack-api-ref https://zuul.opendev.org/t/openstack/build/8f02a640c483438da93bc0a4dfc89889 : SUCCESS in 6m 14s\n- requirements-check https://zuul.opendev.org/t/openstack/build/7775a0a1a5aa403bb4e5bc42403e0a67 : SUCCESS in 2m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/de91927ea1204e92b53f252d1fb2a160 : SUCCESS in 4m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c87e5772dceb4fce8f90450a4b23db63 : SUCCESS in 4m 50s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/7ec1bf48fbf94603b244057962f9e964 : SUCCESS in 4m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c9cabc4790c9429b82b25a1027672267 : SUCCESS in 12m 02s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8d44c052707644c788dcc313f4313a9f : SUCCESS in 5m 18s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/c465dfc1a9d34195a4aa447d07157ea7 : SUCCESS in 4m 50s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/1a037c2af7084945889db5d4debf9399 : SUCCESS in 8m 04s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/231085ef43f14e959041c627ba8fff81 : SUCCESS in 1h 26m 26s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/b08cefc62ab94279b9845bc810fa41e3 : SUCCESS in 2h 08m 47s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/3f6d426cbf634e598287f745d52fea86 : SUCCESS in 1h 21m 39s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/a72b9077ea274abdb53a5d66b02cc5ad : SUCCESS in 55m 44s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/3f42a38dc01c4a3cb16e303a64d09967 : SUCCESS in 1h 01m 23s","accounts_in_message":[],"_revision_number":1},{"id":"dbb0f0747cf28a4bccad406d953fa3cf7e7a7efe","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-19 10:02:48.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"1e04b49da4e819c8a828f845e5d8b624d352b3bb","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2023-12-19 10:04:04.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/ae6b66d8e22a4683b12a26f027c39d1d\n\n- promote-api-ref https://zuul.opendev.org/t/openstack/build/630e9bafd9bd4fa890a93f64e9000222 : SUCCESS in 47s\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ed2235a456da4d4a859c24836cd3cefb : SUCCESS in 52s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c48de7c5d3c54fe18f32f5a86c11aa92 : SUCCESS in 54s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"2a327582f71bf22e32ae6c9f5419d89cb68acf7a","revisions":{"2a327582f71bf22e32ae6c9f5419d89cb68acf7a":{"kind":"REWORK","_number":1,"created":"2023-12-14 12:49:21.000000000","uploader":{"_account_id":9816,"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","username":"kajinamit"},"ref":"refs/changes/74/903674/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/74/903674/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/74/903674/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/74/903674/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/74/903674/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/74/903674/1"}}},"commit":{"parents":[{"commit":"261ee7dcffb1fe634152c4f2ccf79264c1a653b5","subject":"Merge \"Add spec for SR-IOV in Octavia Amphora\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/261ee7dcffb1fe634152c4f2ccf79264c1a653b5"}]}],"author":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-12-14 12:47:34.000000000","tz":540},"committer":{"name":"Takashi Kajinami","email":"kajinamit@oss.nttdata.com","date":"2023-12-14 12:47:34.000000000","tz":540},"subject":"Unpin flake8-import-order","message":"Unpin flake8-import-order\n\n... to resolve the following ValueError.\n\n```\nValueError: \u0027string\u0027 is not callable\n```\n\nThe I202 test is disabled because multiple files need updates. It may\nbe enabled later once all the warnings are addressed.\n\nFixed E721 test errors which appears 4 times.\n\nE721 do not compare types, for exact checks use `is` / `is not`, for\ninstance checks use `isinstance()`\n\nChange-Id: I68880b83c8a555ed30c3ce18b73ae27b8db34dc8\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2a327582f71bf22e32ae6c9f5419d89cb68acf7a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/2a327582f71bf22e32ae6c9f5419d89cb68acf7a"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
