)]}'
{"id":"openstack%2Foctavia~905073","triplet_id":"openstack%2Foctavia~master~Ibe360e1e20bc9854b3df2951577f5163f20184fa","project":"openstack/octavia","branch":"master","topic":"ci-py311","hashtags":[],"change_id":"Ibe360e1e20bc9854b3df2951577f5163f20184fa","subject":"Run octavia-tox-functional-py311 job in Zuul","status":"NEW","created":"2024-01-09 08:33:25.000000000","updated":"2024-01-09 11:18:03.000000000","submit_type":"MERGE_IF_NECESSARY","mergeable":false,"submittable":false,"total_comment_count":0,"unresolved_comment_count":0,"work_in_progress":true,"has_review_started":false,"meta_rev_id":"b3abd323245cfb93e12b9f94cb475408c357e92c","_number":905073,"virtual_id_number":905073,"owner":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2024-01-09 11:18:03.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]}]},"pending_reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}],"CC":[{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]}]},"reviewer_updates":[{"updated":"2024-01-09 08:34:15.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-01-09 08:41:01.000000000","updated_by":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"reviewer":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"state":"CC"}],"messages":[{"id":"d10d9dcf66763682f9f85d62c5c664d4fc959a30","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-01-09 08:33:25.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9663ef5133e00e2601a9a6d638140db4bf10acb4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-09 08:34:15.000000000","message":"Patch Set 1: Verified-1\n\n(1 comment)\n\nZuul encountered a syntax error while parsing its\nconfiguration in the repo openstack/octavia on branch master.  The\nproblem was:\n\n  Job openstack-tox-functional-py311 not defined\n\nThe problem appears in the the \"opendev.org/openstack/octavia\" project stanza:\n\n  project:\n      templates:\n        - check-requirements\n        - periodic-stable-jobs-neutron\n        - openstack-cover-jobs\n        - openstack-python3-jobs\n        - publish-openstack-docs-pti\n        - release-notes-jobs-python3\n        - octavia-tox-tips\n  ...\n\n  in \"openstack/octavia/zuul.d/projects.yaml@master\", line 3","accounts_in_message":[],"_revision_number":1},{"id":"b515c22ef9b1cfac4d4c5e4428edbfa88bf0a14c","tag":"autogenerated:gerrit:newWipPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-01-09 08:40:37.000000000","message":"Uploaded patch set 2: Commit message was updated.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"2331faacd04f35970b8633f4c6c9eeab233a3137","author":{"_account_id":15670,"name":"StorPool OpenStack CI","email":"openstack-ci@storpool.com","username":"storpoolci","tags":["SERVICE_USER"]},"date":"2024-01-09 08:41:01.000000000","message":"Patch Set 2:\n\nThis change depends on a change with an invalid configuration.","accounts_in_message":[],"_revision_number":2},{"id":"b3abd323245cfb93e12b9f94cb475408c357e92c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-09 11:18:03.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d2bf71f96a10474982ec80ed46f5caad\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/aaa02e8406a14e3b883a252a6ed474d1 : SUCCESS in 10m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/86fc7a47e67e465a976301217a4d0cb4 : SUCCESS in 4m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/b8194f21d7e54ca79a4d07f3750187e2 : SUCCESS in 4m 42s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/dde824aec6c64bc38dfe21897df7705d : SUCCESS in 5m 43s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fc7e8bf6072a434b9ac79bbdc494eb2f : SUCCESS in 12m 03s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/378d4235b26c47178bd954116f27d333 : SUCCESS in 5m 04s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/fb23c897e0a24160ac8301efce094dcb : SUCCESS in 8m 31s\n- octavia-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/a6f7281276b845329519e47dcbcaaabb : SUCCESS in 9m 46s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/3d91d4db47414668b610017795db7ca1 : SUCCESS in 6m 51s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/0de957385f05455aa8cd4571cc43cbf7 : SUCCESS in 8m 25s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/439af69e85a24fe6ba8fd116dcc20f0c : SUCCESS in 7m 55s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/da4691572f4842f29890046709060686 : SUCCESS in 7m 45s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/652bc0f20b6d48cfb2a852c7dcaadc13 : SUCCESS in 5m 01s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/b7b35f65f55a433ea94efb465a3daf29 : SUCCESS in 1h 25m 23s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/a6252945a5e14a31bde278b613969ef3 : SUCCESS in 2h 18m 49s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/948cb69f7cc441029724de34b453388f : SUCCESS in 1h 20m 25s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/7df97101590c4e2f8f33686a4884dc25 : SUCCESS in 2h 14m 36s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/633da007bd6843c7905f90072c6744cd : SUCCESS in 1h 17m 16s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/f50bebe025fc46e3856c5206c897c10a : SUCCESS in 37m 13s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/d8633bb0baae456e8e8cbd7fc5bab4ae : SUCCESS in 1h 01m 34s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/2149ea7848b44e0bbe7b66f65844029e : SUCCESS in 54m 41s (non-voting)\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/9ef7f6a1bdeb4af9b4d954e63cfb64a5 : SUCCESS in 1h 19m 03s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d0ff85db78d54145bdc984e24fcf4a7c : SUCCESS in 32m 59s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/e60bc82e24414640b6eb5c95ef187286 : SUCCESS in 1h 55m 56s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/7ffb9b6c28904ab48b5d0a51fc92b74f : SUCCESS in 2h 21m 20s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/9a1af261a0bf4265b69d9db5a524c230 : SUCCESS in 2h 05m 48s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/a40e5204440a4d6384d467155ed5881b : TIMED_OUT in 2h 34m 58s (non-voting)","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"c242d1ce778fc0e8e314f21b4ed5aed54eb962cb","revisions":{"1eb9cb329e2163d258c95a24332d8366faf1467a":{"kind":"REWORK","_number":1,"created":"2024-01-09 08:33:25.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/73/905073/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/73/905073/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/73/905073/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/73/905073/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/73/905073/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/73/905073/1"}}},"commit":{"parents":[{"commit":"490a65fceb6c247f77d47cbcfcb6bc0bbdd6bdd7","subject":"Merge \"Fix issue with certificates with no subject or CN\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/490a65fceb6c247f77d47cbcfcb6bc0bbdd6bdd7"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2024-01-09 08:24:37.000000000","tz":60},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2024-01-09 08:33:14.000000000","tz":60},"subject":"Run octavia-tox-functional-py311 job in Zuul","message":"Run octavia-tox-functional-py311 job in Zuul\n\nPython 3.11 is supported runtime in 2024.1\n\nChange-Id: Ibe360e1e20bc9854b3df2951577f5163f20184fa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1eb9cb329e2163d258c95a24332d8366faf1467a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1eb9cb329e2163d258c95a24332d8366faf1467a"}]},"branch":"refs/heads/master"},"c242d1ce778fc0e8e314f21b4ed5aed54eb962cb":{"kind":"NO_CODE_CHANGE","_number":2,"created":"2024-01-09 08:40:37.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/73/905073/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/73/905073/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/73/905073/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/73/905073/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/73/905073/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/73/905073/2"}}},"commit":{"parents":[{"commit":"490a65fceb6c247f77d47cbcfcb6bc0bbdd6bdd7","subject":"Merge \"Fix issue with certificates with no subject or CN\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/490a65fceb6c247f77d47cbcfcb6bc0bbdd6bdd7"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2024-01-09 08:24:37.000000000","tz":60},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2024-01-09 08:40:16.000000000","tz":60},"subject":"Run octavia-tox-functional-py311 job in Zuul","message":"Run octavia-tox-functional-py311 job in Zuul\n\nPython 3.11 is supported runtime in 2024.1\n\nDepends-On: I647dbac4e80518561b91a1be5b296d0d98395cda\nChange-Id: Ibe360e1e20bc9854b3df2951577f5163f20184fa\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c242d1ce778fc0e8e314f21b4ed5aed54eb962cb"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c242d1ce778fc0e8e314f21b4ed5aed54eb962cb"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"OK","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY"},{"label":"Workflow","status":"MAY"},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{"label:Verified\u003dMAX":"","label:Verified\u003dMIN":""}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{"label:Code-Review\u003dMAX":"","label:Code-Review\u003dMIN":""}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{"label:Workflow\u003dMAX":"","label:Workflow\u003dMIN":""}}}]}
