)]}'
{"id":"openstack%2Foctavia~905097","triplet_id":"openstack%2Foctavia~stable%2Fyoga~I06911f42b9bf29cf9a5f2e76d8333d8a2f1bc60b","project":"openstack/octavia","branch":"stable/yoga","attention_set":{},"removed_from_attention_set":{"6577":{"account":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"last_update":"2024-01-23 16:03:24.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I06911f42b9bf29cf9a5f2e76d8333d8a2f1bc60b","subject":"Fix issue with certificates with no subject or CN","status":"MERGED","created":"2024-01-11 06:59:17.000000000","updated":"2024-01-23 16:04:43.000000000","submitted":"2024-01-23 16:03:24.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":0,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"905097","cherry_pick_of_change":901689,"cherry_pick_of_patch_set":3,"meta_rev_id":"0817d82df133f328fdb63ad79df7fa1c9d09c475","_number":905097,"virtual_id_number":905097,"owner":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-01-23 16:03:24.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":2,"date":"2024-01-22 12:40:55.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-01-23 10:52:23.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"all":[{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-01-23 10:52:23.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-01-11 06:59:17.000000000","updated_by":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"CC"},{"updated":"2024-01-11 10:01:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-01-22 12:40:55.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2024-01-23 10:52:23.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"}],"messages":[{"id":"ec03704b1ce34abc4f0df34a44a271bc1f90c9ff","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"date":"2024-01-11 06:59:17.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"549210ec8c0708fbfbebda17417b9ab0d8eccf51","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-11 10:01:29.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/9a12141fcb51449a82fa5e5f0cd2bf01\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/850cfc5b044b423b896d2323a032ba1d : SUCCESS in 9m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f0ec0ce3cbbc49d087270333d82a3524 : FAILURE in 3m 57s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/f4e8a44d74044adf9e54ff165698e093 : SUCCESS in 5m 34s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/8252d87f4e83465387df17518173c58e : SUCCESS in 6m 08s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/38b47854bb3a443bbc8bf68b6ddf2ff2 : SUCCESS in 11m 30s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/5f56fcf2f9874751abc5e2366a16f6aa : SUCCESS in 5m 28s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/b6bba3bdf73c437e9ac29e8546c6030c : SUCCESS in 6m 03s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/138b08a10fba48ffae8c48708b0e9892 : SUCCESS in 5m 24s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/bd313edbb5914443ae787b1d30470224 : SUCCESS in 6m 25s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/8d1a7ab5f4704454ae3dbf47b690fb34 : SUCCESS in 7m 50s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/a4bb32f8c9e6404fb2e6449fbe225dde : SUCCESS in 1h 36m 27s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/cbc5841495714bfba79bda6005b7da87 : SUCCESS in 2h 07m 35s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/71427f03714a441da6c558f4c7c3b903 : SUCCESS in 35m 42s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/24c8b55e1bc44595ad40bf8d71d16e8c : SUCCESS in 49m 03s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/f83fd3ce41d5456aa0f5e7cc0713dc88 : FAILURE in 1h 03m 05s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f69ff8bab97a439aacccb068182d591c : SUCCESS in 42m 54s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/56ad90a12aba42d1a91d2c89575dc788 : FAILURE in 2h 18m 09s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/93522a056675432183a853e1808e2435 : FAILURE in 10m 13s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/ac37bfdc4127402eb1a50a1928b0abfb : FAILURE in 7m 33s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/41326c42094b424e86ec90649c6bf1b3 : TIMED_OUT in 2h 33m 34s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/3fb07d9fb6344e689b46e4d596be0aa2 : TIMED_OUT in 2h 32m 39s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/44763a2df5884257949ce299653c1463 : FAILURE in 49m 59s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"635259ae44d964761606f00ecc08b5cb19d15930","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"date":"2024-01-19 06:30:13.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"34c3837b0e63f14b679e93b6a5d116cec8148a25","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-19 09:22:42.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/1c8a626755904cbfaee4db1438e5fc43\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/46a67ca641d54bf98f529547211430f1 : SUCCESS in 10m 27s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/29b68200b2a64d249e7734d56069a865 : SUCCESS in 7m 17s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/903177720e3a4b40acb9b9ae6aea6a68 : SUCCESS in 5m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1c8a47961dca4a07baeb03c5b55e61d2 : SUCCESS in 5m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d18e54b164db4654a5b07cf2eba3defb : SUCCESS in 12m 16s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/958e4376ac614adfa7159ed30b10a564 : SUCCESS in 5m 34s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/6f1883e3bfd04e7791c58b15babce31b : SUCCESS in 5m 03s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/90a43ff40b4141c5ae0c01156c095368 : SUCCESS in 7m 56s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/4f6069a4ddb6462187a4515ea57d1e8b : SUCCESS in 5m 00s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/5972cc89a8c64ad9aee0e32ef7e3623a : SUCCESS in 11m 42s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/155c8d0d9c164a8bb9e529b7f8c9cdc2 : SUCCESS in 59m 02s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3975119f8f014e2aa3518a4c87680323 : SUCCESS in 1h 33m 03s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/5633685a0294417faa36e69cadd4929f : SUCCESS in 33m 18s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/9e00c1fb896b424eb493dbb182d5ff58 : SUCCESS in 35m 42s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/f40e571f72394b0bbed4c1ff870aabfd : FAILURE in 1h 06m 01s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/9d3b33c28b64444197556b37bd4ddc97 : SUCCESS in 40m 48s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/97dcf597c4f74c5ca3ff48515db9023a : FAILURE in 2h 19m 42s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/709f8a21de704001bab764a29b00d25e : FAILURE in 10m 21s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/fb9b2d52fe894cfa9b50a9535256120f : FAILURE in 4m 58s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/638dd3f2d1634d568ef4e82c7b761043 : TIMED_OUT in 2h 33m 52s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/fddee96074274b28b81ba66e72e12a4d : SUCCESS in 1h 43m 53s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/7c6f228aeb2b4547abe3bdd05ce0750f : FAILURE in 37m 54s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"3e762db6ab8821a03df6fc9444b675c4cd59d531","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2024-01-22 12:40:55.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"34f2f10c9f8eb3918456f8c5558dd19232633416","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-01-23 10:52:23.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"a6c744f91e8d399e60cad4ea5df962436e5dd2af","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-23 10:53:04.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"a0bd34ab01268b66740c5ab4c9651b263e4fe75c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-23 16:03:24.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/eedc76a958d24a71887de4af4db13f0b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1bf3b65d7544472c8513a73c594174c8 : SUCCESS in 7m 36s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/d8ba3057f99c4353b6b577c68d138249 : SUCCESS in 6m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/81d4a4b0b15e44ac9ed641dfd86f44c2 : SUCCESS in 6m 26s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f2bad68f86ac4445b0b618f94c3a6a3a : SUCCESS in 12m 40s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/615eb643d7b74a589e6807cd1419bc37 : SUCCESS in 5m 30s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/4c43bf2fd8ed4d399fa78b4f6dce1343 : SUCCESS in 5m 18s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/401db51aa9484ce086e2c18a08068739 : SUCCESS in 8m 10s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/efbb9e4b1aab4d2c8d06ea03609171ce : SUCCESS in 1h 40m 57s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/4172499507724f5d8f87e9f2dce35386 : SUCCESS in 1h 47m 33s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/141ac57861bc4463b6e12d8e5289c22f : SUCCESS in 37m 24s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/64f404502da44da694c22122806d67da : SUCCESS in 39m 14s","accounts_in_message":[],"_revision_number":2},{"id":"faa4271cf203e0efbb5fbadbd81d89d2cfe3fe97","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-23 16:03:24.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"0817d82df133f328fdb63ad79df7fa1c9d09c475","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-23 16:04:43.000000000","message":"Patch Set 2:\n\nBuild failed (promote pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7112568cbd5d4c3fa6613f5ef2475fa7\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ab8eb850748e4e15b09c35de8da39c02 : SUCCESS in 55s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/02d49ec2270b44aea42be82f3c67bf60 : FAILURE in 40s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"0398dd6eddd9124257371a9028f98daf1406d112","revisions":{"5b66377243a6c4acb736841f0219406af8bae7d2":{"kind":"REWORK","_number":1,"created":"2024-01-11 06:59:17.000000000","uploader":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"ref":"refs/changes/97/905097/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/97/905097/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/97/905097/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/97/905097/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/97/905097/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/97/905097/1"}}},"commit":{"parents":[{"commit":"5b388dbf3809d2d35234767bc781ec7bd7e05673","subject":"Add Noop Certificate Manager","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5b388dbf3809d2d35234767bc781ec7bd7e05673"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-11-22 21:45:44.000000000","tz":0},"committer":{"name":"Sergey Kraynev","email":"sergejyit@gmail.com","date":"2024-01-11 06:59:17.000000000","tz":0},"subject":"Fix issue with certificates with no subject or CN","message":"Fix issue with certificates with no subject or CN\n\nThis patch fixes an issue where if the user attempts to use a\ncertificate that does not have a subject or CN, we would fail to create\na listener using the certificate.\nPer the x.509 specification, a blank subject is allowed as long as the\nsubjectAltName extension is present in the certificate.\nOctavia will now check for the a valid subAltName if the subject CN can\nnot be retrieved. If both are missing an appropriate error is raised for\nthe user.\n\nCloses-Bug: #2043582\nChange-Id: I06911f42b9bf29cf9a5f2e76d8333d8a2f1bc60b\n(cherry picked from commit 73cdee503ff1cb3419d8db0295b20b2b6ddf30c0)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5b66377243a6c4acb736841f0219406af8bae7d2"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5b66377243a6c4acb736841f0219406af8bae7d2"}]},"branch":"refs/heads/stable/yoga"},"0398dd6eddd9124257371a9028f98daf1406d112":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-01-19 06:30:13.000000000","uploader":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"ref":"refs/changes/97/905097/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/97/905097/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/97/905097/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/97/905097/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/97/905097/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/97/905097/2"}}},"commit":{"parents":[{"commit":"fd5185f8ed4691437ac7053dd9402134ae995d72","subject":"Stable-only: Cap hacking to \u003c 6.1.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fd5185f8ed4691437ac7053dd9402134ae995d72"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-11-22 21:45:44.000000000","tz":0},"committer":{"name":"Sergey Kraynev","email":"sergejyit@gmail.com","date":"2024-01-19 06:30:13.000000000","tz":0},"subject":"Fix issue with certificates with no subject or CN","message":"Fix issue with certificates with no subject or CN\n\nThis patch fixes an issue where if the user attempts to use a\ncertificate that does not have a subject or CN, we would fail to create\na listener using the certificate.\nPer the x.509 specification, a blank subject is allowed as long as the\nsubjectAltName extension is present in the certificate.\nOctavia will now check for the a valid subAltName if the subject CN can\nnot be retrieved. If both are missing an appropriate error is raised for\nthe user.\n\nCloses-Bug: #2043582\nChange-Id: I06911f42b9bf29cf9a5f2e76d8333d8a2f1bc60b\n(cherry picked from commit 73cdee503ff1cb3419d8db0295b20b2b6ddf30c0)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0398dd6eddd9124257371a9028f98daf1406d112"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0398dd6eddd9124257371a9028f98daf1406d112"}]},"branch":"refs/heads/stable/yoga","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
