)]}'
{"id":"openstack%2Foctavia~905099","triplet_id":"openstack%2Foctavia~stable%2F2023.1~I06911f42b9bf29cf9a5f2e76d8333d8a2f1bc60b","project":"openstack/octavia","branch":"stable/2023.1","attention_set":{},"removed_from_attention_set":{"6577":{"account":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"last_update":"2024-01-24 04:37:28.000000000","reason":"\u003cGERRIT_ACCOUNT_6577\u003e replied on the change","reason_account":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"}}},"hashtags":[],"change_id":"I06911f42b9bf29cf9a5f2e76d8333d8a2f1bc60b","subject":"Fix issue with certificates with no subject or CN","status":"MERGED","created":"2024-01-11 06:59:50.000000000","updated":"2024-01-24 09:09:24.000000000","submitted":"2024-01-24 09:08:20.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"905099","cherry_pick_of_change":901689,"cherry_pick_of_patch_set":3,"meta_rev_id":"8b7e95d3518189c2d3e707043af9a9cfc5ef7853","_number":905099,"virtual_id_number":905099,"owner":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-01-24 09:08:20.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-01-22 12:40:11.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":2,"date":"2024-01-23 10:51:53.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":1,"date":"2024-01-23 10:51:53.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"CC":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}],"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-01-11 06:59:50.000000000","updated_by":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"CC"},{"updated":"2024-01-11 09:53:17.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-01-22 12:40:11.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2024-01-23 10:51:53.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"}],"messages":[{"id":"e2e7970eaf849454b7cf871064b2478d6667d150","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"date":"2024-01-11 06:59:50.000000000","message":"Patch Set 1: Cherry Picked from branch master.","accounts_in_message":[],"_revision_number":1},{"id":"184fc89b1a0886581c85312aab2b57bc3535f7d0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-11 09:53:17.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/3d635eec8f2241a29b6dccb7f28439fc\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7a5fbbdd564d4d0a8a0dbd9da8eb60a3 : SUCCESS in 10m 19s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/38765c7b9fc54e3180867c80a4a35cb4 : FAILURE in 3m 51s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0a461ab606814f638d54ccedd1677345 : SUCCESS in 4m 44s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/eb65de80c6634d19af89fcb9fe5ddd68 : SUCCESS in 11m 13s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0796b2b31610468091bd36d88e52000c : SUCCESS in 5m 07s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/b173e472b2754177a28452884aff4076 : SUCCESS in 4m 45s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/89c1cd0a756e4a9fbb920beed2ec9929 : SUCCESS in 8m 50s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/3c237241bc644b029d407bd3c39b9a6e : SUCCESS in 4m 44s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/208da8b801304fd9aedc71e5d3485d6c : SUCCESS in 8m 13s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/f0efefe204fb4a328d0f491910270e22 : SUCCESS in 7m 13s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/58e74d1adab841d5878e1cef6abdf8cb : SUCCESS in 7m 06s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/2aeaf25a77474051aa37f9aee7888e53 : SUCCESS in 1h 24m 51s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/f4d2fdbd70844d548893a8ebade0d101 : SUCCESS in 1h 18m 39s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/99c3662175bc4f5a9d671a1f702e96ac : SUCCESS in 1h 37m 25s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/d57491cbceb84dd182d591d91739d4aa : SUCCESS in 36m 31s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/69654b79a6d8423e87353cda7cffe688 : SUCCESS in 37m 55s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/7e151d9c3b254284aaf0ab320bb1e534 : SUCCESS in 1h 03m 23s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/7ba3888d6994415f9adb2e2478c5b7b8 : SUCCESS in 1h 03m 52s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f4ab5800d211489f9586890b503482d1 : SUCCESS in 33m 21s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/77f1edc42ee54c8ea8e6e28bd5de3085 : TIMED_OUT in 2h 33m 39s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/49c0a073ff5d476793b7c47cf1384dac : SUCCESS in 2h 10m 48s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/b6b48f7d558b4e9fb50d0f8c3505d81b : SUCCESS in 1h 52m 37s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/62a24fd979024c0084ba469554ee9949 : SUCCESS in 2h 08m 00s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/014f8c33a43e4b7883ad1df38ed8f0e3 : SUCCESS in 1h 40m 24s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/8eb926514e3a40abb99b892465cb1661 : SUCCESS in 2h 15m 58s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0872c2e4f33308d28031c0a01fcb960c43996cde","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"date":"2024-01-19 06:29:55.000000000","message":"Patch Set 2: Patch Set 1 was rebased\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"ef48f8389a08c542ceece567a62075252454eddd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-19 09:13:31.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/5ed10bc084c0423abd9ffc57a3a1e1e9\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/72b4e765ec21432983a0f71802b5ed35 : SUCCESS in 9m 59s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eb15a4a66fb24f09ad0a1690c68e8cee : SUCCESS in 6m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/3418b67849c6474b8b18d5c0596dc997 : SUCCESS in 4m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/bedbd3563a2d4877a65c704f2101153e : SUCCESS in 11m 52s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/b05dd4299d874ed39aaec3406140b6d0 : SUCCESS in 5m 14s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/2e0d2317bc324d2cb0b7845667e2e1e6 : SUCCESS in 4m 52s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/c4b874a8f3604ab0936ad8b856aca936 : SUCCESS in 8m 38s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/b61aeacf20a44b8ca8e57b3730865d50 : SUCCESS in 4m 46s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/fdb4e646112345db86481e03bf495379 : SUCCESS in 7m 32s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/8f7913ce65e3453c95d4d631f32bcdb3 : SUCCESS in 7m 46s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/f93cda2b4ec7430fbf4febb4637d4d37 : SUCCESS in 8m 27s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/91366fbae9304468b3aa0824b3b83c58 : SUCCESS in 1h 21m 12s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/e7726ec47808438f84174ff8072e1d28 : SUCCESS in 1h 21m 41s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/a7a07e7c7f154f739eeb08547438d058 : SUCCESS in 1h 00m 21s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/8859011e6cea43dcae0770dbe2c5a6e0 : SUCCESS in 42m 09s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/87d729b91b8d49d4ad6dfc45db22623e : SUCCESS in 47m 13s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/fda5518cf23d4f29bba0f4460e9f480e : SUCCESS in 44m 10s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/e5813bc048844721a44ac2dcab5b3506 : SUCCESS in 1h 14m 33s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7bc0430539ef438b90cd98945358b67f : SUCCESS in 35m 50s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/55f83d8b11fa4e7d937e84021436e07b : SUCCESS in 1h 49m 18s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/222a9bd61f274ca784d25fcfa660a95c : SUCCESS in 2h 11m 52s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/99a6edd1e2724880b72bb2d54d322487 : FAILURE in 20m 47s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/aa39a331989e4b44850a137b899f2b4b : SUCCESS in 2h 03m 07s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/874fd4f8344d45cd92b643ee40b504df : SUCCESS in 2h 05m 07s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/8a8546993caf441ebc36381dd4e65190 : TIMED_OUT in 2h 35m 43s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"92bbd4c952710f05b2bab9021b96d52b72ab9ea0","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2024-01-22 12:40:11.000000000","message":"Patch Set 2: Code-Review+2","accounts_in_message":[],"_revision_number":2},{"id":"0f5cc4ce0c42a5d5ee350b7273ca69e29f4aab30","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-01-23 10:51:53.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1","accounts_in_message":[],"_revision_number":2},{"id":"ce52fa3331bc9834ad5f5d281310fe880c1fe0fd","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-23 10:52:30.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"1d7dde9a4a3265fb8c34d44329946651659e3ece","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-23 13:49:06.000000000","message":"Patch Set 2: Verified-2\n\nBuild failed (gate pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/fff763c2ca704c9593180ace668019f9\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8a107a34043148658f61960780830fb0 : SUCCESS in 5m 53s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/4e8d5af2e69f4800b7c9990b9718a689 : SUCCESS in 4m 49s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/70530b78cc644ff69104370de7f61dc1 : SUCCESS in 13m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/55430ce961214b6783eea9883de6976f : SUCCESS in 4m 55s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/6b30b6f3c7f14fb79552d3ccc1883041 : SUCCESS in 4m 44s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/85423a58f21443d590eb4311df0f36f2 : SUCCESS in 8m 22s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/3547f1c6bb94403983bedf156412cef7 : SUCCESS in 1h 26m 43s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/2c395b4debef4fc6a27b5d69c9d22ad8 : FAILURE in 1h 34m 38s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/f0530bf95996491f9ac279e7756a60bb : SUCCESS in 54m 03s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/a7b4a3876a94477daaa6a1e397c962c2 : SUCCESS in 49m 38s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/981e5c1d7132413da1e0316c6bf2387f : SUCCESS in 58m 39s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/e6759dea7fdb4094a0394ff8db23e23d : SUCCESS in 40m 04s","accounts_in_message":[],"_revision_number":2},{"id":"eb2d40bd927a70a84b5c5ee8e57626318199d539","author":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"date":"2024-01-24 04:37:28.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"94e0956bad95d7a8a79b29bb53b542079353892d","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-24 07:21:41.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cd94e7c3233c4c1c864e5d4c54e3b609\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/756c05e3774e4407bffd882a114233c1 : SUCCESS in 13m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/bc94b234a69f46f787ed387679e976f0 : SUCCESS in 5m 14s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/f1301062573649e5a31ba3a9836362bf : SUCCESS in 4m 41s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8ca837cc59204565adc4a936ee105a6f : SUCCESS in 12m 41s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a0eecda761b646439454e6ac02b8bb61 : SUCCESS in 5m 15s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/e1a6d9ee8fd7442aad063fbe14bdf84b : SUCCESS in 5m 03s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/2e3850bb26fa441192b0dee9b8055d34 : SUCCESS in 8m 54s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/646beba991d5453aa8068c46a1dd3396 : SUCCESS in 4m 48s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/f7e4ccda35cd4a52856c7d464b7f1559 : SUCCESS in 10m 06s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/7ba9d1bd721a43fb8c24f5a6b3dd1983 : SUCCESS in 7m 31s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/2071359f14734f15a4b5bce2576982f8 : SUCCESS in 8m 14s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/ffeb9bf01b0d4fbab7f990ab82af39df : SUCCESS in 1h 28m 32s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/e6ca0d6fcb1b4ee585bb26b8eaf2fac5 : SUCCESS in 1h 14m 31s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/e909696958d7480caa39a6a49779913e : SUCCESS in 1h 00m 04s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/50121eceb97f42fb953bbd24803233d8 : SUCCESS in 34m 35s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/db666471fa8e4c64b4631656834e51af : SUCCESS in 38m 37s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/33f2fc990ec9413c9888c2c098207932 : SUCCESS in 40m 34s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/5cce1f4deec34df4a2d11b748c3c4413 : SUCCESS in 1h 07m 12s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/04a432dd064c4dccbd7605e6fea1d95c : SUCCESS in 45m 41s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/4703c08a3a6944c6ae0fe3d789e8acd6 : SUCCESS in 2h 23m 24s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/8d989b35ca9e4a21a93196d6ec3f0230 : SUCCESS in 2h 41m 28s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/fdbbe322935f40298b49aa772bedc052 : FAILURE in 32m 40s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/ce4f0fb036aa48f59512b4070075207e : TIMED_OUT in 2h 32m 56s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/0c84b892c1ad4beea598120a042a3e5d : SUCCESS in 2h 31m 06s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/8c479b7e666048fcab675dbf0ef61a73 : SUCCESS in 2h 06m 16s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d9c8ef3d3aba442fd4a14f6a069766e50903a281","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-24 07:22:18.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"5b12cbf48c5aa36675c533945e589cff5109a37b","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-24 09:08:20.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9044ac61851c44bc87baa8b518dfdebc\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f9dee71ba6a54b709a2190143f1a7f10 : SUCCESS in 4m 32s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/e38ab647155946a285e6d7a19fe9159a : SUCCESS in 5m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/594238c643714d11a65f4a6e533583d6 : SUCCESS in 10m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0c88b87bd1454513bdc097ec42f2b59c : SUCCESS in 5m 00s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/688901ccdafc40c680c20896f1148974 : SUCCESS in 5m 51s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/d7ecbe59953246a6a5fef33221f41eea : SUCCESS in 8m 40s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/50009e97689947b686fc0b153f08a7f1 : SUCCESS in 1h 17m 49s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/5f0f808f85bd4a26a57ab7f52aefc9c7 : SUCCESS in 1h 44m 58s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/e58498453e3b41f691b8bc51024f0fa8 : SUCCESS in 1h 25m 54s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/f41bd80d9c434775bd8d8ae2278106fb : SUCCESS in 53m 43s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/7c15c7f1bbf04b2fb0d9596fae68f384 : SUCCESS in 1h 05m 35s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/18b950082a2b41cf8a4d1f026cd8ef0e : SUCCESS in 40m 06s","accounts_in_message":[],"_revision_number":2},{"id":"bf4ba6c4d3ed762dc6e2239e3186fb6749bfeb0f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-24 09:08:20.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"8b7e95d3518189c2d3e707043af9a9cfc5ef7853","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-24 09:09:24.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/40c9f12b8ff0426695d1d4d89c5728c5\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/46cdb8d5f7784ef1a124052ba5c790f1 : SUCCESS in 46s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/0c7aa766591048cea12c084431224253 : SUCCESS in 39s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"20a7a26c165edef401eb3ebd775000b9d9ab71b5","revisions":{"945561ec1bdbb1c21d24c627238c90b0fa66921b":{"kind":"REWORK","_number":1,"created":"2024-01-11 06:59:50.000000000","uploader":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"ref":"refs/changes/99/905099/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/99/905099/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/99/905099/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/99/905099/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/99/905099/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/99/905099/1"}}},"commit":{"parents":[{"commit":"ec36152a03bd27a183d189d70effe3f095825e91","subject":"Add Noop Certificate Manager","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ec36152a03bd27a183d189d70effe3f095825e91"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-11-22 21:45:44.000000000","tz":0},"committer":{"name":"Sergey Kraynev","email":"sergejyit@gmail.com","date":"2024-01-11 06:59:50.000000000","tz":0},"subject":"Fix issue with certificates with no subject or CN","message":"Fix issue with certificates with no subject or CN\n\nThis patch fixes an issue where if the user attempts to use a\ncertificate that does not have a subject or CN, we would fail to create\na listener using the certificate.\nPer the x.509 specification, a blank subject is allowed as long as the\nsubjectAltName extension is present in the certificate.\nOctavia will now check for the a valid subAltName if the subject CN can\nnot be retrieved. If both are missing an appropriate error is raised for\nthe user.\n\nCloses-Bug: #2043582\nChange-Id: I06911f42b9bf29cf9a5f2e76d8333d8a2f1bc60b\n(cherry picked from commit 73cdee503ff1cb3419d8db0295b20b2b6ddf30c0)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/945561ec1bdbb1c21d24c627238c90b0fa66921b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/945561ec1bdbb1c21d24c627238c90b0fa66921b"}]},"branch":"refs/heads/stable/2023.1"},"20a7a26c165edef401eb3ebd775000b9d9ab71b5":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-01-19 06:29:55.000000000","uploader":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"ref":"refs/changes/99/905099/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/99/905099/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/99/905099/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/99/905099/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/99/905099/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/99/905099/2"}}},"commit":{"parents":[{"commit":"7d9f2914af0dfb59188caf99d00b4013f69c1c2f","subject":"Stable-only: Cap hacking to \u003c 6.1.0","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7d9f2914af0dfb59188caf99d00b4013f69c1c2f"}]}],"author":{"name":"Michael Johnson","email":"johnsomor@gmail.com","date":"2023-11-22 21:45:44.000000000","tz":0},"committer":{"name":"Sergey Kraynev","email":"sergejyit@gmail.com","date":"2024-01-19 06:29:55.000000000","tz":0},"subject":"Fix issue with certificates with no subject or CN","message":"Fix issue with certificates with no subject or CN\n\nThis patch fixes an issue where if the user attempts to use a\ncertificate that does not have a subject or CN, we would fail to create\na listener using the certificate.\nPer the x.509 specification, a blank subject is allowed as long as the\nsubjectAltName extension is present in the certificate.\nOctavia will now check for the a valid subAltName if the subject CN can\nnot be retrieved. If both are missing an appropriate error is raised for\nthe user.\n\nCloses-Bug: #2043582\nChange-Id: I06911f42b9bf29cf9a5f2e76d8333d8a2f1bc60b\n(cherry picked from commit 73cdee503ff1cb3419d8db0295b20b2b6ddf30c0)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/20a7a26c165edef401eb3ebd775000b9d9ab71b5"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/20a7a26c165edef401eb3ebd775000b9d9ab71b5"}]},"branch":"refs/heads/stable/2023.1","description":"Rebase"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
