)]}'
{"id":"openstack%2Foctavia~905193","triplet_id":"openstack%2Foctavia~master~I5b0f084785f477c855218c54a21caecdecefb62f","project":"openstack/octavia","branch":"master","topic":"rate-limiting","attention_set":{},"removed_from_attention_set":{"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2025-02-11 09:39:17.000000000","reason":"Change was abandoned"},"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2025-02-11 09:39:17.000000000","reason":"Change was abandoned"},"34429":{"account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"last_update":"2024-10-28 11:14:58.000000000","reason":"\u003cGERRIT_ACCOUNT_34429\u003e replied on the change","reason_account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}}},"hashtags":[],"change_id":"I5b0f084785f477c855218c54a21caecdecefb62f","subject":"Add typing information to octavia.amphorae package","status":"ABANDONED","created":"2024-01-10 08:40:57.000000000","updated":"2025-02-11 09:39:17.000000000","total_comment_count":15,"unresolved_comment_count":2,"has_review_started":true,"meta_rev_id":"6355439eb17770785cd805661a752a00e1a36d7e","_number":905193,"virtual_id_number":905193,"owner":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"tag":"autogenerated:zuul:check","value":1,"date":"2025-02-10 18:47:09.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"permitted_voting_range":{"min":-2,"max":2},"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-01-10 11:25:37.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-04-29 15:57:40.000000000","updated_by":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"reviewer":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"state":"REVIEWER"},{"updated":"2024-10-23 18:08:19.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"d8531046adf5c1be639792a760659a49d663aee3","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-01-10 08:40:57.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"5f8428c5a71d8367d6404117937981c5f0600c54","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-10 11:25:37.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/c43d0bb3da334c698c378fc307f8b65e\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/494d1b3e3af94707aaa8e6892b484f50 : FAILURE in 6m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3ac31f0b4c5b423697af6da31a99043c : SUCCESS in 5m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0db8c6fa963d4b178db4bf1ce24fb144 : FAILURE in 4m 39s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/6c2d0b933c3e449ea5a71b34295991cb : FAILURE in 5m 32s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/749094064bf749cda7bf0dbf19141731 : SUCCESS in 9m 49s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/3438ee0dcfb443f2a97465cbf506e26e : FAILURE in 5m 40s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/c34a7296a031470b8fad465de4a2a063 : SUCCESS in 9m 17s\n- octavia-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/12d87bbe9c0d4a90b191361336863b9a : SUCCESS in 10m 26s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/99b02626631a4e55b3d2250b218466c4 : SUCCESS in 4m 09s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/680c3a6b57854684b37df1bc6fe14046 : SUCCESS in 8m 29s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/bc7f79a242c34b3ca23ae3ac4ed7fd1a : SUCCESS in 8m 19s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/7a94872ebdae4a9c88e8cb7dc6dbe490 : SUCCESS in 10m 23s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/5d0ccf3c8bc347bdacce22915a3d3eaa : SUCCESS in 1h 28m 48s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/15bc04f72e354a3a9d9c06bea276b310 : SUCCESS in 1h 51m 56s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/e3ee0650a4864448bd4b011323fb2b3a : SUCCESS in 1h 16m 09s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/8e7d4e7c13574563a863925d6d629a73 : SUCCESS in 2h 01m 09s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/d4391b728b61432bb4d72e34d9bf609e : SUCCESS in 1h 29m 01s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/b99f7369279d4a8aafde1441cb3e1f61 : SUCCESS in 53m 17s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/61cd7e6e28c94c1bb259317ccf2053b9 : SUCCESS in 55m 19s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/c9de50e544c241279c14b5d436e046f5 : SUCCESS in 1h 01m 03s (non-voting)\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/7f658ad029a64f86a2c9e98b4c5896ae : SUCCESS in 1h 03m 02s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/698de54c524645e58d8570d2b3f8f255 : SUCCESS in 56m 57s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/07371c2a86c8453d9403f5411499ab87 : SUCCESS in 1h 54m 38s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/b24f95bcf166457ca8bcb5060258a548 : SUCCESS in 2h 41m 44s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/78e2c8f22b5e4347a601610f906c0d53 : TIMED_OUT in 2h 33m 20s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/a3c930595963432082b80d0e9b34bb1f : TIMED_OUT in 2h 34m 56s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e8ab5366d076e9d47d57102e92bcae8f76d4548e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-01-10 13:22:02.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Verified-1\n","accounts_in_message":[],"_revision_number":2},{"id":"b52a0e74465e59621a07a693cdc2a335a35b2a09","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-10 16:35:25.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/66b495e4bd0b41dcb37d164efc876614\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a8a535aae0334962a12b5c80e0672c1c : SUCCESS in 12m 26s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e611a89db7e14d41b2a041ff9821cd92 : SUCCESS in 5m 11s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d12b270dc201429d9f60f9a21efcee8a : SUCCESS in 4m 58s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/c1f9fd3c201943df97825385b7552a8d : SUCCESS in 6m 54s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/78540bfe3813411796dde37437ae9cc7 : SUCCESS in 13m 09s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/4ed0ecfb898049e3ade50584956fb7f3 : SUCCESS in 5m 43s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/c83b8c1033ad42a2bba0d0d5b82e26b3 : SUCCESS in 8m 36s\n- octavia-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/30f97975d7854fab992b9dbf900a9273 : SUCCESS in 10m 41s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/9719511a2c3a4a139c347d2bbaa8c53b : SUCCESS in 6m 24s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/44229b600e1f441b995d92615c08d2c3 : SUCCESS in 11m 07s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/2d151870a4f540849824ece662a53668 : SUCCESS in 9m 14s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/9c1a7ed9c4ba45c9b209645761369125 : SUCCESS in 9m 54s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/cb3c2dff9d2943c2ad4ecc21eb423c80 : SUCCESS in 1h 16m 15s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/2d5beb57f0a146388eeafb0091c10068 : SUCCESS in 2h 05m 12s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/3c13bdc75ab0410db2ab9d74909ca1f8 : SUCCESS in 1h 33m 22s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/c07245557bf446d7afb8293c5a2e3b35 : SUCCESS in 1h 58m 36s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/a872b46cd99848329294c29f144e6287 : SUCCESS in 1h 42m 04s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/32248e9b1dc7404889416e9652eb639a : SUCCESS in 48m 30s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/b0bbbe8813ec42db8c3ee074ff29ad9d : SUCCESS in 39m 39s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/c9c9741dc8fd4e1a8eeb56c6ffa3dafc : SUCCESS in 53m 05s (non-voting)\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/4d70d7bf93f6408ca64958e09bd5ee7f : SUCCESS in 1h 16m 47s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/bf3cbaad03454b77a3c0d434daed2f98 : FAILURE in 47m 03s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/9510fd06b4044f4497e7af7f556ce59f : SUCCESS in 2h 10m 14s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/cefdf27b79f34521a460a6e8b2ca453a : TIMED_OUT in 3h 04m 33s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/f08502f60b7a4e31a5e104f1cc7d982b : TIMED_OUT in 2h 33m 44s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/3e53b875318c41769a09f794ec522271 : TIMED_OUT in 2h 33m 25s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"4fda48974576ce5dde0e8ba3640169d1c5cbce30","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-03-25 11:43:51.000000000","message":"Uploaded patch set 3: Patch Set 2 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":3},{"id":"d536ad1e88cadfb0fa3ba95846914bfc19133d8c","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-03-25 14:33:27.000000000","message":"Patch Set 3: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b58db04680414f5bb19576e599701c27\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/7ed9bc10de7f448199e7a3be386d9bdf : SUCCESS in 10m 13s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a4e77885d07943b79fec04799e335157 : SUCCESS in 5m 17s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/58cef9dd034246c4b76d25351f2874ae : SUCCESS in 5m 03s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/bbcc4430909a423aaf31797d91b69ef3 : SUCCESS in 5m 56s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9168919d5dc64294b777afce78074101 : SUCCESS in 8m 51s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8f3c17cd97df481baa6355b986b75b20 : SUCCESS in 12m 20s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/261f1ff799c74175b638ee63fa526a23 : SUCCESS in 5m 14s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/05c3a75e56d04468911a90e2ed428243 : SUCCESS in 7m 43s\n- octavia-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/7e411b50f9e94a4dbaba848af3b3ec59 : SUCCESS in 10m 46s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/28da228a6cda4f0086843db15fbd40db : SUCCESS in 6m 37s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/05a591012de34f779468e73d98919158 : SUCCESS in 8m 02s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/d872d844ad5c4a30b7e7293d6418d7ce : SUCCESS in 7m 44s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/5e83a2475ea24f73815629a371f7d118 : SUCCESS in 9m 04s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/72ea8ac250804f82ab627c2c27105258 : SUCCESS in 1h 24m 56s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/0e3cce8ed69b4c1daeee819ef618a275 : SUCCESS in 1h 40m 44s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/6fa216525f3f48be94844f929e937c88 : SUCCESS in 1h 01m 19s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/9bb12ac0b813427781f56755c8195134 : SUCCESS in 1h 46m 01s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/3d18ed43c22f46f683ebcee297b9ab44 : SUCCESS in 1h 12m 34s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/cab1edc5015543a1a0ee3978f993f649 : SUCCESS in 37m 09s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/bc9c13cb642747f282f8a74c5876969c : SUCCESS in 42m 21s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/7ef60cade29a4e1c904e50f19576ee5d : SUCCESS in 40m 58s (non-voting)\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/8893256c49d84b1bb0aa8642db6d5090 : SUCCESS in 1h 07m 36s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/958ce4f0f5494886bb3716af0ec4a67a : SUCCESS in 34m 20s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/ed3765e1121a43bfb75cb8c1adaba4e6 : SUCCESS in 2h 01m 41s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/9fbee65c26d9447f86775a9966f62db2 : SUCCESS in 2h 47m 33s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/250956887e1b43889007fc52f64df72f : FAILURE in 23m 08s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/21a73929983f4f9b81ae75f8823fa194 : TIMED_OUT in 2h 35m 00s (non-voting)","accounts_in_message":[],"_revision_number":3},{"id":"83a977ee585b3788a4cbbcddd6206f0901c53610","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-04-09 08:40:02.000000000","message":"Uploaded patch set 4: Commit message was updated.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":4},{"id":"5324e04bd18854bba66c45974ff75cc70fc5b99b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-09 11:17:51.000000000","message":"Patch Set 4: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e8d848b593e145208009ab6eb6ebd675\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/321952b11f004921914971371474d5c4 : SUCCESS in 11m 31s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/cc61baad2a3b4bca87b68281d6900484 : SUCCESS in 4m 33s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4670f4162d084ff2908a544f4e5e22cb : SUCCESS in 4m 32s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/170870f4ffe24a8abd993e866516f495 : SUCCESS in 5m 06s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c140405134ea4daba8e298a62b9ed5ae : SUCCESS in 8m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1b4547a567bb497cbbbfa4c895c6ac7d : SUCCESS in 11m 56s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/7cf2325ef15b42beb98e49d0776616df : SUCCESS in 5m 05s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/72bc1ba1a1974f95afe5dd981025c502 : SUCCESS in 8m 30s\n- octavia-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/34e9da7014714ca89c04190809229d19 : SUCCESS in 10m 18s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/f7f95e7871024b3f9d63f632c7623d24 : SUCCESS in 4m 01s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/88e74eaa1ec54d329ec3b5518b3e7a45 : SUCCESS in 6m 27s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/79c84e09456c4f90b9e199e62f39f669 : SUCCESS in 6m 12s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/9966f9f4a33748a19c852dcd8a597088 : SUCCESS in 6m 20s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/e558181c96094c7c9c2362d37d6704ac : SUCCESS in 1h 17m 12s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/4ddb437c2e0941b3862f505362f5cb59 : SUCCESS in 1h 42m 17s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/fd3d466b63e24c26a635987ba937ebdb : SUCCESS in 1h 04m 35s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/c578d38b48f5494189f2f30a3b5e6fc1 : SUCCESS in 1h 41m 32s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/3029d76dd5aa4aa0b77d03b6eb5213cb : SUCCESS in 1h 19m 21s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/a6190beb1ac443ad9035cdee03d62ce1 : SUCCESS in 43m 16s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/59b1aa4afd5d40fbbb66e04202d917af : SUCCESS in 43m 55s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/df86d21d49ce4bb2a67ff5a72707b655 : SUCCESS in 1h 06m 08s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/13ff6af6c2fb4610a9a2c2dba6eba214 : SUCCESS in 30m 40s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/8cea57a14f174e13891dda20d60c3827 : SUCCESS in 2h 01m 51s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/d2ee1b3cd8424a87b20532b051143218 : SUCCESS in 2h 18m 31s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/799a9e3a59f64c8fb9f6c6de7624da42 : FAILURE in 23m 01s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/646562f11f164a8fb0f943650f1a103c : SUCCESS in 2h 17m 22s (non-voting)","accounts_in_message":[],"_revision_number":4},{"id":"7ea8a3619cf7cda58560c75181cb2b3419002b46","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-04-15 12:24:49.000000000","message":"Uploaded patch set 5.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":5},{"id":"7e3c00fc6356f33ebac899164ea7d709cef3f3e4","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-15 15:34:10.000000000","message":"Patch Set 5: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/26896c1397a94a558094a94a2b77a215\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/10718d9ad4a54f078a640ceff7c8b40c : SUCCESS in 9m 57s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/e085e05389424526898794bd7eac5eb8 : SUCCESS in 4m 25s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/c0f08bddaa034535b0d6205562c91ab4 : SUCCESS in 5m 00s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/e3bc584253244e43a704b3120424975d : SUCCESS in 5m 22s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b26c1b3eeccd4022872a4cf7556d7715 : SUCCESS in 11m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/65a648468aac402bb5a69e3d11375679 : SUCCESS in 12m 05s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/8a964c8ab81440fcb389f5b7efbaf9a0 : SUCCESS in 5m 26s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/7822b5d53c5a4c639a217c5b17b762ec : SUCCESS in 8m 47s\n- octavia-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/9571ec5a92ba4da680f16f930cb0eeab : SUCCESS in 7m 33s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/da42a4517e8643c8aaca4a18754e27f7 : SUCCESS in 4m 58s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/d8053199e21d48178ef300dadb4559aa : SUCCESS in 8m 31s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/ec6df438c10945fda6009b2520ab107e : SUCCESS in 6m 50s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/eaea97e6661442168fe7e8ca7999ea10 : SUCCESS in 6m 18s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/de74fcd480bf4a9d88a7fc58e75751c6 : SUCCESS in 1h 48m 20s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/b5d9328413184eaab3816b92984e5f22 : SUCCESS in 1h 58m 41s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/ed316a206a514d73a86cd863a67aeb3b : FAILURE in 29m 18s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/df6cc4562f5743e1baa1fc9fc55d1c2f : SUCCESS in 2h 07m 41s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/00f1fd00e3b8411dba888e5215a4befc : FAILURE in 29m 21s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/14da52cde02c41ea8d75faae077023ec : SUCCESS in 51m 51s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/e24c911fbe9440d699ee3186d42f72f0 : FAILURE in 29m 28s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/3ae5c586a6a2452397cf26cb454fe5d5 : SUCCESS in 1h 07m 41s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/72594faf31fb4475b09769b420c6e3c5 : SUCCESS in 47m 12s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/b52675c5b402472bb7e71e2605da646b : FAILURE in 28m 52s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/9edfcc9165194dbf96e9aa9f4901b6e7 : TIMED_OUT in 3h 04m 54s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/b526f13d86954582b72a668242d6d89a : FAILURE in 29m 38s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/caced68078384a989c3c642fdf373ee2 : TIMED_OUT in 2h 33m 54s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"3a49196703b615cc77297688e91ffeb80387f687","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-04-16 06:43:34.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"0ae64f3870a46af957760fa11e4476d4c89382f0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-16 09:37:52.000000000","message":"Patch Set 5:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/882477d47a06479494b4fdd13b00f576\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6cca5ac6dba949678207a05e61a01ff1 : SUCCESS in 12m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3c4c4e6649b146ed839c13fa684c8bad : SUCCESS in 5m 53s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b87a24aaa9d443288648aa742e1913bc : SUCCESS in 4m 52s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/3eeaf1e2f7fb4eab93624a2a14ae2271 : SUCCESS in 4m 17s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/1e5c8dc0bff34c74a7afa86e3a0eea7a : SUCCESS in 9m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5d74d06b77294fa095dbab666eaabe07 : SUCCESS in 11m 46s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/3e523bf95e474047a5863aec1918560c : SUCCESS in 4m 56s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/9d421c963f28496693735d332e797683 : SUCCESS in 7m 28s\n- octavia-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/59a5ad9bc2034ed986ada09eb4aeae4d : SUCCESS in 8m 59s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/3f2cb2d053904c43923bdd92e64f6552 : SUCCESS in 6m 05s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/40c761fdeea7414c8d61b7dc99935a75 : SUCCESS in 6m 23s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/348f732d68944bd3a5c142bb5362f62e : SUCCESS in 8m 57s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/20bcf2a63e5d4adeb19509482fe630c2 : SUCCESS in 7m 38s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/78961f947ace4ff3b11e83c6c88b3ea9 : SUCCESS in 1h 20m 41s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/9110e5fd221c418596cc90c2c8a6d8f7 : SUCCESS in 1h 41m 13s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/0e50f5da181c47748d81cddbd77bace5 : SUCCESS in 1h 30m 32s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/05ecfe5aeb7e40479a2e8bf2d5307888 : SUCCESS in 2h 16m 16s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/0dc0fce7b69848f083a694b145108687 : FAILURE in 1h 34m 27s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/59e789ca6dda43b1a91a96e15553d6f9 : SUCCESS in 48m 13s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/d518073e466945ed833defb545ba2856 : SUCCESS in 57m 23s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/0d486ff1ab8c449c8d7501a11e85f880 : SUCCESS in 1h 07m 50s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c1ada69e160b47d89c41441f48346dcd : SUCCESS in 49m 26s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/fb33c150410e41159eb62dd09c5a2956 : SUCCESS in 2h 19m 43s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/90a3deb7cf164845a72a174a7af2c584 : SUCCESS in 2h 48m 07s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/c5dad59095674920885f51c0d3542f91 : FAILURE in 36m 18s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/85aee19fe4d54ca29ff8f112e6ba39de : TIMED_OUT in 2h 34m 50s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"c82ff3e554adce796bed24ba9b9e2e7b1e2a612c","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-04-22 07:40:19.000000000","message":"Patch Set 5:\n\n(1 comment)","accounts_in_message":[],"_revision_number":5},{"id":"4983cd50384d044309e47968bd95d7f5f5720b95","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-22 10:17:07.000000000","message":"Patch Set 5: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/0ab1acbcdbde4de1906d1519cb54fec8\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/46f59b70557f41d6a27cfff29a2ab8ee : SUCCESS in 8m 37s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/22de9dd8704e4e10aba6055ec3373fc8 : SUCCESS in 4m 31s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6e9840051f8945b49bc5a0f91eff976b : SUCCESS in 4m 52s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/8cf642f473be48b9a81ec7eb4da054d2 : SUCCESS in 4m 52s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/44a2bc81d5bc481cb76a2ace00273f98 : SUCCESS in 8m 43s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/b80f3469e444456fb733d91878e5da53 : SUCCESS in 12m 03s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/d0d9a64865fc4ac785965707549d6208 : SUCCESS in 4m 22s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/1c9c9f19ac6347669fbdb499f11fe135 : SUCCESS in 8m 46s\n- octavia-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/c08f39e42bf34a86a2c6fa3e71a19efd : SUCCESS in 10m 15s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/20ee645957d849a19dcb67b45f04faf7 : SUCCESS in 3m 58s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/3a7d40c694a4467cb73c24e49ed4e27c : SUCCESS in 7m 32s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/cb761827ef404537b94e427f74f600af : SUCCESS in 7m 32s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/f594dad736b54a0eb408695a6ef1503e : SUCCESS in 6m 06s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/fc5fb961a9404595ab07538c2b9ac4ac : SUCCESS in 1h 07m 19s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/a40ef679921749a1b135f69fd3a51dcc : SUCCESS in 1h 51m 16s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/0eff51fd08184c208146b44ea478ae83 : SUCCESS in 1h 05m 01s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/1e505645fe3f4f5f8039cbd96dd9e345 : SUCCESS in 1h 51m 08s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/b0bd4c5c9e1c4dc9b67a6959c72e5917 : SUCCESS in 1h 10m 01s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/4447073dac1e4be3b894bf1d0972ca56 : SUCCESS in 39m 24s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/2d9a2d43664440e6a3891278513d8ebe : SUCCESS in 43m 41s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/ce04fa85f61b4fbb870745ec4215a680 : SUCCESS in 1h 09m 05s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c10a031f137940d8a7af8402c040077d : SUCCESS in 37m 46s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/e11804230fa64e1db3db6f8f58b7f518 : SUCCESS in 2h 00m 44s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/d64cdc080ae643ab8fd1212b29a98035 : SUCCESS in 2h 34m 42s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/89cf46df44c04199a6b179f4e8824d43 : FAILURE in 20m 43s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/43dad3abf57f4d839a8278a6987fb48f : TIMED_OUT in 2h 34m 58s (non-voting)","accounts_in_message":[],"_revision_number":5},{"id":"556fcf09a58ea4766b38c0b6d363c8c55ef02a49","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-04-29 08:19:06.000000000","message":"Uploaded patch set 6.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":6},{"id":"de6444479c39a0f89968bfa32ef6237b6a6a1ce9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-29 11:05:58.000000000","message":"Patch Set 6: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/757948075bd54a8eaec43f1cd44cc9bc\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/ab6426c728b1496aa80c01d80001160d : SUCCESS in 9m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f9c03b34f9594026a273b288f1cf1901 : SUCCESS in 5m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a9ac78906be34d6daccdf00f641039f5 : SUCCESS in 4m 54s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/4e451a24bedc4489825a9e04eff0ecbe : SUCCESS in 7m 54s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c5826bb2e3d34932a0b513b5ad4d4611 : SUCCESS in 14m 54s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/20ee69d46e3748adbe691c9be6f0db34 : SUCCESS in 12m 44s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/cdc53b35e8ea408a98dab5841e95c17d : SUCCESS in 4m 45s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/84965b10efd042d19ef87ef580e4e12f : SUCCESS in 4m 58s\n- octavia-tox-functional-py311-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/3c618a9cd4414b7d9545fad1acdfa2e4 : SUCCESS in 7m 29s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/00d8b8d015c34b86823943254ef87c54 : SUCCESS in 8m 46s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/33d41b01dd5e42cbabb503e78b1315a2 : SUCCESS in 7m 50s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/ddfc1fcb49ae4ea18d5953c7a9c3ef02 : SUCCESS in 6m 54s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/1faba7121931449082e30359356bcb7d : SUCCESS in 5m 31s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/8466ebe7ce664dafa63785280ea17fcf : SUCCESS in 1h 34m 17s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/432942d7a59b4b93b40a033552c16760 : SUCCESS in 1h 37m 55s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/8b7b25cf9fbd479d895b4ab9ac03a896 : SUCCESS in 1h 04m 52s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/9f3eb31343594d29a4208e5dd54ca129 : SUCCESS in 1h 42m 36s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/58eb672f0d31413ca1fdebbdd4730f4a : SUCCESS in 1h 00m 41s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/93ced0a290fb4860a8deca6bb9289976 : SUCCESS in 37m 38s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/774e17bfc50c437ba1c9379495a4d160 : SUCCESS in 40m 28s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/8430e78c345b429f8e00b39f8472b64f : SUCCESS in 1h 11m 36s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a98e5b8069d14ef0a09b50e292e391b3 : SUCCESS in 34m 10s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/ec5559bf13084c3e844eb4fe5d77866f : SUCCESS in 2h 26m 58s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/70b82681729f46a9b55b55b34acbf17d : SUCCESS in 2h 32m 52s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/704feec9094949f8957dfbd822fa678e : FAILURE in 22m 57s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/5032f2aa520f4d338c5c7b0f56458422 : TIMED_OUT in 2h 33m 22s (non-voting)","accounts_in_message":[],"_revision_number":6},{"id":"82f857934865cff140f501d87617db3dfefa1d59","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2024-04-29 15:57:40.000000000","message":"Patch Set 6: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":6},{"id":"4e5ab1c06c27ba6fb65697c729b39e82b6e6305c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-05-15 11:32:34.000000000","message":"Uploaded patch set 7: Patch Set 6 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":7},{"id":"2e5e830ae34171fff020860f3f0a6e30cb3fdfc8","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-15 14:39:26.000000000","message":"Patch Set 7: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9e700841119045c3b9e28644ed51900f\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/aa568865b5a340bc9854388a632489aa : SUCCESS in 8m 51s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/0c1c053459014222aa9d985ca9633c0b : SUCCESS in 4m 21s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d02dbaaad3d44f478cff2ecf44ae92fe : SUCCESS in 4m 41s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/fd261eb326a340c8ad028af7f785f16f : SUCCESS in 4m 15s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c929c7cb02cf4ff9af8d23ae2ec1633d : SUCCESS in 8m 16s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/908f0f1133b6416a898202a2d290cfda : SUCCESS in 9m 54s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/96894d0c882641028f7edf1c6255410f : SUCCESS in 5m 13s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/594704933b29415da7ffbe72d98fd2d9 : SUCCESS in 6m 05s\n- octavia-tox-functional-py311-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/dd47ceb2abb6416787d9935d8539722a : SUCCESS in 5m 32s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/778d24cd8dd24c59a3a30f8eaf09dcb8 : SUCCESS in 4m 38s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/e4a16f0b6b7b44769b70182099716c58 : SUCCESS in 8m 40s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/15d57cada67e4d4a964f321276428df0 : SUCCESS in 7m 27s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/998ff79862224e57b4485c00e55c09ce : SUCCESS in 4m 39s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/e5fd9b81bf5345f08c380cde1c170297 : SUCCESS in 1h 33m 56s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/4c40aefd41e24fc6971c3c26ff632ee6 : SUCCESS in 1h 44m 24s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/e27e98b8918a472daf5a38bbcf99b65b : SUCCESS in 1h 15m 28s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/9c01ca4b8b0e4296816fcd07dd598b74 : SUCCESS in 2h 21m 40s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/40deb06ceed64df59d3f2634e90a9c07 : SUCCESS in 1h 04m 59s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/687f43ba9efd4a949832cb6f7bf8828c : SUCCESS in 48m 09s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/15239a2ba1194256b5483b9d36c45fb3 : SUCCESS in 52m 46s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/8cd8a5093e7340c787e71d9da95a0248 : SUCCESS in 1h 10m 23s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/efbd45e19bc243ec80dd7a94e114d61a : SUCCESS in 45m 39s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/c4ad075a7bcb4e0db6706964ce15ccf7 : TIMED_OUT in 2h 33m 07s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/713408946a9248319473283a06e12efd : TIMED_OUT in 3h 04m 35s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/f86f4abf28d249c3bc31689121da50f9 : FAILURE in 20m 07s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/70d0d11a3b814effb669ddce2610a67d : TIMED_OUT in 2h 34m 28s (non-voting)","accounts_in_message":[],"_revision_number":7},{"id":"ebd19d9ece48fe67ded6b874f7cbb427ec0e6c29","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-06-03 16:25:42.000000000","message":"Uploaded patch set 8.\n\nOutdated Votes:\n* Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":8},{"id":"06fea46bfdbaaf5452cc44aa263b233ae89606cd","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-06-03 19:12:39.000000000","message":"Patch Set 8: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b2acdc962dd94075b70283992617141a\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3e1f16a0d1e84f7a91176e55b399b01b : SUCCESS in 10m 06s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8df9b76ca336459bbb39fe95c919a377 : SUCCESS in 5m 00s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d8ac43d6802b431d91d80d8752ebd36a : SUCCESS in 4m 48s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f668f202439241da975730ba9cc8fdeb : SUCCESS in 4m 51s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9a71db51da9a444faa19fac0ce98d383 : SUCCESS in 9m 18s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/11038494ad474a7f85cff84c93f28011 : SUCCESS in 14m 12s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/991717413c8247b2a8273918f4c9d456 : SUCCESS in 5m 12s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/ff988767362a4cd8aef982c36f3b04fc : SUCCESS in 5m 14s\n- octavia-tox-functional-py311-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/95e560c3da3445bdbbaf59c348ab5eac : SUCCESS in 6m 27s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/d061f5b29fff4292bc2590d6be415364 : SUCCESS in 4m 29s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/3db0a92b25ef4ff18aac92257325fb83 : SUCCESS in 8m 41s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/58e67b0a76214c4c89f3f7b4f3f538e2 : SUCCESS in 7m 48s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/5f4bda69208241bdbf71f59b81271982 : SUCCESS in 5m 29s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/6ce998c50a0c4e099a22760b34dee562 : SUCCESS in 1h 30m 48s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/83c9e0c12959459eb7a84001c6c542dc : SUCCESS in 1h 51m 14s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/4cda5ef70d884c65b52751752e9b5eae : SUCCESS in 1h 07m 32s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/e84b64ab128f482f94ea6a8c24708b9d : SUCCESS in 1h 56m 46s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/88782f0a020f4885837641897582cd84 : SUCCESS in 1h 03m 03s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/359b8456f18745818aae461e69a9f515 : SUCCESS in 36m 34s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/24d69062bbef4991b862ce4f601bec6f : SUCCESS in 40m 11s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/212af75accf74a5589b0e659acad60e7 : SUCCESS in 1h 13m 18s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/3e734246ceb24dcfacf1199ef4914a8c : SUCCESS in 36m 47s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/9e1397c0ff7841c883759add4f2fe21e : TIMED_OUT in 2h 32m 34s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/c26b5aefb422482cbe3dfeb4b0cdce4b : SUCCESS in 2h 44m 45s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/1f953002247345a6836e30c38e721f97 : FAILURE in 22m 06s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/dc697108f8aa4fb4b5425a03d9049600 : TIMED_OUT in 2h 34m 39s (non-voting)","accounts_in_message":[],"_revision_number":8},{"id":"70b3feab47c51c0e463311288319706b761da894","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2024-06-06 12:01:10.000000000","message":"Patch Set 8: Code-Review+2","accounts_in_message":[],"_revision_number":8},{"id":"2e2e5a3d0dab2613b3c7774567cf1159ce8ca8dd","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-08-21 12:50:09.000000000","message":"Patch Set 8:\n\n(1 comment)","accounts_in_message":[],"_revision_number":8},{"id":"44036a35d2cff45cb9b1af396ccf08623611f71c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-10-22 07:41:27.000000000","message":"Uploaded patch set 9: Patch Set 8 was rebased.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":9},{"id":"ef6d9b63820311be64e8f1c07226d384677e5296","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-22 10:28:49.000000000","message":"Patch Set 9: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/daaa6cd3c23648838a10c7f2755026c2\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/c3acbcb1c4064d109466f119e82a02e9 : SUCCESS in 7m 52s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/ebe1c69f24da42a1ba13245e82d5ce80 : SUCCESS in 4m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/b00a37877c5c47528424aad62ea5aee4 : SUCCESS in 4m 33s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/93e3749f414a461c83097fc341c77578 : SUCCESS in 5m 07s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7ace65040e504bd588a0366788ef177c : SUCCESS in 8m 45s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/eea374c7a706415aaeac78fc643f69f1 : SUCCESS in 4m 10s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/b9218584ec79449385f39cff10900b58 : SUCCESS in 6m 34s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/4ad09bccca1d4dc8bd982527650b4d26 : SUCCESS in 4m 05s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/c6e2b5f44b03469da30a1546baa915e6 : SUCCESS in 7m 13s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/dddbb52e5dcf4f87a239046ba720ae98 : SUCCESS in 5m 20s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/834d165ee705404d92501edda638a229 : SUCCESS in 7m 27s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/973a2be1c2f84bb3bb32a01ae9747f8c : SUCCESS in 49m 30s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/85a38185f00c4baab6bf1161167854b7 : SUCCESS in 1h 55m 43s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/dfac5ad5542c4fb1b92bc68fafd50045 : SUCCESS in 1h 03m 29s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/e350ef8bd4f345be93ac0c55ebb4beeb : SUCCESS in 2h 06m 52s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/f6abd1eb8a714d53b7233cf9f7129d4d : SUCCESS in 58m 01s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/59593aea5c76432693bbfa8949f7bb1e : SUCCESS in 38m 25s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/9d5c693ec62542b687f49768a472de22 : SUCCESS in 44m 56s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/e598fccf3c0f43a088410d5237a88f47 : SUCCESS in 44m 42s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/feb9f795451a4a27ae6f9b86de0169f2 : SUCCESS in 30m 12s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/a0b8633ca5044f59b1d2e4f677cc668f : TIMED_OUT in 2h 32m 43s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/cd46f0911dd04a8fb3bd8c1c33ddbc5c : SUCCESS in 2h 41m 07s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/a2ec1c5fdff34be884508949beabab8e : SUCCESS in 2h 21m 58s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/5398ac0a91ca445ba846621c24f1d075 : TIMED_OUT in 2h 33m 18s (non-voting)","accounts_in_message":[],"_revision_number":9},{"id":"42df103938515ad2f25130db6cde26cc0064e77e","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-10-22 14:59:30.000000000","message":"Uploaded patch set 10: New patch set was added with same tree, parent tree, and commit message as Patch Set 9.\n\nCopied Votes:\n* Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":10},{"id":"4a45a9d9d003e500ae3158507ee257bd3dd1b216","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-10-22 16:35:18.000000000","message":"Topic rate-limiting removed","accounts_in_message":[],"_revision_number":10},{"id":"4f8b1b52206cb1cc8f749e409ea061ab5e589c74","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-10-22 16:35:24.000000000","message":"Topic set to mypy","accounts_in_message":[],"_revision_number":10},{"id":"3eb8d69c0619a420a1a6b3b37f76940bf3bd72ca","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-22 17:35:22.000000000","message":"Patch Set 10: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cedca0dc49a14ea6b7fbd47e8a51ed5c\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/0b1314b4cfe74d48865e6b17325f5503 : SUCCESS in 12m 53s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/d2715e7d317a41e4bc21cea8dfa3c054 : SUCCESS in 4m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/60867b0957a344ad8843b73296059194 : SUCCESS in 4m 40s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/b5bcbad8d44a410bb1c4518046299397 : SUCCESS in 4m 38s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/740223db9190497cb14ec08e0383d2fb : SUCCESS in 14m 59s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/9751a92a227a454f82aa1de606de2d8e : SUCCESS in 4m 46s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/36e710b25cfc40d7a967bf359b30c14f : SUCCESS in 7m 56s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/a52df2412f1a45618556e7b34602aeeb : SUCCESS in 8m 50s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/e98825be93cd4705a7b8a5cc73da77a3 : SUCCESS in 9m 37s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/b6e3cc128f884e50a51508ca46654689 : SUCCESS in 7m 16s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/0209ed624aae4c769d365453958edd7d : SUCCESS in 6m 40s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/eeb2adf542d046f7b389657d44e24bda : SUCCESS in 1h 48m 51s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/3e26a139318a49539d4456962b2e3a46 : SUCCESS in 1h 41m 48s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/e69b1cc7a04f49b0b9f37cb20283be44 : SUCCESS in 55m 36s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/8ee81f766f354d769f4a3c698766311f : SUCCESS in 1h 46m 24s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/01468244389240a1abba9c08688fa4ce : SUCCESS in 1h 24m 09s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/7dc52bfefecc4e5d9cc00aca1bf3043b : SUCCESS in 32m 19s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/46299ab2eec14b3cb0df8c942354d257 : SUCCESS in 36m 39s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/3015e042485443ff87e8786f11b2f32b : SUCCESS in 56m 00s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/643309187bed4522941cc18db88616d2 : SUCCESS in 31m 22s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/33a5448fb33e436d8c3879ec91ad513b : TIMED_OUT in 2h 34m 01s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/bb02768d7cb5493da262e12183818b29 : FAILURE in 2h 14m 21s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/59396271cef845cc9fdcb8565987cc52 : TIMED_OUT in 2h 33m 01s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/0b95a6a8e3494a18946d1c88b22dbd17 : TIMED_OUT in 2h 34m 01s (non-voting)","accounts_in_message":[],"_revision_number":10},{"id":"be79b7a6b1ca47ca0876fe09a3b668efa63651e5","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2024-10-23 18:08:19.000000000","message":"Patch Set 10: Code-Review-1\n\n(6 comments)","accounts_in_message":[],"_revision_number":10},{"id":"e70a2e78c932c3c18a3a64953f7b5ce503a9a40c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-10-28 11:14:50.000000000","message":"Uploaded patch set 11.\n\nOutdated Votes:\n* Code-Review+2, Code-Review-1 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":11},{"id":"39b78ead172ee1b9a29ec74835ba8ef14de74c67","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-10-28 11:14:58.000000000","message":"Patch Set 10:\n\n(5 comments)","accounts_in_message":[],"_revision_number":10},{"id":"79a4b2dc623d6a1bcea26bbd41286dd78b1dea9a","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-10-28 11:47:27.000000000","message":"Topic rate-limiting removed","accounts_in_message":[],"_revision_number":11},{"id":"eddd8fdaca5ac0a62f9d34bfa3d549103e47b747","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-10-28 11:47:31.000000000","message":"Topic set to mypy","accounts_in_message":[],"_revision_number":11},{"id":"b05f2603ff1b0c8ff51aa9bc8825cdab918f445c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-10-28 13:21:24.000000000","message":"Uploaded patch set 12.","accounts_in_message":[],"_revision_number":12},{"id":"d7ebd7055248a311286fb8606386b4deec6841e9","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-10-28 14:49:26.000000000","message":"Topic rate-limiting removed","accounts_in_message":[],"_revision_number":12},{"id":"13949a8b1e61431189178bb557dbc64505d78bab","tag":"autogenerated:gerrit:setTopic","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-10-28 14:49:33.000000000","message":"Topic set to mypy","accounts_in_message":[],"_revision_number":12},{"id":"a9f3326e564006e9360887fb216af50762c024c9","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-10-28 16:00:31.000000000","message":"Patch Set 12: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dbbf8697eecd46c08b23fcb842ad7d97\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/1ac82f4390ca4f6bbe4ca1b7cec04b86 : SUCCESS in 10m 30s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eb496c41be974b3394ec5e58646f72e0 : SUCCESS in 4m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/4e64962057394f6bb1911a7867ca3594 : SUCCESS in 4m 13s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/9f87d010319e4f6781d26f71d05dec6f : SUCCESS in 5m 12s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/8216d309314041f7b028c8990edbeba0 : SUCCESS in 11m 15s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/7385079ddfaf49bcbeb1012534fa1ab8 : SUCCESS in 4m 34s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/082041e5a37a4a75bf331619db690b95 : SUCCESS in 6m 47s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/6a17996fdaa94e64a1e33c065c0fb487 : SUCCESS in 4m 11s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/38327ebcf1cc45dab0d487f96cb54884 : SUCCESS in 6m 59s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/a35b1bd6df7242848fcaed42125177ab : SUCCESS in 4m 56s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/d173236bfb3c4f05a667490e0f062cba : SUCCESS in 6m 19s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/d6b5825002fa4a658f1610396e936ce6 : SUCCESS in 1h 43m 20s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/e728542c4a1542f7a1c86b6fe480222d : SUCCESS in 1h 28m 56s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/4ffc86f45d724612a9d43e90acf4068a : SUCCESS in 1h 21m 15s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/7105a1df978248d291f3b090c029aac0 : SUCCESS in 2h 26m 33s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/11d1d6c3abfb4674adc2ded88a5a339a : SUCCESS in 1h 36m 03s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/bfda38d3ff1c4a1baabcae25ccf257d5 : SUCCESS in 41m 16s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/1544d4111e8442d2afad4787cfa5aadc : SUCCESS in 48m 31s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/bd5c9d6b83d94730bcb7166576ad7ac6 : SUCCESS in 1h 14m 58s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0f1d1f5814c74eafb6a2dedfdd58e92f : SUCCESS in 38m 35s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/b22d1664321848db862a351fee98489b : TIMED_OUT in 2h 33m 26s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/84d3cf93c4b64f51bdf5610e4ece2a83 : SUCCESS in 2h 35m 23s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/a8d2612cff3d4b0180cac977c3375690 : SUCCESS in 2h 17m 07s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/a8e0e59451b0444ab4961121dfbba024 : SUCCESS in 2h 33m 20s (non-voting)","accounts_in_message":[],"_revision_number":12},{"id":"bcdf75f02b8d3a05c507676142e7eea0e00ce909","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-11-20 12:07:07.000000000","message":"Uploaded patch set 13: Patch Set 12 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":13},{"id":"ad0cf35ca911c607b978fc48b47f2fc07a4679aa","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-11-20 14:52:03.000000000","message":"Patch Set 13: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/32cad298072345cc9421bf1997873844\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6126fbffe2284e7e9f59f5fcefa8af67 : SUCCESS in 11m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a91ab3478da74da99a1c4d6d1f706c7e : SUCCESS in 4m 37s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6bd9a65a21014cdcb1faa1cb38ca2700 : SUCCESS in 4m 48s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/6bdfe0ab471e4ba6aabf20157a16e026 : SUCCESS in 5m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/60d1eb8f42014497b535df371a5f185a : SUCCESS in 12m 15s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/beb0bb1f224c43ecb08bbdcb12e1671b : SUCCESS in 4m 44s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/41523ec2e60c413cb13050f3d624fe18 : SUCCESS in 6m 08s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/10b91a1602bc4a6cb73ea25a02550389 : SUCCESS in 6m 20s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/546bd7f2b5aa4900a1ae61618653cdc7 : SUCCESS in 7m 37s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/4022d43a2cfa481a84ba74c1adcfd69c : SUCCESS in 7m 11s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/2dfc7396a0704bfaa58eebdb6d02961d : SUCCESS in 6m 41s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/017c3940bd7d4650b19efb3578998997 : SUCCESS in 1h 12m 01s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/7722699a4f65456488f2ab9151636c64 : SUCCESS in 2h 20m 14s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/0826ba7b4e3a4b17a203dc3c1cc81d0f : SUCCESS in 1h 16m 28s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/d318530806f84ca4b01927874acf787d : SUCCESS in 2h 15m 48s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/606086da321149ae8dc717143d72808a : SUCCESS in 1h 00m 53s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/0830f401a0934a34af1ff4c140460576 : SUCCESS in 52m 46s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/77f8ca6ae8fd49a4a599f50fa59c8d52 : SUCCESS in 49m 30s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/8365eb038c82490d9a6aebe49baeb52f : SUCCESS in 59m 16s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/98f7b3e29ec64a0786c22014977488fb : SUCCESS in 37m 29s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/9175900269a6458691edc09b5a9606f7 : SUCCESS in 2h 31m 04s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/d1fd48d765a84fa181ff768273ebc5e2 : SUCCESS in 2h 42m 08s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/b7c9ec870b1640549a421efa8b39d556 : SUCCESS in 2h 11m 05s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/d1fb0902deb5428f9322bc9ff7c11050 : TIMED_OUT in 2h 33m 27s (non-voting)","accounts_in_message":[],"_revision_number":13},{"id":"edf0d5fb01951d78effc1cccac519b40666e93a0","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2025-02-10 15:47:30.000000000","message":"Uploaded patch set 14: Patch Set 13 was rebased.\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":14},{"id":"0fb5b1dbe70caebd8174a0cfd8db671073175afe","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2025-02-10 18:47:09.000000000","message":"Patch Set 14: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e5bab4a661604d8dbaecc9d3548c8fa5\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/166652dce0ce456cbe8d87ca86c16be1 : SUCCESS in 13m 55s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/97d64693373c4ec1986395bc60357133 : SUCCESS in 5m 14s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9840f14f24ad46b1b92dc919c77c8683 : SUCCESS in 5m 03s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/a6e621a2834442d587b6e6490cab63a2 : SUCCESS in 4m 56s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f61daf575db04bcf9cba5e35d916e7fa : SUCCESS in 12m 23s\n- octavia-tox-py311-tips https://zuul.opendev.org/t/openstack/build/8058c3557e1441caa8829aaafa5d4cd1 : SUCCESS in 4m 24s\n- octavia-tox-functional-py311-tips https://zuul.opendev.org/t/openstack/build/0738e83073444611b91c0d3b9a80e275 : SUCCESS in 6m 32s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/22b6348c84674f26b39d277a695f91a2 : SUCCESS in 4m 42s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/527af586901f4297858e8ce47135a9ba : SUCCESS in 6m 48s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/b40547e6810d43808858b460f55ecfcf : SUCCESS in 8m 20s\n- openstack-tox-functional-py311 https://zuul.opendev.org/t/openstack/build/02afb872520d403dadb0d7f15e587941 : SUCCESS in 8m 17s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/2c0507a78efc41a7b0b3e8e9697c6d22 : SUCCESS in 1h 16m 34s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/59989436208c46999db892e38125907d : SUCCESS in 1h 49m 49s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/b32133963370495b8d8a1eda0461f09d : SUCCESS in 1h 08m 37s\n- octavia-v2-dsvm-scenario-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e86779d432b7483f81eee58538d997e5 : SUCCESS in 1h 47m 33s\n- octavia-v2-dsvm-scenario-non-traffic-ops-ubuntu-jammy https://zuul.opendev.org/t/openstack/build/e6b94a8629b645f4ae2e5c06f7ab7dee : SUCCESS in 56m 23s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/a691e459d18c47f09fc1bcd72595891f : SUCCESS in 1h 56m 32s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/fcf6a8c3526c4cf58d504ef3ff3396f9 : SUCCESS in 1h 12m 58s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/6035721ac2bf4186b7aae2b17d6a0a6a : SUCCESS in 41m 56s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/929b0a232b374db3a285eb5528f92bea : SUCCESS in 47m 05s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/b292e241356147e7839bbd6f6e4df3d4 : SUCCESS in 1h 05m 07s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b09aa89c65f2426ea0efbefe1e515a95 : SUCCESS in 34m 48s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/7c56327d14d446aa8aaeaca5d57cb144 : TIMED_OUT in 2h 33m 34s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/731a3492a5ae4ce9b26e4a48378d2344 : SUCCESS in 2h 46m 27s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/58c20256d52349dd9cd74770ae402c18 : SUCCESS in 2h 24m 58s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/4051058150064d4d91b52aee6a1d0775 : TIMED_OUT in 2h 33m 38s (non-voting)","accounts_in_message":[],"_revision_number":14},{"id":"6355439eb17770785cd805661a752a00e1a36d7e","tag":"autogenerated:gerrit:abandon","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2025-02-11 09:39:17.000000000","message":"Abandoned","accounts_in_message":[],"_revision_number":14}],"current_revision_number":14,"current_revision":"5ce452565cbfa463a2e942e947e363094e167451","revisions":{"4ba3705583cdd12ac94607e65daae1f292f335ec":{"kind":"REWORK","_number":1,"created":"2024-01-10 08:40:57.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/93/905193/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/93/905193/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/905193/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/905193/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/905193/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/93/905193/1"}}},"commit":{"parents":[{"commit":"490a65fceb6c247f77d47cbcfcb6bc0bbdd6bdd7","subject":"Merge \"Fix issue with certificates with no subject or CN\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/490a65fceb6c247f77d47cbcfcb6bc0bbdd6bdd7"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2024-01-10 08:30:49.000000000","tz":60},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2024-01-10 08:30:49.000000000","tz":60},"subject":"Add typing information to octavia.amphorae package","message":"Add typing information to octavia.amphorae package\n\nApart from the benefit of finding possible bugs using mypy,\ntype annotations also have other benefits like for instance:\n\n- IDEs like PyCharm and VS Code can use annotations to provide\n  better\n  code completion, highlight errors and to make coding more efficient\n- It serves as documentation of the code and helps to understand\n  code\n  better. The annotation syntax is very easy to read.\n- Sphinx generated module documentation can use this information\n  (without the need for additional type info in comments/docstrings)\n\nThe syntax used here should be supported by Python 3.6, so it should\nnot create issues when backporting patches in the future.\n\nPartial-Bug: #2017974\nChange-Id: I5b0f084785f477c855218c54a21caecdecefb62f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4ba3705583cdd12ac94607e65daae1f292f335ec"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4ba3705583cdd12ac94607e65daae1f292f335ec"}]},"branch":"refs/heads/master"},"ace7b87bc880015d1186b03edf933212a27bb1e1":{"kind":"REWORK","_number":2,"created":"2024-01-10 13:22:02.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/93/905193/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/93/905193/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/905193/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/905193/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/905193/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/93/905193/2"}}},"commit":{"parents":[{"commit":"490a65fceb6c247f77d47cbcfcb6bc0bbdd6bdd7","subject":"Merge \"Fix issue with certificates with no subject or CN\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/490a65fceb6c247f77d47cbcfcb6bc0bbdd6bdd7"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2024-01-10 08:30:49.000000000","tz":60},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2024-01-10 13:21:41.000000000","tz":60},"subject":"Add typing information to octavia.amphorae package","message":"Add typing information to octavia.amphorae package\n\nApart from the benefit of finding possible bugs using mypy,\ntype annotations also have other benefits like for instance:\n\n- IDEs like PyCharm and VS Code can use annotations to provide\n  better\n  code completion, highlight errors and to make coding more efficient\n- It serves as documentation of the code and helps to understand\n  code\n  better. The annotation syntax is very easy to read.\n- Sphinx generated module documentation can use this information\n  (without the need for additional type info in comments/docstrings)\n\nThe syntax used here should be supported by Python 3.6, so it should\nnot create issues when backporting patches in the future.\n\nPartial-Bug: #2017974\nChange-Id: I5b0f084785f477c855218c54a21caecdecefb62f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ace7b87bc880015d1186b03edf933212a27bb1e1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ace7b87bc880015d1186b03edf933212a27bb1e1"}]},"branch":"refs/heads/master"},"a00d7c67782c345b94147a40ddc6950f5244fc4a":{"kind":"TRIVIAL_REBASE","_number":3,"created":"2024-03-25 11:43:51.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/93/905193/3","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/93/905193/3","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/905193/3 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/905193/3 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/905193/3 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/93/905193/3"}}},"commit":{"parents":[{"commit":"fa7cc0178783ad4ab9f3ddce893042d5952d0bd8","subject":"Merge \"Fix fully-populated API with allowed_cidrs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fa7cc0178783ad4ab9f3ddce893042d5952d0bd8"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2024-01-10 08:30:49.000000000","tz":60},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2024-03-25 11:42:55.000000000","tz":60},"subject":"Add typing information to octavia.amphorae package","message":"Add typing information to octavia.amphorae package\n\nApart from the benefit of finding possible bugs using mypy,\ntype annotations also have other benefits like for instance:\n\n- IDEs like PyCharm and VS Code can use annotations to provide\n  better\n  code completion, highlight errors and to make coding more efficient\n- It serves as documentation of the code and helps to understand\n  code\n  better. The annotation syntax is very easy to read.\n- Sphinx generated module documentation can use this information\n  (without the need for additional type info in comments/docstrings)\n\nThe syntax used here should be supported by Python 3.6, so it should\nnot create issues when backporting patches in the future.\n\nPartial-Bug: #2017974\nChange-Id: I5b0f084785f477c855218c54a21caecdecefb62f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a00d7c67782c345b94147a40ddc6950f5244fc4a"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/a00d7c67782c345b94147a40ddc6950f5244fc4a"}]},"branch":"refs/heads/master"},"6ea4bfdab2fc932f9c53eacd7af67f3c457bd514":{"kind":"NO_CODE_CHANGE","_number":4,"created":"2024-04-09 08:40:02.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/93/905193/4","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/93/905193/4","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/905193/4 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/905193/4 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/905193/4 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/93/905193/4"}}},"commit":{"parents":[{"commit":"fa7cc0178783ad4ab9f3ddce893042d5952d0bd8","subject":"Merge \"Fix fully-populated API with allowed_cidrs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fa7cc0178783ad4ab9f3ddce893042d5952d0bd8"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2024-01-10 08:30:49.000000000","tz":60},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2024-04-09 08:38:37.000000000","tz":120},"subject":"Add typing information to octavia.amphorae package","message":"Add typing information to octavia.amphorae package\n\n... and fix issues reported by mypy.\n\nApart from the benefit of finding possible bugs using mypy,\ntype annotations also have other benefits like for instance:\n\n- IDEs like PyCharm and VS Code can use annotations to provide\n  better\n  code completion, highlight errors and to make coding more efficient\n- It serves as documentation of the code and helps to understand\n  code\n  better. The annotation syntax is very easy to read.\n- Sphinx generated module documentation can use this information\n  (without the need for additional type info in comments/docstrings)\n\nThe syntax used here should be supported by Python 3.6, so it should\nnot create issues when backporting patches in the future.\n\nPartial-Bug: #2017974\nChange-Id: I5b0f084785f477c855218c54a21caecdecefb62f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6ea4bfdab2fc932f9c53eacd7af67f3c457bd514"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/6ea4bfdab2fc932f9c53eacd7af67f3c457bd514"}]},"branch":"refs/heads/master"},"3ceb7ac0345d5d36c50b83f581d579cfb932f662":{"kind":"REWORK","_number":5,"created":"2024-04-15 12:24:49.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/93/905193/5","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/93/905193/5","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/905193/5 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/905193/5 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/905193/5 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/93/905193/5"}}},"commit":{"parents":[{"commit":"603d437aefc5234af5790933185c135c50553080","subject":"Merge \"Add pyupgrade hook to pre-commit config\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/603d437aefc5234af5790933185c135c50553080"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2024-01-10 08:30:49.000000000","tz":60},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2024-04-15 12:24:36.000000000","tz":120},"subject":"Add typing information to octavia.amphorae package","message":"Add typing information to octavia.amphorae package\n\n... and fix issues reported by mypy.\n\nApart from the benefit of finding possible bugs using mypy,\ntype annotations also have other benefits like for instance:\n\n- IDEs like PyCharm and VS Code can use annotations to provide\n  better\n  code completion, highlight errors and to make coding more efficient\n- It serves as documentation of the code and helps to understand\n  code\n  better. The annotation syntax is very easy to read.\n- Sphinx generated module documentation can use this information\n  (without the need for additional type info in comments/docstrings)\n\nThe syntax used here should be supported by Python 3.6, so it should\nnot create issues when backporting patches in the future.\n\nPartial-Bug: #2017974\nChange-Id: I5b0f084785f477c855218c54a21caecdecefb62f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3ceb7ac0345d5d36c50b83f581d579cfb932f662"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/3ceb7ac0345d5d36c50b83f581d579cfb932f662"}]},"branch":"refs/heads/master"},"40ce99d06f10ecd673e8143ef3802f4c02a9e2de":{"kind":"REWORK","_number":6,"created":"2024-04-29 08:19:06.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/93/905193/6","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/93/905193/6","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/905193/6 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/905193/6 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/905193/6 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/93/905193/6"}}},"commit":{"parents":[{"commit":"603d437aefc5234af5790933185c135c50553080","subject":"Merge \"Add pyupgrade hook to pre-commit config\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/603d437aefc5234af5790933185c135c50553080"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2024-01-10 08:30:49.000000000","tz":60},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2024-04-29 08:18:55.000000000","tz":120},"subject":"Add typing information to octavia.amphorae package","message":"Add typing information to octavia.amphorae package\n\n... and fix issues reported by mypy.\n\nApart from the benefit of finding possible bugs using mypy,\ntype annotations also have other benefits like for instance:\n\n- IDEs like PyCharm and VS Code can use annotations to provide\n  better\n  code completion, highlight errors and to make coding more efficient\n- It serves as documentation of the code and helps to understand\n  code\n  better. The annotation syntax is very easy to read.\n- Sphinx generated module documentation can use this information\n  (without the need for additional type info in comments/docstrings)\n\nThe syntax used here should be supported by Python 3.6, so it should\nnot create issues when backporting patches in the future.\n\nPartial-Bug: #2017974\nChange-Id: I5b0f084785f477c855218c54a21caecdecefb62f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/40ce99d06f10ecd673e8143ef3802f4c02a9e2de"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/40ce99d06f10ecd673e8143ef3802f4c02a9e2de"}]},"branch":"refs/heads/master"},"233554a8ee88b6f7c30ad3fa6f5b6b8ba010c570":{"kind":"TRIVIAL_REBASE","_number":7,"created":"2024-05-15 11:32:34.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/93/905193/7","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/93/905193/7","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/905193/7 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/905193/7 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/905193/7 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/93/905193/7"}}},"commit":{"parents":[{"commit":"bc259c0bf06a9ff6df2362239ec13e3e028253b0","subject":"Fix new pylint errors","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/bc259c0bf06a9ff6df2362239ec13e3e028253b0"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2024-01-10 08:30:49.000000000","tz":60},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2024-05-15 11:32:18.000000000","tz":120},"subject":"Add typing information to octavia.amphorae package","message":"Add typing information to octavia.amphorae package\n\n... and fix issues reported by mypy.\n\nApart from the benefit of finding possible bugs using mypy,\ntype annotations also have other benefits like for instance:\n\n- IDEs like PyCharm and VS Code can use annotations to provide\n  better\n  code completion, highlight errors and to make coding more efficient\n- It serves as documentation of the code and helps to understand\n  code\n  better. The annotation syntax is very easy to read.\n- Sphinx generated module documentation can use this information\n  (without the need for additional type info in comments/docstrings)\n\nThe syntax used here should be supported by Python 3.6, so it should\nnot create issues when backporting patches in the future.\n\nPartial-Bug: #2017974\nChange-Id: I5b0f084785f477c855218c54a21caecdecefb62f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/233554a8ee88b6f7c30ad3fa6f5b6b8ba010c570"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/233554a8ee88b6f7c30ad3fa6f5b6b8ba010c570"}]},"branch":"refs/heads/master"},"af7734cf83fe1a4b7a8f5c6ee6d13a844042e1f6":{"kind":"REWORK","_number":8,"created":"2024-06-03 16:25:42.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/93/905193/8","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/93/905193/8","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/905193/8 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/905193/8 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/905193/8 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/93/905193/8"}}},"commit":{"parents":[{"commit":"541682314693ab543b148a9ed7703192f1673e00","subject":"Merge \"String conversion to f-strings\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/541682314693ab543b148a9ed7703192f1673e00"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2024-01-10 08:30:49.000000000","tz":60},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2024-06-03 16:17:40.000000000","tz":120},"subject":"Add typing information to octavia.amphorae package","message":"Add typing information to octavia.amphorae package\n\n... and fix issues reported by mypy.\n\nApart from the benefit of finding possible bugs using mypy,\ntype annotations also have other benefits like for instance:\n\n- IDEs like PyCharm and VS Code can use annotations to provide\n  better\n  code completion, highlight errors and to make coding more efficient\n- It serves as documentation of the code and helps to understand\n  code\n  better. The annotation syntax is very easy to read.\n- Sphinx generated module documentation can use this information\n  (without the need for additional type info in comments/docstrings)\n\nThe syntax used here should be supported by Python 3.6, so it should\nnot create issues when backporting patches in the future.\n\nPartial-Bug: #2017974\nChange-Id: I5b0f084785f477c855218c54a21caecdecefb62f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/af7734cf83fe1a4b7a8f5c6ee6d13a844042e1f6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/af7734cf83fe1a4b7a8f5c6ee6d13a844042e1f6"}]},"branch":"refs/heads/master"},"67c6567dcc439c5d0b17e4a3ae32e6cb5e248ab6":{"kind":"TRIVIAL_REBASE","_number":9,"created":"2024-10-22 07:41:27.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/93/905193/9","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/93/905193/9","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/905193/9 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/905193/9 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/905193/9 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/93/905193/9"}}},"commit":{"parents":[{"commit":"d1bb768220aceae137ab439863ef24118d5fdab2","subject":"Run octavia-db-manage in grenade testing from venv","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d1bb768220aceae137ab439863ef24118d5fdab2"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2024-01-10 08:30:49.000000000","tz":60},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2024-10-22 07:40:06.000000000","tz":120},"subject":"Add typing information to octavia.amphorae package","message":"Add typing information to octavia.amphorae package\n\n... and fix issues reported by mypy.\n\nApart from the benefit of finding possible bugs using mypy,\ntype annotations also have other benefits like for instance:\n\n- IDEs like PyCharm and VS Code can use annotations to provide\n  better\n  code completion, highlight errors and to make coding more efficient\n- It serves as documentation of the code and helps to understand\n  code\n  better. The annotation syntax is very easy to read.\n- Sphinx generated module documentation can use this information\n  (without the need for additional type info in comments/docstrings)\n\nThe syntax used here should be supported by Python 3.6, so it should\nnot create issues when backporting patches in the future.\n\nPartial-Bug: #2017974\nChange-Id: I5b0f084785f477c855218c54a21caecdecefb62f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/67c6567dcc439c5d0b17e4a3ae32e6cb5e248ab6"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/67c6567dcc439c5d0b17e4a3ae32e6cb5e248ab6"}]},"branch":"refs/heads/master"},"1aed14add667a6e073724548b727740ca4268b8f":{"kind":"NO_CHANGE","_number":10,"created":"2024-10-22 14:59:30.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/93/905193/10","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/93/905193/10","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/905193/10 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/905193/10 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/905193/10 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/93/905193/10"}}},"commit":{"parents":[{"commit":"d1bb768220aceae137ab439863ef24118d5fdab2","subject":"Run octavia-db-manage in grenade testing from venv","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d1bb768220aceae137ab439863ef24118d5fdab2"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2024-01-10 08:30:49.000000000","tz":60},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2024-10-22 14:57:40.000000000","tz":120},"subject":"Add typing information to octavia.amphorae package","message":"Add typing information to octavia.amphorae package\n\n... and fix issues reported by mypy.\n\nApart from the benefit of finding possible bugs using mypy,\ntype annotations also have other benefits like for instance:\n\n- IDEs like PyCharm and VS Code can use annotations to provide\n  better\n  code completion, highlight errors and to make coding more efficient\n- It serves as documentation of the code and helps to understand\n  code\n  better. The annotation syntax is very easy to read.\n- Sphinx generated module documentation can use this information\n  (without the need for additional type info in comments/docstrings)\n\nThe syntax used here should be supported by Python 3.6, so it should\nnot create issues when backporting patches in the future.\n\nPartial-Bug: #2017974\nChange-Id: I5b0f084785f477c855218c54a21caecdecefb62f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1aed14add667a6e073724548b727740ca4268b8f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1aed14add667a6e073724548b727740ca4268b8f"}]},"branch":"refs/heads/master"},"62983070cee20d6fcdccb25881ce2ff696b658ce":{"kind":"REWORK","_number":11,"created":"2024-10-28 11:14:50.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/93/905193/11","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/93/905193/11","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/905193/11 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/905193/11 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/905193/11 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/93/905193/11"}}},"commit":{"parents":[{"commit":"fd071d3391e5695e50748f74af4529186d63e472","subject":"Merge \"Drop SQLALCHEMY_WARN_20\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fd071d3391e5695e50748f74af4529186d63e472"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2024-01-10 08:30:49.000000000","tz":60},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2024-10-28 11:14:24.000000000","tz":60},"subject":"Add typing information to octavia.amphorae package","message":"Add typing information to octavia.amphorae package\n\n... and fix issues reported by mypy.\n\nApart from the benefit of finding possible bugs using mypy,\ntype annotations also have other benefits like for instance:\n\n- IDEs like PyCharm and VS Code can use annotations to provide\n  better\n  code completion, highlight errors and to make coding more efficient\n- It serves as documentation of the code and helps to understand\n  code\n  better. The annotation syntax is very easy to read.\n- Sphinx generated module documentation can use this information\n  (without the need for additional type info in comments/docstrings)\n\nThe syntax used here should be supported by Python 3.6, so it should\nnot create issues when backporting patches in the future.\n\nPartial-Bug: #2017974\nChange-Id: I5b0f084785f477c855218c54a21caecdecefb62f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/62983070cee20d6fcdccb25881ce2ff696b658ce"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/62983070cee20d6fcdccb25881ce2ff696b658ce"}]},"branch":"refs/heads/master"},"1a2ef66a322ec1619564e694b6b40511057868a9":{"kind":"REWORK","_number":12,"created":"2024-10-28 13:21:24.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/93/905193/12","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/93/905193/12","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/905193/12 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/905193/12 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/905193/12 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/93/905193/12"}}},"commit":{"parents":[{"commit":"fd071d3391e5695e50748f74af4529186d63e472","subject":"Merge \"Drop SQLALCHEMY_WARN_20\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fd071d3391e5695e50748f74af4529186d63e472"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2024-01-10 08:30:49.000000000","tz":60},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2024-10-28 13:20:57.000000000","tz":60},"subject":"Add typing information to octavia.amphorae package","message":"Add typing information to octavia.amphorae package\n\n... and fix issues reported by mypy.\n\nApart from the benefit of finding possible bugs using mypy,\ntype annotations also have other benefits like for instance:\n\n- IDEs like PyCharm and VS Code can use annotations to provide\n  better\n  code completion, highlight errors and to make coding more efficient\n- It serves as documentation of the code and helps to understand\n  code\n  better. The annotation syntax is very easy to read.\n- Sphinx generated module documentation can use this information\n  (without the need for additional type info in comments/docstrings)\n\nThe syntax used here should be supported by Python 3.6, so it should\nnot create issues when backporting patches in the future.\n\nPartial-Bug: #2017974\nChange-Id: I5b0f084785f477c855218c54a21caecdecefb62f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1a2ef66a322ec1619564e694b6b40511057868a9"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/1a2ef66a322ec1619564e694b6b40511057868a9"}]},"branch":"refs/heads/master"},"c8332debebc87129e19f009dda5f780f92575b53":{"kind":"TRIVIAL_REBASE","_number":13,"created":"2024-11-20 12:07:07.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/93/905193/13","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/93/905193/13","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/905193/13 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/905193/13 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/905193/13 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/93/905193/13"}}},"commit":{"parents":[{"commit":"60287ed692f1d37a0279bb52cd262cdd95089204","subject":"Merge \"Honor [neutron] service_type\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/60287ed692f1d37a0279bb52cd262cdd95089204"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2024-01-10 08:30:49.000000000","tz":60},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2024-11-20 11:51:38.000000000","tz":60},"subject":"Add typing information to octavia.amphorae package","message":"Add typing information to octavia.amphorae package\n\n... and fix issues reported by mypy.\n\nApart from the benefit of finding possible bugs using mypy,\ntype annotations also have other benefits like for instance:\n\n- IDEs like PyCharm and VS Code can use annotations to provide\n  better\n  code completion, highlight errors and to make coding more efficient\n- It serves as documentation of the code and helps to understand\n  code\n  better. The annotation syntax is very easy to read.\n- Sphinx generated module documentation can use this information\n  (without the need for additional type info in comments/docstrings)\n\nThe syntax used here should be supported by Python 3.6, so it should\nnot create issues when backporting patches in the future.\n\nPartial-Bug: #2017974\nChange-Id: I5b0f084785f477c855218c54a21caecdecefb62f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c8332debebc87129e19f009dda5f780f92575b53"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/c8332debebc87129e19f009dda5f780f92575b53"}]},"branch":"refs/heads/master"},"5ce452565cbfa463a2e942e947e363094e167451":{"kind":"TRIVIAL_REBASE","_number":14,"created":"2025-02-10 15:47:30.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/93/905193/14","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/93/905193/14","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/905193/14 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/905193/14 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/93/905193/14 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/93/905193/14"}}},"commit":{"parents":[{"commit":"697c89ec234fd8ff42ebb97cf282960cc65683f4","subject":"Merge \"Update gate jobs as per the 2025.1 cycle testing runtime\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/697c89ec234fd8ff42ebb97cf282960cc65683f4"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2024-01-10 08:30:49.000000000","tz":60},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2025-02-10 15:45:07.000000000","tz":60},"subject":"Add typing information to octavia.amphorae package","message":"Add typing information to octavia.amphorae package\n\n... and fix issues reported by mypy.\n\nApart from the benefit of finding possible bugs using mypy,\ntype annotations also have other benefits like for instance:\n\n- IDEs like PyCharm and VS Code can use annotations to provide\n  better\n  code completion, highlight errors and to make coding more efficient\n- It serves as documentation of the code and helps to understand\n  code\n  better. The annotation syntax is very easy to read.\n- Sphinx generated module documentation can use this information\n  (without the need for additional type info in comments/docstrings)\n\nThe syntax used here should be supported by Python 3.6, so it should\nnot create issues when backporting patches in the future.\n\nPartial-Bug: #2017974\nChange-Id: I5b0f084785f477c855218c54a21caecdecefb62f\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5ce452565cbfa463a2e942e947e363094e167451"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5ce452565cbfa463a2e942e947e363094e167451"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Code-Review\u003dMAX","label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
