)]}'
{"id":"openstack%2Foctavia~905474","triplet_id":"openstack%2Foctavia~stable%2F2023.2~Ia681679e24437832e1e23e7399e1a34da8ab54c5","project":"openstack/octavia","branch":"stable/2023.2","attention_set":{},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2024-05-02 18:25:01.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia681679e24437832e1e23e7399e1a34da8ab54c5","subject":"Fix TLS-HELLO healthmonitors in the amphora-driver","status":"MERGED","created":"2024-01-12 15:47:12.000000000","updated":"2024-05-02 18:25:01.000000000","submitted":"2024-05-02 18:25:01.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"905474","meta_rev_id":"3930b5215d0da9c80621e7b3f2b82166c61c9447","_number":905474,"virtual_id_number":905474,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-05-02 18:25:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":2,"date":"2024-05-02 15:51:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-01-23 10:47:40.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":1,"date":"2024-05-02 15:51:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-01-12 18:31:29.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-01-23 10:47:40.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"},{"updated":"2024-05-02 15:51:29.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"e89f6140b111089df50e498cd3de25a524b99110","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2024-01-12 15:47:12.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"8b7040904a1d68434d4fdc0fe5167dd910d3bcc0","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-12 18:31:29.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/771cf9aa679e4d4dbe949411ed1ee05e\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a5d9d4940d2445d8a543ad933f45c8b0 : SUCCESS in 11m 08s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/8eb91769f9fa492d80c8675e57202904 : FAILURE in 6m 18s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d354f40c29b54f128f2de5223d24a0c0 : SUCCESS in 5m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/cc0d9a92a65a45eea42fd29d700713ac : SUCCESS in 9m 21s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/79810d11bee84100b90af156ee9fc0ac : SUCCESS in 6m 05s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/1d76d25ce56f4ff0a42355234bfbc92c : SUCCESS in 7m 38s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/2f13f9aa7cfc4aa3acbddd6172b6843e : SUCCESS in 11m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1e828b92d65441ca8c68a3abc5641beb : SUCCESS in 7m 00s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/d5b51111dc8c4e198103a65e4f1d5965 : SUCCESS in 5m 28s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/d23d37eb47a741f48301fbfa3d2277e4 : SUCCESS in 10m 02s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/a4584b58de834d69985bc9bd549957b1 : SUCCESS in 5m 28s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/0182ba906d69466f9dbd7f3280d81e5e : SUCCESS in 9m 43s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/2512a3ed57a848569830a30a309a617f : SUCCESS in 9m 21s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/b69664b3bab1406f80a59b597f704c4c : SUCCESS in 9m 55s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/0a54198f487d41d385a49c4780f3bca2 : SUCCESS in 1h 24m 13s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/9e8c922d0ba24501b614ff5b89d9a362 : SUCCESS in 1h 49m 38s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/faa07c6021874155a8589dadcd8daa8e : SUCCESS in 53m 57s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/af7e5dfdf8a04d8cab76caeb7107c0f1 : SUCCESS in 1h 18m 36s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/644a331dcf4744b8b87e955f8cb44feb : SUCCESS in 59m 36s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/9d18c243c9ff4a77ad16c626baf473f0 : SUCCESS in 39m 01s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/5373e929919c488f92dd4d07563bf039 : SUCCESS in 39m 56s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/87a1d2ec9ba84afab5a1cf600b569c81 : SUCCESS in 1h 19m 14s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/575947d8daa04390984b76ddc6fe6c1d : SUCCESS in 44m 45s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/af772c6d1c8b4be7840a14b37c9eb2c2 : SUCCESS in 2h 00m 37s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/7faf1a67c5e349528aa5a71e282123ad : SUCCESS in 1h 58m 42s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/60aa40d2c3534ad6987be2f312723c42 : SUCCESS in 1h 43m 29s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/764694aa55384b588307f665343fa56a : SUCCESS in 2h 25m 11s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"e1bd4a580097c4574ad4ce46076053899a7233b2","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2024-01-22 13:24:31.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"f4ef1436fa8c30389c6e3e72678ecf86d52cb9f1","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-22 16:30:12.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/dd3f9f0b8ccf4ffc8a688a42fd9c24f9\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8571994665f844f9b673777d743bb045 : SUCCESS in 12m 36s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/9f39834b95844825813f9cbcc90644be : SUCCESS in 6m 00s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c35edc97a2034b99a822f00c88dc3f4e : SUCCESS in 6m 03s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/1e16b82535454adeb35bdfd8e7fe7906 : SUCCESS in 6m 31s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/a828761788ef4654bec03b666ea58a45 : SUCCESS in 5m 45s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/d4dc7d48d5dc4f50be3b13cf9721bd54 : SUCCESS in 5m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5a1e640e4c084195b023c67961008508 : SUCCESS in 13m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/1e0a3ba4933742c88533b53caec9bb55 : SUCCESS in 4m 56s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/4801a780cec44218ac5e59cbc392093d : SUCCESS in 5m 17s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/59d74469338f4546b702470d292f8c18 : SUCCESS in 10m 04s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/7959417a7b614de1a1350bd29981cda7 : SUCCESS in 5m 01s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/453942bdfa2f411dabfa3a707013a17e : SUCCESS in 8m 45s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/0ac6f717294a4cc891d82f624b4bb09a : SUCCESS in 8m 07s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/a8bfd39d750f421585a0ed633d1175de : SUCCESS in 8m 03s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/c837efffd14d49fdac35c96bac79ffbe : SUCCESS in 1h 18m 39s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/d6a883e5b13b4a4aaa89bc46499786ff : SUCCESS in 1h 42m 24s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/ca3a022b6fd144668192f4d2cc501666 : SUCCESS in 1h 22m 43s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/548befc9712d47e3a02f09df7d43143c : SUCCESS in 1h 44m 49s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/b0112488d2fd4080845a689237a75b00 : SUCCESS in 1h 23m 21s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/07ee931a79a941d7a4fb97919d269ea8 : SUCCESS in 33m 54s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/070c511265574e709fb7022ef4018480 : SUCCESS in 41m 04s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/4bc854eadfb344959dfa266a7f18fcaa : SUCCESS in 1h 18m 02s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/f5483727ceb543e9bba16d6e63d70f1e : SUCCESS in 47m 13s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/556e509292ba49a486fc12614472a7c7 : SUCCESS in 1h 58m 40s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/5a9341f5b22348d58087ca6d8ec1b82f : SUCCESS in 2h 48m 16s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/d1d73ad8ce6d41c1bc14694d1c0b2711 : FAILURE in 37m 03s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/f5f5c2746f89495485c45e5a945b7b7a : SUCCESS in 2h 25m 43s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"ace43ce60340bf7db31cae9a3805db2b90170909","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-01-23 10:47:40.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"6c328450b7afed4afb7b3b867d8d8648ca93d39d","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2024-05-02 15:51:29.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3df7ed2af914bb226dd6c1150ec26db04f0a1705","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-02 15:54:40.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"6fc446c0158639de7020f95b194a6f254accdae2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-02 18:25:01.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/c6fe0704e84f4f2a912a430e54f24f26\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/194ea257a8be40ef926edb96b2e94106 : SUCCESS in 5m 58s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/066b10446b60499895b15f14e20306e7 : SUCCESS in 5m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/a6d4eb83d2b943b996d6a19456db6f67 : SUCCESS in 6m 08s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/b9280ea583094d89a98a7666e8c1520d : SUCCESS in 5m 57s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/ecafab7c75644c86899d2fedf94d7dfa : SUCCESS in 13m 22s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/eeed26a635904c4c9245fde922a08b52 : SUCCESS in 5m 44s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/2c5df71ff1c043bbb6aec12b2dae7e63 : SUCCESS in 3m 58s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/3f96f8b2232840b1a8045e6e176d5ce7 : SUCCESS in 6m 16s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/60a3f0d98b0746b3888b1b3c2341f25e : SUCCESS in 1h 45m 32s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/13ffb6d6de094be69d0ce1164319606d : SUCCESS in 1h 20m 20s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/404066c878ff43db8996f56052060fd2 : SUCCESS in 1h 03m 09s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/5cf83121a54b42d4aa65c9c45908dcdd : SUCCESS in 37m 44s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/2fd78e38853c4bab8630373895d79313 : SUCCESS in 1h 00m 53s","accounts_in_message":[],"_revision_number":1},{"id":"3930b5215d0da9c80621e7b3f2b82166c61c9447","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-02 18:25:01.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"0f9bc3b8c68bd818bc5307186f7e0e78761fcb65","revisions":{"0f9bc3b8c68bd818bc5307186f7e0e78761fcb65":{"kind":"REWORK","_number":1,"created":"2024-01-12 15:47:12.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/74/905474/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/74/905474/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/74/905474/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/74/905474/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/74/905474/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/74/905474/1"}}},"commit":{"parents":[{"commit":"7c468ca4c4d9ea89ba208e8287538ed4d5d1dee3","subject":"Fix health-monitors with ALPN members","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/7c468ca4c4d9ea89ba208e8287538ed4d5d1dee3"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-11-20 18:43:48.000000000","tz":-300},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2024-01-12 15:46:56.000000000","tz":60},"subject":"Fix TLS-HELLO healthmonitors in the amphora-driver","message":"Fix TLS-HELLO healthmonitors in the amphora-driver\n\nTLS-HELLO HMs were based on the ssl-hello-chk option of haproxy, which\nuses SSLv3 messages. SSLv3 is deprecated and most distributions have\ndisabled it, remove this option and rely only on the default checker\nwhen ssl is enabled [0]\n\n[0] https://docs.haproxy.org/2.8/configuration.html#5.2-check\n\nRelated-Bug: #2043812\n\nChange-Id: Ia681679e24437832e1e23e7399e1a34da8ab54c5\n(cherry picked from commit 8e7c2737b6c6c9ca08f9fce0867955c4c3740332)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0f9bc3b8c68bd818bc5307186f7e0e78761fcb65"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/0f9bc3b8c68bd818bc5307186f7e0e78761fcb65"}]},"branch":"refs/heads/stable/2023.2"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
