)]}'
{"id":"openstack%2Foctavia~905475","triplet_id":"openstack%2Foctavia~stable%2F2023.1~I5698558857cbaa585f8a3d7ac37aaa31c0189d46","project":"openstack/octavia","branch":"stable/2023.1","attention_set":{},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2024-05-02 17:49:26.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I5698558857cbaa585f8a3d7ac37aaa31c0189d46","subject":"Fix health-monitors with ALPN members","status":"MERGED","created":"2024-01-12 15:47:53.000000000","updated":"2024-05-02 17:50:40.000000000","submitted":"2024-05-02 17:49:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"905475","meta_rev_id":"f4b51807ce13ead149bf011356b2574780dff566","_number":905475,"virtual_id_number":905475,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-05-02 17:49:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"recommended":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"all":[{"value":1,"date":"2024-01-24 16:51:03.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"value":2,"date":"2024-05-02 15:53:35.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-01-23 10:44:42.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"value":1,"date":"2024-05-02 15:53:35.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-01-12 19:00:11.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-01-23 10:44:42.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"},{"updated":"2024-01-24 16:51:03.000000000","updated_by":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"reviewer":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"state":"REVIEWER"},{"updated":"2024-05-02 15:53:35.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"f905afc01da35de52f1322e9f5d975367781042b","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2024-01-12 15:47:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"00a916887a4d152fd59105ffce6e1e3ce71d9e9f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-12 19:00:11.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/7f2385537cf84eb49956e366ad08949e\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8ce39487af3e4a4ba94086c4c11046db : SUCCESS in 11m 04s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/63bb0f70e0e945d88da4be0fa43f7465 : FAILURE in 4m 28s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/280be3328b3441d1a5257e3c33725d76 : SUCCESS in 4m 30s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/76c47f4fbd724a77ab259781104725aa : SUCCESS in 12m 04s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c85dc8ee35c941e29ca5ee62b3b58ddd : SUCCESS in 6m 01s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/bbcdd99e73664421a474ef01070c4b03 : SUCCESS in 6m 41s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/82a82b83e4304665a8aff39f3d03f321 : SUCCESS in 7m 41s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/909caa1010a547a8a051fa1442911a52 : SUCCESS in 4m 59s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/5a9887f820d9416ebbd27bdfd40c81bf : SUCCESS in 8m 14s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/452449730b194cc79014755d6c5800d4 : SUCCESS in 7m 54s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/191c7e6c978f4670bd8f2eef4e8e31a5 : SUCCESS in 7m 53s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/eb35322bb1d74a4abb48da67505dd396 : SUCCESS in 1h 28m 34s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/7f5f91e9a7244e94b2f2a9eed641dae2 : SUCCESS in 1h 41m 34s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/6ea2e007ca324ea5b6b873e2deb1e88e : SUCCESS in 1h 27m 08s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/e6328c81000142a0a4894871c903b9e6 : SUCCESS in 56m 12s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/9e582cc6058443a3934f78b2bfb19d8b : SUCCESS in 1h 06m 52s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/534973a38594409089a754b9baf442f1 : SUCCESS in 50m 01s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/d5f4f8e9e90b47b79278a938685b7fe7 : SUCCESS in 1h 14m 34s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8df339d5bd0e4394908e7ba9351a4530 : SUCCESS in 39m 37s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/78debe1f5e734b0cbb0ed5863345378f : SUCCESS in 2h 12m 41s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/37c2927e5503422f965640f70828a2a4 : SUCCESS in 2h 32m 07s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/aeb8573281d3496c955d6db9a3bb1f17 : SUCCESS in 1h 47m 08s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/51b0ce8c4c2d44028a3212fb5997b866 : SUCCESS in 2h 15m 52s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/2f7967eee61c4dbba041ec6999785f9f : SUCCESS in 1h 44m 35s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/595a5738b23644868cfb172a7b96f9dd : SUCCESS in 2h 28m 13s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8b6d2e87a7f66df28bbcea8299e39e37454a4e23","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2024-01-22 13:24:15.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"588254173b24ddee9fd7cf2a814b7f1817006a8e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-22 17:13:22.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/603785afb88d45df9bc44f66832d66df\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/6d8f39820233425e8e16cba69dc30fc8 : SUCCESS in 12m 09s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7e5c0413a1bb48a397ab7f1d4b5b818a : SUCCESS in 5m 49s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/58258811d05e44d8997721c9b4e4e04f : SUCCESS in 4m 42s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/7a93d4e14b294214b22ce7c8241c7db5 : SUCCESS in 13m 14s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/47b3d1a641494341972df17d4682fbf3 : SUCCESS in 5m 34s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/506f3bebadd34d74a185742e520bc423 : SUCCESS in 5m 08s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/880383a8b31d4440b2d2fbb97da4b560 : SUCCESS in 8m 24s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/26588351ff5048a8aa1af5929721f0ae : SUCCESS in 5m 10s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/8258df2a255b4b4cae33a0d7657d5b7d : SUCCESS in 7m 39s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/87506f80e0c146df843cb2ea840d35f3 : SUCCESS in 8m 06s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/9b8231ff6c5e4d5780d5283d3c90ea6b : SUCCESS in 8m 51s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/2fb9e57403504e37b7c09f1d2d07215d : SUCCESS in 1h 26m 20s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/f0a549d84daf471187e151fc3c5e743b : SUCCESS in 1h 17m 25s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/01204ab211f2461d89a7d08c0e20a8ab : SUCCESS in 1h 01m 23s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/6b9cdd0cb90242eab9ffe920c664032f : SUCCESS in 36m 41s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/ba9b10a2e764481bb911561c831c157a : SUCCESS in 38m 30s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/7f344f9e51e24149b662c1d5c043cad6 : SUCCESS in 40m 38s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/741dca7c45924798a9c38c45abf9db9a : SUCCESS in 1h 10m 05s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d2d4381b14d64b52a88c11a10e516b81 : SUCCESS in 31m 01s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/5a02a57c3b554e4fac06951a83fa753f : SUCCESS in 1h 51m 49s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/628369e8a97f45318c10c1ef90570d13 : TIMED_OUT in 3h 04m 41s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/85b726b9c1794f9885be5e6bac12311c : FAILURE in 36m 32s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/13e7abae4c8f4782a28a57e4017bd3ee : SUCCESS in 2h 01m 09s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/fdb269df46524acba07adcf00e9c8c4f : SUCCESS in 1h 48m 43s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/f8c4cce4daef410aa99846eef15aee1e : TIMED_OUT in 2h 34m 32s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1cb955bc85cdeb65ca2552802db406ed2fd5b2dd","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-01-23 10:44:42.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"5f6f1272a3379747992082062b1392343fb7ec44","author":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"date":"2024-01-24 16:51:03.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"aa1a4854d4f80086f0180370f4c32e6265d2e0c6","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2024-05-02 15:53:35.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"3e4af2bb1a7f8c22478beb1ed43137985bd7b51c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-02 15:54:11.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"ef34731ba8a37d14bf422471ddb48ad07e28c834","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-02 17:49:26.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/cf68f568d02d4494a45d146baf74027b\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/53866c2b78ca4caf9874cdd75cc4ef0d : SUCCESS in 5m 06s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/37cdf7906937455fb79e051edf43de14 : SUCCESS in 4m 46s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/da5fbcbb67b54dac9cb5870cb6366982 : SUCCESS in 11m 06s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9843a8ce4b3c425480560dd675ec8793 : SUCCESS in 4m 34s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/ac4f17073b7e43a8b8060a4b0226b898 : SUCCESS in 4m 33s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/d0270d4a6191473191387a31438acf0b : SUCCESS in 7m 35s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/29658f8356024687a6f4425804b668b3 : SUCCESS in 1h 50m 11s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/5365e5a748ca401393eb945f1895b4a4 : SUCCESS in 1h 43m 26s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/4f9e36ea379646cd85431c62aa6ea538 : SUCCESS in 1h 19m 10s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/3c03ce780e074c8f99dc57e15ab98171 : SUCCESS in 52m 13s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/64884e75272f412e94e700f975c32fcc : SUCCESS in 54m 17s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/71de9aec299b42c382ce466224561421 : SUCCESS in 1h 00m 52s","accounts_in_message":[],"_revision_number":1},{"id":"04fd65da1ee79d8e9e24584bc5e1f460d3e04da5","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-02 17:49:26.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"f4b51807ce13ead149bf011356b2574780dff566","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-02 17:50:40.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/570ced22b3724ba4b1c2637c2457ca93\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/583fe27b08444c9ab994b73a68ebf590 : SUCCESS in 53s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/8f365f689ed94c2bbfe14ce4d26f723a : SUCCESS in 48s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"121e0533d0cdcf07327ef4894cb32400f5bdde78","revisions":{"121e0533d0cdcf07327ef4894cb32400f5bdde78":{"kind":"REWORK","_number":1,"created":"2024-01-12 15:47:53.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/75/905475/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/75/905475/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/75/905475/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/75/905475/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/75/905475/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/75/905475/1"}}},"commit":{"parents":[{"commit":"ec36152a03bd27a183d189d70effe3f095825e91","subject":"Add Noop Certificate Manager","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/ec36152a03bd27a183d189d70effe3f095825e91"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-11-20 07:16:37.000000000","tz":-300},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2024-01-12 15:47:37.000000000","tz":60},"subject":"Fix health-monitors with ALPN members","message":"Fix health-monitors with ALPN members\n\nUsing HTTP or HTTPS health-monitor on a ALPN pool failed with SSL\nerrors.\nhaproxy doc mentions that when using \"check\" with ALPN servers, the\ncheck-alpn option must be enabled.\n\n[0] https://docs.haproxy.org/2.8/configuration.html#5.2-check\n\nCloses-Bug: #2043812\n\nChange-Id: I5698558857cbaa585f8a3d7ac37aaa31c0189d46\n(cherry picked from commit c28eb84dcafcb066746b04e9e6ef7a51decb68cd)\n(cherry picked from commit 7c468ca4c4d9ea89ba208e8287538ed4d5d1dee3)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/121e0533d0cdcf07327ef4894cb32400f5bdde78"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/121e0533d0cdcf07327ef4894cb32400f5bdde78"}]},"branch":"refs/heads/stable/2023.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
