)]}'
{"id":"openstack%2Foctavia~905476","triplet_id":"openstack%2Foctavia~stable%2F2023.1~Ia681679e24437832e1e23e7399e1a34da8ab54c5","project":"openstack/octavia","branch":"stable/2023.1","attention_set":{},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2024-05-02 18:25:09.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia681679e24437832e1e23e7399e1a34da8ab54c5","subject":"Fix TLS-HELLO healthmonitors in the amphora-driver","status":"MERGED","created":"2024-01-12 15:47:53.000000000","updated":"2024-05-02 18:27:28.000000000","submitted":"2024-05-02 18:25:09.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"905476","meta_rev_id":"a4913f1d6c025215140ffed412ba5f59934e1bc3","_number":905476,"virtual_id_number":905476,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},{"value":0,"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-05-02 18:25:09.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"recommended":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"all":[{"value":1,"date":"2024-04-12 04:58:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},{"value":1,"date":"2024-04-26 07:51:21.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"value":2,"date":"2024-05-02 15:59:29.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-01-23 10:47:45.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},{"value":0,"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"value":1,"date":"2024-05-02 15:59:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},{"value":0,"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-01-12 18:14:19.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-01-23 10:47:45.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"},{"updated":"2024-04-12 04:58:21.000000000","updated_by":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"reviewer":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"state":"REVIEWER"},{"updated":"2024-04-26 07:51:21.000000000","updated_by":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"reviewer":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"state":"REVIEWER"},{"updated":"2024-05-02 15:59:29.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"4857ee02c503ed6cae2675c4ad927cf16e33d2de","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2024-01-12 15:47:53.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"e569fd02639f33af3bb6ef4bd85233fad4f48557","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-12 18:14:19.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/80c8d6e10d794841a84e2be9d34c13b4\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/500ff7c293d444218cc938a9ee464d9f : SUCCESS in 11m 23s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/129ec7e5d69b49878c933ac91ba10972 : FAILURE in 4m 09s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/aed591acbc9d4c4bb98324e5a465e8c7 : SUCCESS in 4m 58s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fc5ea58a9db94021b659dd74a301398c : SUCCESS in 14m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/78faf43d6f374578876d2aafe0e63ee2 : SUCCESS in 5m 31s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/26c6df9b390f46e7a86a83ceee8e610c : SUCCESS in 5m 46s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/7c794f346be14d11bb3b5dcad90bd07e : SUCCESS in 7m 45s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/b02dafb6ac284fda94229646e3f5e117 : SUCCESS in 5m 38s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/16c6b007697246e780151f3ba3a0929d : SUCCESS in 9m 53s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/331d41ab273d4f929470e19c9006875e : SUCCESS in 9m 54s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/2727ec5688874f6c978bc492aea3fb5e : SUCCESS in 7m 11s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/e60137a50a1f4a39a6038425b5a760aa : SUCCESS in 1h 37m 44s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/c76cb616ec634aefbe0e6ada54267e63 : SUCCESS in 1h 47m 18s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/b2041f36c7394dd3802181575c3171fa : SUCCESS in 1h 20m 22s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/42f0c91b005c4673a28cc7014ceb3e8e : SUCCESS in 40m 07s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/031c678d937547d09baee2cc52f98620 : SUCCESS in 39m 56s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/16fd36f9831e448eb69768d1566074d2 : SUCCESS in 57m 33s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/6de402cf833e40c0b97e6213d7832cf5 : SUCCESS in 1h 08m 15s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/8c566c59f29f4c95b54a752b9a2a003f : SUCCESS in 36m 11s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/3024397527fb436abd769e48ee7aa350 : SUCCESS in 1h 59m 42s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/c9783f7259dd4c79ada508583981ba26 : SUCCESS in 1h 59m 54s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/583d711963ba4c4189e448ccb83893f4 : SUCCESS in 2h 09m 44s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/f1afe3e98874484287c4948a7eef23da : SUCCESS in 2h 11m 00s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/4ffec35dae9a45b18074a700314680be : SUCCESS in 1h 49m 49s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/f6627bd47fd848e6bd51198c61ec17a5 : SUCCESS in 1h 54m 07s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"8c46be06bdd746c8435739cc3c7c99e3087d94fe","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2024-01-22 13:24:42.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"0262dda26bedf5574205e67bf35d1fb2f2b8bdec","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-22 16:27:39.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6f3338c60c044cdc8223ebbfbdfbdda9\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/fba80487d9da4b4d8101f42dc0101b05 : SUCCESS in 11m 43s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/f0577d309e5148eaa000bca82cbd6329 : SUCCESS in 5m 22s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/de4d0713ade84873acfeb7b02751a809 : SUCCESS in 6m 03s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a249f78785ab461aa3af18d0469fad13 : SUCCESS in 13m 34s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/45f07b9657364870802902dae942aba2 : SUCCESS in 4m 29s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/3e5e76bc884842e489d359c4f7b1e9fd : SUCCESS in 5m 12s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/3bb48d60011f4518b7a54310f3e89bd3 : SUCCESS in 17m 26s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/27cceaed66c9400698bc41c74e5e97a0 : SUCCESS in 7m 41s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/8a3f0dd910fc47a18f9beed586e9ca6e : SUCCESS in 6m 43s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/a8c3ec1cbfd34996a32c6e0c4fc19661 : SUCCESS in 7m 23s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/cb6255b86da14397b440a3205bdf54b7 : SUCCESS in 7m 24s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/b6d5f3d363d146b099d8c48a50c86830 : SUCCESS in 1h 35m 46s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/e9f490c9d96d47c88a20d7215a4935c5 : SUCCESS in 1h 15m 30s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/52c63dfb925a4c618053ecdbac7f8ce8 : SUCCESS in 1h 22m 53s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/19e4d8b1fb174144957f9272cf817492 : SUCCESS in 49m 37s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/828b84309a1042a7a157ecc5a4613ee2 : SUCCESS in 53m 14s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/3ab087c4ab11478aa21d1a614a6b49cf : SUCCESS in 54m 39s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/5d405e3a0ee34200a319516c07b44276 : SUCCESS in 1h 05m 35s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/1e18c03539714e0ca35833b41b414d98 : SUCCESS in 58m 24s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/c19ee7c7f4a640519b7640367dda7629 : SUCCESS in 2h 26m 14s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/8d2443f31b704afe947348d0b27f4f54 : SUCCESS in 2h 34m 18s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/83988ce97b96426abb19d6ef87e4a703 : FAILURE in 39m 08s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/8a1c36b322c64c6f9d91c4b99ff27b52 : TIMED_OUT in 2h 33m 57s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/28aa13a25417487e91b1c6ea239e1f8f : SUCCESS in 2h 18m 25s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/f375e398b62745be86ea84fff73d10d3 : SUCCESS in 2h 17m 20s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"780f6b776c8ff0a8c88fe0daafd660ada3ef5c5e","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-01-23 10:47:45.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"6abe853a24d755b8230d0eb9bc4b56fcc5fe7ba6","author":{"_account_id":6577,"name":"Sergey Kraynev","email":"sergejyit@gmail.com","username":"skraynev"},"date":"2024-04-12 04:58:21.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"8dec278229eb456d309c7ba698bce23cee12295f","author":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"date":"2024-04-26 07:51:21.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"be12daf45ae4de17bcea1b7786ca42ad38298cbc","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2024-05-02 15:59:29.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6162cec0be926ce9b614d9f8bf8c61b1d27a3a3e","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-02 16:00:14.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"62eb67f94a4517e1e3a8720cf1bbea7b00e56814","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-02 18:25:09.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/616bb6c61de940b7aa8e40f0123d791e\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/81ad1f2ba9f341dc9c278df8679260ec : SUCCESS in 5m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/0d9c84ecb47a4d1691f806c381f72222 : SUCCESS in 4m 47s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/a2f5b6aee0fb4279b6b8467f2aaedcb5 : SUCCESS in 12m 48s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/f2557839224d42848e723eb411aabe4b : SUCCESS in 6m 38s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/379a78d3ca5e42689ed3575dd31831a9 : SUCCESS in 4m 57s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/e70fe7e31c124a06a5c586d8b7b85b13 : SUCCESS in 7m 50s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/1360acd67302452a992f3f8f316d56b7 : SUCCESS in 1h 42m 03s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/a16d7c3207584add848789adedf565b6 : SUCCESS in 1h 50m 12s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/c050f27059d84f51aa0816314cdc6efa : SUCCESS in 1h 25m 58s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/3e83484be5a1436a81cdde56f0df22d7 : SUCCESS in 49m 39s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/552a8886396344fb83526463b7263189 : SUCCESS in 1h 02m 17s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/5b6d2bb025e248208cd1c9637a396676 : SUCCESS in 51m 14s","accounts_in_message":[],"_revision_number":1},{"id":"0f68e0b982b666661b31339103145b2c83d34731","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-02 18:25:09.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"a4913f1d6c025215140ffed412ba5f59934e1bc3","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-05-02 18:27:28.000000000","message":"Patch Set 1:\n\nBuild failed (promote pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/e71a8c9ba81a4e36826ba42dfc59baf9\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e0253ac5b020400ead00adbd1d7572bb : SUCCESS in 1m 00s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/48a04f9a4efb4722962a4191baa6c786 : FAILURE in 39s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"353cafd8d4663357d4ded57a87493d1ad00d2410","revisions":{"353cafd8d4663357d4ded57a87493d1ad00d2410":{"kind":"REWORK","_number":1,"created":"2024-01-12 15:47:53.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/76/905476/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/76/905476/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/76/905476/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/76/905476/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/76/905476/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/76/905476/1"}}},"commit":{"parents":[{"commit":"121e0533d0cdcf07327ef4894cb32400f5bdde78","subject":"Fix health-monitors with ALPN members","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/121e0533d0cdcf07327ef4894cb32400f5bdde78"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-11-20 18:43:48.000000000","tz":-300},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2024-01-12 15:47:42.000000000","tz":60},"subject":"Fix TLS-HELLO healthmonitors in the amphora-driver","message":"Fix TLS-HELLO healthmonitors in the amphora-driver\n\nTLS-HELLO HMs were based on the ssl-hello-chk option of haproxy, which\nuses SSLv3 messages. SSLv3 is deprecated and most distributions have\ndisabled it, remove this option and rely only on the default checker\nwhen ssl is enabled [0]\n\n[0] https://docs.haproxy.org/2.8/configuration.html#5.2-check\n\nRelated-Bug: #2043812\n\nChange-Id: Ia681679e24437832e1e23e7399e1a34da8ab54c5\n(cherry picked from commit 8e7c2737b6c6c9ca08f9fce0867955c4c3740332)\n(cherry picked from commit 0f9bc3b8c68bd818bc5307186f7e0e78761fcb65)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/353cafd8d4663357d4ded57a87493d1ad00d2410"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/353cafd8d4663357d4ded57a87493d1ad00d2410"}]},"branch":"refs/heads/stable/2023.1"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
