)]}'
{"id":"openstack%2Foctavia~905481","triplet_id":"openstack%2Foctavia~stable%2Fzed~Ia681679e24437832e1e23e7399e1a34da8ab54c5","project":"openstack/octavia","branch":"stable/zed","attention_set":{},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2024-04-24 19:11:12.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"Ia681679e24437832e1e23e7399e1a34da8ab54c5","subject":"Fix TLS-HELLO healthmonitors in the amphora-driver","status":"MERGED","created":"2024-01-12 15:52:40.000000000","updated":"2024-04-24 19:13:17.000000000","submitted":"2024-04-24 19:11:12.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":3,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"905481","meta_rev_id":"8022845f487a652f19749526605cc0fd292b9068","_number":905481,"virtual_id_number":905481,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-04-24 19:11:12.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"recommended":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"all":[{"value":1,"date":"2024-01-24 16:54:29.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"value":2,"date":"2024-04-24 16:50:41.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-01-23 10:47:21.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"value":1,"date":"2024-04-24 16:50:41.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-01-12 18:56:56.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-01-23 10:46:25.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"CC"},{"updated":"2024-01-23 10:47:21.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"},{"updated":"2024-01-24 16:54:29.000000000","updated_by":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"reviewer":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"state":"REVIEWER"},{"updated":"2024-04-24 16:50:41.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"958455096a11fcca896f5ca2039a9d0b6d7aa060","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2024-01-12 15:52:40.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"1dec621dcbba78fb7b8be9c9f8e625d5855c40e2","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-12 18:56:56.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/da4c4ff7795b4f2e90646795fb470e16\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/35f3f2b665bc41dc80ae14a089fff28b : SUCCESS in 10m 35s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/1ae6c0b5695a4cc3af912680ef25b990 : FAILURE in 3m 36s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/c67df1fadb814bf09e191bf5bbc9d2c0 : SUCCESS in 5m 19s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/480e759fed9a48dfb58f0b3cb2164218 : SUCCESS in 5m 11s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/5f1c89c3a602464aae15366f83b08c72 : SUCCESS in 5m 17s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/f9d96b92851e4d4c8c4efb488374c443 : SUCCESS in 12m 25s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/d2efbc2596f84ae8824218b12698c999 : SUCCESS in 5m 09s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/ec52fc7fd78e421c89c7a5e1eb71cec1 : SUCCESS in 5m 24s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/22747fffc6e54bd2b2247aa3727e881f : SUCCESS in 7m 44s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/fd5573fcbd84482fbad3387cee702f80 : SUCCESS in 5m 17s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/3e7c84bc2887432eae0af2aec4c8e198 : SUCCESS in 8m 00s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/27510af6c45243ca9d948f5c23cd5526 : SUCCESS in 7m 30s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/0be64aafd8b843eab5eae8d3719032b5 : SUCCESS in 8m 09s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/7dfb49fdc8b546978fd94cd4bc58f2c1 : SUCCESS in 1h 25m 08s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b8986b2123b4460796a2a2bbbb9a7a37 : SUCCESS in 1h 46m 43s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/7de0ac8d0b7c45439fa051dc830055b0 : SUCCESS in 53m 56s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/b636db3fff0c4faaa7c1a14744e0b4a2 : SUCCESS in 43m 00s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/72b941f159df40dd9f5d50232619702d : SUCCESS in 58m 01s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/09b3f6fedf574b9b93d153e06d52330a : SUCCESS in 30m 53s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/1303022513f840738628a1c1ae02fbf8 : SUCCESS in 1h 58m 33s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/cc5883ea913945398e9ceadb7a22c4a3 : SUCCESS in 2h 33m 26s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/05230ba25a9f41c6aa296a6ef69f129a : SUCCESS in 1h 56m 27s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/42b37c94ca3c488784fd566aba3ff123 : SUCCESS in 1h 58m 55s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/57865eb0b7f745c4a791806483bcf99e : SUCCESS in 1h 47m 58s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/0e739d3ffc704cbe89131b36a076403e : SUCCESS in 2h 04m 24s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"3d3be4b698e0de9ef3fe8c5fb99923e4beb19e70","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2024-01-22 13:24:19.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"bed2ea791f2aa29486534e03ec2ac4f35cb0649b","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-22 16:53:42.000000000","message":"Patch Set 1:\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/1ed5ea8e06f84f1aa0b19021a0c95648\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/b439ad518a58434093c4e80de093e925 : SUCCESS in 11m 15s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/237bbb57ca6b4a36a867dc9372011dc8 : SUCCESS in 6m 47s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/70c2ea7f8866426ab02ee3f4e226057a : SUCCESS in 4m 52s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/fbe55a82e0204576922940fc79bcbae0 : SUCCESS in 4m 57s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/eb0d2b42b6e543619bc217bc0b7e45d2 : SUCCESS in 5m 03s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/47125809454a44b88b35d0867aa4cca8 : SUCCESS in 11m 39s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/10ba2eb039734e95acb51afe10684f6e : SUCCESS in 5m 05s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/2e2cb9e7192a48cb9af021dd89a880b4 : SUCCESS in 5m 14s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/1453829b3c5a409fa4eb9d5dfc239337 : SUCCESS in 9m 24s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/0b72be48bef04f1eb1c1b250b29a5474 : SUCCESS in 4m 49s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/9181733edddf40a5af10a7ea3233508d : SUCCESS in 8m 00s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/39102acaa9f54d16b164e216483dc85c : SUCCESS in 10m 12s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/7a19598d66f5484b9fa215e2b4043898 : SUCCESS in 8m 47s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/f18835a44a5947138a2aca1b84e5cc46 : SUCCESS in 2h 18m 09s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/917908f139664bccb680639e7221f3f9 : TIMED_OUT in 2h 34m 10s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/c833307417dc4ed48a22da907b07a66c : SUCCESS in 36m 03s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/13bc356dfc3c42638f109a86aa0a0b0c : SUCCESS in 54m 42s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/76660a8b13e44e928e48068e18b5d59f : SUCCESS in 1h 05m 11s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/ee994a04dc7a4e4180886e0c90de405d : SUCCESS in 44m 34s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/be083568f6dd40f692a2b27f62b61f19 : TIMED_OUT in 2h 32m 53s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/5de26a9f83bd4193a2d5b70cfa79d1e6 : SUCCESS in 2h 16m 15s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/f20b367682404623a914167bd4fdb83c : FAILURE in 32m 24s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/8d61e4dc73ed4f91a1e498c8da9bf50b : TIMED_OUT in 2h 33m 02s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/021274e548274693afbfb66e60ee238a : SUCCESS in 2h 22m 27s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/a5d1e6167bf84dce9bf42f83cf4e5d2c : SUCCESS in 2h 36m 09s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"54084952b3565674db5f58023219e18ed9f04d40","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-01-23 10:46:25.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"53db6e6e2e4514858075d46d8657610f99878f80","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-01-23 10:47:21.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"ee34a620098fff8184a86cbb1a02a6287049a519","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-23 13:28:40.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/f1146e739af34b36832a13e05fad8a71\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/3a625c9a1c214930b3216f2679fa6e13 : SUCCESS in 9m 34s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a5df00d4391849669675df051f3fa098 : SUCCESS in 6m 55s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/dfa81ad4b24c4116bf7557a6967742f8 : SUCCESS in 6m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/d75ab32c8f05445d8cb676ba51870f02 : SUCCESS in 4m 42s\n- openstack-tox-py310 https://zuul.opendev.org/t/openstack/build/c2350cf7d501440f994304f6207c649f : SUCCESS in 4m 44s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/93b216bdbf734c728a4669e57dd9b334 : SUCCESS in 13m 54s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/772ddf95a6a3417cbc24131545f305d9 : SUCCESS in 4m 52s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/c2a412ba2bdc48389a20e818c8d48be4 : SUCCESS in 5m 52s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/d51bafe1f2aa4ad6aa467282cc1cbae6 : SUCCESS in 8m 18s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/ea11957e3ef34a64a853b8f3e4a86ce7 : SUCCESS in 5m 08s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/b709330cc4734cd5883311ca44133606 : SUCCESS in 7m 21s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/16e04d86a1ef49328970f848395c1404 : SUCCESS in 7m 36s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/8c2a4a6099fb4382bcd4f20472536173 : SUCCESS in 9m 36s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/5703d802c9cc44e9bd6de6f18782563d : SUCCESS in 1h 26m 02s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/d71c1f836b1d4370bf5cb0ee3ddb0cbb : SUCCESS in 2h 19m 02s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/81654e08bd9c4bdaa048fbdbd3d4e63a : SUCCESS in 50m 04s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/bda831edc2ee4cb59d42f2545d4f8022 : SUCCESS in 55m 06s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/a44a69e16f594bb6b892e277d820d4a2 : SUCCESS in 1h 16m 49s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/afbae26e606145f6ab554343318ffa57 : SUCCESS in 35m 44s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/6785bdba41534c07962e03d38542c1ff : SUCCESS in 1h 41m 55s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/9212c7871def41babd5a1d008f6750c9 : SUCCESS in 2h 11m 22s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/de67cd41f86d4827850a9852d0dc6c94 : FAILURE in 39m 14s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/a5b85fc591ee48628af7f83af755467e : SUCCESS in 1h 46m 11s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/a261b4257fdf48a1806caddde471f441 : FAILURE in 1h 42m 57s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/b25202212a3b41a191a502b9ce9bf034 : TIMED_OUT in 2h 34m 32s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"2187ab2476352e3673d96cf59700f06c797d0af7","author":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"date":"2024-01-24 16:54:29.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"4b81989fc0e5f7d400e07556cca7b5537abb40d5","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2024-04-24 16:50:41.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"7f5e4c669f3e3199f1e865a6b1f6ed7b48e6064d","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-24 16:51:21.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"1cc19b556cb0c6e2c8188c3e4b2b2adf156d291c","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-24 19:11:12.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/6c996e1548984a48b0c4138dab391aeb\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/3ab6372487124a5cb979edcab02efd17 : SUCCESS in 6m 26s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/d696cf5d6344492da864cf99d5bf9fba : SUCCESS in 5m 01s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/08d43aac2a304deeac7d57008693591d : SUCCESS in 4m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fac013c991f34f7c9124e0f13185a5c8 : SUCCESS in 13m 37s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/cea69bf17efc49b2b2282c0712d525df : SUCCESS in 6m 09s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/adc2b26dfd6b40f9b309a989c01bf7e6 : SUCCESS in 4m 33s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/17b57be60a264613b86721e1dd5bd960 : SUCCESS in 7m 20s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/5f10ff90d6e44ed596d202e075513dd0 : SUCCESS in 1h 46m 43s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a8793ad0abda47bb93959d4778f7a8f4 : SUCCESS in 2h 04m 09s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/c21e3b2a2e914c58ba7efe8fd0f22803 : SUCCESS in 34m 30s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/0cbaa41712864b9dad52a3945f0a70c8 : SUCCESS in 36m 03s","accounts_in_message":[],"_revision_number":1},{"id":"ae0714bf8552a5017dbf018d05382cbd9346493f","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-24 19:11:12.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1},{"id":"8022845f487a652f19749526605cc0fd292b9068","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-24 19:13:17.000000000","message":"Patch Set 1:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/2271286557904e5db48116774de4b54f\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e2e8b23a52034169a4bc96124e8fdf90 : SUCCESS in 44s\n- promote-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/9e5d4364cc1940c88b4c95865d0aef19 : SUCCESS in 36s","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"beb6453413d1ac533283a1837bb093bfa924276f","revisions":{"beb6453413d1ac533283a1837bb093bfa924276f":{"kind":"REWORK","_number":1,"created":"2024-01-12 15:52:40.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/81/905481/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/81/905481/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/81/905481/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/81/905481/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/81/905481/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/81/905481/1"}}},"commit":{"parents":[{"commit":"778859a297299c39d05f69538b50e00c29a7d4d4","subject":"Fix health-monitors with ALPN members","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/778859a297299c39d05f69538b50e00c29a7d4d4"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-11-20 18:43:48.000000000","tz":-300},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2024-01-12 15:52:12.000000000","tz":60},"subject":"Fix TLS-HELLO healthmonitors in the amphora-driver","message":"Fix TLS-HELLO healthmonitors in the amphora-driver\n\nTLS-HELLO HMs were based on the ssl-hello-chk option of haproxy, which\nuses SSLv3 messages. SSLv3 is deprecated and most distributions have\ndisabled it, remove this option and rely only on the default checker\nwhen ssl is enabled [0]\n\n[0] https://docs.haproxy.org/2.8/configuration.html#5.2-check\n\nRelated-Bug: #2043812\n\nChange-Id: Ia681679e24437832e1e23e7399e1a34da8ab54c5\n(cherry picked from commit 8e7c2737b6c6c9ca08f9fce0867955c4c3740332)\n(cherry picked from commit 0f9bc3b8c68bd818bc5307186f7e0e78761fcb65)\n(cherry picked from commit 353cafd8d4663357d4ded57a87493d1ad00d2410)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/beb6453413d1ac533283a1837bb093bfa924276f"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/beb6453413d1ac533283a1837bb093bfa924276f"}]},"branch":"refs/heads/stable/zed"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
