)]}'
{"id":"openstack%2Foctavia~905482","triplet_id":"openstack%2Foctavia~stable%2Fyoga~I5698558857cbaa585f8a3d7ac37aaa31c0189d46","project":"openstack/octavia","branch":"stable/yoga","attention_set":{},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2024-02-01 19:02:40.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I5698558857cbaa585f8a3d7ac37aaa31c0189d46","subject":"Fix health-monitors with ALPN members","status":"MERGED","created":"2024-01-12 15:53:48.000000000","updated":"2024-02-01 19:02:40.000000000","submitted":"2024-02-01 19:02:40.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"905482","meta_rev_id":"37c570c91f11a044a78e9a9edb1d61d8b61f26ec","_number":905482,"virtual_id_number":905482,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-02-01 19:02:40.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"recommended":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"all":[{"value":1,"date":"2024-01-24 16:51:25.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"value":2,"date":"2024-02-01 16:39:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-01-23 10:45:20.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"value":1,"date":"2024-02-01 16:39:26.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-01-12 18:29:00.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-01-23 10:45:20.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"},{"updated":"2024-01-24 16:51:25.000000000","updated_by":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"reviewer":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"state":"REVIEWER"},{"updated":"2024-02-01 16:39:26.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"}],"messages":[{"id":"ef2c05b9e93a2ee081d8e3f23f8266d128902a2c","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2024-01-12 15:53:48.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"b585522143a604230037fabf8b2c59aee35866dc","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-12 18:29:00.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/0499c4488e5743a29dc0d8489c14173e\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/8fa060f4bfb144b7a26232f4662b01b9 : SUCCESS in 10m 56s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/2cf15e9f76ed4aefad1460bace5a4167 : FAILURE in 4m 07s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/7667c546dc1040368f23e30d9e2b5bea : SUCCESS in 5m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/82b9395684244850bed293cf37705b97 : SUCCESS in 5m 33s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/5b0889abd3f3497fa0ffe22444de84b6 : SUCCESS in 12m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/a1b26aabff6d4e8a8486b73a7049bb56 : SUCCESS in 4m 51s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/988f8138bd164607a7b5119bb59e2398 : SUCCESS in 5m 46s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/7e0813e077b148af83560bcc792ff6e8 : SUCCESS in 7m 47s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/13624b92fa4f4a3fafbae68a260eeaf6 : SUCCESS in 8m 11s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/f62304d1e60a47909a48fa3a589a51ea : SUCCESS in 10m 38s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/70f6ca1f01c54c6aaa76166a149db51d : SUCCESS in 1h 15m 41s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/dcb89ce77fe24d4ab79ca74383e5c9ec : SUCCESS in 1h 35m 28s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/b458f9895332447199146185b4df82b8 : SUCCESS in 34m 13s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/9b78691227bf42218b34327481d039eb : SUCCESS in 36m 49s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/47a91a9900484be986a695a958132de2 : FAILURE in 57m 37s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b72a7decf743495ea6ed3ae1d9c6287a : SUCCESS in 46m 34s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/1d6a991b3a6d4bd28ebaa1e03ce2f9bb : TIMED_OUT in 2h 32m 38s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/b49be40582134d95b6842dfe606fa985 : FAILURE in 13m 19s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/ba389ce55f074e34be2e943c3e7950fa : FAILURE in 6m 10s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/b029ca92df084de1a5c4500e47bc795f : SUCCESS in 2h 03m 42s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/43efa18abf7b4818b42587d7fe050d8d : SUCCESS in 1h 41m 38s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/f5053264c36d4b4cb340d054218fdf30 : FAILURE in 41m 41s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1053e520cb43517839720239b8f878789905fb6b","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2024-01-22 13:24:38.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"6b67be3354d80d2c638885b38d55b0fdf7028f7f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-22 15:51:14.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/b0840ffc156046acb87be4ca7c7952f2\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/938de0494bc34c8bbc52dd9558744fac : SUCCESS in 10m 45s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/248ba8c2084c4edbb4f7b4ad0f7103a0 : SUCCESS in 7m 02s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/6b8cc7b76183488bb68ac221e117d7f6 : SUCCESS in 5m 04s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2913235dc8e64d519aec44e8af8a5083 : SUCCESS in 5m 35s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/fbc7b6dab38a4dabbb0ac8f255bd3f09 : SUCCESS in 10m 58s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/c6b406ebfbba49ec8b6a0c0166c5d528 : SUCCESS in 5m 09s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/58daecf6e3704874bee3ba8ac52bb62f : SUCCESS in 5m 18s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/df49c5c654da4dd5af9ebbc4621b2587 : SUCCESS in 6m 38s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/ae09fddb14cc4063858c4e63569d9cb2 : SUCCESS in 5m 29s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/99ea73dadf524644a68afa08542e7cb0 : SUCCESS in 10m 33s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/59c5640b44904087ab4d8349f9a44f71 : SUCCESS in 1h 14m 34s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/b1b6832fe1044113844594d9c0ee9344 : SUCCESS in 1h 45m 55s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/5ce14686275044158453bbe3d7955667 : SUCCESS in 34m 19s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/240b5b45676841b68af2cb135c578120 : SUCCESS in 36m 01s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/e3bfd7cd28f5489babbe2f7a378ebbbf : FAILURE in 58m 15s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7b42c4b12acb417da941278b521a0b48 : SUCCESS in 29m 45s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/0d43f631e8c448b397766fe8784bedaa : SUCCESS in 1h 45m 40s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/ecaefe0b95d34b51a0d55ae456d3589c : FAILURE in 9m 51s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/e86d15a9bd494338b78fd0e12fb176c2 : FAILURE in 5m 47s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/f37c2473999e44ebb689e15819917aed : SUCCESS in 2h 07m 50s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/c92d0de2025d452cacbad5bcd6e08e0a : SUCCESS in 2h 01m 11s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8-stream https://zuul.opendev.org/t/openstack/build/0cdfd662ed7c47da9082131e07c196ba : FAILURE in 42m 16s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"128392631edbd6b959e129ed83185fe0c69a9080","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-01-23 10:45:20.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"83b4aac6e701ffc4e0d2b667d55167bded33431f","author":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"date":"2024-01-24 16:51:25.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"304595753dad15ef790dfb83a6d6baaae3e44ed1","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2024-02-01 16:39:26.000000000","message":"Patch Set 1: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"9ed0d440392f751cd7b6f17dfe27e86dccd0ca16","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-01 16:40:14.000000000","message":"Patch Set 1: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":1},{"id":"b337b83e0cc2a8ff01595c5920320b7101b611a6","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-01 19:02:40.000000000","message":"Patch Set 1: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/49ba024d2e0d43b39f755255256c3f90\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/a5074c1e6d2d4fdca7e53817d8f49637 : SUCCESS in 6m 53s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/a3cbca8f528b4f56902efa13d0d858b1 : SUCCESS in 8m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/e48a3f07e3764453a8340d22e2fee453 : SUCCESS in 6m 11s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/735dc76fad42472899c5259e9cf758d6 : SUCCESS in 11m 33s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/51613f626671421daca7de555d913176 : SUCCESS in 5m 51s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/0247e778368143b0ac787cee3e988512 : SUCCESS in 5m 05s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c5d5a90323fc4dbe95c3bed0d0038605 : SUCCESS in 9m 32s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/13a9a9323b2144108e63508b50d00cc3 : SUCCESS in 1h 33m 41s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7c3204f83ea84ceba9d400c923f4c556 : SUCCESS in 1h 27m 52s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/31eb401b88144e2f877d71d80797ad9d : SUCCESS in 36m 42s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/765162586027405e8c6f4694a9319fc5 : SUCCESS in 37m 52s","accounts_in_message":[],"_revision_number":1},{"id":"37c570c91f11a044a78e9a9edb1d61d8b61f26ec","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-02-01 19:02:40.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"937bb928aa61e51ce38743921a1434770f4d844b","revisions":{"937bb928aa61e51ce38743921a1434770f4d844b":{"kind":"REWORK","_number":1,"created":"2024-01-12 15:53:48.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/82/905482/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/82/905482/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/82/905482/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/82/905482/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/82/905482/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/82/905482/1"}}},"commit":{"parents":[{"commit":"5b388dbf3809d2d35234767bc781ec7bd7e05673","subject":"Add Noop Certificate Manager","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5b388dbf3809d2d35234767bc781ec7bd7e05673"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-11-20 07:16:37.000000000","tz":-300},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2024-01-12 15:53:17.000000000","tz":60},"subject":"Fix health-monitors with ALPN members","message":"Fix health-monitors with ALPN members\n\nUsing HTTP or HTTPS health-monitor on a ALPN pool failed with SSL\nerrors.\nhaproxy doc mentions that when using \"check\" with ALPN servers, the\ncheck-alpn option must be enabled.\n\n[0] https://docs.haproxy.org/2.8/configuration.html#5.2-check\n\nCloses-Bug: #2043812\n\nChange-Id: I5698558857cbaa585f8a3d7ac37aaa31c0189d46\n(cherry picked from commit c28eb84dcafcb066746b04e9e6ef7a51decb68cd)\n(cherry picked from commit 7c468ca4c4d9ea89ba208e8287538ed4d5d1dee3)\n(cherry picked from commit 121e0533d0cdcf07327ef4894cb32400f5bdde78)\n(cherry picked from commit 778859a297299c39d05f69538b50e00c29a7d4d4)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/937bb928aa61e51ce38743921a1434770f4d844b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/937bb928aa61e51ce38743921a1434770f4d844b"}]},"branch":"refs/heads/stable/yoga"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
