)]}'
{"id":"openstack%2Foctavia~905485","triplet_id":"openstack%2Foctavia~stable%2Fxena~Ia681679e24437832e1e23e7399e1a34da8ab54c5","project":"openstack/octavia","branch":"stable/xena","attention_set":{},"removed_from_attention_set":{"29244":{"account":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"last_update":"2024-03-05 19:14:45.000000000","reason":"Change was abandoned"}},"hashtags":[],"change_id":"Ia681679e24437832e1e23e7399e1a34da8ab54c5","subject":"Fix TLS-HELLO healthmonitors in the amphora-driver","status":"ABANDONED","created":"2024-01-12 15:54:18.000000000","updated":"2024-03-05 19:14:45.000000000","total_comment_count":1,"unresolved_comment_count":0,"has_review_started":true,"meta_rev_id":"4e96b0f8ad3afc121ae9a1188c8d3bcfa1e6dba5","_number":905485,"virtual_id_number":905485,"owner":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"actions":{},"labels":{"Verified":{"recommended":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"tag":"autogenerated:zuul:check","value":1,"date":"2024-01-22 16:23:49.000000000","permitted_voting_range":{"min":-2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","value":1,"default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"all":[{"value":1,"date":"2024-01-24 16:52:04.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-01-23 10:47:52.000000000","permitted_voting_range":{"min":-1,"max":1},"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"all":[{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"permitted_voting_range":{"min":-1,"max":1},"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}],"CC":[{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-01-12 18:37:43.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-01-23 10:47:52.000000000","updated_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"reviewer":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"state":"REVIEWER"},{"updated":"2024-01-24 16:52:04.000000000","updated_by":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"reviewer":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"state":"REVIEWER"},{"updated":"2024-03-05 19:14:45.000000000","updated_by":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"reviewer":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"state":"CC"}],"messages":[{"id":"815ccefe04324c360c409b278e88b6a11f41c83d","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2024-01-12 15:54:18.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"f8875a4634e2c91d332cf305d9f3998b89906c6f","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-12 18:37:43.000000000","message":"Patch Set 1: Verified-1\n\nBuild failed (check pipeline).  For information on how to proceed, see\nhttps://docs.opendev.org/opendev/infra-manual/latest/developers.html#automated-testing\n\nhttps://zuul.opendev.org/t/openstack/buildset/547d9272fd3c4cb08a9ebd47742c23e5\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/efab2db2d2da483e97743b28c0380162 : SUCCESS in 10m 12s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/7cc2ebab096841be8fb351ac02a1cf31 : SUCCESS in 4m 54s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/725fd09843e94caf9063bea97646ce9a : SUCCESS in 4m 38s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1af7d40c3c7a4845a641bcb3164b1b28 : SUCCESS in 5m 10s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/2fedc6b4f6c94400a875caab9c70f5fe : SUCCESS in 5m 45s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9807e6537c604fc9a551e83537065a94 : SUCCESS in 13m 45s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/e01d354914e94ef3b1acd460b8ab2154 : SUCCESS in 5m 16s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/c136969d1416461abe191be800b83bb7 : SUCCESS in 6m 01s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/542d6486438345f6a2b5d695ab76befd : SUCCESS in 7m 30s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/51cb3417f23c43dd9d1cfacf07b4e9a5 : SUCCESS in 5m 14s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/c444db52731e47588a675c506f00c09e : SUCCESS in 7m 16s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/333d9719634a4a05bf61c72cf95cf935 : SUCCESS in 1h 47m 02s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/0df0234619cf4f87921a644906380db0 : TIMED_OUT in 2h 33m 42s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/b232a847337c4920a7e6955dfba9bb69 : SUCCESS in 55m 03s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/f8f11887b8df44968fe89ed2f93974f6 : SUCCESS in 38m 58s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/1b0dbf81778b4770b1f9730d630ada87 : FAILURE in 10m 33s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/322a281fe82340d8b762a7a122b34bdf : SUCCESS in 43m 49s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/31757694276d4e3f91544d4058772f76 : SUCCESS in 1h 34m 52s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/e6429b0532e6469680905ee36f41d227 : FAILURE in 15m 52s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/c3f085eee8cd4690889115fc4be660f8 : FAILURE in 6m 35s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/594f67c5e394452aa3b14853df86a0b0 : TIMED_OUT in 2h 32m 28s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/9cd294452404487a962cea374ac3113b : SUCCESS in 1h 53m 16s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/2fdaabf3d2b342ec8d421080719b21e2 : NODE_FAILURE Node request 300-0023292587 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"733d1711e19fd6375523762e51a9f0f90961b636","author":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"date":"2024-01-22 13:24:27.000000000","message":"Patch Set 1:\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ec1270af92348ef3bdbeb2de4a5ca96521c2db53","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-22 16:23:49.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/e05a5dfaed53430b9af95a96717f3fa2\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/03490d557932421cb54d8a17e4147737 : SUCCESS in 11m 05s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/6046ead4978c47e7a21f5b689812943c : SUCCESS in 5m 45s\n- openstack-tox-py36 https://zuul.opendev.org/t/openstack/build/b179439ae0b8484b87beb25c25edbd54 : SUCCESS in 5m 07s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/1ca181ace52c438281b70613488ac453 : SUCCESS in 5m 27s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/db2b707cbeef4f15844a641445959d7b : SUCCESS in 5m 05s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/0f44f542680b476998e903fb601124df : SUCCESS in 12m 35s\n- build-openstack-releasenotes https://zuul.opendev.org/t/openstack/build/fe370f19976241879bf145e6d865d3b9 : SUCCESS in 5m 24s\n- octavia-tox-py37-tips https://zuul.opendev.org/t/openstack/build/3a7c1b76c5204932a65b9675aefec18e : SUCCESS in 5m 11s\n- octavia-tox-functional-py37-tips https://zuul.opendev.org/t/openstack/build/2ad50dc59cb84882b62bdf9c4bfe5277 : SUCCESS in 8m 48s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/de3a948e5aa74293897f24984d8201b0 : SUCCESS in 6m 39s\n- openstack-tox-functional-py36 https://zuul.opendev.org/t/openstack/build/7b2ed27b087d45bd8b3d385523551a4a : SUCCESS in 9m 55s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/cddf966ed52c49fb8160a3038450e985 : SUCCESS in 1h 29m 10s\n- octavia-v2-dsvm-scenario https://zuul.opendev.org/t/openstack/build/63c68efe1aba452d8e5722224e50e172 : SUCCESS in 1h 54m 00s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/de8de695ef584a1a98d27fad9f922cd7 : SUCCESS in 32m 54s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/990f5d3f527d4301829172e7cf2f97f2 : SUCCESS in 39m 41s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/82399189532347b6a30f912fb3e78d97 : FAILURE in 8m 43s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/948bef9376d24f3f854d714c63b5c30a : SUCCESS in 30m 31s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/47ddc48a5e894f80893ffe4496860cc0 : SUCCESS in 1h 46m 47s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/5f11dc4d35ea449384baac0fdf370deb : FAILURE in 10m 11s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/a235eebeb76c49b192cf44396a74c2dd : FAILURE in 5m 26s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2 https://zuul.opendev.org/t/openstack/build/fa43db5ef5e84ff79ada2ca4618f5f3c : SUCCESS in 2h 28m 40s (non-voting)\n- octavia-v2-dsvm-scenario-amphora-v2-no-jobboard https://zuul.opendev.org/t/openstack/build/5c98a12bd6dd4857b89f24497298c680 : SUCCESS in 1h 48m 12s (non-voting)\n- octavia-v2-dsvm-scenario-centos-8 https://zuul.opendev.org/t/openstack/build/ee81e28711c146a8ba7a60612154403d : NODE_FAILURE Node request 300-0023372926 failed in 0s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"1e5669c59f2f0531f4af8c52d187c24a2ec4e264","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-01-23 10:47:52.000000000","message":"Patch Set 1: Code-Review+2","accounts_in_message":[],"_revision_number":1},{"id":"be4457efc45a2f2b8de6d5c029cfb30230ce7d38","author":{"_account_id":27442,"name":"Quentin GROLLEAU","email":"quentin.grolleau@corp.ovh.com","username":"QG"},"date":"2024-01-24 16:52:04.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"4e96b0f8ad3afc121ae9a1188c8d3bcfa1e6dba5","tag":"autogenerated:gerrit:abandon","author":{"_account_id":17685,"name":"Elod Illes","email":"elod.illes@est.tech","username":"elod.illes"},"date":"2024-03-05 19:14:45.000000000","message":"Abandoned\n\nstable/xena branch of openstack/octavia is about to be deleted. To be able to do that, all open patches need to be abandoned. Please cherry pick the patch to unmaintained/xena if you want to further work on this patch.","accounts_in_message":[],"_revision_number":1}],"current_revision_number":1,"current_revision":"d0bb5a80729f9bff59dacf294d5ebcdc6d4f68e1","revisions":{"d0bb5a80729f9bff59dacf294d5ebcdc6d4f68e1":{"kind":"REWORK","_number":1,"created":"2024-01-12 15:54:18.000000000","uploader":{"_account_id":29244,"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","username":"gthiemonge"},"ref":"refs/changes/85/905485/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/85/905485/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/85/905485/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/85/905485/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/85/905485/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/85/905485/1"}}},"commit":{"parents":[{"commit":"b9f57923c81aba0cb187ac21c363d8405f95fad0","subject":"Fix health-monitors with ALPN members","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/b9f57923c81aba0cb187ac21c363d8405f95fad0"}]}],"author":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2023-11-20 18:43:48.000000000","tz":-300},"committer":{"name":"Gregory Thiemonge","email":"gthiemon@redhat.com","date":"2024-01-12 15:54:08.000000000","tz":60},"subject":"Fix TLS-HELLO healthmonitors in the amphora-driver","message":"Fix TLS-HELLO healthmonitors in the amphora-driver\n\nTLS-HELLO HMs were based on the ssl-hello-chk option of haproxy, which\nuses SSLv3 messages. SSLv3 is deprecated and most distributions have\ndisabled it, remove this option and rely only on the default checker\nwhen ssl is enabled [0]\n\n[0] https://docs.haproxy.org/2.8/configuration.html#5.2-check\n\nRelated-Bug: #2043812\n\nChange-Id: Ia681679e24437832e1e23e7399e1a34da8ab54c5\n(cherry picked from commit 8e7c2737b6c6c9ca08f9fce0867955c4c3740332)\n(cherry picked from commit 0f9bc3b8c68bd818bc5307186f7e0e78761fcb65)\n(cherry picked from commit 353cafd8d4663357d4ded57a87493d1ad00d2410)\n(cherry picked from commit beb6453413d1ac533283a1837bb093bfa924276f)\n(cherry picked from commit 714ef1c802d7d9ad16eb0ea726f6deb7575c6f44)\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d0bb5a80729f9bff59dacf294d5ebcdc6d4f68e1"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/d0bb5a80729f9bff59dacf294d5ebcdc6d4f68e1"}]},"branch":"refs/heads/stable/xena"}},"requirements":[],"submit_records":[],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Verified\u003dMAX","label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"UNSATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":false,"status":"FAIL","passing_atoms":[],"failing_atoms":["label:Workflow\u003dMAX","label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
