)]}'
{"id":"openstack%2Foctavia~905591","triplet_id":"openstack%2Foctavia~master~I8560b5440c0c2338e468b12d8c4abcc47a8fb50a","project":"openstack/octavia","branch":"master","topic":"pyupgrade","attention_set":{},"removed_from_attention_set":{"32761":{"account":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"last_update":"2024-04-12 18:03:26.000000000","reason":"Change was submitted"},"11628":{"account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"last_update":"2024-04-10 17:36:14.000000000","reason":"\u003cGERRIT_ACCOUNT_11628\u003e replied on the change","reason_account":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"}},"34429":{"account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"last_update":"2024-04-12 16:09:54.000000000","reason":"\u003cGERRIT_ACCOUNT_34429\u003e replied on the change","reason_account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},"32238":{"account":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"last_update":"2024-04-12 18:03:26.000000000","reason":"Change was submitted"}},"hashtags":[],"change_id":"I8560b5440c0c2338e468b12d8c4abcc47a8fb50a","subject":"pyupgrade changes for Python3.8+ (1)","status":"MERGED","created":"2024-01-15 12:34:37.000000000","updated":"2024-04-12 18:04:38.000000000","submitted":"2024-04-12 18:03:26.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":9,"unresolved_comment_count":5,"has_review_started":true,"submission_id":"905591-pyupgrade","meta_rev_id":"d1db1dacd462194642723e294529598759ce51c7","_number":905591,"virtual_id_number":905591,"owner":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-04-12 18:03:26.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"all":[{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":2,"date":"2024-04-10 17:36:14.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-04-12 16:09:54.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"all":[{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":1,"date":"2024-04-12 16:09:54.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-01-15 15:21:16.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-02-23 00:48:44.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2024-02-23 07:51:42.000000000","updated_by":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"reviewer":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"state":"REVIEWER"},{"updated":"2024-02-27 07:34:54.000000000","updated_by":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"reviewer":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"state":"REVIEWER"}],"messages":[{"id":"c9a29c6eaefb6f843a041749454c109a92163cdc","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-01-15 12:34:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"c5bc905080c7fcaf9549ff1036eee64252c46700","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-15 15:21:16.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/d5a6c3bb617443c698cdac74cb0bba5f\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/24882825d9ec44eea020e8e5b47a2643 : SUCCESS in 10m 24s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/261b09e2b5334708b7fd8f32e651f423 : SUCCESS in 6m 15s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/8939e0fe2c07437888dca474b60303ba : SUCCESS in 4m 53s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/a3b37d9e3ca241f8bc6c815e4a1f419d : SUCCESS in 4m 50s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/78aa6d877f7542099b086c1c77911330 : SUCCESS in 14m 24s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/01eb0c0cde8b458ba8beb2e4e06efb38 : SUCCESS in 7m 59s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/ff8f75ff6f3249d094412f3d62d12be1 : SUCCESS in 7m 58s\n- octavia-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/5cdd1a584ba849a993a877d5d4c467b4 : SUCCESS in 9m 39s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/80c7065b1ddd413b9988f485f9b65317 : SUCCESS in 5m 10s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/1f59ae0795c648bcb9e26034df658f91 : SUCCESS in 7m 58s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/4a44fb9f0c3540a788bb7aad6a1ecb75 : SUCCESS in 8m 23s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/544745a10f454a4285ee41b21cb0ca40 : SUCCESS in 9m 00s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/4026ca4f116b493eba2994dba63c718d : SUCCESS in 1h 20m 13s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/70d64df5f58948e396b907f829ded944 : SUCCESS in 1h 55m 42s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/bb888975965a4da08c01cd4606f777d4 : SUCCESS in 1h 23m 07s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/97ec8b768865434cbd37f6a7b73ed47a : SUCCESS in 1h 43m 53s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/a08a4d7e70604c8f8435d7a152a49960 : SUCCESS in 1h 19m 45s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/4e18259ef1a1435e9587570953469d62 : SUCCESS in 55m 46s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/acb5d5d705324f159bc6e0975fbb85d7 : SUCCESS in 43m 30s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/1451ec1af63249598688d1a43e234985 : SUCCESS in 42m 57s (non-voting)\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/e1b61513b53a4f4c82cf0067fb991210 : SUCCESS in 1h 04m 41s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/a044c50fe8174e35851e6a394891ccad : SUCCESS in 33m 19s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/0f2729436621466eb8c996cdeadcc6b4 : SUCCESS in 2h 20m 18s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/4b9d711bd2e0415b9f77a5c62c518ee1 : SUCCESS in 2h 21m 08s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/9f952697c1284fd9a13a4770ccd9b182 : SUCCESS in 2h 12m 56s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/060fd276a3fa49759f0a2090ec378729 : TIMED_OUT in 2h 34m 53s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"137008b948e1ea20ca69f7632d52485dc67e1536","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2024-02-23 00:48:44.000000000","message":"Patch Set 1: Code-Review+2\n\n(6 comments)","accounts_in_message":[],"_revision_number":1},{"id":"11abd1bdb133028528e7b600656c29ef4f4b6166","author":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"date":"2024-02-23 07:51:42.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"3113c7ce5a92736f5eaa792b4a48c41e412b1e91","author":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"date":"2024-02-27 07:34:54.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"c8fc14bdef389d759c0d60bc4c37ab062caab2ae","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-04-10 08:36:01.000000000","message":"Uploaded patch set 2.\n\nOutdated Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"changekind:TRIVIAL_REBASE OR is:MIN\")\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"4f8841137361079b931e0d471dc39150247a2df0","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-04-10 10:19:47.000000000","message":"Patch Set 2:\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"5758e2144a0e00d7bcaabcd1e0c618121b7fa612","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-10 11:23:13.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/38fcded5e9b147fba45d626b16e3eec2\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/f15f8c91b0c741f79899e78f25828459 : SUCCESS in 11m 33s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/03433a631e5e4ea0a84bc130c7db7f07 : SUCCESS in 4m 41s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/6af6899e6f254b329c8ce42e4a924d9d : SUCCESS in 6m 00s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/f1d9c5ac37af41b4bf3587395b92e014 : SUCCESS in 4m 33s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/bf0c62e754094246b493f54933115af2 : SUCCESS in 11m 57s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/02b7d2056f2d47eebb72dff76f08e284 : SUCCESS in 10m 32s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/87539af046c0455b9b6d3d43f9453b9d : SUCCESS in 4m 53s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/21e5f4f2f959479b9e6b8119540a0334 : SUCCESS in 7m 56s\n- octavia-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/86e786284da746adb405573e7b40a9ff : SUCCESS in 10m 21s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/89281a2ed9484f498a720be404ef8884 : SUCCESS in 4m 42s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/d5276782fe0d4394953a6ca4464538d5 : SUCCESS in 6m 54s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/6530d1e9728a48b0a0268071fb3682c8 : SUCCESS in 8m 25s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/1286a21fbe6c4bbe89bde556e391644c : SUCCESS in 7m 08s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/7b01fc9d22494d0d9893ddeee6c9c801 : SUCCESS in 1h 42m 01s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/225786c693a44eb3b6c1d49180bbe65b : SUCCESS in 1h 44m 53s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/e8b1a1d110dc478e85316581dbdbfd8b : SUCCESS in 1h 03m 27s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/b118d050230046bcb5c7ec2773ba68cb : SUCCESS in 1h 48m 34s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/87a186d080604b33bdb5126d78649094 : SUCCESS in 1h 33m 23s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/a3f3e694be494557a4b7361c9a748048 : SUCCESS in 43m 36s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/37d44b47602d46e1b62d2f9a7233ed1c : SUCCESS in 1h 02m 44s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/25e5401d48e842d39bb7a196e221c212 : SUCCESS in 1h 14m 37s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/5b4c6f48d7424093a7bf527933ddc92c : SUCCESS in 50m 10s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/498126226e264f78879033a5c376c764 : SUCCESS in 2h 25m 32s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/5b4721a570be401482e2a756b8d9a150 : SUCCESS in 2h 44m 26s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/6622b9c8d1704e5d9ba917da6e2b6745 : FAILURE in 34m 28s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/50a592309f294fb4868d8750850b8bd6 : SUCCESS in 2h 28m 58s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"d0f8dc2c38a30fa50e80e9827408dec94d1e2715","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2024-04-10 17:36:14.000000000","message":"Patch Set 2: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"2bb174d26a932006b1b64c81e20e0a4caf0bf395","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-04-12 16:09:54.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"d31048f0f7c6ea548aaca0d896245fcce5453357","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-12 16:10:39.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"c7c3e5fe0e80eaa5a49b2a72c9bfc5a3bcd35379","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-12 18:03:26.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/14bc61da55a24f3c935853dff3fea236\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/89fec57a1a124e3daafea9d0e26d636f : SUCCESS in 5m 30s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/57751f7587d24134b96ed3cca9acdbab : SUCCESS in 5m 20s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/49a1119519424ffd9a9951bbc723d8e5 : SUCCESS in 5m 09s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/9528114ba03940b6beca33e49e6758ae : SUCCESS in 12m 01s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/ef7abf3b4c164a46b6f34e861a7de88c : SUCCESS in 5m 01s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/7282516ba7564c99908470195f2cbb56 : SUCCESS in 8m 44s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/5784914be20a481fa09250b3c0df3e2b : SUCCESS in 1h 41m 36s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/19455e4949f34dc3be34bc43531745a3 : SUCCESS in 1h 40m 06s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/92d816057024492693218d95f4252c2c : SUCCESS in 59m 55s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/057b1cf4c20045d9aa8ca56608886c06 : SUCCESS in 38m 28s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/9fad6c19c2f94e819ef4279fdcd2956a : SUCCESS in 42m 30s","accounts_in_message":[],"_revision_number":2},{"id":"8e608e98e905aa0f0ff00b0d9a37636a4dd366aa","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-12 18:03:26.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"d1db1dacd462194642723e294529598759ce51c7","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-12 18:04:38.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/8d1743a327064b5cb0e20b61a7ce443b\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/c373ba51b60b4bcc91590e3275503ac7 : SUCCESS in 53s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"69552a2082386d0b08da0f050ce05d4106214df8","revisions":{"700c003d686aef522815f2faf37b4aa96aeae22c":{"kind":"REWORK","_number":1,"created":"2024-01-15 12:34:37.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/91/905591/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/91/905591/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/91/905591/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/91/905591/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/91/905591/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/91/905591/1"}}},"commit":{"parents":[{"commit":"5750e4512d622450f6ecab1c7001ebd5637e0d53","subject":"Merge \"Provide Amphora stats for Octavia no-op drivers\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/5750e4512d622450f6ecab1c7001ebd5637e0d53"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2024-01-15 11:52:30.000000000","tz":60},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2024-01-15 11:52:30.000000000","tz":60},"subject":"pyupgrade changes for Python3.8+ (1)","message":"pyupgrade changes for Python3.8+ (1)\n\nResult of running\n\n$ pyupgrade --py38-plus $(git ls-files | grep \".py$\")\n\nThis was inspired by Nova [1]\n\nFixed PEP8 errors introduced by pyupgrade by running:\n\n$ autopep8 --select\u003dE127,E128,E501 --max-line-length 79 -r \\\n  --in-place octavia\n\nand manual updates.\n\n[1]: https://review.opendev.org/c/openstack/nova/+/896986\n\nChange-Id: I8560b5440c0c2338e468b12d8c4abcc47a8fb50a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/700c003d686aef522815f2faf37b4aa96aeae22c"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/700c003d686aef522815f2faf37b4aa96aeae22c"}]},"branch":"refs/heads/master"},"69552a2082386d0b08da0f050ce05d4106214df8":{"kind":"REWORK","_number":2,"created":"2024-04-10 08:36:01.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/91/905591/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/91/905591/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/91/905591/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/91/905591/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/91/905591/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/91/905591/2"}}},"commit":{"parents":[{"commit":"fa7cc0178783ad4ab9f3ddce893042d5952d0bd8","subject":"Merge \"Fix fully-populated API with allowed_cidrs\"","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/fa7cc0178783ad4ab9f3ddce893042d5952d0bd8"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2024-01-15 11:52:30.000000000","tz":60},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2024-04-10 08:23:44.000000000","tz":120},"subject":"pyupgrade changes for Python3.8+ (1)","message":"pyupgrade changes for Python3.8+ (1)\n\nResult of running\n\n$ pyupgrade --py38-plus $(git ls-files | grep \".py$\")\n\nThis was inspired by Nova [1]\n\nFixed PEP8 errors introduced by pyupgrade by running:\n\n$ autopep8 --select\u003dE127,E128,E501 --max-line-length 79 -r \\\n  --in-place octavia\n\nand manual updates.\n\n[1]: https://review.opendev.org/c/openstack/nova/+/896986\n\nChange-Id: I8560b5440c0c2338e468b12d8c4abcc47a8fb50a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/69552a2082386d0b08da0f050ce05d4106214df8"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/69552a2082386d0b08da0f050ce05d4106214df8"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
