)]}'
{"id":"openstack%2Foctavia~905592","triplet_id":"openstack%2Foctavia~master~I4ad020fbedd8106a3a86768c25a0c4dc750ec88a","project":"openstack/octavia","branch":"master","topic":"pyupgrade","attention_set":{},"removed_from_attention_set":{"34429":{"account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"last_update":"2024-04-12 16:13:40.000000000","reason":"\u003cGERRIT_ACCOUNT_34429\u003e replied on the change","reason_account":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}}},"hashtags":[],"change_id":"I4ad020fbedd8106a3a86768c25a0c4dc750ec88a","subject":"pyupgrade changes for Python3.8+ (2)","status":"MERGED","created":"2024-01-15 12:34:37.000000000","updated":"2024-04-12 18:05:31.000000000","submitted":"2024-04-12 18:03:28.000000000","submitter":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"total_comment_count":2,"unresolved_comment_count":0,"has_review_started":true,"submission_id":"905592-pyupgrade","meta_rev_id":"10c956b4602f4056a4c52abb514c96abff31d8ca","_number":905592,"virtual_id_number":905592,"owner":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"actions":{},"labels":{"Verified":{"approved":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"all":[{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"tag":"autogenerated:zuul:gate","value":2,"date":"2024-04-12 18:03:28.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-2":"Fails","-1":"Doesn\u0027t seem to work"," 0":"No score","+1":"Works for me","+2":"Verified"},"description":"","default_value":0,"optional":true},"Code-Review":{"approved":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"recommended":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"all":[{"value":1,"date":"2024-04-10 08:36:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":2,"date":"2024-04-10 08:36:01.000000000","permitted_voting_range":{"min":2,"max":2},"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":2,"date":"2024-04-12 16:13:40.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},{"value":1,"date":"2024-04-10 08:36:01.000000000","permitted_voting_range":{"min":1,"max":1},"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-2":"Do not merge","-1":"This patch needs further work before it can be merged"," 0":"No score","+1":"Looks good to me, but someone else must approve","+2":"Looks good to me (core reviewer)"},"description":"","value":1,"default_value":0,"optional":true},"Workflow":{"approved":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"all":[{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":1,"date":"2024-04-12 16:13:40.000000000","_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-1":"Work in progress"," 0":"Ready for reviews","+1":"Approved"},"description":"","default_value":0,"optional":true},"Backport-Candidate":{"all":[{"value":0,"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"value":0,"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"value":0,"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"value":0,"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},{"value":0,"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"}],"values":{"-2":"Do Not Backport","-1":"Not A Backport Candidate"," 0":"Backport Review Needed","+1":"Proposed Backport","+2":"Should Backport"},"description":"","default_value":0,"optional":true}},"removable_reviewers":[],"reviewers":{"REVIEWER":[{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}]},"pending_reviewers":{},"reviewer_updates":[{"updated":"2024-01-15 15:42:38.000000000","updated_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"reviewer":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"state":"REVIEWER"},{"updated":"2024-02-23 00:52:20.000000000","updated_by":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"reviewer":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"state":"REVIEWER"},{"updated":"2024-02-23 07:50:59.000000000","updated_by":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"reviewer":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"state":"REVIEWER"},{"updated":"2024-02-27 07:34:58.000000000","updated_by":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"reviewer":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"state":"REVIEWER"}],"messages":[{"id":"326a1ed283ab9d40385c47ea22569bcce17b00e5","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-01-15 12:34:37.000000000","message":"Uploaded patch set 1.","accounts_in_message":[],"_revision_number":1},{"id":"9acfc9521f7bfdb7041f5d93400a4034f98f990e","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-01-15 15:42:38.000000000","message":"Patch Set 1: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/a672525d391844afbe1e388e8847ed6b\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/16ab86ca40d9436ab90e43a57a7f4054 : SUCCESS in 10m 39s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/b4537d8e96cb4ab6ba9636619f84333a : SUCCESS in 5m 23s\n- openstack-tox-py38 https://zuul.opendev.org/t/openstack/build/367acb257b334009b59f8e17ac518295 : SUCCESS in 5m 04s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/effeac2b4e9b4487b6b33f3c5ea5e945 : SUCCESS in 6m 31s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/d4a82517815340b28ce591da880dad31 : SUCCESS in 13m 04s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/fe3a22ed05f446958192edda81a4eecc : SUCCESS in 5m 39s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/30d7c1142d8342e5aa4980432ce23900 : SUCCESS in 9m 46s\n- octavia-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/f76f7afc36004f8bb60f3fc7b470d48c : SUCCESS in 11m 10s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/fa9b246e336143679f5f96d640ea54e1 : SUCCESS in 4m 30s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/472cfb15c684460da3149c3221feecc2 : SUCCESS in 7m 52s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/3cfb633f3089433e8f81b7a919aec8ad : SUCCESS in 7m 23s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/ff759f63558b4c4cb23eb9f65227b6bd : SUCCESS in 7m 22s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/eea05dcac10b41dc8d045459fd3b4dae : SUCCESS in 1h 20m 49s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/3defa5b239dc422d908e2c78b97545d5 : SUCCESS in 2h 16m 33s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/6ce403074ecf4488b678aebe8ea39745 : SUCCESS in 1h 18m 24s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/991eaefb6b9d42b88597eac1e5dc9763 : SUCCESS in 2h 15m 27s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/68646c404bbe4e378a314be153e0f6c6 : SUCCESS in 1h 25m 03s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/6594c599aecf4c10a01228384c64c348 : SUCCESS in 55m 14s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/84d3ae0d072e4bf5bffccbadbd2b63a6 : SUCCESS in 57m 25s\n- octavia-grenade-skip-level https://zuul.opendev.org/t/openstack/build/a66173f9b44b49cd84b5461e6829c02b : SUCCESS in 1h 04m 25s (non-voting)\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/854fe06d54834bda8e177b40d75fba26 : SUCCESS in 1h 08m 40s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/c3a9fb76df3445c38e4221ce0a8f9031 : SUCCESS in 33m 52s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/2fc83669d9e7410882052384baa849aa : SUCCESS in 1h 51m 23s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/71a13cd6e5e74b428aac7f8cf5b044e2 : TIMED_OUT in 3h 04m 02s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/cedd6f4420ef4e0c94a7c662e761ec15 : SUCCESS in 2h 10m 07s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/91568e81b76d4d1d842485d76cd1b0b1 : SUCCESS in 2h 27m 23s (non-voting)","accounts_in_message":[],"_revision_number":1},{"id":"0028872d16b831abf13e9f5e29ab34a971eedd28","author":{"_account_id":11628,"name":"Michael Johnson","email":"johnsomor@gmail.com","username":"johnsom"},"date":"2024-02-23 00:52:20.000000000","message":"Patch Set 1: Code-Review+2\n\n(1 comment)","accounts_in_message":[],"_revision_number":1},{"id":"ece1adda294e63cbd64d885dda917739ff26372a","author":{"_account_id":32238,"name":"wdd","email":"wangzihao@yovole.com","username":"wangzihao"},"date":"2024-02-23 07:50:59.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"26e090869c3e67c1d14ad8db78e4ffbe6794684f","author":{"_account_id":32761,"name":"panhongyin","display_name":"panhongyin","email":"panhongyin@yovole.com","username":"panhongyin"},"date":"2024-02-27 07:34:58.000000000","message":"Patch Set 1: Code-Review+1","accounts_in_message":[],"_revision_number":1},{"id":"7c721abc7ad223dba99d564009c87d34561d0538","tag":"autogenerated:gerrit:newPatchSet","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-04-10 08:36:01.000000000","message":"Uploaded patch set 2: Patch Set 1 was rebased.\n\nCopied Votes:\n* Code-Review+1, Code-Review+2 (copy condition: \"**changekind:TRIVIAL_REBASE** OR is:MIN\")\n\nOutdated Votes:\n* Verified+1\n","accounts_in_message":[],"_revision_number":2},{"id":"b3c42b31bcff8be556800ebd0ef3632fe3e6c166","tag":"autogenerated:zuul:check","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-10 11:29:22.000000000","message":"Patch Set 2: Verified+1\n\nBuild succeeded (check pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/9f0b2f3ae6bd4ebea818adf2523a1bb5\n\n- openstack-tox-cover https://zuul.opendev.org/t/openstack/build/a0f5956c7fb74b1198a1ccc498057b76 : SUCCESS in 10m 42s\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/55dcb83b7864488eae945284c873938c : SUCCESS in 4m 39s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/9ce91f40860a4d09981f721cea44591a : SUCCESS in 4m 38s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/631e1637fcb748d59af8fe676f079e78 : SUCCESS in 4m 48s\n- openstack-tox-py312 https://zuul.opendev.org/t/openstack/build/c853a2d97ef14aff90ea5d6d407e6f40 : SUCCESS in 9m 34s (non-voting)\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/e88670b08ef44207b6262449462e8c6a : SUCCESS in 11m 14s\n- octavia-tox-py39-tips https://zuul.opendev.org/t/openstack/build/d823025172744b358e9db9c1ad4bda13 : SUCCESS in 4m 55s\n- octavia-tox-functional-py39-tips https://zuul.opendev.org/t/openstack/build/041dfcfddbf8405d8c74c25acfaffcd2 : SUCCESS in 8m 03s\n- octavia-tox-functional-py39-sqlalchemy-tips https://zuul.opendev.org/t/openstack/build/e7acd6180f30423caf7a980643c87dfc : SUCCESS in 10m 22s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/f55be2bf438f4572b5eb731898d4f898 : SUCCESS in 4m 59s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/63f4f161741643b190a29ba311c2464b : SUCCESS in 5m 48s\n- openstack-tox-functional-py39 https://zuul.opendev.org/t/openstack/build/8843e5ce9aab44d08b27ae4b6b4994e2 : SUCCESS in 8m 29s\n- openstack-tox-functional-py310 https://zuul.opendev.org/t/openstack/build/635dd0d325a14de1ae00ab080c78f1e0 : SUCCESS in 7m 31s (non-voting)\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/d587685bc6f74f6b9c0b08341a29a060 : SUCCESS in 1h 25m 31s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/033f28925ba54240a5f2783ce04e0f7e : SUCCESS in 2h 21m 59s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/fd539effc3bb475aaf7a1e16fbaf8155 : SUCCESS in 58m 24s\n- octavia-v2-dsvm-scenario-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/be5fd2402b7b4ceeaa7a8e1938ae6190 : SUCCESS in 1h 43m 54s\n- octavia-v2-dsvm-scenario-non-traffic-ops-jobboard https://zuul.opendev.org/t/openstack/build/c0882903b8644c3aa6becd5a578c23b9 : SUCCESS in 1h 06m 40s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/91a7be30092940a8b7a90fd70c0cd0cc : SUCCESS in 36m 35s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/58c6263fe40243168692dfa6e804308c : SUCCESS in 54m 27s\n- octavia-v2-dsvm-tls-barbican-fips https://zuul.opendev.org/t/openstack/build/9ccda048b77c4e36aba09c6b5f9abfa5 : SUCCESS in 1h 17m 01s (non-voting)\n- octavia-v2-act-stdby-dsvm-scenario https://zuul.opendev.org/t/openstack/build/7ad815fca4cb4df39f4af0bd259592d3 : SUCCESS in 34m 05s (non-voting)\n- octavia-v2-dsvm-cinder-amphora https://zuul.opendev.org/t/openstack/build/fb087a06e76b40e2ad3584b079230c3d : SUCCESS in 2h 30m 48s (non-voting)\n- octavia-v2-dsvm-scenario-two-node https://zuul.opendev.org/t/openstack/build/78a8cfea22944e299bd842ad928bbd04 : SUCCESS in 2h 38m 34s (non-voting)\n- octavia-v2-dsvm-scenario-ipv6-only https://zuul.opendev.org/t/openstack/build/95b155df81a74c8da344134ef2901d34 : FAILURE in 23m 21s (non-voting)\n- octavia-v2-dsvm-scenario-centos-9-stream https://zuul.opendev.org/t/openstack/build/bfd11263c23a487ab0e5b267ea674610 : TIMED_OUT in 2h 34m 52s (non-voting)","accounts_in_message":[],"_revision_number":2},{"id":"65e40f4bf4d350769e2cd677639f2b7ee00f7d90","author":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"date":"2024-04-12 16:13:40.000000000","message":"Patch Set 2: Code-Review+2 Workflow+1\n\n(1 comment)","accounts_in_message":[],"_revision_number":2},{"id":"f041b9017fd9d1972e0d8d269c1795f9881ab1b2","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-12 16:14:20.000000000","message":"Patch Set 2: -Verified\n\nStarting gate jobs.","accounts_in_message":[],"_revision_number":2},{"id":"79e03a3b79f92a1b79c99cb3431d4796afedebc3","tag":"autogenerated:zuul:gate","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-12 18:03:28.000000000","message":"Patch Set 2: Verified+2\n\nBuild succeeded (gate pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/942e40cafb2142be878809995a6e6620\n\n- openstack-tox-pep8 https://zuul.opendev.org/t/openstack/build/eb4b90ee84874639a8e8b7655e0f2b0f : SUCCESS in 6m 50s\n- openstack-tox-py39 https://zuul.opendev.org/t/openstack/build/59a46816d34342ec83be068a7708dadd : SUCCESS in 5m 15s\n- openstack-tox-py311 https://zuul.opendev.org/t/openstack/build/af018854c2ce4f6198c00a7cc05e846d : SUCCESS in 5m 21s\n- openstack-tox-docs https://zuul.opendev.org/t/openstack/build/1493a657aab044b7a167098ff6f70eb4 : SUCCESS in 12m 13s\n- openstack-tox-pip-check-reqs https://zuul.opendev.org/t/openstack/build/daf3a0662d9f461299f02ef21423eccb : SUCCESS in 5m 42s\n- openstack-tox-functional-py38 https://zuul.opendev.org/t/openstack/build/ec48ac1ec0f54bf2871b2c4d68801ab3 : SUCCESS in 7m 07s\n- octavia-v2-dsvm-noop-api https://zuul.opendev.org/t/openstack/build/c5804c00cca14dfeb9e147ee3028554c : SUCCESS in 1h 21m 37s\n- octavia-v2-dsvm-scenario-traffic-ops https://zuul.opendev.org/t/openstack/build/f8fa085fe76f48aea9428d425e0c202c : SUCCESS in 1h 37m 37s\n- octavia-v2-dsvm-scenario-non-traffic-ops https://zuul.opendev.org/t/openstack/build/bf93a4bb239f4ee8a940d25daa8c6528 : SUCCESS in 1h 02m 27s\n- octavia-v2-dsvm-tls-barbican https://zuul.opendev.org/t/openstack/build/17640552da8a4df68e3b260d345e4e6c : SUCCESS in 36m 24s\n- octavia-grenade https://zuul.opendev.org/t/openstack/build/afa67ab3bc07489ebfec6035702e1f30 : SUCCESS in 38m 23s","accounts_in_message":[],"_revision_number":2},{"id":"1a7336b866f5eb8b60bb0ffd694955f2da667f00","tag":"autogenerated:gerrit:merged","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-12 18:03:28.000000000","message":"Change has been successfully merged","accounts_in_message":[],"_revision_number":2},{"id":"10c956b4602f4056a4c52abb514c96abff31d8ca","tag":"autogenerated:zuul:promote","author":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]},"date":"2024-04-12 18:05:31.000000000","message":"Patch Set 2:\n\nBuild succeeded (promote pipeline).\nhttps://zuul.opendev.org/t/openstack/buildset/33dc7f7759d64eb6bff4ad9b2c8acbf9\n\n- promote-openstack-tox-docs https://zuul.opendev.org/t/openstack/build/125f77c456874fcdaf166debdfdba989 : SUCCESS in 47s","accounts_in_message":[],"_revision_number":2}],"current_revision_number":2,"current_revision":"9b1b8a3b6f121b5aec5ba7a8418078631adec310","revisions":{"4940997d60b704a8b3b4bf5eab9f48b55975a98b":{"kind":"REWORK","_number":1,"created":"2024-01-15 12:34:37.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/92/905592/1","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/92/905592/1","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/92/905592/1 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/92/905592/1 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/92/905592/1 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/92/905592/1"}}},"commit":{"parents":[{"commit":"700c003d686aef522815f2faf37b4aa96aeae22c","subject":"pyupgrade changes for Python3.8+ (1)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/700c003d686aef522815f2faf37b4aa96aeae22c"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2024-01-15 11:53:21.000000000","tz":60},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2024-01-15 11:53:21.000000000","tz":60},"subject":"pyupgrade changes for Python3.8+ (2)","message":"pyupgrade changes for Python3.8+ (2)\n\nResult of running\n\n$ pyupgrade --py38-plus $(git ls-files | grep \".py$\")\n\nThis was inspired by Nova [1]\n\nFixed PEP8 errors introduced by pyupgrade by running:\n\n$ autopep8 --select\u003dE127,E128,E501 --max-line-length 79 -r \\\n  --in-place octavia\n\nand manual updates.\n\n[1]: https://review.opendev.org/c/openstack/nova/+/896986\n\nChange-Id: I4ad020fbedd8106a3a86768c25a0c4dc750ec88a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4940997d60b704a8b3b4bf5eab9f48b55975a98b"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/4940997d60b704a8b3b4bf5eab9f48b55975a98b"}]},"branch":"refs/heads/master"},"9b1b8a3b6f121b5aec5ba7a8418078631adec310":{"kind":"TRIVIAL_REBASE","_number":2,"created":"2024-04-10 08:36:01.000000000","uploader":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"},"ref":"refs/changes/92/905592/2","fetch":{"anonymous http":{"url":"https://review.opendev.org/openstack/octavia","ref":"refs/changes/92/905592/2","commands":{"Checkout":"git fetch https://review.opendev.org/openstack/octavia refs/changes/92/905592/2 \u0026\u0026 git checkout FETCH_HEAD","Cherry Pick":"git fetch https://review.opendev.org/openstack/octavia refs/changes/92/905592/2 \u0026\u0026 git cherry-pick FETCH_HEAD","Format Patch":"git fetch https://review.opendev.org/openstack/octavia refs/changes/92/905592/2 \u0026\u0026 git format-patch -1 --stdout FETCH_HEAD","Pull":"git pull https://review.opendev.org/openstack/octavia refs/changes/92/905592/2"}}},"commit":{"parents":[{"commit":"69552a2082386d0b08da0f050ce05d4106214df8","subject":"pyupgrade changes for Python3.8+ (1)","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/69552a2082386d0b08da0f050ce05d4106214df8"}]}],"author":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2024-01-15 11:53:21.000000000","tz":60},"committer":{"name":"Tom Weininger","email":"tweining@redhat.com","date":"2024-04-10 08:23:50.000000000","tz":120},"subject":"pyupgrade changes for Python3.8+ (2)","message":"pyupgrade changes for Python3.8+ (2)\n\nResult of running\n\n$ pyupgrade --py38-plus $(git ls-files | grep \".py$\")\n\nThis was inspired by Nova [1]\n\nFixed PEP8 errors introduced by pyupgrade by running:\n\n$ autopep8 --select\u003dE127,E128,E501 --max-line-length 79 -r \\\n  --in-place octavia\n\nand manual updates.\n\n[1]: https://review.opendev.org/c/openstack/nova/+/896986\n\nChange-Id: I4ad020fbedd8106a3a86768c25a0c4dc750ec88a\n","web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9b1b8a3b6f121b5aec5ba7a8418078631adec310"}],"resolve_conflicts_web_links":[{"name":"gitea","tooltip":"Open in GitWeb","url":"https://opendev.org/openstack/octavia/commit/9b1b8a3b6f121b5aec5ba7a8418078631adec310"}]},"branch":"refs/heads/master"}},"requirements":[],"submit_records":[{"rule_name":"gerrit~DefaultSubmitRule","status":"CLOSED","labels":[{"label":"Verified","status":"MAY","applied_by":{"_account_id":22348,"name":"Zuul","username":"zuul","tags":["SERVICE_USER"]}},{"label":"Code-Review","status":"MAY","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Workflow","status":"MAY","applied_by":{"_account_id":34429,"name":"Tom Weininger","email":"dienste@weinimo.de","username":"tweining"}},{"label":"Backport-Candidate","status":"MAY"}]}],"submit_requirements":[{"name":"Verified","description":"Verified in gate by CI","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Verified\u003dMAX AND -label:Verified\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Verified\u003dMAX"],"failing_atoms":["label:Verified\u003dMIN"],"atom_explanations":{}}},{"name":"Backport-Candidate","description":"Backport candidate status","status":"NOT_APPLICABLE","is_legacy":false,"applicability_expression_result":{"fulfilled":false,"status":"FAIL"},"submittability_expression_result":{"expression":"is:true","fulfilled":true,"status":"NOT_EVALUATED","passing_atoms":[],"failing_atoms":[],"atom_explanations":{}}},{"name":"Code-Review","description":"Code reviewed by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Code-Review\u003dMAX AND -label:Code-Review\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Code-Review\u003dMAX"],"failing_atoms":["label:Code-Review\u003dMIN"],"atom_explanations":{}}},{"name":"Workflow","description":"Approved for gate by core reviewer","status":"SATISFIED","is_legacy":false,"submittability_expression_result":{"expression":"label:Workflow\u003dMAX AND -label:Workflow\u003dMIN","fulfilled":true,"status":"PASS","passing_atoms":["label:Workflow\u003dMAX"],"failing_atoms":["label:Workflow\u003dMIN"],"atom_explanations":{}}}]}
